)]}'
{"id":"openstack%2Fkolla~880317","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I29dde2696ff81f3f48effdd6821d94d0d84ff536","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-09 15:05:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-08-16 09:09:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I29dde2696ff81f3f48effdd6821d94d0d84ff536","subject":"ovsdpdk: add libdpdk-dev","status":"MERGED","created":"2023-04-13 12:53:13.000000000","updated":"2023-08-16 09:11:25.000000000","submitted":"2023-08-16 09:09:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880317","cherry_pick_of_change":872259,"cherry_pick_of_patch_set":1,"meta_rev_id":"8a0fce105d216fc9e40e96712e8aec2495724527","_number":880317,"virtual_id_number":880317,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-16 09:09:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-09 15:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-09 13:12:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-08-09 13:12:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-08-16 07:03:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 12:53:13.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-04-13 13:48:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-13 15:27:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-13 16:01:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-20 14:43:00.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-08-09 15:05:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"a8237882fe3afcbafac144d3692ff8abca7925d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-13 12:53:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"780d2998334063a9a05941e21f698770f964b8b9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-13 13:48:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ea6a7053634ff8795c5bc16d9e29b91ad2507640","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 15:27:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ba6030f526940728b6fd92d6eb432da\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/951466f5c93f4c849855f0f69fe51c53 : SUCCESS in 1h 51m 47s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ee8b759bb024733960b462eb5dccfc9 : SUCCESS in 1h 03m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/742892c5a0a649cb9e29ed7a31240692 : SUCCESS in 55m 50s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ff1e798751746e181137ef2ca31a6cc : NODE_FAILURE Node request 300-0020967147 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/762df606a51c4599bdb73a8bcadd90dc : NODE_FAILURE Node request 300-0020967148 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3abf32a7849c4ff493ea17d8a09a5654 : SUCCESS in 6m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/05336764ba254084a323bae24244026e : NODE_FAILURE Node request 300-0020967150 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68d19d2d86d206c01a89f37ac46ba6afb25da454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 16:01:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4d2d2997dbe49d6bc10424cc639b15e\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1a680e25309f48ce9de5ef3258c679ea : SUCCESS in 35m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4e0192ba1bbd4c1bb87a8a459c3ec61e : SUCCESS in 37m 01s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/54404cbdd2c24e4495a2e1ab3c0793c1 : FAILURE in 29m 27s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5922afa129df49f48f975cce38aaa641 : SUCCESS in 1h 29m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/84f920d48e0f4210b16925d4b2d58652 : SUCCESS in 1h 13m 53s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/2fb5f3741f8a48dc9a1a014b819a753a : SUCCESS in 59m 30s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/16d32c0289a94b809b6e6e56b37c4e65 : SUCCESS in 1h 31m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9632c2fbca3844d695f31cdf85b87b25 : SUCCESS in 26m 23s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5e5fa6e8dc25412bbfbd8d9972c4faec : SUCCESS in 21m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3aced70505124273b168b70b97754d75 : SUCCESS in 1h 15m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8933d23007b1415e9fbaf718bef34a0b : SUCCESS in 1h 19m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466ae52343a24d2f99c80cefda9d4949 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7449d166deeb41a796c0d20f27ee093d : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e9c55a403d944a588c13b8dd550dcc2 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3854dfceb5684be988ce334a34f96fec : SUCCESS in 5m 11s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/17569d684b4b4cb68e3aa6df16a3a7ec : SUCCESS in 29m 33s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/5b02471fb1394081b43db79e31a6c5f1 : SUCCESS in 1h 12m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/70b2ddad3f5846a6bc49039a414576bf : FAILURE in 20m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/087798f03645441190b00cd8a6a3e3a7 : FAILURE in 22m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f273f3c45977a6f187f91fe9a13944c20a1e78c7","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-20 14:43:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0af925e4d333dfc5dcef8c713f07df5a64cf465d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 08:12:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6323349dc38522f9d84fb823bfedecad477243bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 10:17:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/022b070dc74449c19dd8cfe185777bf0\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b6002a8c5a1483d87c1472fc045820b : NODE_FAILURE Node request 300-0021541909 failed in 0s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f503d5b3e6304080a4a98cfba717f49b : FAILURE in 35m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6daee2f3db304b588ac771d5735d61c1 : SUCCESS in 40m 10s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ba0dd6bba54147a99571f5330bd0f5dc : FAILURE in 1h 59m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/667f6822210041899316d26b44e92c24 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b13d6209a9364cfaa284d0831e58dfc9 : SUCCESS in 4m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/584e1d64b9d8410eb025f8a3f4e4c1d5 : SUCCESS in 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c8d0aa7e56e5b1ca0438a94dec7a9859f072db4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 10:52:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f031a6880e8e45818fe8e9c8c5a30a02\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6d282bcce96b4eb0a49bfa11d1b25a79 : SUCCESS in 34m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/16c4a4f7495440e2a737d9071e490c19 : SUCCESS in 2h 01m 56s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/ff40b42be8294845976ab014468a76e6 : FAILURE in 1h 05m 43s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/20a8bcabe0db4573b51617d0a9a23944 : TIMED_OUT in 2h 02m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d598eca70a3a4628902a45d41517f4f4 : SUCCESS in 1h 29m 10s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/94036222e58748fc800b32817cfd81a8 : FAILURE in 53m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bb4c654faa344c13bfa7b98bd993fd9d : TIMED_OUT in 2h 32m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/820c87b1a0af4360ae7442e0489e0b7b : FAILURE in 26m 03s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9918d8f10e054ab19e2f4802c1821592 : SUCCESS in 31m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0d1269700ee54ebfafbbda0cff66a532 : FAILURE in 31m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bf74d56a598b471eb6a3f71577140eb2 : FAILURE in 24m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10da6dbf4f404c1a8df0d4c50fdf0761 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c529509d10f84b63a866ea8d9d1fd282 : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dbc630f665d4103bcb5aa4de14700fc : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c98347abff24c4cad1604df376620b2 : SUCCESS in 5m 08s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/d594d288e0584598ab5a8afdc01c86a4 : FAILURE in 24m 10s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1f3e721f17f24fa687e372b4958b164d : FAILURE in 20m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/02044c01509c485cb226bfb3d83bed0c : FAILURE in 31m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/003d95b359224dedace4dc4ff53f2638 : FAILURE in 1h 15m 56s","accounts_in_message":[],"_revision_number":2},{"id":"12e987c8019b4e4a49e140ce1ef0c916dea04bb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-09 13:12:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"62ac5be14982b78a440af9fb2bc2ef78c3e579c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 14:47:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f547eaf4844b999e2a7b910d7c77ed\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/481e30bf7e5b4aabbf1092b6f1038c1a : SUCCESS in 37m 28s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a4a64e289a2543d998a6e6ee67c5c1be : SUCCESS in 32m 30s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/090a6ef0a93a4a3fbb1823c9ad43b1c4 : SUCCESS in 27m 20s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e6e33da0339a4931b5f239abb44e09d7 : SUCCESS in 1h 28m 06s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/69234bd56ed649d592a0356ba7a60153 : SUCCESS in 1h 12m 20s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/9e20706f3e6549c8bfbd16e9ecb6289c : SUCCESS in 1h 09m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/aa667a44510546719a9fa77c2158ab46 : SUCCESS in 1h 27m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b681c955254c4b4ba16c0d35ec46734d : SUCCESS in 26m 31s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5a5a7cbeb4f144b0800f97ff9482c4cf : SUCCESS in 21m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eec47888ef2942bcbc84e64556a8e32f : SUCCESS in 1h 06m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/52bee6b7f2044c8491a064b5c1fcb955 : SUCCESS in 1h 03m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14267f1d522840eab500506dbc1e3a06 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84b7b2b859164b4f917d10f1a4a116c1 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ea637bc190748f09c37c105c2a92a8b : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c795ef6f0a454318b6e451eefcb48c22 : SUCCESS in 5m 12s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/c673a61036874994ba237f71dd420434 : SUCCESS in 27m 28s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/fe9bc0a8f1dd4f2f896c3a451913ca9b : SUCCESS in 1h 12m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ba4aedefeefa461bb3c1e622edfa933a : SUCCESS in 22m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed3687bed682407ba84e907903bd14bd : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":3},{"id":"15a2bbbc65e6bd303a02ce2971a8b82be234c894","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-09 15:05:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"367c9b972cda9934e9ccad734b6e71e19dddb610","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 15:24:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53a8624d11e542dcb5285098f3133e32\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/adb9501ced634c158f4d3585f5e4f27c : FAILURE in 2h 09m 43s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4012c46d30d24a3fb7638f1bbdc2538a : FAILURE in 1h 33m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/db108f8342a643b5badb9aa83932bb65 : SUCCESS in 1h 51m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/81fa0f3aa0d84da3a2a4272bdf63fd89 : SUCCESS in 1h 03m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/077e75bc87aa41aca5404f7a1743fb29 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/727dea1b950a46f3868734ddf9ef3898 : SUCCESS in 3m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0119fd8c2854c8f8fd42626b52b1893 : FAILURE in 1h 53m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"96801908b398477e2028e9073e669b1859f68a0f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-16 07:03:36.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"474747099d7f31f0cb8949fd6b09c54f07d7aa06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 07:04:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b0ff88fa6efd76e517a09b51cd8b16747293379f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 09:09:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/799f5d698e0b44bc8e42ba8dcf0c75be\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2ac8b97d39dc4727962eeeedc4f84fd7 : SUCCESS in 58m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6699d1d0e1be4284a86a35732d202ad4 : SUCCESS in 1h 39m 56s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a2a90385f78446d686eac6befa588152 : SUCCESS in 1h 53m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b2ca911f160148eaa8d5306c91549a77 : SUCCESS in 18m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2cb1e4f27d3346aab124d93d28e09c3d : SUCCESS in 1h 29m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e03b96da06824d7dbfe1eefffa1e7701 : SUCCESS in 1h 05m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68359ea1d7f448e68432603eb7e80595 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50192cfa6f25458eb8b0ca2f930877fc : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2af415fd25474971a7b4fde6f64e81fc : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56c6da239aad4a7ea4b5462db5094fa1 : SUCCESS in 6m 00s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/75e2d2401cb14b32b2891070353ffc5f : SUCCESS in 1h 38m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f1a36da243e1422cab9ef224872d2c0b : SUCCESS in 20m 50s","accounts_in_message":[],"_revision_number":3},{"id":"98806a4b53c8548eae06835917f16eb04893f083","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 09:09:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8a0fce105d216fc9e40e96712e8aec2495724527","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 09:11:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50f006c1f13a407598b671958c4837ce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1790c16041f440bb7a86b7cb70420df : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9cdee83987b2587225c96d4b2068bd391ae8ba43","revisions":{"4341e286beb2a482d25051b2e14c601e45a99f39":{"kind":"REWORK","_number":1,"created":"2023-04-13 12:53:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/17/880317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/880317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/880317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/880317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/880317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/880317/1"}}},"commit":{"parents":[{"commit":"867f27e6b6a69128844c240028ee15b80ffa0fa5","subject":"Merge \"Cinder backup fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/867f27e6b6a69128844c240028ee15b80ffa0fa5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 06:41:01.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-13 12:53:13.000000000","tz":0},"subject":"ovsdpdk: add libdpdk-dev","message":"ovsdpdk: add libdpdk-dev\n\nCloses-Bug: #1980328\nChange-Id: I29dde2696ff81f3f48effdd6821d94d0d84ff536\n(cherry picked from commit f7436ee60256687a71a3a89fac2e15b61e9e4859)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4341e286beb2a482d25051b2e14c601e45a99f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4341e286beb2a482d25051b2e14c601e45a99f39"}]},"branch":"refs/heads/stable/yoga"},"23424782f643bb87746bc416d1b4f07427dff102":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-26 08:12:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/880317/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/880317/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/880317/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/880317/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/880317/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/880317/2"}}},"commit":{"parents":[{"commit":"75f9186288e6410935d19f430394410289ecb63b","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75f9186288e6410935d19f430394410289ecb63b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 06:41:01.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-26 08:12:43.000000000","tz":0},"subject":"ovsdpdk: add libdpdk-dev","message":"ovsdpdk: add libdpdk-dev\n\nCloses-Bug: #1980328\nChange-Id: I29dde2696ff81f3f48effdd6821d94d0d84ff536\n(cherry picked from commit f7436ee60256687a71a3a89fac2e15b61e9e4859)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23424782f643bb87746bc416d1b4f07427dff102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23424782f643bb87746bc416d1b4f07427dff102"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"9cdee83987b2587225c96d4b2068bd391ae8ba43":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-09 13:12:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/880317/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/880317/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/880317/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/880317/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/880317/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/880317/3"}}},"commit":{"parents":[{"commit":"d49cca54922507c040dd1003e628938b1aaf5cf1","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d49cca54922507c040dd1003e628938b1aaf5cf1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 06:41:01.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-09 13:12:01.000000000","tz":0},"subject":"ovsdpdk: add libdpdk-dev","message":"ovsdpdk: add libdpdk-dev\n\nCloses-Bug: #1980328\nChange-Id: I29dde2696ff81f3f48effdd6821d94d0d84ff536\n(cherry picked from commit f7436ee60256687a71a3a89fac2e15b61e9e4859)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cdee83987b2587225c96d4b2068bd391ae8ba43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cdee83987b2587225c96d4b2068bd391ae8ba43"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
