)]}'
{"id":"openstack%2Fkolla~880634","triplet_id":"openstack%2Fkolla~stable%2Fzed~Ifa095a4a5a4a2476b356b082b9b57b491c9e29db","project":"openstack/kolla","branch":"stable/zed","topic":"glance-api-multipath","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-24 07:07:58.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-24 08:36:34.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-24 08:36:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifa095a4a5a4a2476b356b082b9b57b491c9e29db","subject":"Glance-api fails due to absent multipath tools","status":"MERGED","created":"2023-04-17 13:46:53.000000000","updated":"2023-04-24 08:37:39.000000000","submitted":"2023-04-24 08:36:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880634-glance-api-multipath","meta_rev_id":"ee79d128ed67eb654ef9f0c1682faec62448f3be","_number":880634,"virtual_id_number":880634,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-24 08:36:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-24 07:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-21 11:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-18 10:14:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-24 07:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-17 15:31:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-18 10:11:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-18 10:14:10.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2023-04-19 15:15:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-21 11:09:04.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"2b8ef3f023a53377e94cf1a7018367470c05b52a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-17 13:46:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81ec77051cb258d1cc16e8b571e128231f746ea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:31:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/009e16a6e58446818369571e243a2f48\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2412744cc98e4a08933e90754371d3a9 : SUCCESS in 22m 47s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/fad74eb02ec84165926ee03c0da91ce7 : SUCCESS in 58m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/16b7c3321d5840cfbf4fb17d5c02a5f2 : SUCCESS in 14m 26s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/582e8df9d94448cea4a3f23b03aedb5d : SUCCESS in 56m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3ae4b8b063ab40459df53b7f773ca39b : SUCCESS in 1h 09m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02a78c10fad14fbe9216836920ca1e08 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5397d3b9303245c7b58b0645bd87c275 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8796c84a78b43e9916e01686fedb89a : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a075da557f64d0c87b81584be644d9c : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17acde75642a45b6b8249e4eca017267 : SUCCESS in 6m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7e2b6d51a0b44e12b80347a773cc4bfe : SUCCESS in 20m 03s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/18f40d2c5b964f0ea43ec4c1cab15b86 : FAILURE in 27m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/43ed58864c584975bd607b51efce8dd0 : SUCCESS in 31m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a13c40fd7924889a79c7fa03ce2d433 : SUCCESS in 1h 13m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/85ab19ab310d4425844e836f62f5e1bc : SUCCESS in 1h 35m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a3407b246136153b97e8ac969749f801ef8d4ff4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 16:53:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb6afef7c349439bb241860dd2d7de22\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7c99662199ee4eddbf48c221509a1efe : NODE_FAILURE Node request 300-0020992211 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3f7b813d10974d55b32237ab9115d62e : SUCCESS in 16m 11s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cbab0fb2e12e411aa04b55b0d0169ab4 : TIMED_OUT in 3h 03m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d9047029e1e4eabaa0bb3397921a574 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b3f9d002aa14255b585fa6fc42d3aa0 : SUCCESS in 5m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4239327b4c9643aa9308126fab23035f : SUCCESS in 17m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f5442de3064043b9a8b73c94f944595f : SUCCESS in 1h 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"702fb5773f6c542e893d65337b5e63938c498464","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-17 19:48:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c2d2a4bfd1bdbe71cc3d1c809760b8c8f38e1bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 21:09:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89b292e49fdd4b339fda89118daf4db0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/af6353edf8be4a1caedb7d9fdea0a3fa : SUCCESS in 1h 13m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f236463c303d4cb1a70b5fdf35bb5151 : SUCCESS in 1h 10m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/67d4c8d33c544a3a93944f9edce06d2a : SUCCESS in 1h 14m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ff95662a9084b5fb28c331db28b181d : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a362df823be5425c8b42317f28777561 : SUCCESS in 3m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/197fabbcd52f4759a1623d79a733d2e2 : SUCCESS in 1h 12m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7aba216731534a3d990c35442a23676d : SUCCESS in 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37d5f3f5f567cffd712b99c2852c4ee567b42d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 21:13:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8252b64bc3124d8ca68597d3b5d3454e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e9d1483c8bc141d8a31b9f3dcd9b3bd1 : SUCCESS in 21m 30s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/198d106b59574d70a3f6443f5e0fdb57 : SUCCESS in 1h 07m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4f7ab968a9b4431eb01cfc5c7c2f14e3 : SUCCESS in 21m 44s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/18507b0317e24a37b32c988c078f9db4 : SUCCESS in 1h 01m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e15ec6df19844248a42c5603b39fc6f2 : SUCCESS in 1h 05m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc3e2c2a195142d2a2e66bc85aa7b5b5 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fba30d7327aa43cca39d29da6a992742 : SUCCESS in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae8fb0f03bef4e52bb0008a45270e5d7 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29016e4dd0254d5d9315f4a885f633ce : SUCCESS in 2m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74fedb07889e4fa49cec8e282d88c1d6 : SUCCESS in 5m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3b888cb2a1574fefa0dfb2dbd57e9eb0 : SUCCESS in 23m 57s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/db1e74de6cfd44d7881ef3ce0bd12bb5 : SUCCESS in 1h 03m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dfb85a093d394a25b31e80e96c4d4808 : SUCCESS in 15m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d40fa7268814ce0bdecfd1780efbe96 : SUCCESS in 51m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2288ad12f7e24cbf9a7b5708e1c1d291 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":1},{"id":"eea000f0331bfa67ae38ac420cafb93b300b8677","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2023-04-18 10:14:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ae3e3f9ed41331838b10fdf664aafb06842a8f0e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-21 11:09:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"16deae55de90ca04fc6eddfe04a99a563670633d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-24 07:07:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e535e58dde391ec7da54c8800e779d6fe2637b6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 07:08:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01e1e0f3895cf29402f83400932e9fa7f32b569e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 08:36:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08b5e4a0cecc4ebf947b6dfed5959392\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6dfb8de7e5ac44e0a3b4bad43ac049a4 : SUCCESS in 25m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cc23dbb0a0f74987a0a39aebdc2bbd99 : SUCCESS in 58m 20s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/70ea0873c2bb465bb333292c7194a405 : SUCCESS in 1h 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bae8a33a0444f3fb57c204b98aa8a80 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d2367a8941f40e281c620f7bcdf6e5a : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe08ab555afe457087c6748fc6214c35 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/674d2837cf8542929d888d516f996802 : SUCCESS in 4m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/350ca85184dc43e8b3d11322ba61f2ce : SUCCESS in 19m 34s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/5004a95fd3454beab096e0b8c7115a96 : SUCCESS in 1h 07m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/508261dc5b5d46d5b8900b3f5ab483d4 : SUCCESS in 17m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b45204692794f3dbb658ed9d105e5df : SUCCESS in 47m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/be3e713733ed47d983ce13e313316303 : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4966e30939438b9a9744b30b709b19b98ed17c33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 08:36:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ee79d128ed67eb654ef9f0c1682faec62448f3be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 08:37:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00ffc5b98f674e9aa0d8cbf9024a8f88\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca522df0e184fd6bc0a566848d9216b : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0c028dfa027d50437f6f0a28b56a7520893dbdb","revisions":{"e0c028dfa027d50437f6f0a28b56a7520893dbdb":{"kind":"REWORK","_number":1,"created":"2023-04-17 13:46:53.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/34/880634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/880634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/880634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/880634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/880634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/880634/1"}}},"commit":{"parents":[{"commit":"bfa8e408972a34d444e9e612f537fbf196d00603","subject":"Merge \"Cinder backup fails due to absent multipath tools\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfa8e408972a34d444e9e612f537fbf196d00603"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-13 11:06:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-17 13:46:21.000000000","tz":180},"subject":"Glance-api fails due to absent multipath tools","message":"Glance-api fails due to absent multipath tools\n\nCloses-Bug: #2015500\nChange-Id: Ifa095a4a5a4a2476b356b082b9b57b491c9e29db\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit d23ace3fa6197d32fc6a883f9f06e54293d2089b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0c028dfa027d50437f6f0a28b56a7520893dbdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0c028dfa027d50437f6f0a28b56a7520893dbdb"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
