)]}'
{"id":"openstack%2Fkolla~880635","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Ifa095a4a5a4a2476b356b082b9b57b491c9e29db","project":"openstack/kolla","branch":"stable/yoga","topic":"glance-api-multipath","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-24 07:08:05.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-24 08:45:32.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-24 08:45:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifa095a4a5a4a2476b356b082b9b57b491c9e29db","subject":"Glance-api fails due to absent multipath tools","status":"MERGED","created":"2023-04-17 13:47:15.000000000","updated":"2023-04-24 08:46:33.000000000","submitted":"2023-04-24 08:45:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880635-glance-api-multipath","meta_rev_id":"e96abe7bc5ab37255c5c9b2d2a751a5c8ed6c642","_number":880635,"virtual_id_number":880635,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-24 08:45:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-24 07:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-21 11:08:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-18 10:14:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-24 07:08:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-17 15:37:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-18 10:11:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-18 10:14:20.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2023-04-19 15:15:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-21 11:08:48.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"aecb2677aabf7a600c967efb3d1c358a68e3e841","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-17 13:47:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a80e9d450e7c82fd8f7f90d159b32caf591181fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:37:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9accaf00c51945898a0a3b79d529b4f4\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/83116c4f351b4a9ebbd9f2f47d519d2c : SUCCESS in 36m 30s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/58a1e63ede5d4b2ca62d9d5c49540ea5 : SUCCESS in 28m 41s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/02fbb64691864a9eb952e60ed2159bc5 : SUCCESS in 29m 25s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a1755a8450914483a6726a3caa2c3799 : SUCCESS in 1h 33m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f27ad0adf07c4898b0973e605ce1d488 : SUCCESS in 1h 17m 20s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/98c457ad02e94e0a86bb76feab491e38 : SUCCESS in 1h 01m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e9b5ee8d95c54f7cb092e0a5aaf4ab4f : SUCCESS in 1h 12m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3098c7d15214446d9d55c55f6420192d : SUCCESS in 46m 16s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8c8e623caa6e4b9883b01714c59f864e : SUCCESS in 22m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bf701e62beda46b8a7aa206be77a93bd : SUCCESS in 1h 28m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e89e7717da0f4be4ab55a7f93abcc460 : SUCCESS in 1h 40m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/007a392ebc0a4df08ef316bbea9db356 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82cd1ef979e24c7ca89ac37b2e303404 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7758abeb98bd4a18a06f327b26be60c9 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8924af6367d47b68b8c936d4e1bb3df : SUCCESS in 4m 35s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/0d7ced33e0764832900d386cbccda8b4 : SUCCESS in 24m 13s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/5cee4e629b9341d0b8148511df4ff50b : FAILURE in 26m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/52de71660ee749c29b8b0ff5839b096d : SUCCESS in 17m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d1475a5f1774be397b55328e0abd9eb : SUCCESS in 20m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2ce6714c44ead426a4e4a73f720c0ca92bdc8c7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 16:33:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93a9c1daad484f7ea06ef4f8392417ae\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f54fddb49154f22a2c75da77c6aa4a9 : SUCCESS in 2h 07m 02s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5583e590f40f4886b32018e1e6d08f6a : SUCCESS in 22m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a2e6559143fe4c56a32e95b33271815e : SUCCESS in 1h 49m 31s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6e0b56f9454f44e883d995f1a20b5949 : FAILURE in 2h 33m 51s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5292358164564b1398a2442c1309e5d8 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c92d08a06e664c6cbd996e66b563e919 : SUCCESS in 6m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ea57f5586c84734827d2960436129c1 : SUCCESS in 1h 22m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f562418a2f31fcc47525d6ae34b3618813cf94de","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-17 19:47:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7910fbbb809a6732136cc6c26e9049399aa4623f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 21:03:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07aa4c5b494f432fa7b4690231c2757c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8cc6163356641c0b741e8c57e94dc0a : SUCCESS in 31m 22s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/93d27ad2154a4ddc96f525622359d920 : SUCCESS in 35m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3664443e32494b4d9bce5c3b6d5fef50 : SUCCESS in 25m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f6f601ca49f466aade053907ff2fec0 : SUCCESS in 1h 13m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4be2852bd4994dd8b66f90b2f6fb1d1c : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8bafdc2f634647bf86bcb24739461ae3 : SUCCESS in 3m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9be883e13a8c4911be07b396b9e3d0a8 : SUCCESS in 1h 08m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1894fa7fcc493d1433c2e7653004b66627c79a6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 21:26:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5932030b12ec455badcc0386c7a40b17\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8238495eef3e4990a95428bbfc4891d6 : SUCCESS in 30m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6f1521f2000a4e5680ee3a1c9c62a4c2 : SUCCESS in 24m 47s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/cb376859002f4543a16acd856163786f : SUCCESS in 25m 08s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3a98a0d8e37544e08b7d817effac976a : SUCCESS in 1h 16m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/20b0fbda63ba4f369c589a65caaedd44 : SUCCESS in 58m 30s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/29cbe479e571412f9a9f338e92a7f119 : SUCCESS in 56m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b72f8c86b8d54ed09e6d2564a55d8ab4 : SUCCESS in 1h 23m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/601bec70fc4d4d2eaff86580c8339cfa : SUCCESS in 24m 03s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/50b3c1296623458b8ca36ad4b061585b : SUCCESS in 16m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/766f00102a8647a1853aa81451dd7c0b : SUCCESS in 56m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/dabc2367b3904b248076cd303cc87383 : SUCCESS in 46m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c1fca2192f4a159a6882625b24ef67 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3d18f0a03a34dcc936182880b91c696 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2ae4147a4774f24b437686ab369e717 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/096c11021fac41acaba54725082a5595 : SUCCESS in 4m 28s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/80b33761178a4691af5a33f43a4077e9 : SUCCESS in 26m 30s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/0abddd5660b9432f94896a42d8722737 : SUCCESS in 1h 06m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e32d267b5ad41a3ae21b9e18e14ea57 : SUCCESS in 16m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7dd94b5e21674e9ba876db0e46ecbff1 : SUCCESS in 24m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a5156c9c80e1ce16ed807251d5b60a0613bffacd","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2023-04-18 10:14:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"226a773e4c23e5de9af00940615dab85d7fde052","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-21 11:08:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8baae48d56d25154caa193136db7764088f8098e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-24 07:08:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7abfb54d788d1353108cb0933631626f340ae6cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 07:08:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41df109701a521b710a8f0258b9e4e4d5f8ed188","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 08:45:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91c449eba35a45508e12b22fed9e53a8\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c0ff9a5d7b0f4f4897133c951bbb8817 : SUCCESS in 34m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f838eb11915245988faf755677fa3478 : SUCCESS in 1h 12m 34s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f1e5b67ecc9f474eaacf7b6c38ac1c73 : SUCCESS in 1h 26m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e9dd03a2c9c245178fc6b15f8de8f0b5 : SUCCESS in 25m 22s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3f6c4053fa9948778e468b74a572df29 : SUCCESS in 1h 11m 20s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/78f6f63a90814960b16facce86266878 : SUCCESS in 1h 09m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216cc7cdd56f41fa9ae8a75896fee48f : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09b14c46dacd42a49f837121dede5b78 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cd964d8093649afb2769bc45bbe96d0 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e252b3d90f1f4a92b33cedf88f6d8b61 : SUCCESS in 5m 40s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/094beecc38b34561abd4f10d7a4266fc : SUCCESS in 1h 08m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b6283fe7d35c46e3b83e993ea1b9f667 : SUCCESS in 26m 39s","accounts_in_message":[],"_revision_number":1},{"id":"eab8f4d9419d1793a24b5ffdf7894905cc242ed0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 08:45:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e96abe7bc5ab37255c5c9b2d2a751a5c8ed6c642","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 08:46:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cad1c9540b1645a1ae291b4329b5f13b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ef768a6ed20469fb3c119327b00d88a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d511369be79574a2e4a7db128e8f2d2e74b13cc0","revisions":{"d511369be79574a2e4a7db128e8f2d2e74b13cc0":{"kind":"REWORK","_number":1,"created":"2023-04-17 13:47:15.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/35/880635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/880635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/880635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/880635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/880635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/880635/1"}}},"commit":{"parents":[{"commit":"867f27e6b6a69128844c240028ee15b80ffa0fa5","subject":"Merge \"Cinder backup fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/867f27e6b6a69128844c240028ee15b80ffa0fa5"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-13 11:06:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-17 13:47:03.000000000","tz":180},"subject":"Glance-api fails due to absent multipath tools","message":"Glance-api fails due to absent multipath tools\n\nCloses-Bug: #2015500\nChange-Id: Ifa095a4a5a4a2476b356b082b9b57b491c9e29db\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit d23ace3fa6197d32fc6a883f9f06e54293d2089b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d511369be79574a2e4a7db128e8f2d2e74b13cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d511369be79574a2e4a7db128e8f2d2e74b13cc0"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
