)]}'
{"id":"openstack%2Fkolla~880799","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Ie067d72bd468467c6ccb8d1ca80695c769ec403e","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-19 10:41:53.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ie067d72bd468467c6ccb8d1ca80695c769ec403e","subject":"rabbitmq: update to RabbitMQ 3.10","status":"ABANDONED","created":"2023-04-19 08:26:15.000000000","updated":"2024-01-29 08:56:18.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"ed7284edb67323fe184b7ba9d858a675e731134c","_number":880799,"virtual_id_number":880799,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-19 09:59:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-19 08:36:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2023-04-19 10:41:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 08:36:10.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-04-19 09:59:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbdad9ca3c4b5cca8a9f79222c55f3895c3d204a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-19 08:26:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8abd82e8a799859e1d59bd4590182de393398ff7","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-04-19 08:36:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1614c9ee74b61e96e56c9b43dfbbe19aa0168f8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 09:59:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b6c5972f7534d83bb104a98aa14de0b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/52724f683ee54a9cb74c899d24180b50 : SUCCESS in 29m 28s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/76c1f701b3d64a148c7409183d366947 : SUCCESS in 30m 33s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/1b78c74cd6854d76b084c9dc6afdde30 : SUCCESS in 32m 09s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f4f1c40ab5be4d1aa8ff15309bee3b25 : SUCCESS in 1h 17m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9035c87e1da9466f953071ff66a7a88c : SUCCESS in 1h 10m 15s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f8eb78fecd6a4792aad3906227bd6262 : SUCCESS in 1h 00m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2bca982ea73b464ebffea36b786627af : SUCCESS in 1h 28m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6cbdf634d6bd465e86195a58a25f52a0 : SUCCESS in 23m 07s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b96e4bbc397a446cb30d08fd9dbb2ccf : SUCCESS in 25m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a49548c755274e48a46ce56d58b86676 : SUCCESS in 57m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/12b202f21bdf4cd0a3796b0b2c637a4c : SUCCESS in 59m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02486871e94418a9f3d03915df2396d : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/045ec74b447243c2963b6567ad20d2ab : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30b30105a2c247e98f170569f0b601e7 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aa3cc2a23c34eeba8556fe77128fa7d : SUCCESS in 6m 03s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/ebaf4eb76877483aaad049b97c9ba859 : SUCCESS in 19m 15s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/7cd71b9bb8864361bb7dc7b67d549a04 : SUCCESS in 1h 07m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/24896d3db58b4c368be4d2aa36a582b8 : SUCCESS in 20m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6cb0e0c7c6784bc487bd9a0ca5a2ad9b : SUCCESS in 21m 54s","accounts_in_message":[],"_revision_number":1},{"id":"26734893f33268605c36d5703b530fb03c31ef83","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-19 10:41:53.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"771cc88aedd4d6565fbeb54dc6b2f55a5716bc50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 11:13:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ee5498afc654ce29af04bdf49f09acb\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad660f1d65e34b76acba83416acb6e8e : SUCCESS in 22m 11s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6930f23e34e94892a788186fe6ec845a : SUCCESS in 37m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f08cd1ccd318475aa02264e1b932746f : SUCCESS in 20m 01s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc4c939e305640638829ae14103c7ff0 : SUCCESS in 2h 33m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9309d914090e4339be1407bc68e76631 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df7f966081104e7a846ae51fa116c35d : SUCCESS in 3m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1a8441661a9d4857aae508fdb2f9d4d2 : SUCCESS in 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed7284edb67323fe184b7ba9d858a675e731134c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-29 08:56:18.000000000","message":"Abandoned\n\nyoga is going unmaintained, you can repropose against unmaintained/yoga once that branch is created","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29d7f6836f126f433dd15c3e7fee77448117a7af","revisions":{"29d7f6836f126f433dd15c3e7fee77448117a7af":{"kind":"REWORK","_number":1,"created":"2023-04-19 08:26:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/880799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/880799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/880799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/880799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/880799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/880799/1"}}},"commit":{"parents":[{"commit":"867f27e6b6a69128844c240028ee15b80ffa0fa5","subject":"Merge \"Cinder backup fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/867f27e6b6a69128844c240028ee15b80ffa0fa5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 08:24:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 08:26:12.000000000","tz":120},"subject":"rabbitmq: update to RabbitMQ 3.10","message":"rabbitmq: update to RabbitMQ 3.10\n\nYoga provides RabbitMQ 3.9 series which is not supported anymore.\n\nRabbitMQ 3.10 is a final RMQ update in stable/wallaby branch.\n\nAccording to upstream [1] 3.9 -\u003e 3.10 migration is fine while 3.9 -\u003e 3.11\nshould be done through 3.10 so we can not move to 3.11 version here.\n\n1. https://www.rabbitmq.com/upgrade.html#rabbitmq-version-upgradability\n\nChange-Id: Ie067d72bd468467c6ccb8d1ca80695c769ec403e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29d7f6836f126f433dd15c3e7fee77448117a7af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29d7f6836f126f433dd15c3e7fee77448117a7af"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
