)]}'
{"id":"openstack%2Fkolla~881236","triplet_id":"openstack%2Fkolla~stable%2Fzed~Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-25 12:15:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7","subject":"https: remove dhparams.pem before running gencerts","status":"MERGED","created":"2023-04-21 15:26:07.000000000","updated":"2023-04-25 12:16:54.000000000","submitted":"2023-04-25 12:15:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881236","meta_rev_id":"caa78bbcc0d7c8bbf09b8ec0b41268648899f6d4","_number":881236,"virtual_id_number":881236,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-25 12:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"},"all":[{"value":1,"date":"2023-04-21 17:21:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"},{"value":2,"date":"2023-04-25 10:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-25 10:55:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-25 10:55:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"},"all":[{"value":1,"date":"2023-04-21 17:21:41.000000000","_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-21 15:26:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-04-21 16:36:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-21 16:49:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-21 17:21:41.000000000","updated_by":{"_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"},"reviewer":{"_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"},"state":"REVIEWER"},{"updated":"2023-04-25 10:54:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-25 10:55:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8ecc227cbeed48233f3afe3b17b5fb8fd6b924d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-21 15:26:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa9ba3f1915e776fa3a43dc63a8809b4137312c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 16:36:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c4a16de72df4989bf65e07e2f7c6a10\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bc135d916b7d4712af4ade7b3670b43c : SUCCESS in 18m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/59b4e7de28e3410b810d8d71cc1705d8 : SUCCESS in 13m 38s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aafce2ae3b8a49759e7bd452b09f644f : SUCCESS in 1h 09m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2842afee00214ad8a2d7b73a5dcd6b98 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/77e2a84eed334754b312e8481cf33f20 : SUCCESS in 2m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6ff44b063252414aa9cfb2e9ba0f2675 : SUCCESS in 1h 02m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/161618480ecd4ca7a3676d95470de439 : SUCCESS in 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da877d3375f0622d96504be1033bf34cebb3a7ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 16:49:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f98213528ccb454282de8d85dbaea10b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f1c490d9c80e4d11a5a76b548bbc5b59 : SUCCESS in 28m 09s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/addb031de90e4502be76264e5f72eb30 : SUCCESS in 59m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dab5591adcd243968f82c8bc8578fadc : SUCCESS in 23m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a6931d0c2f4141ca9ae27c31fb21c229 : SUCCESS in 1h 01m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6996c5e3126a4a539b6f2d3519b0ad61 : SUCCESS in 1h 15m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d29ba687fdb842e0a7e7384627cedfb2 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cd414df58ca427aa502ef0fafc9d290 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c51e197cd2e45cc84dbafb0357a10ff : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/902dbb3224cd42ccae61065b0f071d23 : SUCCESS in 2m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc427c1ed5a4ece9ebf5356724e7587 : SUCCESS in 5m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0440ac0e008b49f5baa0b924eb076e9a : SUCCESS in 21m 08s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/15abca2a8afd4e79813fdd8ed279decb : SUCCESS in 1h 03m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/abe21faeae0841b5ae1d2f14406707cd : SUCCESS in 19m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6531c6a9021f479593893a30b2981c6f : SUCCESS in 49m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/db32c81124404cb59547e788f0db3c59 : SUCCESS in 58m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e427cbb321fbac20513c2c61266f26657d158d92","author":{"_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"},"date":"2023-04-21 17:21:41.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"256e25cf06b6d7e8b9709664ff8546bce42224ee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-25 10:54:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"abbd64283d2c61582a2b49966962d0fd28e8f8df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-25 10:55:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"299dd9715254f6d95743433383b67ad5c69a3199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 10:56:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87cd5bebfbee3987490d98936394e3f1ea9226b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 12:15:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5c6b5fd6856436d958add53a36c6ca4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/17894606bd2b4aed809622e9d76b99c5 : SUCCESS in 23m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1a30011d016d4cfaa346c4a5613b3a87 : SUCCESS in 1h 03m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8745e664cb1d4aa290448f6da83fdd62 : SUCCESS in 1h 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74340fe3ed064d03ad77daa0c5806019 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cbee0e3daef41b1ac2945064983e730 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0528be93b4e14ddfb21f01d8794e10b3 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4222ef87d1844f6084ff1bfd0df7788b : SUCCESS in 4m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/928c9abef33343a190c96b357a6d2c10 : SUCCESS in 17m 57s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/bfe29a73badd4e11a2a85726673f3fe2 : SUCCESS in 52m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/10f421595efa41cfa9ef74926f414bf7 : SUCCESS in 19m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a002ee37b99d49569f67f6582b83d616 : SUCCESS in 56m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9dbd3eeac542443e9e2b0613b68eafd0 : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3f3754d379989ed5a9f422bdd1d3b5c525015cfd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 12:15:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"caa78bbcc0d7c8bbf09b8ec0b41268648899f6d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 12:16:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc52a81d2a64437f9bacac886f2a3dcd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5d3915037f4aa496ea0ed4c73ee0fb : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"003fff5dc1cc94bd8da0fd032e353ddafd0f233a","revisions":{"003fff5dc1cc94bd8da0fd032e353ddafd0f233a":{"kind":"REWORK","_number":1,"created":"2023-04-21 15:26:07.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/36/881236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/881236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/881236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/881236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/881236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/881236/1"}}},"commit":{"parents":[{"commit":"6d1efb092c654650eb2420058ea4d81e7ccc8e68","subject":"Merge \"Replace invisible unicode chars\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d1efb092c654650eb2420058ea4d81e7ccc8e68"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-21 15:23:37.000000000","tz":180},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/003fff5dc1cc94bd8da0fd032e353ddafd0f233a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/003fff5dc1cc94bd8da0fd032e353ddafd0f233a"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29925,"name":"Loan Harrouin","email":"loan.harrouin@gmail.com","username":"LoanH"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
