)]}'
{"id":"openstack%2Fkolla~881467","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-09 07:48:03.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-09 09:06:34.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-05-09 09:06:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7","subject":"https: remove dhparams.pem before running gencerts","status":"MERGED","created":"2023-04-25 10:58:47.000000000","updated":"2023-05-09 09:07:33.000000000","submitted":"2023-05-09 09:06:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881467","meta_rev_id":"0b80d69c7c3500301b6cb9807e517c35724f9bc0","_number":881467,"virtual_id_number":881467,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-09 09:06:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-09 07:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-05-08 14:16:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-05-09 07:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-25 10:58:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-04-25 10:59:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-25 12:18:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-25 12:48:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bd57fff52f66486a8446cc39126f33f45a2fa10a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-25 10:58:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c8b779ffad89f39655996f4bb63894a7c99f69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 12:18:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66771ec0a9e04935bb31a8cf8656272f\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ab80e48a64954a0899a4ef46bf747399 : SUCCESS in 36m 59s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d036ef73bb3348669f004a4eeb8bf519 : SUCCESS in 29m 43s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/4fffde7541f84761a04f2cbf393d5993 : SUCCESS in 42m 05s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/82fac6a154544b069dd6a9301219cb83 : SUCCESS in 1h 16m 25s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/98a127cfad0547aaa3c35a2c8389ada3 : SUCCESS in 1h 08m 35s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ca9ca10d2d354837b3a4b41314c06c0e : SUCCESS in 1h 07m 25s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/41d4acf716d349958501e720fac986cf : SUCCESS in 1h 15m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2f8b58cbee0c400c9b0a05999d9d051f : SUCCESS in 19m 36s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/32c2551a875747208171df5ccf5c59dd : SUCCESS in 18m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5b37e429b7d24a49b54101cc96be3c35 : SUCCESS in 57m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c9215efc0621483c9d7f1a397701f5a1 : SUCCESS in 1h 04m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94bfb7f6943e427fb83a3237c6053d9a : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf3e2f49b72f4f59b91d8a0438eaa28a : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c581baed38d447b8a749e5d286b14115 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3298487d334aafb30cbd22edcf4dd1 : SUCCESS in 4m 45s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/fc5b8dbcdb0c42d78982ad7e0cb7f8cc : SUCCESS in 23m 15s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/287decea3fc74c4e927424d64b5e98f1 : SUCCESS in 1h 12m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ed71c84b2bb843a0bd624c674b286d9d : SUCCESS in 19m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3373b2fe2074a61b432001be45eff7f : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":1},{"id":"33bdf7c7784bbbda8d17073d0a3c8cd6a1166e90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 12:28:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8824ff55e9c44a1c9037721392b651ba\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/235bdb4d58e44bd4b40cde85f0aed1c4 : SUCCESS in 23m 21s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7067fc7db53242db94fb4c5a4c554fb4 : SUCCESS in 23m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/876634f23d9d46f9af9941222b5bb352 : SUCCESS in 1h 24m 39s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5830bc86d71f4a6d8a46c1054dd9342a : SUCCESS in 1h 12m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3845163f96e848b2985d701c4235e377 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a488e45d9a794ce697073412fcf92ec3 : SUCCESS in 3m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8dcbb85036c14ae59307ad8daf7eef38 : SUCCESS in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fba7ee7decbd2ac60c04326ca9356d88a7340311","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-26 10:27:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"551a833d94720eeccb08187d0ff01114f68ca21c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-26 10:30:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"422b5a97668d30f688cc354c3da8208f4c3464b8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-26 10:35:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4d7be315fc940d096d330127001f61000e9cf51","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-27 08:39:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b81aae062a8747cd282c8dffc70c87b7025a5c35","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-27 08:42:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e9edb53f431be175e18fc8f209eff48f14d211a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 08:42:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47b3b15ac10517b968bfbbb9fefcc11cbcc3f142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 10:13:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ddc84fae3944136b569ce6b87a7fb0b\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/579474f3571642f9be40b5295b36c379 : SUCCESS in 38m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/49f0d87c562d438f89ba085722149507 : SUCCESS in 1h 25m 40s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/05db5f90f21041ea84610306f378336e : SUCCESS in 1h 24m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1443c86606f94fca8271cfaa3595018e : SUCCESS in 24m 06s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/00f43e78078f4a819db5092e3ddf4eb8 : SUCCESS in 1h 03m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c14b9e21a61943e89b2e20c567c89a03 : SUCCESS in 1h 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a846e12a9cc04a57808c7e1bca4b8987 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85cd0e85cd6c49158e15bc36d3adcb9b : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/706abcfe565d4cf4ac185e1a71657add : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e199f8c5b64cecab6abe3f3d2230a8 : SUCCESS in 5m 34s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/70aee6d1a67545b59520daa98dab22de : SUCCESS in 53m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7d21cb24ce74b92852c356ef3b1b7e0 : FAILURE in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"84bf14f8cb3f4476eaf85f65ae546d9682abe613","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 11:08:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7de25488cc22098a650842c36b6aa061bd4f0237","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 12:25:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24b41c7e75654bdba4ac994b7b60f04c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/17a00d7baa64468f840fdcc65a514e06 : SUCCESS in 35m 01s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c53e210d54544ad99e3f14d2b5178f07 : SUCCESS in 39m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d219d326f0b4b96b79a24a32c4a2d0c : SUCCESS in 56m 31s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/67524bd6f16e45b98b53df74fce551b4 : SUCCESS in 1h 15m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b3f00229c1c4900917318b0011ceb63 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7efb2ed851e84f7197d72985670a8da0 : SUCCESS in 4m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf433b4e36eb4ed38e7dd5786f150f22 : FAILURE in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45c27cb887373d111062f59835842b63688121f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 12:36:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8756d6fc8df840e2a6c32b7ccf44afb7\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a01836082fbb482b8aa521a449649f2d : SUCCESS in 31m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0fab22e08c6e4d6d97c2188b7bd89b21 : SUCCESS in 35m 45s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/2c5c64486b574114970c30eec71b1f6a : SUCCESS in 32m 55s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b4ed168bb87c4402aa7bc8a04ea803b7 : SUCCESS in 1h 13m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2e540e7c0b1042a48f4c215e89bd0c59 : SUCCESS in 59m 54s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/96a08fc4648e439b97238837e18616ad : SUCCESS in 1h 07m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/75984debdc884c73afcdca858765cba7 : SUCCESS in 1h 12m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e9dc7d6201b148ae880a31503366b549 : SUCCESS in 23m 11s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2b0be5f005504841876bd63ab718510f : SUCCESS in 21m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ae8c3791533c4a60ac6fd23eb9d17595 : SUCCESS in 1h 01m 00s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0436317243e846ffbef4f39c2119ea52 : SUCCESS in 1h 06m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4627a2d4d8a24abebfe7759c371f59b3 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e992a4c80d0b4beeb0c5f3f4b8df4646 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e7162926f6542e3984ca26b887938d7 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be5494beb234e3c9efd3a4979abcfe0 : SUCCESS in 5m 09s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/127979267d8540329d7b8bb4eaf4e39b : SUCCESS in 25m 45s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/eafc050c1c394928b60f51655f3c9c1c : SUCCESS in 1h 12m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1cb8cd1a3b7c4c329702aacb1065edef : FAILURE in 4m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c6024586eab4b4e8c635bad51f92eb9 : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"29fdf92c1f1a095a09b7d58769476e59ebce70fb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 12:37:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c1348ca43c3961f9f8ff3173e4f3067b21fa41f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 14:03:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a39d6d84142344519903bb84290b56ee\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d8a6e47f7acc481098ab3f1d757c269b : SUCCESS in 30m 18s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/38e78f2278174a79901adf86190397ac : SUCCESS in 38m 58s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/e7ac3415c4664273ab9da4128cff2afc : SUCCESS in 38m 39s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ee79be8f0f4c4e4a90b5132660dae4b5 : SUCCESS in 1h 15m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/03cc237500094f76a2a372216d470ca5 : SUCCESS in 1h 08m 36s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/e37103f0f1a4414da51853ed064f365c : SUCCESS in 1h 13m 18s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/76ff7a8653ff44de91f0f96f8f3bb789 : SUCCESS in 1h 21m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4125402416f349c48c95b1ba3500c61e : SUCCESS in 25m 37s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1f82ab9b7f7b405d9292d3e3832f322e : SUCCESS in 16m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/93aa0b503efe485d9975de8b135842ef : SUCCESS in 1h 01m 13s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/543a4b1cb9244d1da47ec995c8e006da : SUCCESS in 1h 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bbec99cb1374d569df1b970bdbb6485 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f36960f49a91448599e45b49efa325d0 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/766f6a51ef73496ca54ced69ce3c78a8 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ec2308323b48f982343fc06aa5006b : SUCCESS in 4m 54s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/90b71abe2b0b49e0a05bd0144b6994e6 : SUCCESS in 18m 56s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/d6298a19bcf8425ebdc436cb5c4fa969 : SUCCESS in 1h 03m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ee14ef77b2a14a1b95496de16980bf75 : FAILURE in 3m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/81b725bb3fe246e5afa79882a639cb3b : FAILURE in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"33247f88702eb53cd9f50401a15ac1b2da1dbd23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 14:24:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b01dfd476b24436cb8e0a6022bdef08b\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1a37693f7fa6495f8660d717116673dc : SUCCESS in 1h 45m 02s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8994e6890e94d5bb0953ab017302aad : SUCCESS in 30m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eead259da7094caf8ff2ce340d774e48 : SUCCESS in 26m 41s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ae1508dc871e45cfaa74f1c36e43f015 : SUCCESS in 1h 04m 21s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89136a3263194d6bac3188f63db07642 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/baa4f3b306ba4635b08680f52a34e9a2 : SUCCESS in 5m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5f276ef5ffce413895442ebf9e7d173e : FAILURE in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cbcd2449f7e59b6215f8888643c14c48eeb4fd7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 15:46:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc8a2c3417047675d84aca3725ae9a838e750e37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 17:06:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceb9eb3ebeea4328b2d23aaba67d0b04\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/785456dfc12641d8b12a1768806c37dc : SUCCESS in 26m 18s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/36b876a053304dc9be05f7d801644871 : SUCCESS in 1h 16m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c21275f6e7c54170a39136a20924407a : SUCCESS in 23m 44s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/71205cd8edd74d718bbd73d49b0194f9 : SUCCESS in 1h 16m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98c50c104e884811842e9578b4cf2b9f : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc0e4ee33d9543e5b8bd650f1ef3fc15 : SUCCESS in 7m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/217b5e13f04f4c1e931578de28aeb005 : FAILURE in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1987ad46c146508f07db0532ad1bcb6e078a2071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 17:46:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f77d5805e874789afcf0d7ebd5f0f22\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1d1f84fc1ade4b36aed1bb36cb138d99 : SUCCESS in 32m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9c918e0b807a450ab3c5257f02cbe0ec : SUCCESS in 32m 42s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/fea34dfa311747069ff51ba96f0392c9 : SUCCESS in 43m 49s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/25c285cd45884779877253c64c33c43b : SUCCESS in 1h 20m 38s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0f438042455840b1896ea67dbe3b6512 : SUCCESS in 1h 06m 17s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a1408a42899b4ddcbe35d0623ae171f7 : SUCCESS in 1h 09m 13s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/750871f3dc7d499c8149e655e7be1959 : SUCCESS in 1h 47m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bc73eb535ad84ebab99546cb997456f6 : SUCCESS in 24m 41s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1855f12181af4dfcb27a37e566d59f1e : SUCCESS in 15m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/43f59bc629c74910b014a415f536d586 : SUCCESS in 55m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9db4122179d64e988ea25b0bcb4aaaab : SUCCESS in 1h 03m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/733a19ab24234017a39b2f724d1ecb84 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b70160eb6d8d4f86bec4dd88b6528ca1 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/091e3d6b3e1f4af5a8be80a0cc74a8da : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52b33fad30904f11bfe534a4ee0a2c46 : SUCCESS in 9m 16s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/159d2103256344f185c8d370cfffad6b : SUCCESS in 29m 52s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f4c98db398e74c02a1a98250f780a219 : SUCCESS in 1h 04m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/06b52cce554c42f388f4784d9348e59a : FAILURE in 3m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5825fb3972545748d794398c1ee04a1 : FAILURE in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d53c4474fd2c95e8b527e065c3b24f487af7c71a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-28 08:18:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a8e17ff8408ce019c85e50b44da5dac0d7ec8d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 09:46:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c28cb80994a4478c9d3c53ce24ce9fda\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/77f8a0f14dfe4af4b015c84d26083d60 : SUCCESS in 37m 05s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef7332f0218b4679abc3fd9c402620f5 : SUCCESS in 33m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf6221d18c64483e89a026078cb671a6 : SUCCESS in 29m 17s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/47c47258100b41a2bdd668a2ae0118c4 : SUCCESS in 1h 27m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9782d75de3f460a9ba49ee9a29ffc62 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48404509c4d14b8e9d7db494d9cd6c0a : SUCCESS in 5m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7ff46d406fe404fb693c4d766ac363c : FAILURE in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d94a75d30997782e199511e7e56f232ec8d36b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 09:50:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fad7441b5aa49ccabe4f83a90da9148\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1f8bcd95bb32485dada6e946537b524e : SUCCESS in 29m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0fb3b0d2f5e7426e8e031d3f4d500759 : SUCCESS in 28m 06s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/3c63b3be242246bdb19772e66d585464 : SUCCESS in 30m 22s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/89eea6b1a07641478666c6e8d84ecdb7 : SUCCESS in 1h 09m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b8b545659ce547e8a4f6d0153412ad8c : SUCCESS in 1h 14m 12s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ba89744b6f154382874229e2b96e9dca : SUCCESS in 1h 25m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f2a10afad3e745fab5afba44ca37ba05 : SUCCESS in 1h 21m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4ae6aa75319448bd9ed5bc56cc40650c : SUCCESS in 26m 10s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ed0b074c02a340b8839c840f24026b4d : SUCCESS in 19m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cbe23dede0264bc888daf95753e7ff32 : SUCCESS in 59m 50s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/faeb94ced9314690a0d20bde9dde4dce : SUCCESS in 1h 19m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2399d4d72f147d2a973ddf2047bc378 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2320602d680b4b8b9d85aaa5a88810f8 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5873f61a5cd3417197791adfbcabcb34 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d60434fc9149998344f22ee5568fda : SUCCESS in 5m 24s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/f121b537c1744fa48f9f35bbff7f59ce : SUCCESS in 22m 18s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/e86bc36814524cc094361320c83bfd68 : SUCCESS in 1h 17m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/81b5bd04cdcb4ec8ad583c09bbf29cab : FAILURE in 4m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4bed7f2a787147eba68ef2f5c9394caa : FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"083ceafbb717f38d315fad8c8b3850f366a7631f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-28 11:21:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b1414691b9b915364af81e03c042e1487eacd84f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-28 11:29:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fb54200b2d41b803e6ed0fa4471387c96f086e9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 12:47:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd5e2ebb8d464fcb8ef0045dde34dba4\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b9cf9d5066c4ac1a5da3f19c969618a : SUCCESS in 33m 10s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/07d886f6a6a84aa79ac863243f9c41d9 : SUCCESS in 34m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/56030461e3e344a38f9c56a79af09402 : SUCCESS in 29m 14s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f25a8b9dd1714d0eb7deae8f2de3fb57 : SUCCESS in 1h 15m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/304f7ad445374bebb90355f42917267c : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb331736b2724cabb252d87d94863fec : SUCCESS in 3m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f4c881b5226420c935440cfdaa56c6f : SUCCESS in 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d52c48746e2a116bbb290482975ecb4c5f0b9303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 12:57:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ec0b78f57fa4a069c6bd9c83976e97b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8ce362e9ef504433847c959d344643e5 : SUCCESS in 45m 11s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f049e440a88f4b2db345bfa9741eda1a : SUCCESS in 48m 38s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/e794bbe7960b4fb694852791fbc793d8 : SUCCESS in 48m 21s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/812cde31bd3e44cb92f9e29c755403ff : SUCCESS in 1h 07m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/248639a73c8943c3b3156b08b4261d51 : SUCCESS in 1h 19m 50s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/531ba61630d54135830c8c30336394af : SUCCESS in 1h 26m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/69ebb371088b40068921b34ee4249034 : FAILURE in 20m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/acf3f29751a44437bff34942376d2bd8 : SUCCESS in 25m 34s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7efce202cad748b39399e49d0fbbaa07 : SUCCESS in 21m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/761f5d06fd1c4e57854eba5ac01c1b75 : SUCCESS in 1h 02m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/60b91a07bcf04c21b0b3d441c3206832 : FAILURE in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a50e2e6dc1c4980a4acdc13599507a8 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/446e7d8784684258af0df738bd11d2ad : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/379e7cb36ba44faab18dde23191f59cf : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd05f16ec43247c0bb33a33ee85bc939 : SUCCESS in 5m 54s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/29f19d318e40434e92d9272a826c9c72 : SUCCESS in 27m 58s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/bdab765029c743c9b9b6522432a0a2c7 : SUCCESS in 1h 11m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/98e44922362b4dffaf36c7d0da2be2ab : SUCCESS in 42m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/72981a6b72ee49aa94ce5889e1330801 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a5acf2430b2bb2452d20b330d7cff47db94702e7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-28 13:40:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6d74835573180104bafb95abe9457916238ec82b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-28 14:07:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b195cc38de7b5eb5b7699d50ffae0e6d5a582518","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-28 14:09:30.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"2b2a20eee91db9a1d2dd8f0f888497093dcf03ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 15:45:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/637ea6519bbe47f6b7fc1da8f8739cd3\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4310e9a46ac1487c8d53ac26e07d88fa : SUCCESS in 52m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/248b23fea98e436ebd6bbde978b35166 : SUCCESS in 42m 53s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/fe434e9dfedc4a0183a7598cd29d6b3b : SUCCESS in 46m 37s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d2f2042496e64aa2966b8472f2c400d8 : SUCCESS in 1h 26m 06s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d3fc557aad57420f877136d157d17ee2 : SUCCESS in 1h 07m 51s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/8fbcdc2585e14011b8fdc936d2eabb88 : SUCCESS in 1h 13m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c09a5553b21848b09a3daac801f75da6 : FAILURE in 21m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f52a7fb11eb940479d9ef0662a704ba7 : SUCCESS in 30m 56s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1e17f18ab89a4f558b5a3a2b4e514612 : SUCCESS in 20m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cc930ddc4838491694de61511338d019 : SUCCESS in 1h 03m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ebfcb3b473874d18b03789171ef513ed : FAILURE in 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cedbbace1e644cba1ea65d9b472fd96 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef7eee4e6c2d485aa636cf6af3580e4f : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d018db628154e9a927b97f51ba3e79c : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec35d554e6b644ad8f1509ad14462d1c : SUCCESS in 4m 57s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/cf629b647ac849e1bfd153c102d97187 : SUCCESS in 23m 58s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f3657f02ae0a4335b54ba25af82b0a06 : SUCCESS in 1h 04m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8fd0a618c45b4cb180692c252d598210 : SUCCESS in 34m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e44aeb989a9c42c48547e5072fcee296 : SUCCESS in 21m 16s","accounts_in_message":[],"_revision_number":5},{"id":"127e7046929460e71c1ffece051c0ccc66b52da6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 16:06:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd535ecbb5a5496a96f134d414134b61\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7014cff55a2f4bd9866ed3759022317a : SUCCESS in 1h 53m 19s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed427be1a2da4046968d0eebd3d88c75 : SUCCESS in 22m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6f12287cc5f343deb73fdd212367e1c4 : SUCCESS in 1h 31m 44s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b6a3c235eef4d819a1abba4bb3830b8 : SUCCESS in 1h 31m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/311b628e48d742f7b213ee75f3939b36 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67093cbc245b4472a3cde2b1a51ec26a : SUCCESS in 3m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fc29748faa56434ba208a2f66f5569af : SUCCESS in 1h 03m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"341d05762c6bfc1081f8543c980c63097f9431cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-03 09:23:07.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"672b28a59517e7491b5b950c3660751bd3e9895a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 10:37:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/202ddab49d844a80a2009c8bb729c1b3\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e2b566c58184f86bd131e5a88332571 : SUCCESS in 28m 43s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/71e7e2f1c1df46e19ba509b9e15e2e31 : SUCCESS in 28m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/40eda2680e2f4b5f8129187247f1266b : SUCCESS in 21m 12s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/988534a1777743acb802baa3a5053c5f : SUCCESS in 1h 12m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/883f763cc7594f26b6275d3658bcb326 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b14f91fc6fd64275a1fd4de1ae798c82 : SUCCESS in 6m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/acfab6fff24b45e18babaa665af216f9 : FAILURE in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0de75d597d98fe2cc5f016d0857450a9d8ac40e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 11:07:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4772786a737b4b3f81344d847bf8c42b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1cf022ad3b8e4a738902e8c4febb4284 : SUCCESS in 35m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/19aa09e5ece34c5c85029df3bc3b8dd7 : SUCCESS in 31m 43s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/ba465eef7daa475dadd17a8c6e8f2144 : SUCCESS in 42m 12s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d10198097d0a4ee0bcaec30b348393cd : SUCCESS in 1h 16m 27s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ce24c18a60764948a87f5f954bf4f8f9 : SUCCESS in 1h 20m 23s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f25a4bb7c9584557a46fd500c6a7d124 : SUCCESS in 1h 10m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0a30304c6d4f491d92f9c71cae9374d8 : SUCCESS in 1h 42m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8816f7c6448447d7b30a89c9491aacee : SUCCESS in 26m 50s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e6316557fa2b41428f806651c0818149 : SUCCESS in 24m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c0c37d1348714579bed0b2d5e664fbb7 : SUCCESS in 1h 15m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7631fb4c73af4329a556c702252b0d0d : SUCCESS in 1h 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a0aa35fd9374543a1ed5c354230158b : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a174cd065e6748c0bc0bc7b37e351989 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a85818e6818c4fc4a4d5c8f97ddea3ed : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64a420dca891406e9dda725f70728c51 : SUCCESS in 5m 34s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/b1b8d9985b044cdd92bd8825266d687a : SUCCESS in 24m 53s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f2b31d9aebbd46c388de06bb78405b19 : SUCCESS in 1h 02m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ad313153f9094e24b7d4da3869429fa2 : FAILURE in 3m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8807e510ba94b7f8c9a08151c374e33 : FAILURE in 3m 49s","accounts_in_message":[],"_revision_number":6},{"id":"8057805c0e51467f2fa5b9acd4f01b72fc8f260c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-03 13:27:12.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"303882fb184f374d3ea4b2108286728869eb9a65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 14:53:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1121c018fbbb4c6d8a036ee896d2d6f5\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fd15091a26424b0794285df98a0737da : SUCCESS in 33m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5a4955262eb541e8b61b952097987a62 : SUCCESS in 35m 29s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/150d971eab544786a7ac2edb1cfe5226 : SUCCESS in 42m 53s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ffb135431598430a9defa15468afd656 : SUCCESS in 1h 08m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/71f2846767b34426be55cb129233dc0c : SUCCESS in 1h 11m 55s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b3bc3ee12ecc4db7b132c77d4b380e70 : SUCCESS in 1h 21m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4c11abaf416a420289e9efd7fcf40298 : SUCCESS in 1h 22m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9838cc4b94bc408db2bd08db1fd9490d : SUCCESS in 26m 43s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9315a2359a264415a56781fb350721fe : SUCCESS in 17m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e28e9aadd55f4d1f8a3502d7e5e7785d : SUCCESS in 58m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9db4936ee7f245a7b638dd53c18a6763 : SUCCESS in 1h 16m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/520b95e0e9fb4fbc86b3cd8d90d9dd46 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3bc5f4c605343b9a3cc418ec1a00b60 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d538185f3b554631bba47d1f6ffe9b08 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a6d59817b72461e8ab2ad52e86c28a9 : SUCCESS in 5m 37s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/e10b141ac6c84c11ad39d344c9e2ccea : SUCCESS in 23m 36s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/4189098a66124ccc88c3177e443dfcd2 : SUCCESS in 1h 00m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/45629c9521864a1a916a1f2629768106 : SUCCESS in 18m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/01f3bf1524d44e29980fdc8fe1062832 : SUCCESS in 28m 26s","accounts_in_message":[],"_revision_number":7},{"id":"d8ebcb0df7311de08cae9d575af5ebe03bc9c749","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 15:05:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50390e31849a4f9baa31a59fac250733\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1cb632770b84f7ebf5e8a3ece1f9349 : SUCCESS in 28m 02s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b1cb405be8df48aa84e7e95505d1ca73 : SUCCESS in 31m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b17ddd63af24c24bfccfaf42d741fcf : SUCCESS in 1h 34m 43s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/37d8843da34f408ca10727a314b454b6 : SUCCESS in 1h 05m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36c63632023241edaf52c1ee36757e81 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02e6cdfb94534636afead4ce185ea34b : SUCCESS in 3m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a6f4ae8b4544c35806a5cc439af37b3 : SUCCESS in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"105f1f9fd09feada69cc0ea218cdd29b48d64db3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-03 15:27:04.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"d6a1f242e7238c6f9aa9c8f5ea2a51596c4981a8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-03 18:01:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a74bc90df2b2cf928972f4c492b2958d998d93d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 10:17:26.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":8},{"id":"25689e949bb870b285c66fc00b19ccac5a4d7d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 11:49:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d45b814a2944adbaa56c454ed412cf9\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0abcc032ffb2401299fd6e8e73d6aaec : SUCCESS in 29m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/84540651fe2f4d5e81b91e05032a18a2 : SUCCESS in 32m 17s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/53292751ef284c79a1710d93ed5afbb2 : SUCCESS in 36m 50s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6e8fc280fb074c16986d5bdd34c89e2b : SUCCESS in 1h 11m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/91ec04d54b0e440cae1ee05091ba0ef1 : SUCCESS in 1h 10m 00s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/c8030bec1272444798b7f679cde21ce2 : SUCCESS in 1h 13m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ead4adde938541d0b1167324e601f1fd : SUCCESS in 1h 30m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/310b13a0d31f472baf3ba20c43a37f35 : SUCCESS in 23m 18s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5322d0de782d4cdba0108301c0971d99 : SUCCESS in 21m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/886ed39ece4c4f10b0cb6a9c4356326a : SUCCESS in 1h 04m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/735319bb388242809e3b5408dbeaff4c : SUCCESS in 1h 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e679a04d6742c98b3b581ee2ac1992 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74aa1cf535394c4db672b7b49eee0736 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfad27eb9f334fc586bf6b3ba0e3390b : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c4f548d12646bc94d4004ad6855167 : SUCCESS in 6m 36s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/4fb62e540682484289f926eea0034dc9 : SUCCESS in 24m 58s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/97e35ecb155e4a619e7f88930d2058d7 : SUCCESS in 1h 07m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2e5a7216df134a359286255fe05844a6 : SUCCESS in 36m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4b42d948d8d84d24ac04689941860e92 : SUCCESS in 20m 40s","accounts_in_message":[],"_revision_number":8},{"id":"2253ff3d1a499a93c267d9fccc8b1e7f4dc2c2a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 12:57:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5232bf8e441d4d06859bf87b2b0dee6f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a4bee4a181184b55891402cb214924b4 : SUCCESS in 42m 54s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2fda584b910e4bbaa711f51cbcc45583 : SUCCESS in 40m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c40c12d2b6344fe8c4d8ebc8c93d82b : SUCCESS in 37m 48s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0efee93dbd704c108fbc5bdc5193966c : SUCCESS in 2h 37m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/84105d56211c4f0383974f96c843b161 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75630daf591d4f6090233df0c1d39be7 : SUCCESS in 5m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/be57f05c3c2c4689a0c8387f73a4b98c : SUCCESS in 34m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"124f7240883cfdd60ad97a306819388258c92340","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 13:37:14.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8d79292dba43f7d056fdda591e2bde448b4f4be6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 19:36:36.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":9},{"id":"9d56809adf292e7a1b4783af0ad7f90bf635b923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 21:06:00.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/21cf9c4da52d488ab4f40cf293d46316\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/87b2df08308e46a9b9937e7e588a2992 : SUCCESS in 29m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4cfeb9aadb244aeea31cda0956325751 : SUCCESS in 28m 29s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/26f8022b5dcf4beb8a4e7302292f9a74 : FAILURE in 12m 48s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4cd6868b84c44c04b4883b396fcf3cf0 : SUCCESS in 1h 19m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/29a993f59ab94cd49b14a8d7b8b70fd1 : SUCCESS in 1h 06m 07s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/70f35bf7924e4acbbb72d99d90d5f995 : FAILURE in 17m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1e97ecfc350a44c9b091f916b8035de1 : SUCCESS in 1h 28m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6c3232e917404e8991bc6350e85915fd : FAILURE in 3m 49s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/69b4c339389c4411a6db187b4fa72f81 : FAILURE in 4m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/948f496c89d34fe69ef6ae96863a2341 : FAILURE in 16m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ca5b0ec9751a481aa2a1a2d353d64ba7 : FAILURE in 17m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/415d706b2253482cb3feacedcd5c8946 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bf37fed59504a9597a5f5b84a2f07eb : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/264ed97ed3ea4081b1dc8fdc8e75d48b : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e68fc60ac14c43a5ecbb3176d50713 : SUCCESS in 6m 27s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/523343481f654abbbdb66534dfc4d3f6 : FAILURE in 4m 47s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1e6a1dfebcec4a7e8cd175027c2c423f : FAILURE in 10m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/abfe9e5e2f46495ca0149126a343d8e7 : FAILURE in 3m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4f1a2ed03f24e94a9f604ce1b24428b : FAILURE in 4m 24s","accounts_in_message":[],"_revision_number":9},{"id":"bd8b784bd22a4d66064cb45c348d0fbdcaeec05f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 21:06:43.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aca250ae669a4cb9a82f16bd8f5c24d7\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/44482a9aa57f40f39e3c3dbd01b7df10 : SUCCESS in 1h 27m 32s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/041660862d624abc9dcda267b5b8551f : FAILURE in 20m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d0d97dac086e4a03aec68e714c944a9f : FAILURE in 4m 58s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a77962a8c52d46a5a56026763226c20c : FAILURE in 18m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f19681297a544f3801fab34c2a4791a : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22e7134f4a554c518e2c168dd3bd2ff7 : SUCCESS in 5m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9533296b50b44b479cd524dd78511964 : FAILURE in 3m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"36d13c89643ec706ccd6c8c5ee222e1af01e9b22","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 21:17:07.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8af5ce4b0e156e980162113aec5733d0824b16c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 21:41:34.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ccc6eedde114e3c811c0343844f9434\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/83d6b255f5fe4a3a82c7a5b69103c69b : SUCCESS in 23m 16s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e45d2d49d29d4931b07aed3deda6d7e5 : FAILURE in 6m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a9f821cd6c844449b81dacf22ba44fb : FAILURE in 2m 59s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/73d4ddb31ddc46a79a2a0416bbb6a56f : FAILURE in 12m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0898c2e687544f45840afa1478ecaafb : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba21048ab2fe4a0c93dbd71a07b4ca7b : SUCCESS in 5m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6ef212b53dc9486cbfc30c8d977400ba : FAILURE in 3m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"685cf01da17322043de732ff0dc1abfc4d82a0af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 22:56:05.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47f938218435442280da947b466478c5\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/82f7506250dd41c9ae3e5cfd2a353bfe : SUCCESS in 36m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b28e448674a446e8be5de3dba0f5f57d : SUCCESS in 42m 04s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/144b53059e474a5d81c8973532dc2997 : FAILURE in 12m 25s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fa9515b6b91140e99dfe8fa41787cf6f : SUCCESS in 1h 14m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/34f9a4d9f09841968ed39b1ea001ff82 : SUCCESS in 1h 04m 50s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/3fa14b4023de4f1986303a7bd9025638 : FAILURE in 20m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1848a5a72b7c4bc4a862d35906549acc : SUCCESS in 1h 32m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0fa740d8b1dd4604930c6df204409074 : FAILURE in 4m 38s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b8406d40297b462e8d5da42cb7d8d38f : FAILURE in 3m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3ba9ffe31a464ab0a0a020511b43a99b : FAILURE in 13m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/61e71b8f82514e6792dc121a2020a324 : FAILURE in 13m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55fb89fe3f964e4f96e330f31f0fd84f : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79639ab89330449b9c464b11b11dbbe3 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d07c2e55655409386e34f3eebc7cf89 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed44679d236547fb80458c7afdf01170 : SUCCESS in 6m 11s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/c89406799d1343e79b06fdf32cd8b2b5 : FAILURE in 5m 08s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/698f369cd7c744f08fd0f420aa932ad3 : FAILURE in 10m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dc759dfd0a6f49b5b007c8c7531ead6c : FAILURE in 4m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/589d65f6b5e14c84aeaf458b4ad7d46c : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":9},{"id":"416a11ca4b74ddb8f5f1575c4280684706c2efea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-05 12:02:22.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"23ebcd6ec6d9cca5fcc1ab257e5b653f03704ebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-05 12:23:35.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":11},{"id":"415595e0861a40765a8de49a0590caee4d4389be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 13:54:14.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e777ac8b60064c73add237043b5b64ff\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/37b01b0e3f77474dadb83c5e833e78c9 : SUCCESS in 31m 08s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4c02e095ec1d4ff98d97d31b7ee81bcf : SUCCESS in 43m 13s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/4a0d23d84b6547c9a12d387487dc739a : FAILURE in 12m 06s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8af76428a7454c03ad6ee8f7a2dcd847 : SUCCESS in 1h 10m 29s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/58a6b6bdf7e24ca89223d38dab27847a : SUCCESS in 1h 11m 26s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/96744dd3b05749938460f98ff836f1ec : FAILURE in 17m 46s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/26f989840883476aa5af8bc050beab70 : SUCCESS in 1h 24m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3e71aa7cc1f14b7da22778b76995b9cf : FAILURE in 4m 09s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b543ee72c5f54011bacb92fab4e054ac : FAILURE in 4m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1b10fa9b58b449d4954fd82f5399786e : FAILURE in 18m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2da5332eb7024d17974169170d49b080 : FAILURE in 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcaf4b6436b94f5bbae58f6de0a18168 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b97ef371d5324759bbe1e5c3d5fec7b2 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef3fa98f30f84a3b903549ddea1fcb31 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90603892bd054b5e979d544413a6dc5f : SUCCESS in 5m 13s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/fefe5a1397504a689278477c72d378a5 : FAILURE in 4m 43s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/201d5386019c45cfa0137b42a03bb369 : FAILURE in 11m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2725783742c84148b8a65e3e140e5d9a : FAILURE in 3m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c5b5f161ddd44dc96f791aeb143caf4 : FAILURE in 3m 32s","accounts_in_message":[],"_revision_number":11},{"id":"624b1c617c35821a2681b8eb447ec08fb6438e56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 13:55:04.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/301a63771c424fd4897edcd291fad8b6\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7afb1361f4124473b6adecf08fb5d899 : SUCCESS in 1h 29m 01s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/726308617fd3416abdfb2b9d55297370 : FAILURE in 7m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/89b3f60f2ceb4ada802ec2beedc56cce : FAILURE in 5m 01s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/985a70723e5c48cd9ecd1cf8a21a8efe : FAILURE in 16m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb79344fdf174a36bb0b9f79c395b696 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66abf2ef54d645afb1d086907e2b0a1c : SUCCESS in 6m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/76022f65e6ab433fbd7a5b8509ac232f : FAILURE in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8d59e489c54d8b9632fb6f4213e71a8ffbccc1ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-05 16:43:45.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"e205e8bd745bdb7ba86803c79a8952b4319770bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 18:08:08.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02853aea66f9497c83097c44e660db0e\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/84248fcfd94142a0aacb1bd6b1d53024 : SUCCESS in 33m 22s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fc75ccd36bb5456980667f9c6ee722cf : SUCCESS in 34m 44s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/d24f57f0d65842c484d6b6ba607648d1 : SUCCESS in 36m 33s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/481bb6e0aaea40e1b2b2ded980172006 : SUCCESS in 1h 07m 35s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4a7d51b562244f06a15486a73a8ac42b : SUCCESS in 1h 08m 31s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/fea71e775acf4a7db8d77995c92e4cf1 : SUCCESS in 1h 14m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c6e5f2e5e32e424a949b0c7e43ee1637 : SUCCESS in 1h 19m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1d22960f4b724c1298f9181f55318242 : SUCCESS in 26m 57s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5ed337f5d0024580a19e7ff8a506e62b : SUCCESS in 19m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6f3fc97e9ba24ccca13b3110da50d468 : SUCCESS in 1h 05m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/77bb6ace18904648bce878ae1de698e7 : FAILURE in 38m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4962fd40b3f4a789b2e57d461e62098 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c4b95355cd74eddaba299bd70cf5a24 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9604e182694943e2a2254dc8c07b40f8 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a83bcfcfc24af9a4b73ea5be411b6a : SUCCESS in 6m 05s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/958e0796540a441fb33db4bd3f5416aa : SUCCESS in 23m 56s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a34c6362320445c1abd813c0b6522fa5 : SUCCESS in 1h 12m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f0944242e3474810bc78e38bb997b7db : SUCCESS in 23m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c41c8140d69498eb7f643943d1dd889 : SUCCESS in 26m 15s","accounts_in_message":[],"_revision_number":12},{"id":"8bf3b4d7e921411c750135b35f5f1172ebd4c0a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 18:22:36.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a11d61e55bb14a26bbf3a81a616246c8\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c703fa89b9e41c3a130f9d888b0c9ff : SUCCESS in 21m 44s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7524df9707fe457aacf8889511f1623c : SUCCESS in 1h 35m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e55660c7ca3d4f0bb251b34cb5cfde07 : SUCCESS in 1h 19m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aa51ccbb9ac44211900441f6b2041ee4 : SUCCESS in 57m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcaf6a830b2c49f9909b3de801a15ab9 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d983036522c049c09467a44501d4077c : SUCCESS in 3m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0807355ff3214d62bd883cfba078b1f4 : SUCCESS in 1h 09m 01s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f7ae97a285359847d8298f01252349383a28e50d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-08 14:16:43.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"0a751d4460fc01141caa594be124831e17049ea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 15:42:06.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c87c90110004f56be13053659f0aa51\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/eef906d373ab46fdb1fe27b7ec75cc3a : SUCCESS in 38m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1591cac39ce94441bfadc67d2f98732e : SUCCESS in 44m 20s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/392ee86aaf70496694d52a1fd92d297b : SUCCESS in 46m 10s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ca371b59c164466b8cd91c845296b0ab : SUCCESS in 1h 16m 29s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9c1d67656d744daf87bf419d6a652552 : SUCCESS in 1h 08m 57s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a47500dcca3442628b4dc8b47fd3490d : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d194294979774285bc54911ff27b8ad5 : SUCCESS in 1h 24m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/86a5cf89cb5a44e9a193e3b83588ce6a : SUCCESS in 23m 11s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/aaef10f8f5684e27ade030389c93b9aa : SUCCESS in 15m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6b43e95d056f4a5f8b043b2e3ced9b56 : SUCCESS in 58m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ee492961a6484b7c929328a28d7bbdf1 : FAILURE in 25m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6dfde26e25a48098d7985a784c75cc3 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3814a0107624a438278159f90ec06b5 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8066bed159c41739e7f79a318d1a1ef : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2822b11ff11c401fbf64ba78d7af1e31 : SUCCESS in 5m 13s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/48bf0d602e85473cb3ee963c7d60ded2 : SUCCESS in 27m 58s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ff663309446343408519428ae6708a23 : SUCCESS in 1h 00m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b29db57a71a437882aaa0641e9c348d : SUCCESS in 21m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0705bc163c048ec8901ff488a928abe : SUCCESS in 21m 20s","accounts_in_message":[],"_revision_number":13},{"id":"9f6a8341e1dc8805f02e912ac37023d50ed99f61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 15:54:26.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01c9e74252d64fe18ab325581a9da6fc\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e28c940932074ed0b1c7dd25c9d8601e : SUCCESS in 1h 36m 47s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e1fcdf411ee41fa96e4cd5a5dec3704 : NODE_FAILURE Node request 300-0021164587 failed in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ce13de94c7142b3b5bac91534600504 : NODE_FAILURE Node request 300-0021164588 failed in 0s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f7c87ffc601344f08a6908540a70b7c1 : NODE_FAILURE Node request 300-0021164589 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a779b68186144400b28c335ab16e3cb3 : NODE_FAILURE Node request 300-0021164590 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00045ac84cad445b909e508b009c78d0 : NODE_FAILURE Node request 300-0021164591 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/09819105e6034578aa8254d950e3b1ba : NODE_FAILURE Node request 300-0021164592 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"5912f671efbcf17b6378b3e1c6e708e8ddb2090b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-08 17:29:06.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5e1b79a358dbbc4e62cbbf9dd9edf7652479e2af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 18:45:04.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f3dfc9adbd54afe9f3028947dde6710\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f8b6ad96e424dbda6058c4c00c2b421 : SUCCESS in 30m 04s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ff3375a775804af4b7a39a187dfecbb3 : SUCCESS in 25m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/785a5b91e5f54191865c1fa3fb4de8d4 : SUCCESS in 17m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/52770b5470f24a2bad48eee237eeb96e : SUCCESS in 1h 14m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4782e3557ef48328d4a4b585d4514e2 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d9720f4e4c543e79d5606ceffdf530c : SUCCESS in 3m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9249408e89914fbfaa31bbbcbbb7f083 : SUCCESS in 53m 55s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"878d24770d704f7d9cd8c54ace4beece4dd319aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 18:46:36.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40db34c813f149bf9066d4b448850b01\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3503e0306d02499cbedb46acf3ca31d0 : SUCCESS in 30m 49s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4290c685c8024651927afe76586b96b5 : SUCCESS in 41m 29s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/4b30382609dc4aabab164ab3f2587ebb : SUCCESS in 41m 50s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7c35094af5db4286b6d2f90b0e8665e1 : SUCCESS in 1h 08m 28s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2f6b7e2ac1ac4cd0bb773ede2281eeb0 : SUCCESS in 1h 12m 40s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/d1f766130c5548df882c86fdaa50b482 : SUCCESS in 1h 04m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e4aa1816357045ed8f1fd9ba9f8e218d : SUCCESS in 1h 08m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/63c201b50cc34e2390be136f7b3b6443 : SUCCESS in 23m 05s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/be3a3401140e446fb4a360918c3dadbb : SUCCESS in 15m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/464fc99e39fe4565a7386ad8adabbbbe : SUCCESS in 54m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6b4b4db19d724848a2f1ea2435c807f6 : SUCCESS in 53m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6d59136ee604fa9ad4f7b5f0b8d0bfd : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0226d417e8f54798aebf2160626f0a6d : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e548f717cdd433daf129c7287fbdfe9 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/609b2e4fe32f4f5db04623c52fe02a64 : SUCCESS in 5m 13s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/84a0e6a728f64ce08692fe8294158230 : SUCCESS in 27m 08s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/e8b280ef67fa4bf19005981b0e595f71 : SUCCESS in 1h 07m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f5b76dca8684a0992920e048c680200 : SUCCESS in 24m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/be807733016847428cc05ac003f419e5 : SUCCESS in 28m 28s","accounts_in_message":[],"_revision_number":13},{"id":"b34d02ac7f3aa302d8dba45602204dc518e64d9e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-09 07:48:03.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"5ccd6e0e33e2806e8af974d63710d4d0e961f535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 07:48:25.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"81da6a8c92cc63907076d818d216ceceddb1f056","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 09:06:34.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3ad4abeba0a4402be2c0386d8b90b76\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e9c5665ccfff4b878c9ec19678e16b3d : SUCCESS in 29m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9c3a3473e72d4c3c9a60ec7bddaccaf2 : SUCCESS in 1h 10m 55s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1efe114222154f55a94615a5c0e6d7ea : SUCCESS in 1h 11m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/118ba083047c4d27a55d5c1a45b6501b : SUCCESS in 23m 34s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/48ca613169ed4979bdaaafad928c9121 : SUCCESS in 52m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d74a19aefbb646fda5f4fa9521a66cf0 : SUCCESS in 1h 07m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2a52b07648148ef9c8f54568f86b8a4 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8800662c01814732a37b8ef73d909aad : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d20aa1dec7a40d680ab78468cdc5d96 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d2709440c5f414f832da830631ca44b : SUCCESS in 5m 05s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/322e45c107244e3fa95ef4d812e28ac5 : SUCCESS in 1h 10m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/676d91f12c244a3f86bbd0c0d32b5e58 : SUCCESS in 23m 11s","accounts_in_message":[],"_revision_number":13},{"id":"abcb9ce82aaea5d051a3727cfa1c56452de1116f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 09:06:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"0b80d69c7c3500301b6cb9807e517c35724f9bc0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 09:07:33.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41c88388aa5741ebb0e3a802ac196ef4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08456c86017c4e9c8d4037acced80264 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"e590ec353841db16a96ad68f7ad5a13bbf7ee168","revisions":{"6b6ffb02820e86be2c5fc827ecf57fb2b59a6490":{"kind":"REWORK","_number":1,"created":"2023-04-25 10:58:47.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/1"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-25 10:58:21.000000000","tz":180},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b6ffb02820e86be2c5fc827ecf57fb2b59a6490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b6ffb02820e86be2c5fc827ecf57fb2b59a6490"}]},"branch":"refs/heads/stable/yoga"},"7fefe07d86e62d2361a0b55f9ba07d6f49e62cbc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-28 11:21:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/2"}}},"commit":{"parents":[{"commit":"4895e6d8e19284b821e64e133368f08c3c03a386","subject":"[DNM] urllib3 test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4895e6d8e19284b821e64e133368f08c3c03a386"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-28 11:21:27.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fefe07d86e62d2361a0b55f9ba07d6f49e62cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fefe07d86e62d2361a0b55f9ba07d6f49e62cbc"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"4b605f2417d901761985c5c95a01e605f1697189":{"kind":"NO_CHANGE","_number":3,"created":"2023-04-28 11:29:26.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/3"}}},"commit":{"parents":[{"commit":"1b4ed130bc145497bf80bf656036590fffe6e658","subject":"Pin incompatible urllib3 2.0.0 released on Apr 26","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b4ed130bc145497bf80bf656036590fffe6e658"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-28 11:29:26.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b605f2417d901761985c5c95a01e605f1697189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b605f2417d901761985c5c95a01e605f1697189"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"f63745a0c0b8f332689beddcbc14bfa66134dc94":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-28 14:07:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/4"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-28 14:07:27.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f63745a0c0b8f332689beddcbc14bfa66134dc94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f63745a0c0b8f332689beddcbc14bfa66134dc94"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"90ed51648d64c5727a4c7afd18fd38c239038480":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-04-28 14:09:30.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/5"}}},"commit":{"parents":[{"commit":"1b4ed130bc145497bf80bf656036590fffe6e658","subject":"Pin incompatible urllib3 2.0.0 released on Apr 26","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b4ed130bc145497bf80bf656036590fffe6e658"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-28 14:09:30.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90ed51648d64c5727a4c7afd18fd38c239038480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90ed51648d64c5727a4c7afd18fd38c239038480"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"ff7247b4267c1309c6f3d82e6b4280845eaffae3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-05-03 09:23:07.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/6"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-03 09:23:07.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff7247b4267c1309c6f3d82e6b4280845eaffae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff7247b4267c1309c6f3d82e6b4280845eaffae3"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"ce6afb752ba128e3652561d2148a5b93598a87ab":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-05-03 13:27:12.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/7"}}},"commit":{"parents":[{"commit":"5e546567825ceba9fff96c24cecac3253253c895","subject":"Upgrade pip/setuptools/wheel for tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e546567825ceba9fff96c24cecac3253253c895"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-03 13:27:12.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce6afb752ba128e3652561d2148a5b93598a87ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce6afb752ba128e3652561d2148a5b93598a87ab"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"4eeca30be491f867ddf0a8b1652f17273ec47c25":{"kind":"NO_CHANGE","_number":8,"created":"2023-05-04 10:17:26.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/8"}}},"commit":{"parents":[{"commit":"605e331176a34f1ef480d9cf842e8e1591e869a6","subject":"Upgrade pip/setuptools/wheel for tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/605e331176a34f1ef480d9cf842e8e1591e869a6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-04 10:17:26.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4eeca30be491f867ddf0a8b1652f17273ec47c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4eeca30be491f867ddf0a8b1652f17273ec47c25"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"bd89a6df750527dc617a903c6ff7d92d4cbcb564":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-05-04 19:36:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/9"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-04 19:36:36.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd89a6df750527dc617a903c6ff7d92d4cbcb564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd89a6df750527dc617a903c6ff7d92d4cbcb564"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"ad867a8674bbd69946f872f6a9ee4119c9380cbc":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-05-05 12:02:22.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/10"}}},"commit":{"parents":[{"commit":"e802c772c90da39d5693ecc4c0d72c218dcf0b7b","subject":"CI: Work around docker-py requests/urllib3 issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e802c772c90da39d5693ecc4c0d72c218dcf0b7b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-05 12:02:22.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad867a8674bbd69946f872f6a9ee4119c9380cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad867a8674bbd69946f872f6a9ee4119c9380cbc"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"ae9f3473cf6419667a89f2deda731e2e998a6ca5":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-05-05 12:23:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/11"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-05 12:23:35.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae9f3473cf6419667a89f2deda731e2e998a6ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae9f3473cf6419667a89f2deda731e2e998a6ca5"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"980caea88b23dce3f73ee82b7137be6a59c09c2b":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2023-05-05 16:43:45.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/12"}}},"commit":{"parents":[{"commit":"d90cd97668272240fdaf5bf2f32125950a6135b8","subject":"Work around docker-py requests/urllib3 issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d90cd97668272240fdaf5bf2f32125950a6135b8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-05 16:43:45.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/980caea88b23dce3f73ee82b7137be6a59c09c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/980caea88b23dce3f73ee82b7137be6a59c09c2b"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"e590ec353841db16a96ad68f7ad5a13bbf7ee168":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-05-08 14:16:43.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/881467/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881467/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881467/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881467/13"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 17:27:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-08 14:16:43.000000000","tz":0},"subject":"https: remove dhparams.pem before running gencerts","message":"https: remove dhparams.pem before running gencerts\n\nOn EL9 the file exists and script errors with wrong\npermissions to it.\n\nCloses-Bug: #2011716\nChange-Id: Ib32baa3208e6bfc5520ff8537193c9a4d6cbada7\n(cherry picked from commit 12d431e3994023ce735a2352554a99a5cc9431c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e590ec353841db16a96ad68f7ad5a13bbf7ee168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e590ec353841db16a96ad68f7ad5a13bbf7ee168"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
