)]}'
{"id":"openstack%2Fkolla~881567","triplet_id":"openstack%2Fkolla~master~I9868af5d00977750d0673fc48f8863a99f878dc8","project":"openstack/kolla","branch":"master","topic":"wallaby-summary-json","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-27 12:35:26.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-04-27 12:29:19.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-12-21 15:45:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9868af5d00977750d0673fc48f8863a99f878dc8","subject":"Summary JSON file","status":"MERGED","created":"2023-04-26 14:10:42.000000000","updated":"2023-12-21 15:46:23.000000000","submitted":"2023-12-21 15:45:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881567-wallaby-summary-json","meta_rev_id":"12bc5fdb5190daf1693962c02bd2b7598788ab89","_number":881567,"virtual_id_number":881567,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-21 15:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-11-22 06:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-27 12:35:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-04-27 12:29:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-11-29 06:07:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-21 14:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-21 14:02:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-12-21 06:46:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 14:21:24.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-04-26 15:50:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-26 16:37:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-27 09:32:00.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-11-22 06:47:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-29 06:07:46.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-12-21 14:02:31.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"a842b39e5bb2410495394ccc5b8c38ac40c96cce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-26 14:10:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eef57533e1ccb4dd50e22d4390d90c5b4d43514c","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-04-26 14:21:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a56f661bee46182eb55b0816e0e969c185c260d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 15:50:34.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4558ab2f790244ecbb3e1335fc9c5c8d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/226f632fbc994a8db3522112328d595b : SUCCESS in 18m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f925553475fa4986abf5361bf948a9e4 : SUCCESS in 1h 15m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7028df0bf24b4ac1ae7594929259f487 : SUCCESS in 15m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1563faf6b00142cd8ad5292f5cbfd437 : SUCCESS in 58m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ec2ce23c89e9404d8ee6405a410d1b5c : SUCCESS in 1h 05m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3083aca5fd46098cdbbe166edef6d1 : FAILURE in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b7262a96bee4851bf0be665ddad0dbb : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1211e591bc14d7597a74edd8ba4b2be : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/259941cf730a4d4f9c877dd20c7d9368 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7406c8104fa04ebdaab3bd1053c00b4c : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee23f38d578843829b643ff9f6f6b4eb : SUCCESS in 3m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5926644ab27c4a2897740c31cad0f149 : SUCCESS in 18m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/daad64de5d8e46c8838fe709fa0256ad : SUCCESS in 1h 16m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ba9b6c0f341c47a7a242ab2474a58e59 : SUCCESS in 1h 32m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/727bb3adfadb4932bf091c564e2ef4a2 : SUCCESS in 20m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8a5fcf8bc930411d95d84fa69f95e55f : SUCCESS in 54m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/687aad87715b4e99a6760b0e6220faf4 : SUCCESS in 1h 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"838bb822c0767b84e490734d1d84ce565a8eca1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 15:54:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d18e777fc4a4fb583066efe516461b3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a26a7789bb904f05a6a5ccb2d86a705a : SUCCESS in 36m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c5afa12134054feeb304f4c1cd7a7ec1 : SUCCESS in 35m 11s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/abea2ff7eb9040c3ac85ee18fadd2d79 : SUCCESS in 1h 40m 48s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdba9f4d355d4ccb877165e1d22fb874 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10cdc78cb5644d1ea0f60d8cb79a02b7 : SUCCESS in 4m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8328e767ed434ff7bc791f3cfe9baa62 : SUCCESS in 37m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bbc3cf7521fe4d5b806bf16d66cafa2a : SUCCESS in 1h 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5dbaab2870d25be3b9cfb14915306e9801bf252c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-27 08:19:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f36c2e79252899b2f35607838eb41efdd38c5699","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-27 08:19:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd7d2e38cff468c08f1dd8f26e6ded812ee95dcf","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-04-27 08:58:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e4b6e38a307af2d0209371362b311ee1f780be8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-27 09:29:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"41ce7a851ebe659ab0b5ab6c4efdcba55b51cf24","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-04-27 09:32:00.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"36a4195041f20044d885f2d4bf03a319efdec709","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 10:21:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6cd454cd8475b59dc6a11d635d26d75d0f0d6fe3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 10:56:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d35686742c1c4398965d7602a574c48c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ebc16557dcfb4306bb8e225b2a4c677f : SUCCESS in 20m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b2a43c5046b24a7dac908dfd0e2599a0 : SUCCESS in 17m 11s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ac1d88c22e74369a8e39f4a9e791b56 : SUCCESS in 1h 21m 44s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4215e2433a5e478b9251483012acd43a : SUCCESS in 2m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b6f03a47da2841b5b9ff0c5c8d05a09d : SUCCESS in 3m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/29d9c39a09cb4675ae00035fa414a09e : SUCCESS in 19m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/26af6ed732464463b7b3f03222e43c94 : SUCCESS in 1h 23m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c1e93791a066d5d17c966727c0d77aa8f37221f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 11:01:55.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e31b231f087421c97c0638c2dbe6d8a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ddecd196ef61471a8f708c6b65f58792 : SUCCESS in 26m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8942a41c5ec94846ada80d07d661ab56 : SUCCESS in 1h 10m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c5ec5941fd7049408e45e42228d7a4e9 : SUCCESS in 17m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/350c67be149c4b24901360e435578ec7 : SUCCESS in 1h 13m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6ad93b50945746f9890781b1f5236fb0 : SUCCESS in 1h 27m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c14d0651ca1941e98712c5b74d664fb0 : FAILURE in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7da99a9474642e4858d627591a07f92 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a76ac36a2f7146bdb690cff4071844cc : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f6cc08a912c4029b8d6afaa265c6d2b : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11352376d4754c1795a867353c141143 : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9908203152b84346bce59966eaa899a9 : SUCCESS in 3m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/87c0cc953090418e859ce54a0683163a : SUCCESS in 16m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8068a3e5e8b4441c98df8e50689963a4 : SUCCESS in 1h 15m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/77782ebc47b74cf7ba7985ff347e1b01 : SUCCESS in 1h 19m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/203a46e3850f49ffa7b571119f355def : SUCCESS in 20m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ef5140d48d474f49967304e95100ea78 : SUCCESS in 43m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f9945bd30855479ea63801c56cbc222a : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":3},{"id":"5a0e87a2d3a4ef52d364ed9fba00946729f1244b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-27 12:19:31.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2f736fafd641eb0f5a0e5463a64813b9afe853b1","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-04-27 12:29:19.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"338cf2ca926db20e85c8aa957e13ecfe4cbed1ee","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 12:35:26.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"fde3b45665c056061cfcfd6db91f25f888b27295","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 13:16:58.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16f183427e694957aac2ed7f7675f3f0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9f679fc72ae640f79b3f1d774680ff1d : SUCCESS in 24m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/989e266741e841358fe488f887dbc7ab : SUCCESS in 23m 19s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4d38156b62fd49ddbfc091970735b3bd : FAILURE in 54m 16s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/edba3668ff704b12970219620a2fb551 : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42f08f9b34cd440480d48e7e5f07fc5f : SUCCESS in 3m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/21cfee193522452d9f17af411297c665 : SUCCESS in 23m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/34210a0329de484c99abed8133a9435b : SUCCESS in 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"51e0484e45baf2c168fcf24b7b59f09bfa3ab01d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 13:52:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bfa68644ab3435f90c9f677f2511d37\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c6125395fab34b3f9cda3409577615de : SUCCESS in 25m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6074f9928f014ffa94d98bf9387d70d4 : SUCCESS in 1h 25m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0df7b58c0fda469d8bb303aab49ead24 : SUCCESS in 17m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a3cbe5f037294267ba3487fdae0e24aa : SUCCESS in 1h 06m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/776b7a88a4be4b47b32f4fc05fea173a : SUCCESS in 1h 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c97c9ec990844168ae74b6740aa1e742 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e41cd1be992b421995f4b894391a7dae : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7471d7a18864c2e9bad3fb6d1852c5e : SUCCESS in 4m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/234cf4af21fe403787b9637733eddf8d : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ba7a66c82946d3a2f5328de2344a2b : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3a589b6371a409cbbda071ece5e91b0 : SUCCESS in 3m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/957194f75f404baeb857bee84e34ef46 : SUCCESS in 16m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/91451583461b4f39a8e96010a0ce2868 : SUCCESS in 1h 07m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ccac3d23130e40d1a41ffb3f696ce050 : SUCCESS in 1h 26m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/43dc1ab390de4b54bb87c556ca064192 : SUCCESS in 14m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6855ead2efd643419ec532b9605be5ed : SUCCESS in 57m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c3ed6e65993f4bc98464146c8abdaab0 : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":4},{"id":"99b5e628dfbe5654062c73c563950feadf1b87b3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 06:47:08.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6056fc5aeecd664eb6689680a100ef34c064a581","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 06:47:12.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"54bf905a140affd1d6e222468bbb71c8fb8f99a7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-28 10:04:17.000000000","message":"Patch Set 4: -Review-Priority","accounts_in_message":[],"_revision_number":4},{"id":"1cd2c8e47d1976827bcf7f3e984776f9bb9df7be","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-11-29 06:07:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"786acf2530a263e6489c9192a62ea0b242a7a303","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-21 06:46:04.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"d33fab66f5575743f3b577602b13575b7aae9295","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-12-21 14:02:31.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fdce0f1bc0a18e230441c1a9ec5549c20f86a88b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 14:02:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"84a3501be28255337db44a877f6193a89eb6c403","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 15:45:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a7fc7fdfb314deca9645d42cc49474e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9bb0a9178b5e42658dcd2b2204c39767 : SUCCESS in 21m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8d17890c28a84bb5b471429a23b21563 : SUCCESS in 30m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/10f3a3701c3a45e0a66e2e8612e68308 : SUCCESS in 58m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c06df632c77949eb842c93e1f9f23483 : SUCCESS in 1h 30m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a39a7debf6254f84a2f010ceb81c5ec7 : SUCCESS in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b530396a6b84e35a6441a960f57a975 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0e44962ba09431ca3e714a329a52917 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37489e6b23cb4c15875f2ca360705449 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/820ece2f37204f5b9f879721a6ff8101 : SUCCESS in 3m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3456a32581244c97b995b89b491d6fa0 : SUCCESS in 18m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e4c043d1cfb045a1a6e1ca239c9fe67b : SUCCESS in 27m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/89387d930ceb41beb0c14ec22d9a0e09 : SUCCESS in 1h 01m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/575e0bb7551949dbb9dd79505633cadd : SUCCESS in 1h 25m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7f31ea7fce3c400081f4a46afc7cdbde : SUCCESS in 20m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fd2b2abceae84c94a097e6c1ebde6338 : SUCCESS in 1h 03m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9bb481544c934033beb38ed899ac6080 : SUCCESS in 1h 06m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c11edf28930e477c83aa070b1076cb1e : SUCCESS in 1h 40m 50s","accounts_in_message":[],"_revision_number":4},{"id":"8ae7ec30b2aa8c87d755773bcf65b8fa3db8e7b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 15:45:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"12bc5fdb5190daf1693962c02bd2b7598788ab89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 15:46:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c899d1e0734646c49061dfa69f32473d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/536768441d89453597a8841e0e49b2c6 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f36d7a6939814594b580820606b79560 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a5351473e08ee18cbc86edfb52da708ad558fe85","revisions":{"1ff6ea0c2d5b317214dcf490ab9a98ed5b689341":{"kind":"REWORK","_number":1,"created":"2023-04-26 14:10:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/881567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881567/1"}}},"commit":{"parents":[{"commit":"bc2d3e4ef4aed12a5329f697967bd325339a7bff","subject":"Merge \"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc2d3e4ef4aed12a5329f697967bd325339a7bff"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-20 15:32:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-26 14:10:10.000000000","tz":60},"subject":"Summary JSON file","message":"Summary JSON file\n\nAdd support for writing the build summary to a JSON file. This is more\nmachine-friendly than the current approach of writing to standard output\nat the end of the build.\n\nThe file is configured via [DEFAULT] summary_json_file.\n\nChange-Id: I9868af5d00977750d0673fc48f8863a99f878dc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ff6ea0c2d5b317214dcf490ab9a98ed5b689341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ff6ea0c2d5b317214dcf490ab9a98ed5b689341"}]},"branch":"refs/heads/master"},"a9c757ce7d4269d660a85e92edcfb9b0b6dbcbbc":{"kind":"REWORK","_number":2,"created":"2023-04-27 08:19:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/881567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881567/2"}}},"commit":{"parents":[{"commit":"bc2d3e4ef4aed12a5329f697967bd325339a7bff","subject":"Merge \"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc2d3e4ef4aed12a5329f697967bd325339a7bff"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-20 15:32:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-27 08:19:37.000000000","tz":60},"subject":"Summary JSON file","message":"Summary JSON file\n\nAdd support for writing the build summary to a JSON file. This is more\nmachine-friendly than the current approach of writing to standard output\nat the end of the build.\n\nThe file is configured via [DEFAULT] summary_json_file.\n\nChange-Id: I9868af5d00977750d0673fc48f8863a99f878dc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9c757ce7d4269d660a85e92edcfb9b0b6dbcbbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9c757ce7d4269d660a85e92edcfb9b0b6dbcbbc"}]},"branch":"refs/heads/master"},"f75dafd8d809f873faa21db3d8350030dd31bf86":{"kind":"REWORK","_number":3,"created":"2023-04-27 09:29:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/881567/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881567/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881567/3"}}},"commit":{"parents":[{"commit":"bc2d3e4ef4aed12a5329f697967bd325339a7bff","subject":"Merge \"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc2d3e4ef4aed12a5329f697967bd325339a7bff"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-20 15:32:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-27 09:29:20.000000000","tz":60},"subject":"Summary JSON file","message":"Summary JSON file\n\nAdd support for writing the build summary to a JSON file. This is more\nmachine-friendly than the current approach of writing to standard output\nat the end of the build.\n\nThe file is configured via [DEFAULT] summary_json_file.\n\nChange-Id: I9868af5d00977750d0673fc48f8863a99f878dc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f75dafd8d809f873faa21db3d8350030dd31bf86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f75dafd8d809f873faa21db3d8350030dd31bf86"}]},"branch":"refs/heads/master"},"a5351473e08ee18cbc86edfb52da708ad558fe85":{"kind":"REWORK","_number":4,"created":"2023-04-27 12:19:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/881567/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/881567/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/881567/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/881567/4"}}},"commit":{"parents":[{"commit":"bc2d3e4ef4aed12a5329f697967bd325339a7bff","subject":"Merge \"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc2d3e4ef4aed12a5329f697967bd325339a7bff"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-20 15:32:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-27 12:19:31.000000000","tz":0},"subject":"Summary JSON file","message":"Summary JSON file\n\nAdd support for writing the build summary to a JSON file. This is more\nmachine-friendly than the current approach of writing to standard output\nat the end of the build.\n\nThe file is configured via [DEFAULT] summary_json_file.\n\nChange-Id: I9868af5d00977750d0673fc48f8863a99f878dc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5351473e08ee18cbc86edfb52da708ad558fe85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5351473e08ee18cbc86edfb52da708ad558fe85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
