)]}'
{"id":"openstack%2Fkolla~881729","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746","project":"openstack/kolla","branch":"stable/yoga","topic":"upgrade-pip-for-tests","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-04 19:36:07.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-04 10:14:51.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-05-04 19:36:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746","subject":"Upgrade pip/setuptools/wheel for tests","status":"ABANDONED","created":"2023-04-27 14:58:02.000000000","updated":"2023-05-04 19:36:07.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bb50ff015ef8c4f2e906da0d6ffccc1216bdf867","_number":881729,"virtual_id_number":881729,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-04 11:51:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2023-05-04 08:27:33.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-28 00:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-28 01:14:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-03 18:00:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-05-03 18:00:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"4123483cbb7344a0c276c19ee6824dbafdc00f02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 14:58:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff6eab0d3f84f41bb88fdb41463470de2591147b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 15:44:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"a9a01099268f5c0e845baf249deacfa251ea14a0","tag":"autogenerated:gerrit:restore","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 23:27:58.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"36ff992b4396d8518acab33143fea11e745b4898","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 23:28:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"04f503345693275834a5b63025ca14a10a28fbcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 00:40:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fd4da90750a4d2fa506b2a0ded14649\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b08844109dfa4ed58127635e91cbeb6f : SUCCESS in 28m 53s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9feac22f7e6443339dd3cce7aad619b2 : SUCCESS in 25m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b79587623c3e46a6ae6891f125ef938a : SUCCESS in 17m 55s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7d2e0403f0b1467797fa51f4f051da52 : SUCCESS in 1h 11m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c83704f22d9b4609b591d554807ad994 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b551ddd969245cab46a105eaec56172 : SUCCESS in 3m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b27837290f45406086c916ea443ef8a9 : SUCCESS in 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ad753d0ce5f2388ee9989dd1c3bb40bfb2dcdb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 01:14:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7e0fca26f234a74add1ac111b2a9f5a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fe3a0a23042e4c63a5d9f0753c5e8cc1 : SUCCESS in 26m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c17809dfab8a4a24b40390283f3fc6ab : SUCCESS in 34m 02s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/a89d63bdbdb047cabd4852533720b98f : SUCCESS in 25m 01s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b34409bf9c3f40dfa93074096a400839 : SUCCESS in 1h 35m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cf38cf19e61d43d19cc9cc1330556d7e : SUCCESS in 1h 12m 39s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/2ef5703776ea4f6a99cda21aa3c19402 : SUCCESS in 1h 12m 14s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/dba343dbd7f34b0694673123b7c13a08 : SUCCESS in 1h 30m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c2811c31bd724cb589ac999d5c082f66 : SUCCESS in 24m 35s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/37d6e7fa350346b5aa11917fc72f8268 : SUCCESS in 21m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0c55fcdee1ab494c9337b8e470dace6c : SUCCESS in 52m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/66d2dfc8549d4ef18ff5a5faee8bb4c8 : SUCCESS in 1h 05m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f5ca61aa8747848c81c996e3f9f527 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ab7a1315dcc4a5cb360561e9f8157f2 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a01ef7c4bfe4478a66861534b0610a8 : SUCCESS in 3m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0bb1a9f8c4a946659bbc8acb11d7ea56 : FAILURE in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e693f24bc2444d88eb2dcb1aa6fb196 : SUCCESS in 5m 36s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/063570570bf64667be3aaf2b0c650c5b : SUCCESS in 23m 35s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/d2a6cc12f8a4484d89553aa5a0c10e12 : SUCCESS in 1h 00m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/741e23d87de84291bc3a99741add56c7 : SUCCESS in 19m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/be34960fde3847219047e3a23c1e521f : SUCCESS in 25m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f2b2c4085dee597ccaeed11110a9c8224b2b005e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-28 11:09:16.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"30099f6d1cb50ffdf4e12f7aeb845dd1606549d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-28 11:19:36.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"35f2fe5fa6dfc36fb989266b6ef457c083872d30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-28 11:28:25.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"b7cb629804f92cff24a62a7787b2b2b42f8631e9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-28 11:52:02.000000000","message":"Topic yoga-urllib3-test removed","accounts_in_message":[],"_revision_number":5},{"id":"ec479095200d750a1c786a1cbb8b721d595c5941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 13:00:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5af6c7ca4c7d4789a70f50d7c77611db\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/40fc7d8af18747f4b0badfa64d7762bf : SUCCESS in 33m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6d82ad9716fe403d89d34fae666e3257 : SUCCESS in 34m 05s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/6e071e7e3a0e44b09b82c6d5b161606f : SUCCESS in 30m 25s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f7313ffd06b349edb38586e2c9bbf26f : SUCCESS in 1h 20m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d3304f3a14a44515929c3090031822d8 : SUCCESS in 1h 09m 10s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b695ae4a92dd40a094e78b8f796e6f94 : SUCCESS in 1h 16m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ff3c4a10ffa043e682c32107c6513803 : FAILURE in 24m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6df5840c7dd04cd981086d5ab75ee7f2 : SUCCESS in 26m 09s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1676be9ad5164e9abe461a58d7cf306a : SUCCESS in 18m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/017584ff3fa541d58234cbdf91f43c61 : SUCCESS in 1h 02m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/24a8cc01b8ab40bb90b22d1d8ef3a5c5 : FAILURE in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91bcc11e7ff04cfa9953394d13f29c18 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52680c0913ec471394e81aa708c4685f : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e7d2cba7d2141d3841e38537f6720e2 : SUCCESS in 4m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/899aae223288460a856e4cab97ba8235 : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b750eb146ae4ee5a1335a7ece004d37 : SUCCESS in 6m 37s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/f992c98b4ccf4dd2be04a9235a42443f : SUCCESS in 23m 22s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/c9b8fbb3e3524c0e9f741a16453305f4 : SUCCESS in 57m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a5ac9333fc5345a7bf79aab2f50919a5 : SUCCESS in 39m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/014607ac2d8b4cf387c3fb259910a5e7 : SUCCESS in 44m 26s","accounts_in_message":[],"_revision_number":5},{"id":"dfac6fa218ba3311fcb9b6a6eaff604fc4e128cd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-28 13:42:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9db2e08c2f1c43b92fbe9ce6d641d046a1df0656","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 13:45:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3346bdbea83452e856786d49db3e49c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/60a32412aaed40a790de5eda3c432063 : SUCCESS in 33m 53s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/46ba14539dbe42158954d4a0c026d2b6 : SUCCESS in 37m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a7dc04987244801a7f3818788ab14e4 : SUCCESS in 30m 33s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bbd8c5f577b84db597289cade1633fe9 : SUCCESS in 2h 15m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5158d6aa84cd453d87ef2d097adff3c8 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e99b6da53e6842dfba4d1841504c2ee8 : SUCCESS in 3m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/88e130e148e845e5900bf9016c0c9249 : SUCCESS in 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d693047e1b353c9f9966eba60bcce99d2e07a760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 15:23:37.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49c02ac41e0a4425beae0a63d61ed12c\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7e27ace85b9f401db8f9c417896e4e67 : SUCCESS in 44m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a57cc8f157f6489b8c4852aeb4dcbba8 : SUCCESS in 1h 07m 33s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/3c979b66932a43519ef6d7cd1d7bbf8b : SUCCESS in 54m 40s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ec4215fee0ad4a0e8cb6338f85bfbac9 : SUCCESS in 1h 20m 09s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2e1a6a77260d46b5b11eaa1fff41e0af : SUCCESS in 1h 36m 19s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/02722c66285a4fa29651e14b67c801a3 : SUCCESS in 1h 04m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/647f8d7ec676438485cccd5dea4e26c6 : FAILURE in 22m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6b2d0b4a08d740f39e9e352418a11f01 : SUCCESS in 25m 22s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b300f260b8664bb1984d2f8911a4cbb4 : SUCCESS in 22m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/36a2a55231f74b5e9bdd4ae20915f3f0 : SUCCESS in 1h 01m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9a2c49101b644aae9ec1f0925c328e67 : FAILURE in 41m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6edd38f4af9347b483eee0ecc73becb5 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0ed7c3969f14cd596811288ce2d47ed : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/583d31cbf71f4bdc9ef888ab216347a5 : SUCCESS in 3m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/db51208076544c2e889251592580d631 : SUCCESS in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9c50782519f445287bae33e3230ea5b : SUCCESS in 7m 06s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/a9b7e7c1350f4d249d5fbffc32e2e1c8 : SUCCESS in 27m 27s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/62127660d186416e916029b40e2c3e74 : SUCCESS in 1h 07m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c2cca42197674a479de2c2a02475f6c8 : SUCCESS in 19m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eb49f149285849b1a846131ee5f95761 : SUCCESS in 24m 54s","accounts_in_message":[],"_revision_number":5},{"id":"2836c5f5288782185eab790c4075528b063dcc27","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-03 10:24:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"30ebb4f94f2dcc4f9d96dd4168973c0ecc8c23b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 10:35:50.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/881814 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"c8ba4e7feeff1c5ec186694b757ddd4056e25030","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-03 12:23:35.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"8a2ee91c7bccf85d8cbe18b21b894350bbea784f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-03 13:15:29.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"ce452f50c036149ed8e73a60598304811507ea91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 14:51:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc849c55019f4431911e20bcc491be72\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/23018271ca294bf8b2f06c2b5c8478c9 : SUCCESS in 31m 03s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/25ceb21be700432f95a1297a63a1e740 : SUCCESS in 31m 00s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/177998d3f3244b56ba519a1a52094608 : SUCCESS in 42m 51s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e536949c6f574835bacc58fd7244a219 : SUCCESS in 1h 16m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d43c653e86c246478c6856675c0131aa : SUCCESS in 1h 09m 49s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/80875112c1d44e059a275dd12800bbd5 : SUCCESS in 1h 23m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/58ca23e3f9614ec2a5c388216b75ae5c : SUCCESS in 1h 30m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2453336a9a3346b29b2120aa77b343f0 : SUCCESS in 20m 49s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e2831eaf5c27428d839b06305ebd771f : SUCCESS in 23m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0e9a0aa4d8904649a9e1f14983c65482 : SUCCESS in 1h 00m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7be671e8a72b4187bffad74c08b277b6 : SUCCESS in 1h 29m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52c592b61deb4eb498b81245db9ae695 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbb28f8da4684e9c8a2d8b5892a59bff : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f405bffe6952483c9f5177f04ebb2e33 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e79ca74f3ba44efb11afce8ae1c7b0c : SUCCESS in 5m 34s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/2c56b3cfb7d04a63a3c51cde3ffd9f28 : SUCCESS in 27m 39s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/97a2afa63ab14358823a59af67dbe80a : SUCCESS in 55m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7111b0bd940b4f2994f47c63f524b115 : SUCCESS in 19m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9859e7451fb34bf69a029643832a8125 : SUCCESS in 23m 26s","accounts_in_message":[],"_revision_number":7},{"id":"bdd3da86b44b1f73854df9ee273cfe059d162730","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 15:03:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4563b4360fef41729da91d15710c0f9f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0aca1678e1364e11a61b4456c157076e : SUCCESS in 26m 45s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ba350c93f774ef4b31cebe7477005f5 : SUCCESS in 29m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d58c212dce3b462ba321c77df315a0aa : SUCCESS in 19m 58s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/05d3dbb4d4af4849af1a779732e34d0f : SUCCESS in 1h 11m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/baf8b626dfc24b758958267638747ea0 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/352d8d19a98e48209cc8b5b0fd756cc3 : SUCCESS in 3m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f924336eec27425c8bba46b15d9a4455 : SUCCESS in 1h 44m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"58ad9807ea92e26a84d9e53b896ba3f132cd8274","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 06:32:19.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"86fbb5b6b1bc483b4c16bd64d3022344839af10f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 07:55:03.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dbcb375fa0f47764a71d632de8e4ac8ec8745d80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 08:23:22.000000000","message":"Uploaded patch set 8: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"76f07f73aa7113da6fe6931064d02cffcf5fb56a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 08:27:33.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"e15cc33631a5cc96a541e4f49e28d6c4bba789dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 10:00:03.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d32b254752e04b21a9180869b058019c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9355e1266de340f58837b68a54fa9e5f : SUCCESS in 38m 14s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4e1753d58cfe4b3d9fb7befa29f3f9bb : NODE_FAILURE Node request 300-0021131890 failed in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a253c98c94d44fe96a5f1667ef22c04 : NODE_FAILURE Node request 300-0021131891 failed in 0s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b658e710b4594fda90e50802e3a44a5e : FAILURE in 1h 32m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/834b06296fad48c58343c051b1e48de1 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a68c1e17e646486f9c6b033dec2bc6d6 : NODE_FAILURE Node request 300-0021131894 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bcab63e9f41042398ff4436a17d60bd0 : SUCCESS in 38m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"80e1520e478a7df9c51f50208825583c2a314d91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 10:03:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac83887faf494989abe050a0d54ab9e7\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/77d8a69dec8e4a29b72e6049be868740 : SUCCESS in 37m 23s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/af2501f907cc4a6087ab25da10791512 : FAILURE in 57m 22s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/0238f9370e3546309eb639a0f0d67591 : FAILURE in 30m 32s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/58ba654d15a94ffb9f1784a80782ac30 : SUCCESS in 1h 26m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e9f253f86a34407e86a7a71e0d815e38 : SUCCESS in 1h 22m 17s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/85e1dbd823374f9d91b48e470c46bc31 : SUCCESS in 1h 18m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/90d6e604bde2423d8f90e5c56e1c31d3 : SUCCESS in 1h 27m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/35d23de0aa27401a91c49c936be3280e : SUCCESS in 26m 27s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a2763346c8834d099446d80bf20eee42 : SUCCESS in 20m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ccc4e61e67ee42d8b351f9eca0d8c37e : SUCCESS in 56m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/52a8540f0ae0495587ccb4beabd3d3aa : SUCCESS in 1h 24m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b6ebfe980f4ec7b2350013b062855d : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daeed940f47c45de8499461029eee34c : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40eebff98a464c93881bd561bb714cae : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b9218e5bfc4f46a61de1a3aad737c6 : SUCCESS in 4m 58s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/f6f42f44de68455b86a47055c41ff584 : FAILURE in 39m 53s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/e0730ea08dbc4ba6bfa2d8fc07400953 : SUCCESS in 1h 07m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c090f99625943c8995c94deafffc2d7 : SUCCESS in 18m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f6921f97e9b0414a86793948fd18c900 : FAILURE in 42m 03s","accounts_in_message":[],"_revision_number":8},{"id":"0d6832ed2d32a241fcd5cc0af8929d989b9ae26c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 10:14:51.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7f53d4a73f40fe576fe5de359637c8ab80dcad6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 11:42:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e2c5c481e494ebcb11c80289a1a74d7\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c218ccefedc4eedbeda09e6889422fa : SUCCESS in 41m 47s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/583dfd241dcb471794a16a66a65f68f8 : SUCCESS in 43m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/afe3279c5d0148b380251906f0212841 : SUCCESS in 37m 07s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/54da1d5869ae4129934b7f41d95453db : SUCCESS in 1h 26m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d7f224c275e4b06b26fbb018da209b1 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dcd5349ada5c4c159ddd425dd50bb6b4 : SUCCESS in 4m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e26e5ee4121d447880e6b8456c8f6c15 : SUCCESS in 1h 14m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f4a6f4547037cb1416dcce4a5abb29e2b12afaa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 11:51:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cab0544dd91743e8b7a8e8bdedf2acc9\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c72f203f5075440b8e1fdc5dea1b029e : SUCCESS in 31m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/651c3ef632d745f5ba5fbd2d045f3884 : SUCCESS in 40m 21s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/1015de65a897425d9b8e5e3a43f9cb1c : SUCCESS in 56m 09s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f0c57e282cc2471e85edbc1a0bb2787c : SUCCESS in 1h 20m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a839c0cea6544c939b6481f7350e53dc : SUCCESS in 1h 12m 58s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/51bbfc5de94543589fa57d2566b3cb15 : SUCCESS in 1h 18m 46s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6e0599ac637b4641bf3d0a557ab054db : SUCCESS in 1h 36m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/16239dba8ead4cce8d056130572e6b8b : SUCCESS in 20m 12s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e0ba3783460044488d666a27c3cbd8d9 : SUCCESS in 18m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2e2a18fe71fe4c27915aba042c0ebf98 : SUCCESS in 1h 04m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/30efda5b312945ada9e1a7cc5cb53c86 : SUCCESS in 1h 28m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e837c80b2c14afea1512a8a94f22c84 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/948ddd5e83124477a2e9359dab9b5e04 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7af3a36ae42945698c013aad398384f5 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66efa34e500a4e01a24dd01c25e9f464 : SUCCESS in 5m 28s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/5f977b395fa9447e917f586a7fcd2c8b : SUCCESS in 23m 28s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/85ccf6a73dc847a0b6031fcf2bae0633 : SUCCESS in 1h 01m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e0262d5c428441e4b75e65c8ed56e16d : SUCCESS in 37m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a62a32638dc48df862dd0811fbb2178 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":8},{"id":"bb50ff015ef8c4f2e906da0d6ffccc1216bdf867","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 19:36:07.000000000","message":"Abandoned\n\nwill create another bug report and solve the issue other way","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"605e331176a34f1ef480d9cf842e8e1591e869a6","revisions":{"704a4a22124150d476ec52ffab009978e7cc3fb3":{"kind":"REWORK","_number":1,"created":"2023-04-27 14:58:02.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/881729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/881729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/881729/1"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:57:29.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:57:29.000000000","tz":180},"subject":"[DNM] urllib3 test","message":"[DNM] urllib3 test\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/704a4a22124150d476ec52ffab009978e7cc3fb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/704a4a22124150d476ec52ffab009978e7cc3fb3"}]},"branch":"refs/heads/stable/yoga"},"dc9c0323ee9adb452633b76cf2cbd0f2d5eca8a4":{"kind":"REWORK","_number":2,"created":"2023-04-27 23:28:08.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/881729/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/881729/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/881729/2"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:57:29.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 23:27:28.000000000","tz":180},"subject":"[DNM] urllib3 test","message":"[DNM] urllib3 test\n\nChange-Id: Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc9c0323ee9adb452633b76cf2cbd0f2d5eca8a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc9c0323ee9adb452633b76cf2cbd0f2d5eca8a4"}]},"branch":"refs/heads/stable/yoga"},"49bdc410ccf30bf2ad59ee3a8c189458d72c76cc":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-04-28 11:09:16.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/881729/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/881729/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/881729/3"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:57:29.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-28 11:09:16.000000000","tz":0},"subject":"[DNM] urllib3 test","message":"[DNM] urllib3 test\n\nDepends-On: I25fcb81da7f2bbdbb02a283e4e787d04f933d1bb\nChange-Id: Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49bdc410ccf30bf2ad59ee3a8c189458d72c76cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49bdc410ccf30bf2ad59ee3a8c189458d72c76cc"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"},"4895e6d8e19284b821e64e133368f08c3c03a386":{"kind":"REWORK","_number":4,"created":"2023-04-28 11:19:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/881729/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/881729/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/881729/4"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:57:29.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-28 11:19:36.000000000","tz":0},"subject":"[DNM] urllib3 test","message":"[DNM] urllib3 test\n\nDepends-On: I25fcb81da7f2bbdbb02a283e4e787d04f933d1bb\nChange-Id: Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4895e6d8e19284b821e64e133368f08c3c03a386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4895e6d8e19284b821e64e133368f08c3c03a386"}]},"branch":"refs/heads/stable/yoga"},"1b4ed130bc145497bf80bf656036590fffe6e658":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-04-28 11:28:25.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/881729/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/881729/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/881729/5"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:57:29.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-28 11:28:25.000000000","tz":0},"subject":"Pin incompatible urllib3 2.0.0 released on Apr 26","message":"Pin incompatible urllib3 2.0.0 released on Apr 26\n\nFollowup on I25fcb81da7f2bbdbb02a283e4e787d04f933d1bb\n\nDepends-On: I25fcb81da7f2bbdbb02a283e4e787d04f933d1bb\nChange-Id: Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b4ed130bc145497bf80bf656036590fffe6e658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b4ed130bc145497bf80bf656036590fffe6e658"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"},"24e5a9465b49b03b99e3055939b8d1db29ad723c":{"kind":"REWORK","_number":6,"created":"2023-05-03 12:23:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/881729/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/881729/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/881729/6"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:57:29.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-03 12:23:11.000000000","tz":0},"subject":"Bump docker library to use 5.x version","message":"Bump docker library to use 5.x version\n\nSince incompatible urllib3 2.0.0 released on Apr 26 the\nlatest version of the docker library can\u0027t be installed.\nWe should install pip\u003e22.3 which uses new dependency\nresolver to use docker\u003e\u003d6.0.0.\n\nChange-Id: Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24e5a9465b49b03b99e3055939b8d1db29ad723c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24e5a9465b49b03b99e3055939b8d1db29ad723c"}]},"branch":"refs/heads/stable/yoga"},"5e546567825ceba9fff96c24cecac3253253c895":{"kind":"REWORK","_number":7,"created":"2023-05-03 13:15:29.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/881729/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/881729/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/881729/7"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:57:29.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-03 13:15:26.000000000","tz":0},"subject":"Upgrade pip/setuptools/wheel for tests","message":"Upgrade pip/setuptools/wheel for tests\n\nSince we use virtualenv on the several different OSes\nto get the same results we should use the same version\nof the libraries used. At least Ubuntu 20.04 uses the\nold pip library with bad dependency resolver, the issue\ndescribed as a warning in the pip 20.2 and later fixed\nin the pip 22.3.\n\nChange-Id: Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e546567825ceba9fff96c24cecac3253253c895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e546567825ceba9fff96c24cecac3253253c895"}]},"branch":"refs/heads/stable/yoga"},"605e331176a34f1ef480d9cf842e8e1591e869a6":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2023-05-04 08:23:22.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/881729/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/881729/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/881729/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/881729/8"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:57:29.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-04 08:22:57.000000000","tz":180},"subject":"Upgrade pip/setuptools/wheel for tests","message":"Upgrade pip/setuptools/wheel for tests\n\nSince we use virtualenv on the several different OSes\nto get the same results we should use the same version\nof the libraries used. At least Ubuntu 20.04 uses the\nold pip library with bad dependency resolver, the issue\ndescribed as a warning in the pip 20.2 and later fixed\nin the pip 22.3.\n\nChange-Id: Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit ae06639dcfbd125233f67cd615839f149ea85da3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/605e331176a34f1ef480d9cf842e8e1591e869a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/605e331176a34f1ef480d9cf842e8e1591e869a6"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
