)]}'
{"id":"openstack%2Fkolla~881948","triplet_id":"openstack%2Fkolla~master~Ic06f3c0e42fb7854e5b198c9b868f10cd507cd24","project":"openstack/kolla","branch":"master","topic":"for-debian-bookworm-upgrade","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-28 16:22:31.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-06-29 14:11:16.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-22 11:27:55.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-06-29 14:11:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic06f3c0e42fb7854e5b198c9b868f10cd507cd24","subject":"base: install adduser on Debian/Ubuntu","status":"MERGED","created":"2023-05-02 09:01:07.000000000","updated":"2023-06-29 14:12:17.000000000","submitted":"2023-06-29 14:11:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"881948-for-debian-bookworm-upgrade","meta_rev_id":"22ea1269991fab9f70bef6853182268ae4e8cb40","_number":881948,"virtual_id_number":881948,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-29 14:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-28 16:22:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-06-29 09:04:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-28 16:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-28 16:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-06-28 16:22:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 09:44:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-02 11:51:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-02 11:59:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-22 10:16:10.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-06-28 13:16:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-28 15:01:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"9ba6bba650c8bc5c076d72645fbef943b18b18b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 09:01:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0dc7b7c9f9934b50e29d22b3995390ca4b44b04","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 09:01:20.000000000","message":"Topic set to for-debian-bookworm-upgrade","accounts_in_message":[],"_revision_number":1},{"id":"dcf092c52247127c44e87b7af123df138526dc19","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 09:11:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a37001b5c3a4bc7992ebc1306ad552110f3d173","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-02 09:44:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88d414d182e32f6cb88263b1fd2977cc151628fa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 09:45:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c8d5470f83c5c2fa4191130dffa3682b10c8cf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 09:50:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f3276574ebc0d5c84f272114fb4df332db785b88","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-02 09:52:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6ae5ea859fdffb48061c495994ae836aaa40f652","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 09:56:30.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1fbdfcfb55b5a50ddf2f5f43e678740f3823c11d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-02 09:57:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"91b66265e929a1bee4c012b57683fe8eadaa701b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 11:51:18.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b45c4cdc966d4e11b3c5a59d9ccd9aeb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d5100381ad6048b5b3d088fe7602b073 : SUCCESS in 26m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/66c6a98163a145319faa22201aab3d77 : SUCCESS in 1h 13m 54s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/db2b928dfdea4481a0820991da20779c : FAILURE in 1h 53m 33s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4a4b0feaa8f4bd0aa164b1df0b6f449 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3e2cb7ec0a2f48088fa8a0b3af4323cf : SUCCESS in 2m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f1f0e2dae0c243a29178a0448c7f4a81 : SUCCESS in 1h 08m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/180b5b56ea2e4408b51d138a3f0df054 : FAILURE in 21m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"65d9cba4d9b500945d1c030c862893f476a0989f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 11:59:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fec752faf0d49de8821ae0ff3955313\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/390deb40cd5d4856b6b42e6a2080af2e : SUCCESS in 24m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c2183876dd314eb1b494830568037529 : SUCCESS in 1h 47m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/15c90780751f4479918f3d6a9bfdd5ad : SUCCESS in 16m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f9ae5e8a736043b297d2ced5d008477c : SUCCESS in 1h 04m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/245618251f4e443da48eba6506136aa0 : SUCCESS in 1h 19m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7974f7fb7fb4562958e5b742e3c47bd : SUCCESS in 2m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0938fa8a2532473ab33c7d8d9dd79d64 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70a24189ef964fa0b439dc5b8d248247 : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/147d94fac7b248e1be6a5d8c9dab0e44 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f7089347ab4f3895874fe1f7180bbb : SUCCESS in 3m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6b956c7fe50246b8bb0ce93e03a1a042 : SUCCESS in 17m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9d75c2c4b6c746bd98507716f66a312e : SUCCESS in 1h 21m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8a37cdd6b3a54d8e969fa1ea314443fa : SUCCESS in 1h 19m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/62da922e41024567a775dce3d6a720fe : FAILURE in 12m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/20f270f7f788408ba67190700d11c6ae : SUCCESS in 47m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ee2b22f6b47b4e8aa04f8265a169d06d : SUCCESS in 1h 06m 38s","accounts_in_message":[],"_revision_number":3},{"id":"fb84b246a0a491861034d9a02c4055fe03193612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-14 12:07:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8a2612d35f453a59fb02e95812f5dfa7677fc88a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-14 12:34:01.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4311bb7d5ef2ffdf97f13838dd3972423230f106","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 13:42:18.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f71f0ff6b81405194e6fe7d6d134235\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ca2876644b4442bd91dcd0fc23c44229 : NODE_FAILURE Node request 300-0021458757 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/06aa7d3b3b2643058b6f5e51dbbeeb0f : FAILURE in 1h 30m 38s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f9e432da1f0449809a0056566ddad277 : FAILURE in 26m 51s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/355e198335cd448898f3280be0f5e017 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d85f5d60b8db4edcbcc2f3a7a5194ecd : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7814efdb75ab48a59608a54648817d96 : SUCCESS in 3m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6ebc2934311c4e63a3c375e021e4a9a8 : FAILURE in 25m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/351a9e16c10645d38e7359669a639961 : FAILURE in 25m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"48734146a955068fc0c074f9b387c2361f9c4ba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 14:38:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64cf97364f484c75852586f3a13638c0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dab0de08f1c8460091c5fdbc09e6c12c : FAILURE in 27m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a42487b2eaa746c8928c693c7287c210 : SUCCESS in 1h 16m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/24fe1bbc22f5452d96f26930748486d3 : FAILURE in 24m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/34cb58cd02ba4f9595493646d5f508c9 : FAILURE in 13m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/299193060e8948f98c69737136730fdb : FAILURE in 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5414713de304ef4b83e6765e4c44592 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daa4d6d844564ad8a75c94996e20ac97 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0656d411a074597b8538e217839ecb5 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a7b9fc1dc384330bc631159b6c5389e : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2230f6d33daf4f8db463d09dbb404bde : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2687e411650947798b5a43d5726c0d8b : SUCCESS in 3m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d164f832dc694d319dcb7647070e41f8 : FAILURE in 18m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a42a9be6e9ad4f64ae2125654ff1ef8f : SUCCESS in 1h 05m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bdfbfe99a12c4a65983269f4ded2365e : FAILURE in 1h 00m 12s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/22e5a4ab0ecd46698e3c8c856fc7e673 : SUCCESS in 1h 30m 11s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d2d9882fca734961879864ebc0fe9aa0 : FAILURE in 1h 15m 23s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/cab6d369682b4d649f742da925a40cf9 : FAILURE in 1h 12m 26s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/0c00f4dd895b4ce788cda3853c410b7c : FAILURE in 48m 36s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/f02d3b97dfe64087b561f6c5c04f23a1 : SUCCESS in 1h 27m 43s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/cfe191a22b7d47838bb8209bf07fd2bb : SUCCESS in 21m 22s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f3ce2fc5b6df40819a5e7d97c8243afb : TIMED_OUT in 2h 02m 17s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/87e38481f5ea4aa4bdfdbb32ba318d6d : SUCCESS in 1h 09m 02s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d0cdd9b6c35745c291b45c8733ce23b2 : SUCCESS in 1h 23m 15s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/079793ef19004e698ecf01643af85fde : SUCCESS in 33m 31s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a248b73302e14e1683d6445c36846241 : FAILURE in 18m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8669f56a31204e69b160cc51b9c295db : FAILURE in 17m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/14fda79d247b41ad9985b96636cbd191 : FAILURE in 17m 16s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7c171891ff11419aa4c66cd669a3a529 : FAILURE in 20m 15s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/03e1529ea2c94c09928433fd6b1ce6d5 : FAILURE in 15m 28s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/b0dd878441c3474987a849009ca13812 : FAILURE in 19m 38s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/17c8b2d5f4024e618a01e22430ced061 : FAILURE in 16m 31s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/5d74bb55475a4607b489152f3ca36193 : FAILURE in 16m 47s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/dba522311137490799591c0854a48327 : FAILURE in 12m 25s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/4e217295a9e141fea72c2364d30d4593 : FAILURE in 18m 01s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d565dfe6060e400bb323b741d102cfcf : FAILURE in 14m 38s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/39f67d2f68434bd6b33113c7e78ea1a2 : FAILURE in 16m 36s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/57a400269f3e46b1a7ff493ecdc09daf : FAILURE in 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1b851c76b24272a8ee0c46983f19f18bb54fe9fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-22 09:39:20.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c9254d16ee59473663214c8bdb1d0ac1bb08a494","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 09:46:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65266a1770124b2ab93d5eff2588aeae\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7de59458ca894ba4aad6c88ee20999d9 : NODE_FAILURE Node request 300-0021515845 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/96bbe734cc03423caa5df93398792d0b : NODE_FAILURE Node request 300-0021515847 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eea1eb99ff2941e9ab48b4f7cb58529f : NODE_FAILURE Node request 300-0021515849 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62ec318af4604e02a36ada65ec1466c7 : NODE_FAILURE Node request 300-0021515850 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09be988459f64c36ba2f4ec16dd7867a : NODE_FAILURE Node request 300-0021515852 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74eb63dfbe9e471ab74b1e32edd12c78 : NODE_FAILURE Node request 300-0021515854 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/833cc6a961b24ee58df31bca50023cf5 : NODE_FAILURE Node request 300-0021515856 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/811d800b5f23438982d9ab2fa50bc829 : NODE_FAILURE Node request 300-0021515858 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"408013ccb8383eef5ba205c5b78e11548621c464","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-22 09:50:03.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"72b48c2e1f3f57d6357247ba7dc942b04f4caec6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 10:14:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb477a6f91784bc5afd57ab27ea8a091\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/738e6dc6d0c54afbbe3c05f98c2dd61d : NODE_FAILURE Node request 300-0021516117 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7942730dda494f9fa3107c8453b4ae40 : SUCCESS in 22m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/406831c7a024412fb9896b54751c0c8e : NODE_FAILURE Node request 300-0021516119 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31a866be588943c9bdf9d15f5d137b4a : NODE_FAILURE Node request 300-0021516120 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06b506fa4fe64ea8b6c266822a25f005 : NODE_FAILURE Node request 300-0021516121 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/37b7b73a4a564745a80127edc4eeea62 : NODE_FAILURE Node request 300-0021516122 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b1fb2fca686142c2aa572c6180cf37f3 : NODE_FAILURE Node request 300-0021516123 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b375a19f6d044531b6441e515d719a1a : NODE_FAILURE Node request 300-0021516124 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3603a708c463dcc185f1921be5b2a3eaa2ff8c77","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-22 10:16:10.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"162e27fe4ae7294a3699249a5d8b4c3fed4dc51a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-22 11:27:55.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"70d33f656410bb2709c5a4480f76923713bb24c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 12:51:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c54ce9a5c5094d68bc892e1ba72f60fe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ef01fac4712e434fa67f1f1b07170ace : SUCCESS in 18m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b3f894ff85c5468aba8509f1690070ca : SUCCESS in 1h 02m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0130529d3bc34ca0bbfa38f93e3c93fe : SUCCESS in 15m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1b850b9360514e609ba87820fb85a66f : SUCCESS in 50m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0cc1f24c5008461ebcf63fed2ae47793 : SUCCESS in 1h 30m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791c12d28dc144348599f8998f887e55 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3c096e20c364b79bf48521593c886c4 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87e49cbed2334049b8202c7652869a15 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb44374093e44c84bf7c0e31549f3bcd : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf05f7e771324d5ca705cd2f3e522b24 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7350a92421764e10b9d1b1056785ee67 : SUCCESS in 5m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b68c1657193d4fb683c94d4e691c91c2 : SUCCESS in 41m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/94181e1495cd4e15bffe39c808315449 : SUCCESS in 1h 02m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/96c8e3ff999b476092169327dd09ecbe : SUCCESS in 1h 33m 02s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/4079889acfd14953bd6851ca1bcf83aa : SUCCESS in 2h 01m 13s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2a80d7f12cd44670aff7af94642c5d11 : FAILURE in 1h 30m 09s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/d6cdf4f04fa64ac995478e7deccd0d7d : FAILURE in 1h 31m 16s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/5abf8c7d99bc4f66adc7a045eda312c0 : FAILURE in 48m 15s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/a4611065049e40deb0432e169b99970d : SUCCESS in 1h 43m 36s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/fab73e6eac234116855e833a52554615 : SUCCESS in 24m 46s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/b8404da4c9ec433e86a8d8a831702967 : SUCCESS in 1h 21m 16s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/00f4769d275540969d0cd1c9df02a7f9 : SUCCESS in 1h 35m 40s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/83c517b9a12646888b321b33a340b97e : SUCCESS in 1h 52m 43s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/41070aeb27a3477798b41279a760131e : SUCCESS in 38m 02s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6c9eabc7ce7747819a45da6eec9f502a : SUCCESS in 30m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/328c105a8f2f406ea0f5eae927fd3499 : SUCCESS in 57m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/da4eb901044846f880b82ba7d3477f2a : SUCCESS in 1h 28m 16s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/32a37a1ec2614d38a16cff45b2be63b8 : FAILURE in 43m 02s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7a60948c50f8427b822f19e170e82ade : SUCCESS in 1h 49m 16s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/16a58aeda1f94f83b881cb6c2aaeac3e : TIMED_OUT in 2h 04m 30s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/357bda0081cc40f296608033b7feffa5 : SUCCESS in 1h 06m 04s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2beeea89b6f14a78a1fc34aae5beaa91 : SUCCESS in 1h 14m 00s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f9aa7280f1434d81926a2114bc794ecd : SUCCESS in 28m 59s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/be464783925d48e6addf69709a024eeb : SUCCESS in 56m 02s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6cfc52db4b4f47299dac6f6e8adbcb5d : SUCCESS in 1h 00m 16s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2657ee33485b44afb9bcfe9c7e99454b : SUCCESS in 2h 00m 33s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/87b767be85d346e6bdfd814d9d9c450a : SUCCESS in 30m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d4b924eb9f958060344853d82def680e832f5771","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-28 13:16:39.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"4129e3fd197c2ac3800bca4f3c6b2fb9cc0354e4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-28 15:01:28.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cb25f556577e83c2c7f9f529530c3f38e60b1284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 15:01:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4e019cbe33cd54496aa110471e2e84a2eff60be6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-28 16:21:04.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7f60ab0b9e70119e96a8d65ddbd2a95fbacad553","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-28 16:22:31.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"7b9f6d60c7df185c44e3711b4d7ce36895933910","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 18:39:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7f5b058e5d8459f9869e08bcff12a0f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ec1634d5b987454bb60971d35444293b : SUCCESS in 36m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2caf161ecfad4185963e73cae41d3fd5 : SUCCESS in 35m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79447044232d49e08c9e3eeacb284557 : SUCCESS in 2h 16m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c56fc8f2dbd14434be0cf86cbcd79c4c : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f009ba9f01c46e599db6e46c9f0a1c7 : NODE_FAILURE Node request 300-0021566499 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f28ed71936a14d10a987e5d2e5d77703 : SUCCESS in 5m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4ebfd3aaa508496391c63e43a802d7bf : SUCCESS in 32m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c8c90344280244fdaf3f2eafdb6ecc5b : FAILURE in 28m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"adf19ee7c1d6fe2fb37997b08235eab3e0161be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 19:04:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff93f8e196c54b9a892ad07197082ffd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1e86a6fc5cc44810987adf8ce7447d29 : SUCCESS in 19m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f765b72e9aaa4c97b879a0577bc0b2c1 : SUCCESS in 57m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5b7210a47d8742448519dff6ed79632a : SUCCESS in 18m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ed5579b0337b44cbad1136b205f213a4 : SUCCESS in 1h 03m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/017d820db2f94debaf93fb0a053e980b : SUCCESS in 1h 04m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12750b4687a4263983f4fdd393614fa : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b50109bc95c1471290efb49940dc4944 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/235e7edf29a24f1c9a90875059f6eadb : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d643939f66f462d9973ef688d9ec096 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36d3ad5f01df4366b297cff3e2d76b31 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4c4e1d83914d14ad9633a27be0ba6d : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/de540c6027f64417bb5a54647584b069 : SUCCESS in 16m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0eedca5390d84d44b5a3d97f3d8fd77d : SUCCESS in 1h 07m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ffdf2d70d04e426ab2388197b5f522cc : SUCCESS in 1h 09m 36s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/71ed94013c93495a807607e5f449deea : TIMED_OUT in 2h 32m 46s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9424e84fe1aa4563be586ca1991a5c4f : FAILURE in 17m 59s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/9019f1428452446da4602c4574af87cd : FAILURE in 28m 06s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/9e23ad1336974231b2948bfdc443a89b : FAILURE in 41m 45s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/30b705e8224040ac990fdf1fb7b415a1 : FAILURE in 25m 51s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/7ed8f9186e97419cb4a088f83067971d : SUCCESS in 25m 23s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/5e30b36876524d4e837bd2d8b8b621e0 : FAILURE in 21m 03s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/1cc44768623348a09d5dd4c03006601a : SUCCESS in 1h 13m 55s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/7ba8066d28944d7cab08089e81e2460d : FAILURE in 51m 31s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/aeaebebabda44664ad1bb13995307d46 : SUCCESS in 30m 14s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/df9d7d743c8a41b598bb343339b87404 : FAILURE in 51m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e7f7b0d07d6a4bbf92a7487686a0f7e2 : SUCCESS in 1h 08m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/58b333d6db20438ea282780fcc15b245 : SUCCESS in 1h 25m 17s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b300fdf681e640578a8b2f4a0efa11d0 : SUCCESS in 1h 32m 52s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c8b9998f5c384ba6a114df9591670bbc : FAILURE in 1h 33m 18s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/174f09d667804227b3c073fb4fff89ce : FAILURE in 1h 03m 05s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/6109e093fa0b474cb072c463261f3b93 : FAILURE in 52m 45s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/7f9c065400254bbda2a31b8a8db3eae2 : TIMED_OUT in 2h 04m 20s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/3086439fdd504f58b99ceb446d065b3e : SUCCESS in 34m 04s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9b0f074650da4e3ca7a66411f4b6f1f0 : SUCCESS in 1h 34m 32s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c333333ed6764b5f959b926593245f9e : SUCCESS in 1h 27m 37s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/10961e64271642afae66a1aa313b0c68 : FAILURE in 1h 37m 44s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8339eacfd6064d3c848532ac80e8c66c : SUCCESS in 37m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"de242a5b01c15c2a32f3e17f38de396f3b8baef5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 09:04:09.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3e709c39f501eae2dc88f1adcf1fcc9cfe3b2d4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 10:27:19.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42f60931f9b9467794c88e9693020c8f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9148c84a16f84c93b6e3f44dce79315b : SUCCESS in 1h 19m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bdeb444a1c2645f58023c681380c78c5 : SUCCESS in 21m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/60a2d03601fc42e49369c25cb9093c4f : NODE_FAILURE Node request 300-0021574433 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b2b78df7c1e45198f3eb5d850c999e6 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a808365deb3b4781b02397c39ca7e69d : NODE_FAILURE Node request 300-0021574437 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6643b7df13c04524a5093db3587d289e : SUCCESS in 3m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b821d69cee8d4f7cbab30d131417ac35 : NODE_FAILURE Node request 300-0021574441 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8cb79328295448db862b15bc35394f28 : SUCCESS in 19m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ef7c23a13395824bfd7955189bca6a2e6a47f7be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 11:36:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11f6629c95f7459795075db1ee623f45\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f26b96ff617745bb825cb8778cc1ab88 : SUCCESS in 31m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9862f8fe84ce4f1bb78b0fbb15882fbe : SUCCESS in 1h 16m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dcb4a908877047ebbc4028cb3634fcb8 : SUCCESS in 19m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ccfe06d01a354a7489501e8ae97916ac : SUCCESS in 1h 08m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/12e387c6bf2146929d3e74a69a36598e : SUCCESS in 1h 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ac295168f994580874cc0c3b70ce2d0 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a93c111f0c6446a8f9feeb4e55a9066 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54272d2d80ec4952b1c28382f80707f7 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/129917b5023d425c9d0f88f2dd556025 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c4216ea2d464388bb3c101581fecb7f : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc73e4573414b45a1ded8c133fc2408 : SUCCESS in 4m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0bf28c1eedea4a509c278a8740e7e986 : SUCCESS in 19m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7188084635a64fd28fb45cd246d6a761 : SUCCESS in 1h 12m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0026eefbb5514bf292873e8693df75d9 : SUCCESS in 1h 29m 05s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c84b1aa03bcb49568b4fa149024f3cb9 : FAILURE in 52m 55s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/52c2ef4727c747eb9a8bd47346f7268f : FAILURE in 1h 56m 30s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/fe4c924756344e8da71643aa60740226 : FAILURE in 1h 13m 46s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6a29eec6063a4f8087cf0566046299e1 : FAILURE in 1h 09m 48s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/65f223e0f5444942b9f6f95187066470 : SUCCESS in 1h 32m 42s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3f53731df07b42a4a42a92f20bcbebdf : SUCCESS in 26m 36s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/7c19d17785494b4db64e058c0cfb1418 : SUCCESS in 1h 05m 36s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d14e72be77294fcd9e8fbfb7b93b8436 : SUCCESS in 1h 08m 34s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/5fd6ed274e7e4fe9a3de21f0870f6930 : FAILURE in 50m 52s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d89bdb9b957a481eb1c5ddf139df2ddd : SUCCESS in 31m 23s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0ff57135e8eb43c7a4eab84896fd0c43 : SUCCESS in 28m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f552a737033b4485a13f950e46d13959 : SUCCESS in 56m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c0087cfd5a8d4ba491f8480cad2f433c : SUCCESS in 1h 21m 27s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/1c81b074b0db4bb88d3b2232840da7fa : SUCCESS in 1h 47m 52s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/232964fbae7449d99c61b69aec601b22 : FAILURE in 54m 37s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/95eafbbbc4fa4bf0b84a2d3b0baa597a : SUCCESS in 1h 41m 22s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/52db546486a4439984f7985e15189c53 : SUCCESS in 1h 21m 34s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/aacc9759ea054d0aa33f32df34d86c80 : SUCCESS in 1h 34m 08s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/030828678352417884abed965c264507 : SUCCESS in 22m 46s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/3a2f4de212564101a9159bc1b0e55c4e : FAILURE in 47m 54s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c21cbed801264fc182cdc8ae082461ce : SUCCESS in 1h 05m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/496320edf8d545df903e6f3cb14145ad : FAILURE in 1h 03m 52s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/742a06e12b9a4ff5bfe286177374959d : SUCCESS in 29m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bfc67a6a65cd41e9e73a53d886cdd97dd9f3120b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 11:36:33.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"fba795aaa8201b23369bf5e690046e532ddb99ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 14:11:15.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0449dec9a8e54640a845ce052d11a3cc\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a0873273496e42e1a3d45460910ba392 : SUCCESS in 20m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/74cbf763613248269979efd4a1d7692d : SUCCESS in 1h 03m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b3cc196ea2f84d158c0e891c1de45d5a : SUCCESS in 54m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d78509617984521b86f80618c5c3885 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ae6d5dac2f445c19f38ed437999a18c : SUCCESS in 12m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90f96502ef934256a0f41460a06e5c5a : SUCCESS in 12m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd383bcc6c59401ca67b41c80024d5de : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef15c68c6ea94bc48cf3ab432f7aa682 : SUCCESS in 3m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6e8fe0c6f4ec4666adc4a7f90fc6e472 : SUCCESS in 16m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ecf4aaa3ec5d4fe190b8dab52b472caa : SUCCESS in 1h 04m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/da70363794574183bfe0c55ddcaf0f1a : SUCCESS in 1h 20m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/254d18b2966f4407b0c591debc3d3140 : SUCCESS in 22m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8cb51abfa29a461dac37754d0124528d : SUCCESS in 1h 04m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/60a60e8849104c1c8e855951791e85a6 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":7},{"id":"b2310a5d9deb602cb5e8825c09f659dd81925d3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 14:11:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"22ea1269991fab9f70bef6853182268ae4e8cb40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 14:12:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59a6aae4788344ca87c8117752d8c6cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3905ac6769c64360b1fab702bf957c4c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"3de04436fa0e497a08d5d4312e25bf4b1049cac2","revisions":{"e202c02bd4a49df2546e1ce5d5eafc5e5ae257f0":{"kind":"REWORK","_number":1,"created":"2023-05-02 09:01:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/881948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/881948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/881948/1"}}},"commit":{"parents":[{"commit":"1aeb351d6f2410982fbca65044cb81be583d7344","subject":"base: install adduser on Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aeb351d6f2410982fbca65044cb81be583d7344"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2023-05-02 09:00:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2023-05-02 09:00:05.000000000","tz":120},"subject":"base: handle Debian bookworm adduser.conf","message":"base: handle Debian bookworm adduser.conf\n\nIn Bookworm configuration file for adduser changed.\n\nChange-Id: Ic06f3c0e42fb7854e5b198c9b868f10cd507cd24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e202c02bd4a49df2546e1ce5d5eafc5e5ae257f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e202c02bd4a49df2546e1ce5d5eafc5e5ae257f0"}]},"branch":"refs/heads/master"},"eabc74d1541c0b24710ed3799fa8509dc8051fbc":{"kind":"REWORK","_number":2,"created":"2023-05-02 09:50:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/881948/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/881948/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/881948/2"}}},"commit":{"parents":[{"commit":"20eb6d1ccb0e2957788d07d6d20a77e0e70689e0","subject":"base: install adduser on Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20eb6d1ccb0e2957788d07d6d20a77e0e70689e0"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2023-05-02 09:00:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2023-05-02 09:49:14.000000000","tz":120},"subject":"base: handle Debian bookworm adduser.conf","message":"base: handle Debian bookworm adduser.conf\n\nIn Bookworm configuration file for adduser changed.\n\nChange-Id: Ic06f3c0e42fb7854e5b198c9b868f10cd507cd24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eabc74d1541c0b24710ed3799fa8509dc8051fbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eabc74d1541c0b24710ed3799fa8509dc8051fbc"}]},"branch":"refs/heads/master"},"f28d435cffc56be9bd28d3fd2d793aaec2e6d3ef":{"kind":"REWORK","_number":3,"created":"2023-05-02 09:56:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/881948/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/881948/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/881948/3"}}},"commit":{"parents":[{"commit":"20eb6d1ccb0e2957788d07d6d20a77e0e70689e0","subject":"base: install adduser on Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20eb6d1ccb0e2957788d07d6d20a77e0e70689e0"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2023-05-02 09:00:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-05-02 09:56:30.000000000","tz":0},"subject":"base: handle Debian bookworm adduser.conf","message":"base: handle Debian bookworm adduser.conf\n\nIn Bookworm configuration file for adduser changed.\n\nChange-Id: Ic06f3c0e42fb7854e5b198c9b868f10cd507cd24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f28d435cffc56be9bd28d3fd2d793aaec2e6d3ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f28d435cffc56be9bd28d3fd2d793aaec2e6d3ef"}]},"branch":"refs/heads/master"},"0ef3bde9192f623ee8fb92acfabf43650d77df97":{"kind":"REWORK","_number":4,"created":"2023-06-14 12:07:49.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/881948/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/881948/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/881948/4"}}},"commit":{"parents":[{"commit":"6815f420acde00e3a383a930fbddccf583eef183","subject":"Revert \"neutron-mlnx-agent: pin pyzmq at 21.*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6815f420acde00e3a383a930fbddccf583eef183"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 15:39:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:56:01.000000000","tz":120},"subject":"base: install adduser on Debian/Ubuntu","message":"base: install adduser on Debian/Ubuntu\n\nDebian:bookworm container does not have \u0027adduser\u0027 installed so\nwe need to do it before we touch /etc/adduser.conf file.\n\nIn Bookworm configuration file for adduser changed.\n\nChange-Id: Ic06f3c0e42fb7854e5b198c9b868f10cd507cd24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ef3bde9192f623ee8fb92acfabf43650d77df97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ef3bde9192f623ee8fb92acfabf43650d77df97"}]},"branch":"refs/heads/master"},"c4901c55a18a0d98b4a56b3e6be824a053ba7e1c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-06-22 09:39:20.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/881948/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/881948/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/881948/5"}}},"commit":{"parents":[{"commit":"406022d0721b69dd7bf77182beab791f5dbcdc68","subject":"Revert \"neutron-mlnx-agent: pin pyzmq at 21.*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/406022d0721b69dd7bf77182beab791f5dbcdc68"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 15:39:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-22 09:25:04.000000000","tz":120},"subject":"base: install adduser on Debian/Ubuntu","message":"base: install adduser on Debian/Ubuntu\n\nDebian:bookworm container does not have \u0027adduser\u0027 installed so\nwe need to do it before we touch /etc/adduser.conf file.\n\nIn Bookworm configuration file for adduser changed.\n\nChange-Id: Ic06f3c0e42fb7854e5b198c9b868f10cd507cd24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4901c55a18a0d98b4a56b3e6be824a053ba7e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4901c55a18a0d98b4a56b3e6be824a053ba7e1c"}]},"branch":"refs/heads/master"},"036d963d0edb47906aaf57515d4f7a65ef32e654":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-06-22 09:50:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/881948/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/881948/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/881948/6"}}},"commit":{"parents":[{"commit":"9abe6d7997ab578d536fd215d6571487de7a09c6","subject":"Revert \"neutron-mlnx-agent: pin pyzmq at 21.*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9abe6d7997ab578d536fd215d6571487de7a09c6"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 15:39:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-22 09:49:50.000000000","tz":120},"subject":"base: install adduser on Debian/Ubuntu","message":"base: install adduser on Debian/Ubuntu\n\nDebian:bookworm container does not have \u0027adduser\u0027 installed so\nwe need to do it before we touch /etc/adduser.conf file.\n\nIn Bookworm configuration file for adduser changed.\n\nChange-Id: Ic06f3c0e42fb7854e5b198c9b868f10cd507cd24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/036d963d0edb47906aaf57515d4f7a65ef32e654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/036d963d0edb47906aaf57515d4f7a65ef32e654"}]},"branch":"refs/heads/master"},"3de04436fa0e497a08d5d4312e25bf4b1049cac2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-06-28 16:21:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/881948/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/881948/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/881948/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/881948/7"}}},"commit":{"parents":[{"commit":"70f74eb64101431e23d56c6a7df96d7aab37ce2f","subject":"Merge \"Revert \"neutron-mlnx-agent: pin pyzmq at 21.*\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70f74eb64101431e23d56c6a7df96d7aab37ce2f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 15:39:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-28 16:19:31.000000000","tz":120},"subject":"base: install adduser on Debian/Ubuntu","message":"base: install adduser on Debian/Ubuntu\n\nDebian:bookworm container does not have \u0027adduser\u0027 installed so\nwe need to do it before we touch /etc/adduser.conf file.\n\nIn Bookworm configuration file for adduser changed.\n\nChange-Id: Ic06f3c0e42fb7854e5b198c9b868f10cd507cd24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3de04436fa0e497a08d5d4312e25bf4b1049cac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3de04436fa0e497a08d5d4312e25bf4b1049cac2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
