)]}'
{"id":"openstack%2Fkolla~882238","triplet_id":"openstack%2Fkolla~master~Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746","project":"openstack/kolla","branch":"master","topic":"upgrade-pip-for-tests","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-04 19:35:33.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-04 17:22:32.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"34505":{"account":{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},"last_update":"2023-05-04 19:35:33.000000000","reason":"Change was abandoned"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-05-04 19:35:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746","subject":"Upgrade pip/setuptools/wheel for tests","status":"ABANDONED","created":"2023-05-04 08:21:42.000000000","updated":"2023-05-04 19:35:33.000000000","total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"5a0d8e440da4405a27c8c1e2b7ca6ca910d12384","_number":882238,"virtual_id_number":882238,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-04 15:14:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2023-05-04 15:57:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-04 15:28:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2023-05-04 17:34:01.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-04 08:28:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-05-04 08:28:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-05-04 10:02:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-04 15:28:26.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-05-04 15:38:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},"state":"REVIEWER"}],"messages":[{"id":"f6382f0c3bfbce72f3492fb7b9a2821c57df48a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 08:21:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f325935a6cf21ec86353cd1545d33f27b6424b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 10:02:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9da79a788ad447378052cb4b71f75fcc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7e9dc1c27f7a4fad89aa9103c9cf1ba2 : SUCCESS in 31m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7bdcbf4ebc454ff6ab8495a820b72660 : SUCCESS in 1h 34m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/703584a4643a42859ba3469fccffd213 : SUCCESS in 21m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7c6784f4b6274c79a6edd34d858ce605 : SUCCESS in 58m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/88e0a762f8a84a2ea453dcf8e9253cf5 : SUCCESS in 1h 22m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61b5ff9d3e1349feab1b55c1bb2e08c1 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29e91d686a164ba89d97c61c5c37fba9 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c47e6e115214b7597e0c741170f33e5 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23fe3ae4e7464b41a41ec3977075d027 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d390e284d664083872037091770b994 : SUCCESS in 4m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f625fdb4ab2342b8b496d709093b4504 : FAILURE in 39m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4b45c381d75e4a469bfed0dd09d9db4f : SUCCESS in 1h 04m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/10ae9f5e6a814e3b87197315cba25086 : SUCCESS in 1h 20m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8ea452247fca42e88e19b74dbbbcf27a : FAILURE in 38m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/59d0c906d0974c4392313b769f1b4654 : SUCCESS in 53m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0605a17b3ecd45b39ded25d4f6d13685 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"24d5dcc14985878a2cd401cd753f6462e4bd693b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 10:05:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9b2ef5f2e97b30455a92ecec9a52579114e4142","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 10:32:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2a60aab6d89480d8c3563cd353804c1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0a9a7c0689394ff2bbbe7315c5addf66 : SUCCESS in 35m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b3a9fd0c1b0d4ead9aff9415fa182972 : NODE_FAILURE Node request 300-0021131843 failed in 0s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/25887169855f429fac3227b4042375f8 : SUCCESS in 2h 05m 22s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fdb3120d9cbb427494f0f0b2386a9139 : NODE_FAILURE Node request 300-0021131845 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c53ce9f2a2f54937a9bacb9217f199f5 : SUCCESS in 3m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fc929d62299446b2b38536380bce0e48 : SUCCESS in 32m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/352bd2f916c44b6ca23d77c4430cff00 : SUCCESS in 23m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3dd66f38542ec36e687be241dda7495dc85458a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 11:33:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/235306ebe5d14ac2955afa840a2e7f7b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6ab32606b1cd44f1952d9b81daaf3469 : SUCCESS in 26m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3a5daa4c8f8d47d38c6e81d968593ecd : SUCCESS in 1h 21m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9b94a1168f2f4e128db25471eabc5810 : SUCCESS in 16m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/088f307d25d1462598ed1c036932264c : SUCCESS in 1h 13m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/700cefa769834a62930118f09ab2c861 : SUCCESS in 1h 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68a04a6af5844492b8f6da0f84edd982 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a56689671a1645679333ab7c2f5e52be : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e03a912e6269495bae47efb96b482b87 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f723d8ce362d43c4bbfbfebaa4d4c0f9 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6cb09f897d48bdb55eab0aa6cbffd6 : SUCCESS in 4m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/022fa083b9dd490099303fad481b74d5 : SUCCESS in 18m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a66bada99ad44b879d94bbab7144ea66 : SUCCESS in 1h 16m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/59d6bd5a501342cba4ebda036b96427b : SUCCESS in 58m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e9855a44e8e448568cd73224496e10ee : SUCCESS in 18m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bd00903330554cf485554db0b4437955 : SUCCESS in 1h 05m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/00f6cb05591541868ef6454685a6664b : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a6d622c3132f9bc449ebdf4082b1ac785609977c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 11:57:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50d2189dba7a20750ba9f2656e59db765807b753","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 12:12:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b416ac49b2c75ca2655f0c739b54cb5921b32c4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 13:30:52.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0cafec943033c30dc1dba06bbbd8645084622fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 15:14:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a37795a424924c8db98702908f536dbe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7a80e813d51c41a9bed325bb51a97f8c : SUCCESS in 29m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f998af418b6d46e4983931fb4097f276 : SUCCESS in 1h 32m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/64a33efd49964729a6123cd408bd3a5e : SUCCESS in 19m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a16b3f6458934463ab01f6d5ce3a36fc : SUCCESS in 1h 02m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/920f085a27dd4fbfb4fbe7ae3f9cc803 : SUCCESS in 1h 08m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/802706350ee14555b8c18aeafdbf9aa6 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/817570a77a0f4115b5a0466c90939334 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8710b074d6949629063d4752d4fb723 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ed482d64b16400798f9029405733b6a : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b370f1960fc9485280617cf51cca5ff7 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/61248422b4d5451aacc9c2fca6938f8c : SUCCESS in 17m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f501d104a8c0479cb4caba300a92e17c : SUCCESS in 1h 07m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/abccd7e374834287be024acd360a6c4a : SUCCESS in 1h 26m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7bed490b07774a30a35acdcbed7158c6 : SUCCESS in 40m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d88f5c8122fe4043b9ac432c1b1cac30 : SUCCESS in 53m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9d1ca895a4724943a062b4a035d7d4a4 : SUCCESS in 1h 12m 38s","accounts_in_message":[],"_revision_number":2},{"id":"89d352ed40d1375147b16f3306f5ec97c19c8371","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 15:16:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9c4b4eb23854a6bad83a562ece3ec64\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/63cf4a56967743528bd9182c2e70f3ac : SUCCESS in 31m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/288fcbf50d404f55891ee18cab2f08e0 : SUCCESS in 1h 04m 26s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4f4434a1d88245349ee62d77f8629412 : SUCCESS in 1h 44m 16s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91623206ed13460283fc918edf4b6349 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89b22134ca984805be58a9218cab91d2 : SUCCESS in 3m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6567732d45284b4b8c4c57aedc78b470 : SUCCESS in 31m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/499712250a1c461f8f4caeaeb29687e5 : SUCCESS in 26m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd43ecc390205567a6ea64b8d7a8e53687ec12bb","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-05-04 15:28:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1ee090354dd63c52fcef1df6666028263a94b9e4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 15:57:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"157569c32197cf9d291dfe45649b36b5259b6a8e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 16:05:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"608766e714d52d1dc917c663848a149555237aca","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 16:09:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"830326ffe30495d7398c776fc568a884b56ce0fa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 17:01:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba5db4dd4ebebe1dbca1119e09eef501cb6e006d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 17:04:53.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bd24e66e01263bca3b1ed5e9519510709a21877e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 17:09:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3348823c00c66d91260381492675a760bd4ace43","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 17:22:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22dd03c68fcfe9e428ab9dbcd7005c6629c224a9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 17:34:01.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"5a0d8e440da4405a27c8c1e2b7ca6ca910d12384","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 19:35:33.000000000","message":"Abandoned\n\nwill create another bug report and solve the issue other way","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dae16a49fc8efd4bb6d121662d0f1b338038924d","revisions":{"ae06639dcfbd125233f67cd615839f149ea85da3":{"kind":"REWORK","_number":1,"created":"2023-05-04 08:21:42.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/882238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/882238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/882238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/882238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/882238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/882238/1"}}},"commit":{"parents":[{"commit":"3be1b71948154c80f3ec247084b5049b520602b8","subject":"Merge \"opensearch: move to yum/apt repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3be1b71948154c80f3ec247084b5049b520602b8"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:57:29.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-04 08:21:13.000000000","tz":180},"subject":"Upgrade pip/setuptools/wheel for tests","message":"Upgrade pip/setuptools/wheel for tests\n\nSince we use virtualenv on the several different OSes\nto get the same results we should use the same version\nof the libraries used. At least Ubuntu 20.04 uses the\nold pip library with bad dependency resolver, the issue\ndescribed as a warning in the pip 20.2 and later fixed\nin the pip 22.3.\n\nChange-Id: Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae06639dcfbd125233f67cd615839f149ea85da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae06639dcfbd125233f67cd615839f149ea85da3"}]},"branch":"refs/heads/master"},"dae16a49fc8efd4bb6d121662d0f1b338038924d":{"kind":"REWORK","_number":2,"created":"2023-05-04 13:30:52.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/882238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/882238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/882238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/882238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/882238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/882238/2"}}},"commit":{"parents":[{"commit":"3be1b71948154c80f3ec247084b5049b520602b8","subject":"Merge \"opensearch: move to yum/apt repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3be1b71948154c80f3ec247084b5049b520602b8"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:57:29.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-04 13:30:49.000000000","tz":0},"subject":"Upgrade pip/setuptools/wheel for tests","message":"Upgrade pip/setuptools/wheel for tests\n\nSince we use virtualenv on the several different OSes\nto get the same results we should use the same version\nof the libraries used. At least Ubuntu 20.04 uses the\nold pip library with bad dependency resolver, the issue\ndescribed as a warning in the pip 20.2 and later fixed\nin the pip 22.3.\n\nCloses-Bug: #2018506\nChange-Id: Ia81bc0b864dcc3e94124442fc5ae6e9c9daa6746\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dae16a49fc8efd4bb6d121662d0f1b338038924d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dae16a49fc8efd4bb6d121662d0f1b338038924d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
