)]}'
{"id":"openstack%2Fkolla~882333","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I3112980b4b7cefc165ba6c1b31ccbf182b9a1b77","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-05 12:22:54.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-05 12:22:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3112980b4b7cefc165ba6c1b31ccbf182b9a1b77","subject":"CI: Work around docker-py requests/urllib3 issue","status":"ABANDONED","created":"2023-05-04 15:42:45.000000000","updated":"2023-05-05 12:22:54.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1e9bb50a304586f7c634ac014e2cadcce21d1b21","_number":882333,"virtual_id_number":882333,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-05 11:00:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-04 15:49:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-04 17:08:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"412d091d1b71a328124e6751adfdb49b8521eb20","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 15:42:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79fc97b60128e5ac62b80927260e14d3f07e4904","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 15:49:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"416665cfcbdf78f84a171f1f606c22a2c88b16f4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 15:50:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0da9caf06af86b9a8f5e3f4e46ef1a3a15c7df54","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 15:51:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b453c19f147c4c7f7f5344377603d02fcbe5df59","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 15:52:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85ce23bb856c612a781ba809305422c283ed6e95","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 15:54:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2bff3cef1c03c0c30ee16963809063a7acd44be","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 15:57:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12375a6dca86218adc5f16f62d07e8252dc5a55a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 15:58:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3ee268286b5b6baadb38cf13920a8a42077a976","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 16:26:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c316a3b20ad8e1ebe9b498547c18bb57595aca42","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-04 16:26:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"723c933332173cb0c5670aed090c723b24d404b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 17:08:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d85a971b6dc149eca5a1aa16c23fa014\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/81f27a1a01574537981ce752383a8c21 : SUCCESS in 37m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2c45cd0d96d54085a05b9ef84e305122 : SUCCESS in 58m 55s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/0b961b6e27e346ed8f1bb11f6ce92d16 : SUCCESS in 52m 11s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a00b6740a67b4cfdbc7480977e533c90 : SUCCESS in 1h 09m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c621bf1f394c4572a8b04e3113db8255 : SUCCESS in 1h 05m 58s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f73aae04c7604014b18ea14c2ce2b50b : FAILURE in 45m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6ad15c4786a541bcadfd6794e3d63ba3 : SUCCESS in 1h 20m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d79f1fab679e4896a22a6a9cb69e8bc8 : SUCCESS in 19m 44s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dd1fe7d720b14cb7b69632399d20628b : SUCCESS in 19m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8e9e066cb20c4a819c91f198b9543ea2 : FAILURE in 23m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3eeea807c58a4d9190e75cec2500fdf4 : FAILURE in 35m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51391252c43a4fcb93519e3b31e2a2b8 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d1b916cd102498b93bf63f498ef45c1 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a733f2d5f9143c1a6569b587a8d24e1 : SUCCESS in 3m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e50d3aac05e4ac8b9eadb8896bfff6b : FAILURE in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/193be961c6534d629de6946703dd4a4a : SUCCESS in 5m 39s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/f53eff3905fb45f1864c1fc8274924d0 : SUCCESS in 22m 20s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/4c7a7d2e046544b59c33d4086e2e2e3d : FAILURE in 20m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/785d5231b6db45a6bc764516360bfb7e : SUCCESS in 19m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b879d1a950de44be993ac6a652856da0 : SUCCESS in 22m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c19a069efcae71e03e127ccc33c324d116a172a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 17:37:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fce9063a5f74a5b93f0a19678d789e8\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/194b41bb2bf64e73951fce8df80cc6ae : SUCCESS in 24m 58s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bada1c45d85b496486e8173dacf50217 : SUCCESS in 29m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8bd9e27a7c54dbb8f70ab6d0decca85 : SUCCESS in 1h 50m 56s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bba75a9d5fe14a9fab3cd292ef3d19df : FAILURE in 28m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63a0b894cd3f4e1f94fd70c3042b9b04 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d69aa7b56b443b2b7dd256bf5dd56fe : SUCCESS in 3m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d531e31e76a14aa5b0470666d5833c2a : SUCCESS in 1h 52m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2563ba6809c20b8dfe55d107099803ab6b4c1970","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 06:21:39.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"07d1a0efc8387a749fa196e175492ab7312689eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 07:57:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c08c399d55c4141ab2677f2c75a8a5f\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fe23d1d3ec3d4b649c1f7960aa9b8e68 : SUCCESS in 30m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6081f87f36ad41e1918568526caf052a : SUCCESS in 33m 50s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/73ea6f52a66e457abacc75fd373cf489 : FAILURE in 11m 34s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/937fd2db949144ad85e5515482099051 : SUCCESS in 1h 19m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9ba617d4561b4f54be36c8fdb3a7fcc8 : SUCCESS in 1h 16m 19s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/0272a08974f04d358f4e900312cd6fb0 : FAILURE in 18m 54s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/dd65afddd14b4b8385309ba1c6395e12 : SUCCESS in 1h 28m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/67df2c56ed764e2ea631234fefecee34 : FAILURE in 4m 27s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/80fe3392529c4c7990bb212821f1659d : FAILURE in 3m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ac6af2be4772483c9d727b536d2d1140 : FAILURE in 16m 08s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/70f3f66561f14f55a0fe0935948471d5 : FAILURE in 13m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7135ca2232b04470b67746d0ddbb5552 : FAILURE in 2m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02e2f3861dbf4946ad7a8a537007f2f2 : FAILURE in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3fa188af5dd47d7a761c2da06de94cb : FAILURE in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8f2fa269a2464536a8427dff88a03855 : FAILURE in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21dcbfa70f6409abc0e2503b42165ff : SUCCESS in 4m 37s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/c1712c0f082745799b8c2a9ccc3ae5f3 : FAILURE in 4m 09s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/74677a1dfeb64def9578dc36cff8087e : FAILURE in 9m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9fbf1b36568b47e693faf087fd6b3173 : FAILURE in 4m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9bba382270134528bad33d8356250b3b : FAILURE in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"492b36d7db3c95926212497e78995917de2162cd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 07:59:14.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0e2c2e1a53cc843a6ca9acb162a4c4b90358f81c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 08:11:47.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a22b1b554b194212a9209077da5f9c1fcbc71443","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 08:29:14.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"eb05e3ada5d1ca77ae51d8e3b3398e0547084e9d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 08:30:07.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"a74bbae4cc8c2eea97898fcf410b115ad5545f5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 08:56:56.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5677e72bf1b427987f664ad1d8e58c7\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3049ecc553d8473dada2c60e0b149d53 : RETRY_LIMIT in 5m 18s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/de3c8d35655a439487a454f411420372 : RETRY_LIMIT in 3m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8416b6c4c9cd4a0d828f43775ada48b8 : RETRY_LIMIT in 3m 29s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/78950b5b994f4cf6bc6cdc3438c57f0c : FAILURE in 16m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a5649483ccc44ccbd3f245e20e12f26 : FAILURE in 2m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e431d997804346548726d0230d509e3a : FAILURE in 2m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2a4295190f524f448d4f0dbe31418bd1 : RETRY_LIMIT in 2m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"37ae5157c776e2d7308a1e6b79175b94f6a1aba0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 09:03:13.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"f2712ed004cffb854b407104eba0c3aa51171df6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 09:13:30.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"67b81aafb19cd8a731a80cf6a5941122d9c86bb8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 09:27:12.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"ddef6b67b751c572642502a93012245596162f1f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 09:27:22.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":9},{"id":"a9c0ffab30ef792ccf24aae6a4efac7f7865768c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 10:29:04.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff39ecd0021c40ec87dcc858fd17d63c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4724c67b9c294cec8b8aea7a894fc023 : SUCCESS in 23m 45s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ce5e403c1f247d493777712e94c18b7 : SUCCESS in 26m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9efd83cbe8f4a10bf18e10fb5d2d54d : SUCCESS in 59m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b385c9cf9b434c3abc60d3e650589039 : FAILURE in 31m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/253d63411c4b47e7a20b87b2b2a18a8d : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/500f0b6a11794e3c86ac1906e249ecdf : SUCCESS in 3m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5267b23f51f74ae3ab93c23d20765c22 : SUCCESS in 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ccbee8cd263b0fb006c338a0175e7ce58f5ccc2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 11:00:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95df29a7f83543af9a8dbd5f4ea3cf53\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f00664532a2b428593b93513d81ff262 : SUCCESS in 31m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b88b1f8f412940b4bc2a62779bca19be : SUCCESS in 38m 33s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/c130f2eaabde493dbe6f5393dd17f5eb : SUCCESS in 36m 31s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c0138e2b7ac342848f4fca1729aced97 : SUCCESS in 1h 19m 53s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/54984ed89e254db189b971fcae65b807 : SUCCESS in 1h 32m 12s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/d07b15722daf447ba42cf55570df759a : FAILURE in 34m 25s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0757dfee33cc45ca8774bd614acfe34d : SUCCESS in 1h 23m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/015b6aaf8d4548e0bee5616f9780c31d : SUCCESS in 19m 52s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/81c79a80e3fd480083b6ac4f7efcf6b3 : SUCCESS in 17m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3481e6e0ebd2463cb77fd8b5f05073b5 : FAILURE in 21m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f5df38d49e234462b5c7128fac8a1d0b : FAILURE in 25m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5abc40eda8f47ab9d987a5a95d7aec2 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ef9a8e5628c4f7c900e3dfa54d3c31f : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fb3cd94385548b18150e309cfcdf006 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5f16321f85d4f3d879a2980580280d8 : SUCCESS in 5m 06s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/c24403b8d4b941b6a8d73d31af23c688 : SUCCESS in 24m 35s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8187aac7633c40b795de7fbe87097c75 : FAILURE in 20m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2967b40ec4ca4676b307e50249643c3d : SUCCESS in 40m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/700641cd36e241f5808c5e35615ae0cf : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":9},{"id":"1e9bb50a304586f7c634ac014e2cadcce21d1b21","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 12:22:54.000000000","message":"Abandoned\n\ngood enough for a workaround, needs to be proposed first to master (and ansible-collection-kolla)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"e802c772c90da39d5693ecc4c0d72c218dcf0b7b","revisions":{"4abcea57aebedae787618633c1b7f64f8dc33ce1":{"kind":"REWORK","_number":1,"created":"2023-05-04 15:42:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/882333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/882333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/882333/1"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 15:42:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 15:42:43.000000000","tz":120},"subject":"DNM: pin urllib3\u003c2.0 to fix docker","message":"DNM: pin urllib3\u003c2.0 to fix docker\n\nChange-Id: I3112980b4b7cefc165ba6c1b31ccbf182b9a1b77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4abcea57aebedae787618633c1b7f64f8dc33ce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4abcea57aebedae787618633c1b7f64f8dc33ce1"}]},"branch":"refs/heads/stable/yoga"},"7b26537bf417004419b99171959ce1d79e304f9e":{"kind":"REWORK","_number":2,"created":"2023-05-05 06:21:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/882333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/882333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/882333/2"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 15:42:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 06:21:39.000000000","tz":0},"subject":"DNM: pin docker","message":"DNM: pin docker\n\nChange-Id: I3112980b4b7cefc165ba6c1b31ccbf182b9a1b77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b26537bf417004419b99171959ce1d79e304f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b26537bf417004419b99171959ce1d79e304f9e"}]},"branch":"refs/heads/stable/yoga"},"d13d3c5046cf13e2b7c67ee169290ab97fd8f540":{"kind":"REWORK","_number":3,"created":"2023-05-05 07:59:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/882333/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/882333/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/882333/3"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 15:42:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 07:59:14.000000000","tz":0},"subject":"DNM: pin docker","message":"DNM: pin docker\n\nChange-Id: I3112980b4b7cefc165ba6c1b31ccbf182b9a1b77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d13d3c5046cf13e2b7c67ee169290ab97fd8f540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d13d3c5046cf13e2b7c67ee169290ab97fd8f540"}]},"branch":"refs/heads/stable/yoga"},"d2a477a94a6560bf13c45b60b6dd3393f53d55a9":{"kind":"REWORK","_number":4,"created":"2023-05-05 08:11:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/882333/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/882333/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/882333/4"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 15:42:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 08:11:47.000000000","tz":0},"subject":"DNM: pin docker","message":"DNM: pin docker\n\nChange-Id: I3112980b4b7cefc165ba6c1b31ccbf182b9a1b77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2a477a94a6560bf13c45b60b6dd3393f53d55a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2a477a94a6560bf13c45b60b6dd3393f53d55a9"}]},"branch":"refs/heads/stable/yoga"},"0fd3a3c3f01653ae119a0067367f875a2bb1fa04":{"kind":"REWORK","_number":5,"created":"2023-05-05 08:29:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/882333/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/882333/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/882333/5"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 15:42:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 08:29:13.000000000","tz":0},"subject":"DNM: pin docker","message":"DNM: pin docker\n\nChange-Id: I3112980b4b7cefc165ba6c1b31ccbf182b9a1b77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fd3a3c3f01653ae119a0067367f875a2bb1fa04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fd3a3c3f01653ae119a0067367f875a2bb1fa04"}]},"branch":"refs/heads/stable/yoga"},"87b7d30d8124878d0bb0e59511ed0c75605b7bd8":{"kind":"REWORK","_number":6,"created":"2023-05-05 08:30:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/882333/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/882333/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/882333/6"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 15:42:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 08:30:07.000000000","tz":0},"subject":"DNM: pin docker","message":"DNM: pin docker\n\nChange-Id: I3112980b4b7cefc165ba6c1b31ccbf182b9a1b77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87b7d30d8124878d0bb0e59511ed0c75605b7bd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87b7d30d8124878d0bb0e59511ed0c75605b7bd8"}]},"branch":"refs/heads/stable/yoga"},"f79d5613465cda8870640d3eb8acef34cbe0a1ab":{"kind":"REWORK","_number":7,"created":"2023-05-05 09:03:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/882333/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/882333/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/882333/7"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 15:42:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 09:03:12.000000000","tz":0},"subject":"DNM: pin docker","message":"DNM: pin docker\n\nChange-Id: I3112980b4b7cefc165ba6c1b31ccbf182b9a1b77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f79d5613465cda8870640d3eb8acef34cbe0a1ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f79d5613465cda8870640d3eb8acef34cbe0a1ab"}]},"branch":"refs/heads/stable/yoga"},"2b6b2fb3bd63d54b0fbbca94ce3d479138a984e2":{"kind":"REWORK","_number":8,"created":"2023-05-05 09:13:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/882333/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/882333/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/882333/8"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 15:42:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 09:13:30.000000000","tz":0},"subject":"CI: Work around docker-py requests/urllib3 issue","message":"CI: Work around docker-py requests/urllib3 issue\n\nRevert this when the required patches get merged in\ndocker-py\n\nChange-Id: I3112980b4b7cefc165ba6c1b31ccbf182b9a1b77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b6b2fb3bd63d54b0fbbca94ce3d479138a984e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b6b2fb3bd63d54b0fbbca94ce3d479138a984e2"}]},"branch":"refs/heads/stable/yoga"},"e802c772c90da39d5693ecc4c0d72c218dcf0b7b":{"kind":"REWORK","_number":9,"created":"2023-05-05 09:27:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/882333/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/882333/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/882333/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/882333/9"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 15:42:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 09:27:12.000000000","tz":0},"subject":"CI: Work around docker-py requests/urllib3 issue","message":"CI: Work around docker-py requests/urllib3 issue\n\nRevert this when the required patches get merged in\ndocker-py\n\nChange-Id: I3112980b4b7cefc165ba6c1b31ccbf182b9a1b77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e802c772c90da39d5693ecc4c0d72c218dcf0b7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e802c772c90da39d5693ecc4c0d72c218dcf0b7b"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
