)]}'
{"id":"openstack%2Fkolla~882343","triplet_id":"openstack%2Fkolla~master~I5ec0fbfdaf18cf53a6b5bf91236d3e040daa9d7e","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-18 10:47:07.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-17 13:36:19.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-05-18 10:47:07.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-05-18 09:01:14.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-05-18 09:08:30.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I5ec0fbfdaf18cf53a6b5bf91236d3e040daa9d7e","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","status":"MERGED","created":"2023-05-04 17:00:43.000000000","updated":"2023-05-18 10:49:01.000000000","submitted":"2023-05-18 10:47:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882343","meta_rev_id":"4f6943997f4ff216ae82b32c5deb54d26dfdf5e7","_number":882343,"virtual_id_number":882343,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-18 10:47:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-17 13:36:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-05-18 09:01:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-05-18 09:08:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-05-18 09:09:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-05-18 05:43:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-04 17:24:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-04 17:38:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-10 13:15:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-10 13:15:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-05-10 13:16:22.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-05-17 13:24:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"ca36816f735d82a0125124f918e07758d4f7fd65","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 17:00:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d532c4aad812e175f17a9bf2b96030e940981622","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 17:24:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60232e16558347f1a3b3bc0f432d6a17\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/134f522663214ed0b6133d2b2e0af6b9 : FAILURE in 19m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c193cafdb31a4b81b1959ceb20d85714 : FAILURE in 3m 20s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9d6737df718346cfb0739554d5b5cdaf : FAILURE in 21m 32s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6984e3b1594f48fd933d04aafe9eab35 : FAILURE in 2m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fad441962c134bbbbfb2ffc86c649d68 : FAILURE in 4m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0cb10afa2ace4f4aa17cde44ab5643d1 : FAILURE in 2m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/49e1d53f525a45dd9717835abb91f180 : FAILURE in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbde8f264a5885de59f997aab09d6dd997cb42a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 17:38:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/015d3df361c844f1899a87f25ac034fa\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/849db0d3084d419f8b2000a3b475590f : FAILURE in 13m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/462be3f4a3cf4925874ac67ca2ad644e : FAILURE in 17m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/36c2fd76cc5d4b59bad3ff08b6cca969 : FAILURE in 2m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/da179aac8ac44828a0c7678f84996113 : FAILURE in 14m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0ad28a48c69740d49a2316c487271e51 : FAILURE in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b1e553a234544b88d1b3ecefc968826 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bd1eaee3c4b4a45876106a04841f0cd : FAILURE in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19e69000a5654c80a08ef8c2fc52b9e7 : FAILURE in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/583e87989ca643359a9f2c16758d264b : FAILURE in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bcf1074783c4559a8ef0dbb6fe44036 : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/eff36664330f4faa8f1cbe935bfa73aa : FAILURE in 3m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fc02749cd9704b73b6097411e9dd923b : FAILURE in 11m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/61f474d4b63b41daa16ffcdaff671bf6 : FAILURE in 11m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2e7fee33ab5b466cac6a0fe1f67af92d : FAILURE in 2m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bb6b94da59ba4afdaf10cd0d19c31335 : FAILURE in 6m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dc86533cccea44898f12443d1ca69ca7 : FAILURE in 9m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b60258f92c6dc09f1332f0e71032f1ec1921edc5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 06:20:28.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b75a5cd25e3ffb53d7e40decbee72feb6a613983","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 06:36:53.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39fbb63c4d2148b7a07572bdfc712e16\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e10be775c9b847b597d9a0802455d631 : FAILURE in 8m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9176a9585bd049ae9f3ac417ee274748 : FAILURE in 6m 58s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e69519af498b4732a5beaa7f4079a7d7 : FAILURE in 15m 16s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c3c4127dee247e69ead110af5271921 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1d003ad3ca93426c93c212770095a876 : SUCCESS in 3m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0b292c6141d74250b975ff2be7455811 : FAILURE in 8m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6321816d3e4545cfb62b85abd40ea272 : FAILURE in 6m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5cfebeb60e060311c24894c20729527e06329491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 06:56:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec868dd796f543f9b6a758cf055e8dcd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/991b5ca9e83c4c96b9819495a5db26c0 : FAILURE in 12m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/598aa454841347aeb39caa2d534b1639 : FAILURE in 18m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/81c3c1b9c613407985baf0b9ce0d69a8 : FAILURE in 3m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6d5a8c2d8e474ff88c315b987bd8f458 : FAILURE in 14m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4a9e81b164574838a9c75eb6e45482cf : FAILURE in 14m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f561865110c44b2bad168e77e8e7deb6 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ff463dd49bd45f5bb47017763410137 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b26556386214033aa0113dc72c4b912 : SUCCESS in 2m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41f4cd605d464b269f9997ce10ec2c4a : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fb6c9913da74df88a18429a23f6fa7a : SUCCESS in 7m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/37ca8a10619f4d7bb165003bc9f732cd : FAILURE in 4m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/efd98ec6da1f4a1c98b66bdbb27cdace : FAILURE in 11m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/aea40e35238442b78f6d682ceb0a88ed : FAILURE in 12m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bf193c4350fa4c50941a2e259f851702 : FAILURE in 3m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8272e1c1ffdb426d90ccc64be703ebb7 : FAILURE in 7m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a93a0b2d170447718dd78d900defb138 : FAILURE in 9m 30s","accounts_in_message":[],"_revision_number":2},{"id":"9bf244bbc6707c43a7c1aaed6fc0517852b4743f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-08 11:14:33.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"33f4639394c3f85ea1149cbbde5bfbf111e92d8c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-08 11:14:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a105aaeb1d68e326cbc11351cd3983af441e5807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 13:10:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/800a5ac30293425bbc9cd4c1768216ad\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bf0071387b2e41fcbaa472ca0973a908 : SUCCESS in 50m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/844ab457a1e14634b635f12682c79ba2 : SUCCESS in 1h 47m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a14ee3388c904f819fbdc529794c98ee : SUCCESS in 24m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/64aa73e947e54bacbcbaeaa3799555a1 : SUCCESS in 1h 06m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8adac067365641118f274609ba3f9a3c : SUCCESS in 1h 12m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/031b9cd36db14acea44045f191dc85f1 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aca5ee233a114b0aa46d3ac80a563ab2 : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f76490c78ce2415f9927ad5cd22d623b : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06fc1e95d7424599985faa94c5413492 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e03b4132f54b78aef6101f89dcf095 : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7e1f9225f8094835bd6baed24ab99b68 : SUCCESS in 18m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/87a55e04871049f88836b4523bc21e29 : FAILURE in 25m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5ec8bebe7c1c441ea60769828ec97242 : SUCCESS in 1h 17m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a19382ba48ab4135ae60cfa15b858b64 : SUCCESS in 18m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3ea54be44990484198e65537e2fffb66 : SUCCESS in 1h 14m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/490695ee34504775ab9dc0eb77de988d : SUCCESS in 1h 03m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d2f7a3c37e2cdcdf941ec755f2694cbf639d3e84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 13:36:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/628e6ca4efd64574ae4eba0338320fb5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8dcb38050f844841be0e2f1d2cbf6c7c : SUCCESS in 37m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/14ae1f902b884f1ea188bf8891592d1e : SUCCESS in 32m 33s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/385bac6d939c4480b9bd32ffe3180dc1 : SUCCESS in 2h 18m 54s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6896861c02bc4306962f8b6c83803897 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/51ae3bf6413e44ab8eb84d16b1628ad6 : SUCCESS in 3m 32s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/309b0604f67c4cbd82612fc0d42de0fc : SUCCESS in 32m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eae113f11a944e41a823a145dcb77e0d : SUCCESS in 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"da6af0777d63f58a27dd4712cffc18a6fb60f218","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-08 14:06:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb3f4742395a65577491d8fdfac89bc163176608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 15:46:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98de164027b840159341d148b15b2e89\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a45127b3f4754ca1b4fabfdc22f0cd61 : SUCCESS in 28m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/438f4aa4b9894aaea4966e945f5a9a7d : SUCCESS in 1h 19m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c497a406444a49a3952def12b22b4378 : SUCCESS in 20m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6a546ac9b2184363b575436b403147a5 : SUCCESS in 1h 05m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/52fd0c48345d4e00af7e3853fd3b7dc0 : SUCCESS in 1h 29m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d8b7a28c7be486fb5fcc0dac4a07f52 : SUCCESS in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbd84127b77243d5b6f0ca5269bf2827 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/beb75714b1e5427e98b7e1ad5ae676fe : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6a710980bbd49cb98c2f2a532d81b3a : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68e105d06bc041a4a6fa3763e7c521cd : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/55dae011fc3e43a38f508d88b386c63b : SUCCESS in 15m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6dc546cd74284e849b98a8bc1bcf6be5 : SUCCESS in 1h 28m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b2da58f84c48483d99050c55154536c7 : SUCCESS in 1h 28m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a8249c919cc142578628f80f1ec7783d : SUCCESS in 23m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8ad2ed836dee40c89fe4f11eb975ac24 : SUCCESS in 59m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0c12ee10cf914f2da637768037bddc83 : SUCCESS in 1h 30m 01s","accounts_in_message":[],"_revision_number":2},{"id":"6ca806d8c61adc7168070b8dc42999543fe26fa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 16:39:10.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a90054231b5e45db9647b59d82eed520\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b46d6e478977418ab75ba4a704c9ef5d : SUCCESS in 1h 24m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2f5993074b99485d99d6b26fdac2fb5c : NODE_FAILURE Node request 300-0021164485 failed in 0s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f19eb46ff67b4b449774c1f83aaacc71 : SUCCESS in 2h 31m 52s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2b97d479f4945b5b1d06e17b161ae08 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4895ea56ebe463887063cc84ba11146 : SUCCESS in 3m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ae79f29b85ee4a17a893f47154e01a8d : NODE_FAILURE Node request 300-0021164489 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/db156a8e5eef4f0bbdb38d9efdd05ee2 : SUCCESS in 33m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6725a8cb65b8b8284f4b58222f4ff08b0fa10b0b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-17 13:36:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"713962ce92f6e0dab630300e423556cb62f62695","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-18 05:43:20.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"51a88538ec94df7883da24cb4b00ca3999379d56","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-18 09:01:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f021f1e3571e1a3a2a6dde7b6ebfc5ca3728bee4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-18 09:08:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a8630be13babb31da631edabee494ca661dc23a8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-18 09:09:14.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8f5e4a6488ee4b61f39bdb5bdd5693c3a278ab83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 09:09:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"12636a7d03423dbcd5061cec1a59867cea4691a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 10:47:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6163b464dcfe46a6af0f7ac846856dc7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d7dbad3f02f642aa8de9296804f5b37e : SUCCESS in 18m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bc2441d16e0b4bf0a371d3b5db2057a6 : SUCCESS in 1h 07m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a5c85bd822aa454084361e6ff202139d : SUCCESS in 1h 09m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49833d597c9d4dde838b51419d7dc7c6 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3f902f807fd46f68cd114a4ab0dcd2d : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a2f6f43b7c64fa689c47c840d61b6ee : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4540410bcad44a7ba2fe4337dbb4295a : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05553ff14d6b4a9d8d6c9f5487103ecb : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/43d62c0eb6034e03ae7bb35ebd779afe : SUCCESS in 22m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5c018b08d19a46eb993550c8e9a5e958 : SUCCESS in 1h 11m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ab731011c07f4e5bbd52253f6b070cc4 : SUCCESS in 1h 18m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d4c98efa43304bee81b96520fdfa2ce9 : SUCCESS in 24m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b64325d83bd84b64aa94d92af3cb5508 : SUCCESS in 49m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8d898c8e79f64ac5b3eb00f3e202cb03 : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b7403a5ec1ace3f8340d61eb89ea6073a169fc8e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 10:47:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4f6943997f4ff216ae82b32c5deb54d26dfdf5e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 10:49:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5542903dd8f146d288f8a19f91697dbd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a84bfc34f8c0442f95eee8265af0e936 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"050e26ed89b307a2037f118e314d0a44249a0f89","revisions":{"d3060ae13f8e251e5f17478679aba0e254699933":{"kind":"REWORK","_number":1,"created":"2023-05-04 17:00:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/43/882343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/882343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/882343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/882343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/882343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/882343/1"}}},"commit":{"parents":[{"commit":"115b46f76f1b7cf10ce496a16c0f01ccb5cdd385","subject":"Merge \"CI: Ensure quay repo is public on publish\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/115b46f76f1b7cf10ce496a16c0f01ccb5cdd385"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 16:55:52.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 17:00:40.000000000","tz":120},"subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","message":"Fix docker version check after docker-py\u003e\u003d6.0.0\n\ndocker-py reworked their version function and now it does not\nreturn a string [1].\n\nExample:\n\u003e\u003e\u003e import docker\n\u003e\u003e\u003e docker.version\n\u003cmodule \u0027docker.version\u0027 from \u0027/home/cloud-user/kolla-venv/lib64/python3.9/site-packages/docker/version.py\u0027\u003e\n\n[1]: https://github.com/docker/docker-py/commit/cd2c35a9b699522b282cc4f024efa5699df24896\n\nChange-Id: I5ec0fbfdaf18cf53a6b5bf91236d3e040daa9d7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3060ae13f8e251e5f17478679aba0e254699933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3060ae13f8e251e5f17478679aba0e254699933"}]},"branch":"refs/heads/master"},"050e26ed89b307a2037f118e314d0a44249a0f89":{"kind":"REWORK","_number":2,"created":"2023-05-05 06:20:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/43/882343/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/882343/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/882343/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/882343/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/882343/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/882343/2"}}},"commit":{"parents":[{"commit":"115b46f76f1b7cf10ce496a16c0f01ccb5cdd385","subject":"Merge \"CI: Ensure quay repo is public on publish\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/115b46f76f1b7cf10ce496a16c0f01ccb5cdd385"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 16:55:52.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 06:20:11.000000000","tz":0},"subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","message":"Fix docker version check after docker-py\u003e\u003d6.0.0\n\ndocker-py reworked their version function and now it does not\nreturn a string [1].\n\nExample:\n\u003e\u003e\u003e import docker\n\u003e\u003e\u003e docker.version\n\u003cmodule \u0027docker.version\u0027 from \u0027/home/cloud-user/kolla-venv/lib64/python3.9/site-packages/docker/version.py\u0027\u003e\n\n[1]: https://github.com/docker/docker-py/commit/cd2c35a9b699522b282cc4f024efa5699df24896\n\nCloses-Bug: #2018559\n\nChange-Id: I5ec0fbfdaf18cf53a6b5bf91236d3e040daa9d7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
