)]}'
{"id":"openstack%2Fkolla~882394","triplet_id":"openstack%2Fkolla~stable%2Fzed~I1dc32c01c2a3996b7e51a4930ac1d60a0cc7e613","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-09 09:02:45.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-09 09:02:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1dc32c01c2a3996b7e51a4930ac1d60a0cc7e613","subject":"[stable-only] Work around docker-py requests/urllib3 issue","status":"ABANDONED","created":"2023-05-05 14:14:21.000000000","updated":"2023-05-09 09:02:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":882438,"cherry_pick_of_patch_set":1,"meta_rev_id":"b46217ce469fb0cbc2aba850df004bc95be90fe8","_number":882394,"virtual_id_number":882394,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-08 09:43:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-05 14:14:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-05 16:42:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"215604aefca4caff1c990bbe82c2bd87fec41bc9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 14:14:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* .zuul.d/project.yaml","accounts_in_message":[],"_revision_number":1},{"id":"e2a62ba16c7b08b2458dedfea15a8bac3e3f255f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 14:14:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1420465f95c44dca9c06486fe8874d7f37b39eff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 14:14:46.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/zed.  The error was:\n\n  while parsing a block collection\n  in \"openstack/kolla/.zuul.d/project.yaml@stable/zed\", line 2, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/kolla/.zuul.d/project.yaml@stable/zed\", line 4, column 1","accounts_in_message":[],"_revision_number":1},{"id":"86f4e64d1a21d909bae7afe99b18117cdafaf46a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 14:14:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"87c2e0e38fc0ecc2332420050f520741f11f8e8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 14:14:52.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/zed.  The error was:\n\n  while parsing a block collection\n  in \"openstack/kolla/.zuul.d/project.yaml@stable/zed\", line 2, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/kolla/.zuul.d/project.yaml@stable/zed\", line 4, column 1","accounts_in_message":[],"_revision_number":1},{"id":"8eff783acdb852d98aa88243de9dd198e66fd63b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 14:19:08.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ebec3c71fd81c09be317fd9b9f67b4d187b7579f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-05 14:28:32.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"961771272d02fee6b50b2bd9d66ac941948be08b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 14:49:29.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23314a80409742ed928d4d6489ed79f5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3a133dc324b6437893148dfb9a8fe318 : SUCCESS in 29m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/95164824dbc04b5299489aa4ff521b4c : NODE_FAILURE Node request 300-0021147495 failed in 0s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/843450e373414061a0f11e30de70b8de : NODE_FAILURE Node request 300-0021147496 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/861f2a7738e149bf94ae261e1b464adf : NODE_FAILURE Node request 300-0021147497 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05a77b1f8d714bcf8bc1a6ea7feca1dd : NODE_FAILURE Node request 300-0021147498 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/af0c83685cec4fd3bd8d6ede2a9d5384 : NODE_FAILURE Node request 300-0021147499 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d81fbff9540541f89c897e68fcfb641b : NODE_FAILURE Node request 300-0021147500 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"611edc4d66e9abb52902ed99c4a8e4ca44687524","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 15:43:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a675113b837e4d11a37f0cf7821b7011\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ecd3d9fbf36349189e204e15667a57ff : SUCCESS in 1h 04m 17s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ff67490ea5d54ca29935ca45f7e306e1 : SUCCESS in 1h 16m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/535a9f9c5c284d3c8148dca48183efeb : SUCCESS in 21m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/87dcfe81c06d4222b520d9476c1ea202 : SUCCESS in 1h 17m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/97176801b446408f8e31fbfe5cf99936 : FAILURE in 35m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b2319b5a81d4d60a6b6679d2155dc56 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7aa70185aae449d89aa6a75e3daffab1 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2d0573e76974cd38e0ebd904917e9e8 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ba14226e03340bb96b3398e329b0d14 : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53fc65ee4ea14fa28082a67f6a040e83 : SUCCESS in 4m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b86ca7b966104d20abe9525b7a5aa933 : SUCCESS in 26m 02s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/77e6889edd674f5ba83d61f58c33a203 : SUCCESS in 1h 17m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6c1fa715bb904b53bb45be626ed563e0 : SUCCESS in 14m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/126a0f703ddb4bf39f3c85b264aa5cef : SUCCESS in 57m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/49b2a434d29e4ac1b275633054247e70 : FAILURE in 17m 24s","accounts_in_message":[],"_revision_number":3},{"id":"fcd925bf74d8e99c5be2de200318321c69ced681","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-05 16:42:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b358f23adbe748070ce0e8bd858e35806929090b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-08 06:46:45.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bc576a2a5b56bca82e86ec321e050a0f769885e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 07:45:01.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/744e80c5c5ac4f99beb7e6f25dc3e5de\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d34537a7dd46418fb7c20d62bb437f36 : NODE_FAILURE Node request 300-0021159768 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dcd375d1c2f94398a9bf1679361068c0 : NODE_FAILURE Node request 300-0021159769 failed in 0s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b135d864aa247d6a295c99df5f46a73 : NODE_FAILURE Node request 300-0021159770 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eab9a6c112a24362b8f74a09be55c76e : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/07584e0cdcfc43399d82990eb42b1ae2 : SUCCESS in 3m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f52e71ee9cf34fa4931dd7a09ecfd1fd : SUCCESS in 34m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2c2efd4b2af3495d88d4628421b1a6a3 : SUCCESS in 24m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d8e5d73ef3038bb4761a3345c2a5b5ca3fcdbdaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 08:21:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/050eac52fe744deb8e74299f0f7000d7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/71bff517549c447a953dd7902184aad5 : SUCCESS in 30m 21s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/c8c3beb658634e95a1bf1d3a46447169 : SUCCESS in 1h 03m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f5cc19c7ac024ff298269fc4fb4f9e43 : SUCCESS in 15m 08s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9e15599c9dd4498eaf53d2e835a1fc92 : SUCCESS in 1h 06m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1a89ee03ddfc4f7bae0f8a2e4a90f118 : FAILURE in 27m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d2e71a4ce5741af9e187aa53dc7729a : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c58ce4196af49439fd43c8f5a6a14e8 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0461c27b2af416592eb60c5e0f6e744 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48f83b15ecd9459cac5700dd3f0ffcca : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c064cef38d774cd4b5604c8c5142a3fd : SUCCESS in 4m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/05c3ca746cff4329874d1b3d5fcd0475 : SUCCESS in 21m 05s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/954dd49ad8364b2f8d29bb7976455539 : SUCCESS in 1h 06m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/68fb91ef623044758a53bcd5c550de62 : SUCCESS in 13m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76311e151e2a430a8b30e02c0c582b5a : SUCCESS in 48m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1d59136ad9cb4c0692d4cc908635358f : FAILURE in 16m 05s","accounts_in_message":[],"_revision_number":4},{"id":"45f27413fd162752ab3c002c672943a6daca27fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-08 08:27:07.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1d6c3b5aafe6fe6a8aedf37a5ceb1e4d81445d3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 08:36:51.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad977cbe689444cc83fd4b04aec1406a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f41e63e838da41278ed8a0d1126ae785 : NODE_FAILURE Node request 300-0021160589 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dad574ada60740bd8ced054d7b2c94aa : NODE_FAILURE Node request 300-0021160590 failed in 0s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f69a08f39ef9467eaa1b28234d921b28 : NODE_FAILURE Node request 300-0021160591 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d48811c01b3249dfba788c6faa2109c1 : NODE_FAILURE Node request 300-0021160592 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1a2d84085c94af99f887fe9090ea0c9 : NODE_FAILURE Node request 300-0021160593 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1758a2b0a4fb4e88bc405ffc691ec28a : NODE_FAILURE Node request 300-0021160594 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/70733c17379449719d7ba1f7bb402b29 : NODE_FAILURE Node request 300-0021160595 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"948d719d0b32fa5b53c071b95caa79f6203e253d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 09:43:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6a03da4107441dca7ed189d57d4f0a0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/205d1ebbbc3d4743accfad0b26dbb2c0 : SUCCESS in 29m 36s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/2386fe06a03a44bebdbd457807173995 : SUCCESS in 1h 11m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7c88ec883cf94a19810e519ed7340260 : SUCCESS in 38m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9ab7290b32df4a868dcf11ceda6650d2 : SUCCESS in 56m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/362418a9f6c4492796eb44c00108b2b8 : FAILURE in 22m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eec5aaef06074cec9e2fda04aa100051 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a9a3e54d3ae4294abb9db5a1dfef704 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7ffc4d7b8e2427eb72261b0757bea0f : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48b6c4ea31894e57ac803ae54878a86b : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb975d10d404ee6ab56819fbb3f3bd0 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d7ece05ddcd54a34b771dd7f624a0012 : SUCCESS in 21m 01s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/e98313ebd28e4a94be11f93ca6b1028c : SUCCESS in 1h 02m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/17f2888dd5d24b889324e8709cd7a9ec : SUCCESS in 17m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/006e36ca501d46b4b8a334949dbe3301 : SUCCESS in 47m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/63e9cb287d7d40e68867321cc68656df : FAILURE in 36m 58s","accounts_in_message":[],"_revision_number":5},{"id":"b46217ce469fb0cbc2aba850df004bc95be90fe8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-09 09:02:45.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"76dde53c5117beeb800696bcbc9aa7503791fc19","revisions":{"10bdd29c2f189090652e6c548ce16cdc3514f3f2":{"kind":"REWORK","_number":1,"created":"2023-05-05 14:14:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/94/882394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/882394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/882394/1"}}},"commit":{"parents":[{"commit":"5ae0803da344612b81e82c785688e35f9eeb73bc","subject":"Merge \"ovsdpdk: add libdpdk-dev\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ae0803da344612b81e82c785688e35f9eeb73bc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 12:28:05.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 14:14:21.000000000","tz":0},"subject":"Work around docker-py requests/urllib3 issue","message":"Work around docker-py requests/urllib3 issue\n\nRemoving requirements-check job because it will fail and\nthis change will not get into global-reqs because it\u0027s temporary.\n\nRevert this when the required patches get merged in docker-py ([1])\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/882437\n\n[1]: https://github.com/docker/docker-py/issues/3113\n\nChange-Id: I1dc32c01c2a3996b7e51a4930ac1d60a0cc7e613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10bdd29c2f189090652e6c548ce16cdc3514f3f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10bdd29c2f189090652e6c548ce16cdc3514f3f2"}]},"branch":"refs/heads/stable/zed"},"00b88b2cd26ac4a177b87c081515eb6c52f21983":{"kind":"REWORK","_number":2,"created":"2023-05-05 14:14:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/94/882394/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/882394/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/882394/2"}}},"commit":{"parents":[{"commit":"5ae0803da344612b81e82c785688e35f9eeb73bc","subject":"Merge \"ovsdpdk: add libdpdk-dev\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ae0803da344612b81e82c785688e35f9eeb73bc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 12:28:05.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 14:14:43.000000000","tz":0},"subject":"Work around docker-py requests/urllib3 issue","message":"Work around docker-py requests/urllib3 issue\n\nRemoving requirements-check job because it will fail and\nthis change will not get into global-reqs because it\u0027s temporary.\n\nRevert this when the required patches get merged in docker-py ([1])\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/882437\n\n[1]: https://github.com/docker/docker-py/issues/3113\n\nChange-Id: I1dc32c01c2a3996b7e51a4930ac1d60a0cc7e613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00b88b2cd26ac4a177b87c081515eb6c52f21983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00b88b2cd26ac4a177b87c081515eb6c52f21983"}]},"branch":"refs/heads/stable/zed"},"3e636c83adf09c74b009a65cda37fa3c2dcdaf4b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-05-05 14:19:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/94/882394/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/882394/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/882394/3"}}},"commit":{"parents":[{"commit":"5ae0803da344612b81e82c785688e35f9eeb73bc","subject":"Merge \"ovsdpdk: add libdpdk-dev\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ae0803da344612b81e82c785688e35f9eeb73bc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 12:28:05.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 14:19:07.000000000","tz":0},"subject":"Work around docker-py requests/urllib3 issue","message":"Work around docker-py requests/urllib3 issue\n\nRemoving requirements-check job because it will fail and\nthis change will not get into global-reqs because it\u0027s temporary.\n\nRevert this when the required patches get merged in docker-py ([1])\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/882398\n\n[1]: https://github.com/docker/docker-py/issues/3113\n\nChange-Id: I1dc32c01c2a3996b7e51a4930ac1d60a0cc7e613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e636c83adf09c74b009a65cda37fa3c2dcdaf4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e636c83adf09c74b009a65cda37fa3c2dcdaf4b"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"},"e395467e79d1c9313f7930296e6fd4864be4674b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-05-08 06:46:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/94/882394/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/882394/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/882394/4"}}},"commit":{"parents":[{"commit":"5ae0803da344612b81e82c785688e35f9eeb73bc","subject":"Merge \"ovsdpdk: add libdpdk-dev\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ae0803da344612b81e82c785688e35f9eeb73bc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 12:28:05.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-08 06:46:44.000000000","tz":0},"subject":"[stable-only] Work around docker-py requests/urllib3 issue","message":"[stable-only] Work around docker-py requests/urllib3 issue\n\nRemoving requirements-check job because it will fail and\nthis change will not get into global-reqs because it\u0027s temporary.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/882398\n\n[1]: https://github.com/docker/docker-py/issues/3113\n\nChange-Id: I1dc32c01c2a3996b7e51a4930ac1d60a0cc7e613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e395467e79d1c9313f7930296e6fd4864be4674b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e395467e79d1c9313f7930296e6fd4864be4674b"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"},"76dde53c5117beeb800696bcbc9aa7503791fc19":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-05-08 08:27:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/94/882394/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/882394/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/882394/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/882394/5"}}},"commit":{"parents":[{"commit":"5ae0803da344612b81e82c785688e35f9eeb73bc","subject":"Merge \"ovsdpdk: add libdpdk-dev\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ae0803da344612b81e82c785688e35f9eeb73bc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-05 12:28:05.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-08 08:27:07.000000000","tz":0},"subject":"[stable-only] Work around docker-py requests/urllib3 issue","message":"[stable-only] Work around docker-py requests/urllib3 issue\n\nRemoving requirements-check job because it will fail and\nthis change will not get into global-reqs.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/882398\n\n[1]: https://github.com/docker/docker-py/issues/3113\n\nChange-Id: I1dc32c01c2a3996b7e51a4930ac1d60a0cc7e613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76dde53c5117beeb800696bcbc9aa7503791fc19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76dde53c5117beeb800696bcbc9aa7503791fc19"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
