)]}'
{"id":"openstack%2Fkolla~882520","triplet_id":"openstack%2Fkolla~stable%2Fxena~I3d10973a9857355e68007d70258e8d7b0ab36b85","project":"openstack/kolla","branch":"stable/xena","topic":"qemu-block-rbd-source/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2023-05-09 13:50:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3d10973a9857355e68007d70258e8d7b0ab36b85","subject":"Support qemu block rbd in nova-compute (source)","status":"MERGED","created":"2023-05-09 09:07:14.000000000","updated":"2023-05-09 13:51:05.000000000","submitted":"2023-05-09 13:50:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882520-qemu-block-rbd-source/yoga-stable/xena","cherry_pick_of_change":882679,"cherry_pick_of_patch_set":2,"meta_rev_id":"f76735a80603d48984aa2388a7293fd98467543e","_number":882520,"virtual_id_number":882520,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-09 13:50:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-09 09:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-09 09:09:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-05-09 09:30:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-05-09 09:31:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 09:09:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-09 09:30:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-05-09 09:31:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-05-09 10:57:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-09 12:24:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"174b04426b625851f72a7579ef73e6869e52b34d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-05-09 09:07:14.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"660ce8a58d7a42b2415ec1b2632479b91332c3aa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-09 09:09:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"caceb064acc3c6c095b48c4e1d141717ee815779","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-09 09:30:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4147ac63187c4aba284b23505d107cc5d906b560","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-09 09:31:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72afeec9f8fc829d6210ae2840a6ef2a11bd8ab6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 10:57:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84ed11ac87c44f74a50a1f866f56b39a\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf905880e2fb46299b035beb5c458f22 : SUCCESS in 49m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8373a2cd32c4966a9627aa7935a37c4 : SUCCESS in 42m 28s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f3977b617c19455b8ff4e19a2f1a56c2 : SUCCESS in 1h 46m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7cb866f51ba406baf7f027e999d07c7 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe0265aaf1b04cf1b9ca127bce3fd32b : SUCCESS in 4m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b566388a37a4f12863e4f40f50163ad : SUCCESS in 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3992a83386a1fca3793d69ab9ef4a5360221c3e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:24:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4985c33cbd9c4012a1c7a5128d7d0de2\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5054bcd66b6e4d3e993d704163a482d9 : TIMED_OUT in 3h 02m 23s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/56f7e7ffddf844df99391be7432df62e : SUCCESS in 29m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ed63e74de3c74851b2f75739050e9ba7 : SUCCESS in 1h 14m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/47689665fc9e471bb4541470783216a6 : SUCCESS in 1h 06m 13s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/85210f69db02457b9b94157be776a233 : SUCCESS in 16m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e301b6aaa2d74bf1a25552e0849a16de : SUCCESS in 15m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b9f53339d9274e7aae6beac58ab6362c : SUCCESS in 1h 10m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12f140cdf655446bba425423a5254771 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9e86b20d2fb408fa42684c170be4382 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/666ba938f1d64164a1506d8ff6e21a89 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8a80d3c0da04fef9a066face336ebcd : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3011595989d4e279f1eeb07a0f0bdf1 : SUCCESS in 4m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/68e89b0176db4cf1937465dc9e2b9fec : SUCCESS in 33m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c92cad9ad8cc48249a9f447a9ab5de7c : SUCCESS in 29m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ce4beddf1444d8fa6f58f79e0218950 : SUCCESS in 1h 00m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b030bd06ec334e17927791233b5d9369 : SUCCESS in 45m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3547f3946e5b4e098847828659bd5bea : SUCCESS in 1h 04m 04s","accounts_in_message":[],"_revision_number":1},{"id":"26cc9080af2ada5c65c0ef835af34105ff0bcf24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:24:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f32e5304593b8a9adfaead4c6ebef1376dd5acfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 13:50:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/939830710c0e41e4864fa224c7b8267c\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3f5eccca2c9f421b93a775cd796b14c0 : SUCCESS in 27m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/db84824dd4f9489290004f21e95332d3 : SUCCESS in 1h 07m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/641ed09028954da28cf9c61ca6772b5e : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eec150fa5f743d3b251049b860fe0bf : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44cec96446214abfa2c6d682fc4fb23e : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11b92e0f5194641897dff64d879569c : SUCCESS in 5m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0642c1b2978545af861a7ffcbc542e3f : SUCCESS in 15m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4cab87788d84897a6e542b6753e53c1 : SUCCESS in 56m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f032db8a0b974007a5a64758ae4cd599 : SUCCESS in 1h 19m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a00b3b112a13dc60ce5f9e0fe2ee7f27751f4ccb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 13:50:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f76735a80603d48984aa2388a7293fd98467543e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 13:51:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f65c4679a93469890f21ca3efc9d3a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c797bbaeca34b8b836a9f55e2ac6099 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6aeeb092b6d2fa5a3b9df07eb84f7d22b9020ee","revisions":{"b6aeeb092b6d2fa5a3b9df07eb84f7d22b9020ee":{"kind":"REWORK","_number":1,"created":"2023-05-09 09:07:14.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/20/882520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/882520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/882520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/882520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/882520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/882520/1"}}},"commit":{"parents":[{"commit":"ab72fd7430e2b70764b20589a14f0010c84f4ca0","subject":"Merge \"ovsdpdk: add libdpdk-dev\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab72fd7430e2b70764b20589a14f0010c84f4ca0"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-05-09 08:58:59.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2023-05-09 09:07:14.000000000","tz":0},"subject":"Support qemu block rbd in nova-compute (source)","message":"Support qemu block rbd in nova-compute (source)\n\nCorrects backport of change 860230: Support qemu block rbd in nova-compute | https://review.opendev.org/c/openstack/kolla/+/860230 to also consider source install type.\n\nChange-Id: I3d10973a9857355e68007d70258e8d7b0ab36b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6aeeb092b6d2fa5a3b9df07eb84f7d22b9020ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6aeeb092b6d2fa5a3b9df07eb84f7d22b9020ee"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
