)]}'
{"id":"openstack%2Fkolla~882565","triplet_id":"openstack%2Fkolla~master~I4c86bf69f2f7ab3cd57ae8d7b6fe8966cd34033a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-08 08:57:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4c86bf69f2f7ab3cd57ae8d7b6fe8966cd34033a","subject":"CI: Use centos wheel for rocky","status":"ABANDONED","created":"2023-05-08 13:46:09.000000000","updated":"2023-09-08 08:57:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c7a8383ad596f9c07c65acf12434aa2b956efd4","_number":882565,"virtual_id_number":882565,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-08 15:16:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-05-08 14:15:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-08 14:15:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-08 15:16:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a0d8b34a0029b660490087d4c07bbc01e0cd74b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-08 13:46:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6796c88ce76e7a4a9d80429172e0f1e117022dfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-08 13:49:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba64355db6c9a1973718b69758ddf10d0c1bcf07","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-08 14:15:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ced37f07d3b26ef3cd060b3d62a370d33bfea4e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 15:16:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e88705e20f7d4576ab8a92b3bdb39730\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/44fc51b7a5814da2a9be4a68b68e2983 : SUCCESS in 24m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/776a391b37874d4b9ae82286ac8c19f9 : SUCCESS in 1h 12m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6a711403d2974c9185c8502eb095bb82 : SUCCESS in 26m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/59619e72ded04350a654b842655919c8 : SUCCESS in 1h 09m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/eefc488252114fa78771d703df52da93 : SUCCESS in 1h 08m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/662f84b260444a44bb3492d6c493eaee : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c4fca5305134f42b5f1686670718efe : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15c97efae2504729bb96360caea87229 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36dbf250475848d0ad002b57b9411c37 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4463590c7aad4ae68a531f39a8293d43 : SUCCESS in 3m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/907216e333134d51ba98f60ed7aa1448 : SUCCESS in 17m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/96d7f3a5c13c4247b7b41650ae575683 : SUCCESS in 1h 13m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7708d5fe5fbb4465bb9caf75dd721276 : SUCCESS in 1h 23m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9b46431ec93546c68fb279fcf5f25752 : SUCCESS in 23m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6a004fca0d374d2da7c4e9456acc8b8b : SUCCESS in 54m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4021308d3a884b31938dd252e177fd68 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":2},{"id":"8feff5c4d7e5c62b1a89907a287e4a657ee52f30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 15:34:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0027ad4a4d34cc897e239436e0ca627\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9305f4b88f0d43819375672118eaffae : SUCCESS in 48m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/29ec23e468544402b758d733f96738c5 : SUCCESS in 39m 57s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/99f5fa6036f24d45bebc75f6c7b6c1e7 : SUCCESS in 1h 38m 35s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/054d42a5f7954ffeb9934ac24245945d : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3c7d4454780b4fdc9e031b69ecd1d05b : SUCCESS in 3m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7212f5fccb9f4e42878751c59af25580 : SUCCESS in 37m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9a3784d647c2413d844722226ba27503 : SUCCESS in 39m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c7a8383ad596f9c07c65acf12434aa2b956efd4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-08 08:57:20.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3a27dae0bb6b184ffe005592ae61b4753ac6b878","revisions":{"24a7866a9bdee54ec2f85742dd48d3ea3962c790":{"kind":"REWORK","_number":1,"created":"2023-05-08 13:46:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/65/882565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/882565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/882565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/882565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/882565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/882565/1"}}},"commit":{"parents":[{"commit":"115b46f76f1b7cf10ce496a16c0f01ccb5cdd385","subject":"Merge \"CI: Ensure quay repo is public on publish\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/115b46f76f1b7cf10ce496a16c0f01ccb5cdd385"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-08 13:45:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-08 13:46:06.000000000","tz":120},"subject":"CI: Use centos wheel for rocky","message":"CI: Use centos wheel for rocky\n\nChange-Id: I4c86bf69f2f7ab3cd57ae8d7b6fe8966cd34033a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24a7866a9bdee54ec2f85742dd48d3ea3962c790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24a7866a9bdee54ec2f85742dd48d3ea3962c790"}]},"branch":"refs/heads/master"},"3a27dae0bb6b184ffe005592ae61b4753ac6b878":{"kind":"REWORK","_number":2,"created":"2023-05-08 13:49:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/65/882565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/882565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/882565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/882565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/882565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/882565/2"}}},"commit":{"parents":[{"commit":"115b46f76f1b7cf10ce496a16c0f01ccb5cdd385","subject":"Merge \"CI: Ensure quay repo is public on publish\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/115b46f76f1b7cf10ce496a16c0f01ccb5cdd385"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-08 13:45:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-08 13:49:50.000000000","tz":120},"subject":"CI: Use centos wheel for rocky","message":"CI: Use centos wheel for rocky\n\nChange-Id: I4c86bf69f2f7ab3cd57ae8d7b6fe8966cd34033a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a27dae0bb6b184ffe005592ae61b4753ac6b878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a27dae0bb6b184ffe005592ae61b4753ac6b878"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
