)]}'
{"id":"openstack%2Fkolla~882679","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I3d10973a9857355e68007d70258e8d7b0ab36b85","project":"openstack/kolla","branch":"stable/yoga","topic":"qemu-block-rbd-source/yoga","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2023-05-09 12:34:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3d10973a9857355e68007d70258e8d7b0ab36b85","subject":"Support qemu block rbd in nova-compute (source)","status":"MERGED","created":"2023-05-09 09:03:23.000000000","updated":"2023-05-09 12:36:20.000000000","submitted":"2023-05-09 12:34:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882679-qemu-block-rbd-source/yoga","meta_rev_id":"68072091d678c76813b33e765b74f56e2a5bafaa","_number":882679,"virtual_id_number":882679,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-09 12:34:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-09 09:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-09 09:08:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-05-09 09:22:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-05-09 09:30:02.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-05-09 09:31:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-05-09 09:22:28.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 09:08:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-09 09:22:28.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-05-09 09:30:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-05-09 09:31:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-05-09 10:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-09 10:53:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79f7f0d1fcc7b6240ae26918572e3a78b73751f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-05-09 09:03:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4020cd57e9166032307e2fe17eec1ab02eb4a7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-05-09 09:04:50.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a7b39c6f6dcc71446729e6d9f614a026fb60e802","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-09 09:08:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c56b503c68b8add54c0067534f4b194baa277bf8","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-09 09:22:28.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7e82c8c5bde3aa96d518fff09ef54d9ff7009426","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-09 09:30:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9789245dbe420bcf380bf44aea11d25dcdba8d76","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-09 09:31:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0e2fd4a952114b4c2d3c41c5b5d90f27349f4549","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 10:49:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f0a7fbf47464eb3a5df0764cf38453d\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c55086e334114938b476496d051718b3 : SUCCESS in 38m 00s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/13e6d86682e4458baa021ed2659d0bbe : SUCCESS in 47m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5eb791ca9fb54640bc34ea72382b13fe : SUCCESS in 42m 36s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/97e6d5bdccf7496788b9104f3f936b39 : SUCCESS in 1h 39m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38fc3bd85e784619aa267ec3ecd6e604 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bd76d55524b46eb833d128cfdd889f9 : SUCCESS in 3m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2cc46cbd0ccc44b08ea6f7ed9ad6cbb8 : SUCCESS in 41m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5808a19933104cd3d7db2c08ee9ed15c56e765e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 10:53:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5074889d39ed4ac4bfc7bbc6d5f8484a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6d947dec7d8841d087c2f410cb77e953 : SUCCESS in 29m 41s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e90c9ea3a3d64b22ba831fa100fa72b8 : SUCCESS in 28m 08s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/f39104ef7d01432488a91dc45dc08577 : SUCCESS in 43m 09s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7b98b19f115349678a32c39105cb9b96 : SUCCESS in 1h 04m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f49e9316577547ffbc34aaae38a1439b : SUCCESS in 1h 23m 34s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/4ad1515cf07d49a388fea46f7783e32e : SUCCESS in 1h 29m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3d19508c99044f7abccff430d44459a0 : SUCCESS in 1h 37m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7bf139d8e7eb4efabd48da62beaa1f42 : SUCCESS in 23m 06s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/27915b878b504510bd071652c2803db6 : SUCCESS in 24m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0196a76ee65a4dbd996fd77eadae26ba : SUCCESS in 50m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/93cc073382e640b0a417f3360e4c5342 : SUCCESS in 1h 23m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/035cc666ab4e4f89aef8ebeea90c8485 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f96ab0246c734bf781a720b1f72e49f4 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9e71539abb44a62a4478ae3104bc668 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/020c74ee6da5473b82357817eb53b3af : SUCCESS in 5m 09s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/deba561b239840b68f05fc9d131ebf30 : SUCCESS in 30m 12s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/187aa4a902b9443a893cee23a4dc5453 : SUCCESS in 59m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7ad724d3393144618cd3a85c6cb47cc1 : SUCCESS in 21m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/28e5da5373514e2697f0644589561914 : SUCCESS in 33m 37s","accounts_in_message":[],"_revision_number":2},{"id":"33d38d008ded69cc9851cf113ffd02581b4941d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 10:53:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"db87aa74e371b5df6870f5d5f6047c7a8aecfeec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:34:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4426b8a63b964fdf858baeb7a583e5ed\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5e55b1e9a2004ec2bd8e54c15533e843 : SUCCESS in 39m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2329a5e90e3e404381efe714eb4872c6 : SUCCESS in 1h 14m 58s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e266fc4279174f8292abb6aa7eeaf30b : SUCCESS in 1h 22m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/af4e4db45bc742e9b9c6dd853e9ecbe5 : SUCCESS in 22m 06s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/37a239e7954b403493e0a9ec5ceb920e : SUCCESS in 1h 07m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7630f798eca848448d79805d7dc82f19 : SUCCESS in 1h 15m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/929bcf5de65c42d0b8620588acabc876 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acaf58fc9fd54364ab35f4c5e99b95c9 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31e191aa09ba466b8ff8aa8a8aac41ec : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f9ea6b1d4a4fcea1db56c5cfe2ec91 : SUCCESS in 4m 30s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/7c22faebb5c54c3cbfb3561b8973d34e : SUCCESS in 1h 01m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4cf1cddd2b3a49f2884aef92e359a421 : SUCCESS in 24m 56s","accounts_in_message":[],"_revision_number":2},{"id":"2dd11507839fff43528f9bb134698cea792f5c7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:34:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"68072091d678c76813b33e765b74f56e2a5bafaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:36:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19c6b694a2524f47961b3866eacbc69f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf97d2cd475c42f7876a09822dbf6130 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a28b7d7a0bac0f265a997f7576db9cb82b2022c1","revisions":{"1dd38c3ea05ac7609bf6b75d1ef238a08d12a60b":{"kind":"REWORK","_number":1,"created":"2023-05-09 09:03:23.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/79/882679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/882679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/882679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/882679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/882679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/882679/1"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-05-09 08:58:59.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-05-09 08:58:59.000000000","tz":60},"subject":"Support qemu block rbd in nova-compute (source)","message":"Support qemu block rbd in nova-compute (source)\n\nCorrects backport of change 860230 (878835) to also consider source\ninstall type.\n\nChange-Id: I3d10973a9857355e68007d70258e8d7b0ab36b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dd38c3ea05ac7609bf6b75d1ef238a08d12a60b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dd38c3ea05ac7609bf6b75d1ef238a08d12a60b"}]},"branch":"refs/heads/stable/yoga"},"a28b7d7a0bac0f265a997f7576db9cb82b2022c1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-09 09:04:50.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/79/882679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/882679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/882679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/882679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/882679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/882679/2"}}},"commit":{"parents":[{"commit":"6c4813ca5134e3afeee3f3cee19320c56ebb5290","subject":"Merge \"Glance-api fails due to absent multipath tools\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c4813ca5134e3afeee3f3cee19320c56ebb5290"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-05-09 08:58:59.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2023-05-09 09:04:50.000000000","tz":0},"subject":"Support qemu block rbd in nova-compute (source)","message":"Support qemu block rbd in nova-compute (source)\n\nCorrects backport of change 860230: Support qemu block rbd in nova-compute | https://review.opendev.org/c/openstack/kolla/+/860230 to also consider source install type.\n\nChange-Id: I3d10973a9857355e68007d70258e8d7b0ab36b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a28b7d7a0bac0f265a997f7576db9cb82b2022c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a28b7d7a0bac0f265a997f7576db9cb82b2022c1"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
