)]}'
{"id":"openstack%2Fkolla~882772","triplet_id":"openstack%2Fkolla~stable%2Fxena~If5bec23cb4945fea7eba87a4c9252b1ae201ce05","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-10 09:14:47.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-10 08:24:05.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"If5bec23cb4945fea7eba87a4c9252b1ae201ce05","subject":"fluentd: cap elasticsearch again","status":"MERGED","created":"2023-05-10 07:04:34.000000000","updated":"2023-05-10 10:40:30.000000000","submitted":"2023-05-10 10:39:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882772","cherry_pick_of_change":882289,"cherry_pick_of_patch_set":4,"meta_rev_id":"85d695813f217cce0b8a83b730160afdc47c0f47","_number":882772,"virtual_id_number":882772,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 10:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-10 09:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-10 08:24:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-05-10 09:12:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-05-10 09:14:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-10 07:28:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-10 08:29:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-10 09:12:13.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"411e188918eede1f15863616038bd61d5c052001","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-10 07:04:34.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.\n\nThe following files contain Git conflicts:\n* docker/fluentd/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"c044270d5bfd1cf9a065c2fbf1c918b1bb984ac2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-10 07:06:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7193560c575d2f9b400bd91aedefeaca8caff11c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-10 08:12:21.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"a604522252dea8a4fb16d224540162012791b864","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-10 08:24:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d25f65059dd99e2e64451c54fa69eaacc48e52d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 08:29:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b6d17d31bde40c984b56b18ede6e241\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d80253c7d4f5417b996f17f3fb071167 : SUCCESS in 33m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d0636b67f03a4ddd8fee685dca9338f6 : SUCCESS in 31m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3dd1cec6a2964d42bc2f3f2228124878 : SUCCESS in 1h 13m 34s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d89042a274a8470cae7a2e63961fd6a7 : SUCCESS in 1h 09m 39s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/de5b2b8c66fd42f59200e91b56d85790 : SUCCESS in 16m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b202baa488b3463982478481ab07dd1e : SUCCESS in 17m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0aacef2a2ea0431085c2481e786cb07d : SUCCESS in 55m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9758e1754e594e318d46c9c3a2498c97 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/982b3f547b3947708d6bb6ab1774c7e7 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c414dc0245af4a1faf7ffac0aa420985 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91b51f498fe24ce7a13671fceec10e4c : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/356c4a135e1b4f36844cba9e1693c99a : SUCCESS in 4m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43c9f9a9a55b44aabb5b762eba4eb271 : SUCCESS in 19m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/286556aab517467dba120faa2de17c66 : SUCCESS in 15m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae63de33d7c7479dbd08adea2a8ad2b7 : SUCCESS in 50m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc5c0045d70c4ad6901a1020e6f4f063 : SUCCESS in 40m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/76aa63821c4d49b4bab4e6df3cf9783e : SUCCESS in 57m 05s","accounts_in_message":[],"_revision_number":2},{"id":"f4aa5858ed376b25b959ba741014ed3ec28dd011","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 09:01:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a17f69a3cbcf4cb99a33e9077ef7348d\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fdd88149852e4acdb5b143bd90e42936 : FAILURE in 22m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb7a7050a3344a319256b18fae56f8dd : SUCCESS in 18m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/11effd8947aa4468b21e81f191a3595b : SUCCESS in 1h 53m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8ffbe18ec23471ca54c633113cf5eee : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa55879d404f44f790eec9e378a6ce7b : SUCCESS in 2m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5de97668bafc4a56a677c414d40d61e3 : SUCCESS in 1h 03m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6a8632632ef5079e826aca9f531682445a513c75","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-10 09:12:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"49f532e110920a1e3f239e65ce13dfec00b5e748","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-10 09:14:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0ae4d54294b3fa14e8a19c6eb49809c12a83615b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 09:15:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"22da652f8cec87f4ecdff628702425b3f977a95a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 10:39:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/008dd57670bb49e3b0010fe194601631\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a001964d40474c00bf87300fcd6d8b67 : SUCCESS in 31m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3f95d1c2300e4fd0b21d9c8239d5108c : SUCCESS in 1h 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/279c596c48b74d529f9ec75ebdf229e6 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bd2e0c5e31c4f4490470d8e42f74168 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41caf74e62574a27bb05e4a2909c070b : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db29cdb9abad4660a1fae7a7ba9018ce : SUCCESS in 5m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9fbeb3afe92444b8054fe643fe48dc7 : SUCCESS in 20m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0af2573543f843fbac4eb8a0b8cc8190 : SUCCESS in 47m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e509674bffb74022a26c0a4036c99875 : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d5be54e7c78079f91f478fb1c115c2a97da72a35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 10:39:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"85d695813f217cce0b8a83b730160afdc47c0f47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 10:40:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31e4ca75e60c403691706bb1a452ae79\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5ae7144a88420d9d7eebd9720c8e23 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fed3e69d6c8b366c7698f6ffc9a322ebaf437e34","revisions":{"72d0024691ead358a34c9b31d872b4e128230900":{"kind":"REWORK","_number":1,"created":"2023-05-10 07:04:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/72/882772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/882772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/882772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/882772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/882772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/882772/1"}}},"commit":{"parents":[{"commit":"b6aeeb092b6d2fa5a3b9df07eb84f7d22b9020ee","subject":"Support qemu block rbd in nova-compute (source)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6aeeb092b6d2fa5a3b9df07eb84f7d22b9020ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 13:09:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-10 07:04:34.000000000","tz":0},"subject":"fluentd: cap elasticsearch again","message":"fluentd: cap elasticsearch again\n\nCloses-Bug: #2018463\nChange-Id: If5bec23cb4945fea7eba87a4c9252b1ae201ce05\n(cherry picked from commit 8e8ef4c38999b95cd989fc4ba7580874807e706d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72d0024691ead358a34c9b31d872b4e128230900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72d0024691ead358a34c9b31d872b4e128230900"}]},"branch":"refs/heads/stable/xena"},"fed3e69d6c8b366c7698f6ffc9a322ebaf437e34":{"kind":"REWORK","_number":2,"created":"2023-05-10 07:06:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/72/882772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/882772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/882772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/882772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/882772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/882772/2"}}},"commit":{"parents":[{"commit":"b6aeeb092b6d2fa5a3b9df07eb84f7d22b9020ee","subject":"Support qemu block rbd in nova-compute (source)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6aeeb092b6d2fa5a3b9df07eb84f7d22b9020ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 13:09:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-10 07:06:07.000000000","tz":0},"subject":"fluentd: cap elasticsearch again","message":"fluentd: cap elasticsearch again\n\nCloses-Bug: #2018463\nChange-Id: If5bec23cb4945fea7eba87a4c9252b1ae201ce05\n(cherry picked from commit 8e8ef4c38999b95cd989fc4ba7580874807e706d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fed3e69d6c8b366c7698f6ffc9a322ebaf437e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fed3e69d6c8b366c7698f6ffc9a322ebaf437e34"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
