)]}'
{"id":"openstack%2Fkolla~882785","triplet_id":"openstack%2Fkolla~stable%2Fzed~I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-18 09:58:26.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-18 08:44:16.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-05-18 08:43:38.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee","subject":"opensearch: move to yum/apt repos","status":"MERGED","created":"2023-05-11 12:59:04.000000000","updated":"2023-05-18 09:59:30.000000000","submitted":"2023-05-18 09:58:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882785","cherry_pick_of_change":880809,"cherry_pick_of_patch_set":18,"meta_rev_id":"4012bbcdf758870f48cb00cfb3bf17b8a035f096","_number":882785,"virtual_id_number":882785,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-18 09:58:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-18 07:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-18 08:44:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-18 07:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-05-18 08:43:38.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-11 12:59:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-05-11 13:20:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-11 13:21:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-11 13:53:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-18 07:56:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-05-18 07:57:14.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"e1c8da6e835682407408487cdfe9039556636af0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-11 12:59:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/opensearch/opensearch-dashboards/Dockerfile.j2\n* docker/opensearch/opensearch/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"b129c49e1bb4e3a6fb5666fcc96cec29e3cc9b70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 13:20:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b42004c35864708b5c431c01f2e0e96\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d2251148c07346e099a935b755a70877 : FAILURE in 5m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b3a7de16f1e64554bb933d3614119230 : FAILURE in 2m 44s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f01c01980d3481db0a6261fc120d941 : FAILURE in 20m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f48612d0463470f9b5bc7262323ac1b : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a48a3e52885483c99ab760cb3721a53 : SUCCESS in 3m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c50251e8d33c43779a259a71751a8770 : FAILURE in 7m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6be8a52cc8f54a3bb7e77174a9c57431 : FAILURE in 3m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b45269c710c00988e2c2478426d0ee660ca15a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 13:53:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/308143ba62574639b4bc2bed8d27b3f1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c190dcdb7e04493693a8faeac3e8a6b0 : FAILURE in 4m 35s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/06c8b902aa4a47a397c5b59a387b7a63 : FAILURE in 13m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a60d2d77b81f4542876caec140ae8371 : FAILURE in 5m 08s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/88e5cc0351074062b0c6bd549a740e43 : FAILURE in 16m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/59d5b2036564485babe8ddd25ae01597 : FAILURE in 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d3ea8dd3234312a22930e23ca20a72 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4e08fab233147789ebd6dc000b5588c : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/849cfe8144314b7a89fd38f3050d86cf : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eef768d8d38c4d95b1fd21eb4803f018 : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26c594309870452bad6a9c302207d0ed : SUCCESS in 5m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1b2ba8faece84ebdbdffe458e4037fa9 : FAILURE in 4m 20s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f6fc31f93f61465386366b118dbb35ec : FAILURE in 10m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7cf6c8668b7e4239b63cf6efbcaca4c2 : FAILURE in 4m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/407f88d7c4af4d79a7b4e1c6d498b638 : FAILURE in 8m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/75480c0582dd459aad1d4758dbb26b8d : FAILURE in 9m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a09f7da1b13020fb481f5f967aa5021372d58751","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-18 05:23:12.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1fc406796417247db2be8eab0534b243a61345e8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-18 05:23:38.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"6cea988e0c97e27931c9336a582c9ea2f5aa3f55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 06:49:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0618525424f54d738c8ef1fe24037205\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c93edaf33e684159aeb46f323ef60ee9 : SUCCESS in 24m 07s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/67f55304e6b04c45ab38995f6fbad351 : SUCCESS in 1h 04m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/05912657cd214d4ab731dce1b9f7deaa : SUCCESS in 22m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/91a9d681f03d4c1dbbc9cc0fdcd51052 : SUCCESS in 46m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0e5ae866c0e444bebe3b1a96616b5dd7 : SUCCESS in 1h 20m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/546326be104e413b9b92f809b356e981 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/012aa228f91a4d44926eb56baa05891b : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f81d5b6daf1a43e1a3722fbb637f3098 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e650543d02442f8aabea378402fe55c : SUCCESS in 2m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1be08328075c40429a90b37478f61edd : SUCCESS in 6m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/617b975478464209bd47e0163538f92f : SUCCESS in 21m 02s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1cb61fc06fc6445a95eb91f4dfd5dc00 : SUCCESS in 47m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b0f2b510a659495986af667ef1e4d9ad : SUCCESS in 19m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d856f34073e54d03bf9063795ffb90a9 : SUCCESS in 45m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9cda2592c74540e9a70e0b25ebcfefaa : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":2},{"id":"7bcac94c2940dcef9404fd0fcbe7ea56cfe1baef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 07:36:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6aae1bc5c154bfe955668cb276894d1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/dbbcd755d6d7487193ad5321ed0798d6 : SUCCESS in 1h 13m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/61e04e1892794ee0a27a9c7ed15b08ce : SUCCESS in 1h 14m 58s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2380b5c66e5e4185a2c315a25f8c9c81 : SUCCESS in 2h 11m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/baccc18835724344b64ecae0118ccb7a : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71a9660cc9684fff8381cfa73d2051bd : SUCCESS in 3m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9e8396b48eee43f0b79ff9080a311289 : SUCCESS in 1h 17m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4b5c35159b1848c2ad27de5ac976f88f : SUCCESS in 1h 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"81e7fe207d0ae59579b4cde0022857b921a68966","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-18 07:56:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f137ac628f7baf06c7a51c1e1ec2b6399f68e105","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-18 07:57:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"29be9995e78143f1418abddbf68228aaa26c19a1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-18 08:43:38.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"49852452ed0c5579e676538edf01965e2139b0d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 08:43:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb078cacb853978187fe85c3c25900f999267ae7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-18 08:44:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"29f8e6c22166239b78c7961e510ce81a138d891f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 09:58:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20eb6c46f61547be86eaa9d1583be1d2\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d168119e1d25403bafa2e87e015cab2b : SUCCESS in 25m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c1f0752035e847ddbcb79b9529303c4e : SUCCESS in 1h 00m 30s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9f883bad7c32456c8e079c1ae3527f5a : SUCCESS in 1h 07m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb21401092ad440186752c76b331c9df : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fe44a348ec34a93b14e0333c0d0cb4b : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bbc9e75b6dc4ec39b8e12105616e548 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43185134b65d4eedbf9f4b5ad51c3b4f : SUCCESS in 4m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/43dc2670084445ea8c821e86f774c359 : SUCCESS in 26m 58s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/aa41fb079ae04f5eb8c049680c3e5d81 : SUCCESS in 1h 04m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/60c69b606d66468a8a5e331864b951e0 : SUCCESS in 15m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/28f8af91a6884955ac2e5cfe89165e74 : SUCCESS in 56m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b490deb46fa7422191b78b5e3b83f898 : SUCCESS in 1h 08m 18s","accounts_in_message":[],"_revision_number":2},{"id":"8275043a2f7e6cab2648138de234cc0726db054f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 09:58:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4012bbcdf758870f48cb00cfb3bf17b8a035f096","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 09:59:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c59252a18cbf4da6b0eee651f75c4e55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0807c992704d4310a10c544d4470e111 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aab455c5eb063b5e843bc0919bbae8f1687cf1d5","revisions":{"cefed65fe9eb43dc2cc9b6cfb2cf502c5d341de0":{"kind":"REWORK","_number":1,"created":"2023-05-11 12:59:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/882785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/882785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/882785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/882785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/882785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/882785/1"}}},"commit":{"parents":[{"commit":"5ae0803da344612b81e82c785688e35f9eeb73bc","subject":"Merge \"ovsdpdk: add libdpdk-dev\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ae0803da344612b81e82c785688e35f9eeb73bc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-11 12:59:04.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cefed65fe9eb43dc2cc9b6cfb2cf502c5d341de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cefed65fe9eb43dc2cc9b6cfb2cf502c5d341de0"}]},"branch":"refs/heads/stable/zed"},"aab455c5eb063b5e843bc0919bbae8f1687cf1d5":{"kind":"REWORK","_number":2,"created":"2023-05-18 05:23:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/882785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/882785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/882785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/882785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/882785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/882785/2"}}},"commit":{"parents":[{"commit":"5ae0803da344612b81e82c785688e35f9eeb73bc","subject":"Merge \"ovsdpdk: add libdpdk-dev\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ae0803da344612b81e82c785688e35f9eeb73bc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-18 05:23:04.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aab455c5eb063b5e843bc0919bbae8f1687cf1d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aab455c5eb063b5e843bc0919bbae8f1687cf1d5"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
