)]}'
{"id":"openstack%2Fkolla~883381","triplet_id":"openstack%2Fkolla~master~I61af58fb564a60df1bc2630d7e1c73308838fe28","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-22 06:46:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I61af58fb564a60df1bc2630d7e1c73308838fe28","subject":"debian: Use Antelope extrepo","status":"ABANDONED","created":"2023-05-17 14:32:17.000000000","updated":"2023-11-22 06:46:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e1b0f15228a4523567a8cc6a9e115b7ded1a68b3","_number":883381,"virtual_id_number":883381,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-17 16:07:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":1,"date":"2023-05-17 15:18:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-17 14:34:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-17 14:34:21.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-05-17 15:18:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-17 16:07:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dae3bfa5a9040a598a91b0dc0c6605771f67e70b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 14:32:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"717796b3e240bcef3b1e726eb4f86ee907a0e521","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-17 14:34:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04d9cb36b95f1c19f265376429052b3a5191bc7a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-17 15:18:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"30f9eca2e4b7477bdf908eb4babe38d44bde2df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 16:07:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50f1a08342c44482a81f4366bc1c2c3e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/23e6b92bd01840849d38b09a9403da52 : SUCCESS in 17m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/efd90c1df620428c99752516384844d9 : SUCCESS in 1h 20m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0e5f34a9de804d56992ff47956410efa : FAILURE in 5m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8d3ed71ae3784052bc117089c1a7df9b : FAILURE in 14m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4269a9898dea43c8bb849359c092efe5 : FAILURE in 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/478843ac896d4a5d900dcd8aa7dee3ff : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/549a7aa762d54ed8b33111f98825ab67 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c1d8eac8f9247668250c5fad9161bf6 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abe52c0a353a40a89420af8f904bb37b : SUCCESS in 3m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f380163be2a24041a7ef942f44864cae : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46407b151ac842cba023e09cb855717f : SUCCESS in 4m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/becc3c7bb1e64cca9086a56d406d336b : SUCCESS in 15m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bea584b1738b4f29b4604d7baaff1708 : SUCCESS in 1h 06m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/89055bfad4af454b8355a0592fde50c0 : SUCCESS in 1h 28m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9c84b4d0e0af49bf91e80928d2aebe6d : SUCCESS in 17m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/650e07224b9f4e2e816b388da947460c : SUCCESS in 58m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/96d01e93254b4eee9c2890adda3c488c : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3a18b65998f59ec9d455f90f8b495a878a6df82e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 16:10:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28378b1c93804fb8a8cdd01d35019651\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8c6f14425755493686f83302f7b181a8 : NODE_FAILURE Node request 300-0021234150 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/346a1125917f4018a53895544df69d37 : FAILURE in 9m 26s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f36a28215e8b4cc382c4dce6941d6abe : NODE_FAILURE Node request 300-0021234152 failed in 0s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3444f9af3c4a4dc2821631e6f5e2a5b6 : NODE_FAILURE Node request 300-0021234153 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ca0bc1db30c45e79f0344b97a0feedd : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f9583f2a1c14ff6808360d2b27d9820 : SUCCESS in 4m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/abbca6eb15db4b8ea79d622be8cdd268 : SUCCESS in 1h 30m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/12db54046b7c4872bf9d47da744f5695 : SUCCESS in 1h 30m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1b0f15228a4523567a8cc6a9e115b7ded1a68b3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 06:46:32.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c36aceef705b39c6c9adbe40604cd05da5428d5","revisions":{"0c36aceef705b39c6c9adbe40604cd05da5428d5":{"kind":"REWORK","_number":1,"created":"2023-05-17 14:32:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/883381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/883381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/883381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/883381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/883381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/883381/1"}}},"commit":{"parents":[{"commit":"d965b1b3da9b8c2780a91781834a46f5886bd9c2","subject":"toolbox: pin ansible.posix to \u003c1.5.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d965b1b3da9b8c2780a91781834a46f5886bd9c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 14:32:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 14:32:14.000000000","tz":120},"subject":"debian: Use Antelope extrepo","message":"debian: Use Antelope extrepo\n\nChange-Id: I61af58fb564a60df1bc2630d7e1c73308838fe28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c36aceef705b39c6c9adbe40604cd05da5428d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c36aceef705b39c6c9adbe40604cd05da5428d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
