)]}'
{"id":"openstack%2Fkolla~883536","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I13fe5b46fed3f1ad14eab9fae513617a3d2053c8","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-24 14:53:14.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-05-24 13:33:58.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I13fe5b46fed3f1ad14eab9fae513617a3d2053c8","subject":"fluentd: pin fluent-plugin-elasticsearch to \u003c5.3.0","status":"MERGED","created":"2023-05-18 14:39:41.000000000","updated":"2023-05-24 14:54:23.000000000","submitted":"2023-05-24 14:53:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883536","cherry_pick_of_change":883408,"cherry_pick_of_patch_set":2,"meta_rev_id":"142e425570644c6b991519784421a8cf6b20f792","_number":883536,"virtual_id_number":883536,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-24 14:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-18 15:52:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-18 20:11:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-05-24 13:33:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-05-24 13:33:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-18 15:49:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-18 16:07:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-05-18 20:11:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"d0fa75088fc5c76ffe621de1475fcc5c0637ab39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-18 14:39:41.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"0d6feac4c423d0ef5667665366925ed4261d31d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 15:49:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aae4076f40f44738d5e74ab633bc404\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d64b24b834e941f5b8b7e553b1ad101e : SUCCESS in 46m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e141cb2219d54cde88d1f08b09e84168 : SUCCESS in 48m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f389a45f077146039847e6a5c1097bdc : SUCCESS in 1h 05m 06s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/de78c3521f46491a9fecbc0f3e98b8a5 : SUCCESS in 1h 04m 42s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/48e32483008441cba15e4fcd49d1e005 : SUCCESS in 42m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/06ab2a8842284b04930fdaf1ceecbfc8 : SUCCESS in 34m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7ffb11da19574a4e817291384827aacf : SUCCESS in 58m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/195db87048a642e197079008d05e2194 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6097f8e1722e463f95c59697391bef28 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7790bb6850cd4e37b94f79a8b9187e0e : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/accd6baa671749c59222ea6be486b606 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6300e2584ce43f79dea773f0a32ccf8 : SUCCESS in 6m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/15237e5007944be3944b5fe42da48fa2 : SUCCESS in 35m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5aeb7fbfa7d24e6995371b51c2349251 : SUCCESS in 37m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8532469691114533a63e7653df7d55aa : SUCCESS in 48m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/144111631fc04a68a1d837a3047d0a4a : SUCCESS in 58m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b9d5c9cec2b511490015f2eba82ddc2e9009a73","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-18 15:52:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c60a599e143f8ca6a2cbdc2c0894f3166ce4af85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 16:20:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc61f7eae7fc4496b636a083e17892f1\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/da9f270eeb2544bf8a12e2e588d1bc17 : FAILURE in 40m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/99158015a8d542bd8987009723f0cca3 : SUCCESS in 1h 40m 29s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/79a3f0ebcb5647bb949f842c1050c4f8 : FAILURE in 59m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dda1153a6e1d4f4bbdce677361e7479e : SUCCESS in 1h 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"257c8130febddeaa68b63656d4d3a8fa7cf9a0a5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-18 20:11:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e668ae8e88630f5962d9970488237ccd53159803","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-24 13:33:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2efa56793115e66b27ae719a790392bf8259605a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 13:34:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a81d3737ced494ba4ef6c6bc2a20a98f844a6af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 14:53:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eef7e7a1652423b8a6935c7c914cb19\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e6bd7527ae1a452b93968583e131cc61 : SUCCESS in 29m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ef245c47342c48f48162bffcc1de57bf : SUCCESS in 1h 04m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e0a22c4d984fdba1ab131aa0fa50c6 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1527819a4824741a5f0685a530bc18c : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f40c8c50b13f444ab83973f4e8b582db : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8705e35dc46240259f0ba976e5077c56 : SUCCESS in 4m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef57e3d3b5914c7092969b05bf2c4d81 : SUCCESS in 23m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4460fbd7eb2a49b684360c2499fd94db : SUCCESS in 54m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6f5caf2990a005ab76000481e1a1d2fdbe6b9d59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 14:53:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"142e425570644c6b991519784421a8cf6b20f792","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 14:54:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35aab3bd2f494a019f44385703efa2bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30664086eba743f8a1632cea6fe2ed50 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fccf664b4b64f2711c4eea616114ae60aa4f78c","revisions":{"4fccf664b4b64f2711c4eea616114ae60aa4f78c":{"kind":"REWORK","_number":1,"created":"2023-05-18 14:39:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/883536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/883536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/883536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/883536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/883536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/883536/1"}}},"commit":{"parents":[{"commit":"6f3fe16dc2d8b1198aff0f1ba32fa5622821ed6f","subject":"fluentd: cap elasticsearch again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f3fe16dc2d8b1198aff0f1ba32fa5622821ed6f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 11:47:50.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-18 14:39:41.000000000","tz":0},"subject":"fluentd: pin fluent-plugin-elasticsearch to \u003c5.3.0","message":"fluentd: pin fluent-plugin-elasticsearch to \u003c5.3.0\n\n5.3.0 requires faraday\u003e2.0, let\u0027s pin stable branches requiring\nElasticSearch support.\n\nCloses-Bug: #2019557\nChange-Id: I13fe5b46fed3f1ad14eab9fae513617a3d2053c8\n(cherry picked from commit 64ab2e983de8bec2da394110334c88a7b35d5773)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fccf664b4b64f2711c4eea616114ae60aa4f78c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fccf664b4b64f2711c4eea616114ae60aa4f78c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
