)]}'
{"id":"openstack%2Fkolla~883716","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-17 08:54:17.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-16 11:34:39.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-08-17 08:54:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee","subject":"opensearch: move to yum/apt repos","status":"MERGED","created":"2023-05-22 09:06:22.000000000","updated":"2023-08-17 08:55:22.000000000","submitted":"2023-08-17 08:54:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883716","cherry_pick_of_change":882785,"cherry_pick_of_patch_set":2,"meta_rev_id":"3b50ed86bbe91f30dddf4d46c89a1b6009f91bc4","_number":883716,"virtual_id_number":883716,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-17 08:54:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-16 12:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-16 11:34:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-08-17 07:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-08-17 07:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-22 09:06:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-05-22 10:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-24 16:18:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-16 12:17:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-17 07:09:48.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"8d4205f84ab1b975bf4301e803e624e9dffa311e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-22 09:06:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2\n* kolla/template/repos.yaml","accounts_in_message":[],"_revision_number":1},{"id":"55fd7e7e7182686281ea685ca6191d77eb90dba4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-22 09:08:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2857fc2990b36c9e04b657f7d4d908969df45429","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-22 09:10:24.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"786aeac0a7db8f4b5866044e1906e074ca0dd353","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-22 09:11:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"40285b6edf7bc4dd655ffa3e5726f2c5a8be6e8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 10:42:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06e0ce46f2254096aa6cb33066ec4f47\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3cfc0dcf26eb42cf809166a6463053de : FAILURE in 30m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5e5b313900cf48cca5b95c8bbb18fa8f : FAILURE in 33m 23s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/14b11b849e54419aa8ff0822039ed99f : FAILURE in 32m 40s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0aa8fbdb3fbe4c48bc00258d45137ed8 : SUCCESS in 1h 18m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/caeaff48cf86451f973f595faef37b79 : SUCCESS in 1h 11m 20s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/13421e01c2df43d49907c6a49e467740 : SUCCESS in 1h 03m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/481f16852d4b4568a31876d9c9697ad2 : SUCCESS in 1h 30m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a9e7ab1e121744afb43eca7b898ee021 : SUCCESS in 25m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7d4d26d77d3145c1a225f8c654fb0112 : SUCCESS in 18m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/24c3e6e7a08d449a82565ef4c7930d29 : SUCCESS in 54m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2e7cdd74fea840cb935f48fe840b26ef : SUCCESS in 1h 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a438fc9ac72045109a67d8de9da477ec : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60acc48d8ba3447d9972dd7f1db5758c : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca76947d0587485bafd65a15f23b5caa : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2992a666a1774198a12adebe6ec224e7 : SUCCESS in 6m 09s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/4be85101d7924e8984d9c6c341d8a528 : SUCCESS in 24m 40s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/da23a76d0c104f02b8b97fc3a33ea4a9 : SUCCESS in 1h 03m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/96b55a5e5c9b4aa8ae7a597f726d0428 : SUCCESS in 16m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/943811042a124fa192769bd2887f6cbc : SUCCESS in 20m 49s","accounts_in_message":[],"_revision_number":3},{"id":"25a5a2c60be2af9c7fc674587424b6987a237d23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 11:25:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65e4ab59453e4c408c416cb7000ce872\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/961ce91561b74955aff3c1f9cfe48d41 : FAILURE in 23m 56s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d11e039fc3f2493bac33de43f2292c21 : FAILURE in 2h 14m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/72a716d83af946a8b76dcce97e251ffb : SUCCESS in 1h 46m 41s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e91c3055f190467dbe14e0384d220ff0 : SUCCESS in 1h 05m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ab4b68b27cd4395b46f28959bf0e9ab : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/553224b94d884907ad81f453d38bb38c : SUCCESS in 3m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/081a965713c44aa48fd34342c640b48d : SUCCESS in 1h 36m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"48b84ba9595cbad1621a9686e6a06ebccb467579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-25 15:46:00.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d0d03c628ce6961f73aadfed49c9695bc0c4817a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 17:11:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/329bbf71fcdb4fd2a22fabcb018ced7d\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2db711428a3491eac9b62ae9578aebf : FAILURE in 41m 26s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7694c118cf204415b38ae12b0995e48a : FAILURE in 42m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2cbd6ce613544f12af7b043288889ea0 : SUCCESS in 41m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/df0ef23f57f646e196843860cc5a33e9 : SUCCESS in 1h 24m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5655bc9c5ddf4c2592571667da127bdc : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5831c272a41a4bfd8681774bebe5d88c : SUCCESS in 4m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e232112b043c44eda12c517426dd1b34 : SUCCESS in 41m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5794839e2ba58cdadd2837f7baa916b94c5e9368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 17:17:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd4b2f0ab06743efa84facb3439213a1\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1ffdadab8201462c8d72b08500779498 : FAILURE in 33m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/775fb9b434304855b8745dad922bda35 : FAILURE in 31m 05s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/1fec2b2bf6cb4e08a9149419ab2b8387 : FAILURE in 29m 57s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9d3b2ec561b04a6faa85f6c695d1775b : SUCCESS in 1h 08m 38s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/56fa34e3c4854683ba842e34bf4f7da2 : SUCCESS in 1h 10m 13s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/993978857d864f738d7cb600d3ebaab3 : SUCCESS in 1h 06m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f4cdf2d7a38345949ad705b1ff19170b : SUCCESS in 1h 20m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f80b65c7697a406e9b4cddc503c08b55 : SUCCESS in 32m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5ea45532861b4ad18eb0bde4487d62d5 : SUCCESS in 24m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4ebbeee7b4344818b1cfbc4551091402 : SUCCESS in 58m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/269599c5fead4639b748f1b0e96ac11d : SUCCESS in 1h 25m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6161d1e543145bebc2eaaea63889200 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0a734d101a64f7b8b2e4d48e7077daf : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbe9634966274f1480e3a16e64921fb2 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/650d34111e564bdfbc62ae4c93dd5862 : SUCCESS in 5m 19s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/a9aaf501c60a4af7ab17362c89631a00 : SUCCESS in 31m 37s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8f51084089564ea38dca62039e1833fc : SUCCESS in 1h 19m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b09a7f4245174d5ab14219bca722bc9a : SUCCESS in 21m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ded791baf1747c49f1b9026bc776ec4 : SUCCESS in 24m 20s","accounts_in_message":[],"_revision_number":4},{"id":"11fbf61a425755225c7a395df6936aa29624c7fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-26 12:18:18.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ddda1014ca833182ef8deca097e499f9723b7ee8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-26 13:01:57.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f1c78ea4c09f621261f96f6a9a4d379cdd5683fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-26 13:31:31.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"2b0166bf18ae03cfd1b9e05eff45aa223c40d13b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-26 13:40:55.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"d2e18fe538bcde917257559077b215302b53c6cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-26 14:15:19.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"a2e268e575b37209b15da1efd9d49d994dacb901","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 14:24:50.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a273c92998a44c7f922a9ed34f427143\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5f720cc93a9741e6963506922a061670 : FAILURE in 5m 36s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a5ba32cc3c24480cb71de7b830dd36d3 : FAILURE in 7m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed94e256a1aa4107a06d8bf2010f1eca : FAILURE in 4m 24s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/772fd35e73c24f5bb77bbf96161aba2b : NODE_FAILURE Node request 300-0021322629 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7313d77ec7934156944a6d2c901e4694 : NODE_FAILURE Node request 300-0021322631 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cec16c8611af475692eea63f3f948c98 : SUCCESS in 4m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9bbda1c36723481a900bc6bbf2f5712f : NODE_FAILURE Node request 300-0021322635 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7d582b6056ce303591db04053bff4c976db41a10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 15:40:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1517f8c1e3c84b34b016ca77f53a75b1\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ae9c202a42be43c996831e77d8fa7d90 : FAILURE in 4m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ecb2ab94f8be450b8aa835a3ffc9a676 : FAILURE in 4m 46s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/25f5f233dfb94d72aba43d92a6fb56df : FAILURE in 26m 48s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d9317f68d02e4eb3b4866d967d048b0f : SUCCESS in 1h 05m 29s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8ec9b5a72ab342d49d313b20a10a08a5 : SUCCESS in 1h 13m 53s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/d3de1326949e451293c1485085922cd7 : SUCCESS in 1h 13m 47s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/be3244fb12df46cd9180629f929fad6e : SUCCESS in 1h 20m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f3e863055acd4063906f16f9d1ddf92a : FAILURE in 2m 55s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bc02aaeb98f04f8790e98ea8cfab914a : FAILURE in 2m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5423fcc4b1bc44909c7990f597fedcab : SUCCESS in 1h 06m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/96cc4d6d091745ccb6373c312e06715d : SUCCESS in 1h 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd294d11fbad4119ad3523eefa82e5e3 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48fbb170e5c7445486fce9e3704539cb : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84e045ed64f140a5a11986e42244af56 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f686f4c692fd4bc9a793e7ad73510386 : SUCCESS in 5m 25s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/54b76be8198a44ca8566f9c96dd5157d : SUCCESS in 31m 18s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/4e7bcc0d07c94c5e8ffabe732c2ea97e : SUCCESS in 1h 01m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ed463a92ce4d47279c1f307fd42a7b66 : FAILURE in 4m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/af2532e42a594b9d9f3921995ab89deb : FAILURE in 2m 53s","accounts_in_message":[],"_revision_number":8},{"id":"6884135c8392f517303378fe6dc22b5d2300bb2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 08:52:32.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"15671189b09a0c9ff0d27433a28a5f7238a6e91e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 10:14:29.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fb39b290a8244c7aa21cc2b5996a99c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/06b02e63b2744cd0810bf0a0660601a6 : FAILURE in 30m 58s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e34e75131bae41339fc449000a06ee20 : FAILURE in 10m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2e3bc62914543f1b168906feefdda1a : SUCCESS in 29m 23s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0086beb6fa44d5ca9bb4e788d10f7ec : SUCCESS in 1h 17m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/997f85b7a6be4c678a477d5b11c75057 : NODE_FAILURE Node request 300-0021334655 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d5224356bd94e0e97b91cc2c979c589 : SUCCESS in 3m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c226f6085ed4d7fa5183c5882d6843b : SUCCESS in 28m 36s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"46030952f0ea93605ba6c8de138c56bf21d076ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 10:25:37.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22aa1591052843c3b4a6de6dac007921\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9da7e436c4fe4e898537847a0f0fe181 : FAILURE in 29m 41s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3d866154f59e4eabb198ad84918e96da : FAILURE in 40m 25s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/0504befcaa3b4c7797a1ee138b1a2d2e : SUCCESS in 20m 31s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2db084f43b86429fabef87e555e27e35 : SUCCESS in 1h 08m 58s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/992f1515aa954587898c25a602bfd0f3 : SUCCESS in 1h 12m 09s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/79dcfd8da01a4d86b49fe5e4f62cadff : SUCCESS in 1h 08m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/84d24825c5e54c48a1799e8676c2a2d1 : SUCCESS in 1h 27m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5f19726947e6471fbd6abec72c951adf : SUCCESS in 25m 13s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/43633fce27a84e35b74c407c9fbda60a : SUCCESS in 22m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/392b3a0cc15d4716854927b731fcd31f : SUCCESS in 1h 04m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a7315d0f03aa47d69d85b11c466f06ce : SUCCESS in 1h 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44838b8da3b641c88e0bf610b232b3db : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f447ae59e3f4505a0aa7797fc676644 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60a5b743d45444a8a53feedc0f40c9e8 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9de866927ff406e97101006ccca5f13 : SUCCESS in 5m 14s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/067270bc365f452bac563d74318a5187 : SUCCESS in 24m 08s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/18fe49f8052547839f8152095236aa42 : SUCCESS in 1h 06m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b23138180d741648a70cad922dec559 : SUCCESS in 21m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1304b7d824324f4796c92197515f5aec : SUCCESS in 25m 41s","accounts_in_message":[],"_revision_number":9},{"id":"08bfdefd90f6ce96f66af10b6ad358d664760c6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 11:10:58.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"9a2119ef8524526cc2d1e3fc6e46ecfb6db6ade9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 11:11:35.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 883716,10","accounts_in_message":[],"_revision_number":10},{"id":"97bec403da180d4068755d699af484b725459b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 11:11:35.000000000","message":"Patch Set 10: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 883716,10","accounts_in_message":[],"_revision_number":10},{"id":"1ca9521fe0faa730ed912f7c620969d0a3dfe5d6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 11:12:01.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"63f4d5431116183d7fb30943505e54724ae926bf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 11:12:35.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"c2af9d5cb2e34032941e926caf648ec770cc6253","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 11:12:38.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":12},{"id":"0913da6bdbed2f2614c8808fb936d3fb01947835","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-29 11:18:23.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"95190b53b0cc91609f694d243123fdcbb818d25d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 12:41:57.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a097c2a170584e149d47d0e3cc47deb7\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/46a2db206de84b158169b1c18660d79b : FAILURE in 34m 09s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/091743d5a13d4775ad8dfabd49589389 : FAILURE in 31m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/90f9b7c805f143d6b3c1d435fb3d3244 : SUCCESS in 33m 31s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/65aaa1ec64ac4ca4b0cf9cf548876f68 : SUCCESS in 1h 26m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46092c8eedde4faba54d938487f5cb7c : NODE_FAILURE Node request 300-0021335574 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f40f0007c241433c886eff35267b4ca3 : SUCCESS in 4m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3230d689fb344a8fbc4cc8172994ec02 : SUCCESS in 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0ac26477c627f92fa3cdc5de2c4cedcd9a23420b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 12:42:36.000000000","message":"Patch Set 12: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 884487,10","accounts_in_message":[],"_revision_number":12},{"id":"00f2b9eeac5792c152a2f005be25958ec2bcf3fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 08:17:42.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"17e1dd6bd55f44824b227ad21e9dead4a0478ba3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 08:18:11.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"720213bca1673c72fc87ff3601da9d74d72337ab","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 08:18:24.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":14},{"id":"704bd0b305fb9765b95fb159e6a84cf1fe42abeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 09:04:41.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f49def436bae4c56886e7d7bf67eca36\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fcae8ba3373a4b99b266135fc657affa : FAILURE in 43m 33s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/36d66a2db3d54ccea3a2082a70781ea3 : NODE_FAILURE Node request 300-0021542025 failed in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b2b9e0b3728143b9a2ecbb190861621b : SUCCESS in 35m 51s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c71b8e54c454d41b2b4f0e138098873 : NODE_FAILURE Node request 300-0021542027 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d058629e5f054db5ba7c4eda8b36a20e : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22eb51e4e1de4c2988ec5ea0c69a8c01 : NODE_FAILURE Node request 300-0021542029 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/61e2b7bd3237415f9377efc4498561c2 : NODE_FAILURE Node request 300-0021542030 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"683cc481e4f02d67b9c3dd854c8c11c8c47425b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 10:56:37.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/83f1e83d1b0b40789428dae89d697b7c\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/45a73c84e9774377b812ceb7ce61eab7 : FAILURE in 33m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/06fa10b92e004fd8b695f42947c26a87 : FAILURE in 32m 20s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/3550e69b22d74c038aa2181349ac3089 : SUCCESS in 23m 34s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9f418a6399ec4ef18f3354fca580c1f0 : TIMED_OUT in 2h 02m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cfef2057cee44793880619909063e0e4 : FAILURE in 31m 09s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/37960210da6345c187179973c0106e1a : TIMED_OUT in 2h 01m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bbd9a2c30fbd4959be3e2804d0aad8e4 : TIMED_OUT in 2h 31m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/12b931173dc946a98740d0d38261fa57 : FAILURE in 23m 31s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e07e5b351530459f800551f9c0918b8a : FAILURE in 27m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/29c91f95b9714a48bda20bb8e0033af6 : FAILURE in 33m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/12d4c74e92d3444d8d85f2ce052ae3f3 : FAILURE in 29m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b2452cb7583426ba03312ee563faaa1 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9271fc6fd7e646eabf2d3f35adab0373 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb0e43dea1ed4986ba42f5fc8e6e300c : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7979baa3d0a84b5bbca05f41c299f9b2 : SUCCESS in 5m 51s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/a967ebb9a4f64c7a8aea1a49c6860208 : FAILURE in 12m 46s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/3e99c9a188b8490db53acb38022d0ce5 : FAILURE in 19m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/480803307d714873b5cf0f392d097b4e : FAILURE in 32m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1318290e19b84dbf964fa836af75f8ff : FAILURE in 52m 40s","accounts_in_message":[],"_revision_number":14},{"id":"586716cf529c91f838feb255deb333be545e17be","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 11:49:31.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d7b808b364836765f9ff78d0d161dca966de1f00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 13:27:42.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d28c24e97d9f4caabfaee0e1addf9787\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/92f2684ff8914ae0af6a821fe9aba550 : FAILURE in 36m 49s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f02e9d40aa9b47b68513b88a8575cc2c : FAILURE in 28m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f5010b5d194442c2b11b5bae3573a5b8 : SUCCESS in 27m 13s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d73dd061b2094992b6d34fd37a798a02 : SUCCESS in 1h 37m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa0fcafa712e4c30bf36febc52ba3f20 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3e6c46281304cc794b8e3f2b32538f6 : SUCCESS in 3m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4cddddde2eb64b898f63203ad2bf841f : SUCCESS in 1h 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"5a9bc5b9b8a5ca5a84e0c5e4211f75bd291fa3bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 14:08:15.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bae369ffccf4a5991cc7eb1b39195cf\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/376b2309d7084b9aab27c72e14898e93 : FAILURE in 39m 21s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fef81e9acd5041ecb5022141f3290a83 : FAILURE in 43m 43s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/bca6e610032643d7abd68d9a3a6fb479 : SUCCESS in 29m 39s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2714eaf18352479a849fa9a45b7a4323 : SUCCESS in 1h 32m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c958c695c84a4fccb651da127c8edbff : SUCCESS in 1h 09m 17s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/0b3af3de3c9446328956db8997565ca6 : SUCCESS in 58m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e7966e56979040eca27f5bf61ddfc801 : SUCCESS in 1h 36m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/71fc1cea8f5e4d4e816ad167eea9fee8 : SUCCESS in 27m 52s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e31ab75d8df4448ea83cd5ce2ab021ab : SUCCESS in 22m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/db7d39bd3f3c483cac1195c7b606bd58 : SUCCESS in 1h 03m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/89544606c1d6448c82ad63b97b062063 : SUCCESS in 1h 44m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b41d526cc5124090bc7280892e544a7a : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f8203a590a54cd1832123ee7dc2bb64 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e40c42ffa46460ca39dccf3def10da9 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d69278f522db4323ba495641148714ca : SUCCESS in 7m 27s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/7a52f1920ba54528803a187074f84630 : SUCCESS in 40m 30s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/421d152538dd40f8a85742c63a864258 : SUCCESS in 1h 17m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b1412475832647de857effdc1c95bbba : SUCCESS in 34m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0bde27b9a54e476085efad24a4113848 : SUCCESS in 30m 30s","accounts_in_message":[],"_revision_number":14},{"id":"ee2ae1ab7eb4342ca905a7d5f7cc4254168fdf0b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-07-13 07:29:57.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"512d21ad99b6e5c4dbcbb4cf28b298e639825680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-07-31 13:37:49.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"19d6ee23958c19bc8dfbdd263af3b4409a205f81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 15:42:52.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6d0bd193db34e7b9f6d04385e6ce67e\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/eef778afdb5b4d13bc426412028b3597 : FAILURE in 34m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/110d57aff8884ce5a1a6ec915cce372a : FAILURE in 47m 37s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/605adccf022c4e1eaeac15bafb9fc0cf : SUCCESS in 24m 22s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b077cf92205746e9805fac2544fbf760 : SUCCESS in 1h 13m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8110d098f3b9443b8812166bc93993f3 : SUCCESS in 1h 23m 12s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/2c2a375280c84e3ba2acecde9883c167 : SUCCESS in 1h 24m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0a65ebf44f2e41ed949544ca521aa409 : SUCCESS in 1h 57m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0bea403726244e2896632d900fcb53d7 : SUCCESS in 28m 22s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/345076465151479f9554e999cd229ad5 : SUCCESS in 21m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1f0c6ac1cd924ada9b7608093535a799 : SUCCESS in 57m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/83d598ae1d0a4ae489f8028eea3e01bb : SUCCESS in 1h 03m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849c3517e6b84338a8a3ad2c8ce58e72 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b594ca33b8440bb89b6a8210526c3ed : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bde2e180fcb144c1bcbfd960ee3ec7aa : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1fff36bf261443b87ec2929290740ae : SUCCESS in 9m 23s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/a26aabdbedaf4410b2a53be1cb021eee : SUCCESS in 19m 37s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/0582d3c0349548eea047381f94cc6a5a : SUCCESS in 1h 22m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/423fa1b0244349db847d047516ad1b2d : SUCCESS in 24m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/151f37015ada4a6fbf9d3f2cd3762180 : SUCCESS in 29m 41s","accounts_in_message":[],"_revision_number":15},{"id":"ad2418bd4acaf507fe7b4c0449ce674b025ef3b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 16:08:21.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e38b30e22fc48c484394a98ae490b79\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/45e4d8e18f424635aa8a13dd0535007a : FAILURE in 33m 38s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f46a8df6a39d4a8c8ea9519b3853e83a : FAILURE in 29m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bbfe50cf8b584d32bdb6d0a603ac4e12 : SUCCESS in 30m 09s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b42f9c817464ba2a87b720c2637bf8e : SUCCESS in 2h 28m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/672204d7f3a44e30beaae9309817280f : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc274e6521914ad88d52cdcc1a9f09e7 : SUCCESS in 5m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/566ccc6bb7e4423e9c30e7bd4f74c112 : FAILURE in 28m 03s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"bbf4b1a74a36e077661133edfc17285d15911135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-16 11:30:07.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"64b155011f3cba319d2f578e01c43ffdc37a2e5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-16 11:32:12.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"f143c19cdf75eebd06e19522211a12f1c130a21b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-16 11:34:39.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"1c23228481b3b642ec2635dd1f8e1b1de1194281","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-16 12:17:57.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"d5b917c3fa1328a59cc86ff019b2df826b02a4ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 12:54:09.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1295aeb31da843768a6f408c4f86fe30\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f01ec489a97545eda16c8e107e6b84ed : SUCCESS in 35m 30s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e039ebc111344744a550e34dacbe19a6 : SUCCESS in 41m 59s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/0e5dc2a1568246ecb9daf3e9a9a1ef93 : SUCCESS in 22m 57s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8bfeda2999bc497ba3c7e1c31654a4dd : SUCCESS in 1h 13m 53s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1c037b2e3ae046ddb1706771afe219c7 : SUCCESS in 54m 13s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/31ecd39b3fc34e48a0ebd70f1d9c99f7 : SUCCESS in 1h 09m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f44fda1d8319466a97f445b648fabc79 : SUCCESS in 1h 10m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/79d284a90a2b4a31aae7792cdede5c2a : SUCCESS in 21m 54s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bbb1eab7738d47bcbf36dd155d23bfd8 : SUCCESS in 20m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1d05bb17c82945908d45780f5362e5f4 : SUCCESS in 1h 05m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a98debd7c7614e0fb6314efe465bd52a : SUCCESS in 56m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b95d6f776aa42aeae3d921e80cf18c9 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d76dbf9e7cde4ee4af054227687fcbf2 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82e93fd178294cbeab60bcf851b82686 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41762454c7da43ce8bc0c1cb6ee9d03f : SUCCESS in 6m 07s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/dc0035e4087245ef8c624b089e498abf : SUCCESS in 24m 04s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8516dade07b8412ca32626e87dff6243 : SUCCESS in 1h 07m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a1f5b7c68ee488d91f165e9e2f90377 : SUCCESS in 26m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f578faa64e314521a35352ebfdf6085f : SUCCESS in 20m 10s","accounts_in_message":[],"_revision_number":17},{"id":"398c70ac5f05a8ffee7c90d1bd9116c4e95b0498","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 13:52:59.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2680a334fef14ddfa43d73ee491a7b56\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/813925e47ce44004b10edf0159940284 : SUCCESS in 2h 18m 08s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/906d00bf8c4b4cd2bc7a7933be474606 : SUCCESS in 1h 16m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bfc16463eadf4041aedd844b793399ce : SUCCESS in 21m 06s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/048a32b677654d9587bd8e3524531872 : SUCCESS in 2h 18m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ce485fdd78e49a9b6533bb8fe5b7994 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90251aeb6451445098ded3cb5a48ed52 : SUCCESS in 5m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d4a9f6af4b3477b8f05d401bce0c0c8 : FAILURE in 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"2cb9a67a114c280b857006df09945b8cdbd11587","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-17 07:09:48.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"915868d77b2da38240ab22c1d071d1e0eb178457","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-17 07:09:58.000000000","message":"Patch Set 17: Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"8804a2a19e46ec570d285ee841efd6e06755f1ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 07:10:28.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"104fda9404bd71ba255c5b5722cae982580a4adb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 08:54:17.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01a3b4c3acad4b6fba62034bd941a5ba\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7b74a4bab8114839aea0bcb0494c4455 : SUCCESS in 51m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/09570d32503e4285b5ea610b3ef3762c : SUCCESS in 1h 20m 44s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9087dad5eb0f458581b4a345ba6ba832 : SUCCESS in 1h 37m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5d598762ee234a41a461d85d37b8fa61 : SUCCESS in 24m 23s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7c7b4f24d60f4c52add0e558f9d3d376 : SUCCESS in 56m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7c057e5e60ae4ce6b8eed7b13f0b13df : SUCCESS in 1h 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acb49fe48c37423b8a2fba9ce648197f : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ae6c7d1ef3c4b4b9453c59196bd07c7 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2223e9123f444f9a22ca98436ff2fcb : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52fce2ec8e944d8bac625738b97c40b6 : SUCCESS in 5m 57s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/155031dc0d2c4b3eafa53241f58d33f5 : SUCCESS in 58m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/70b9325ab86448129d903d9ae72659a5 : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":17},{"id":"d99ecc68932cd4e67c6089c23626ae17a5931ce9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 08:54:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"3b50ed86bbe91f30dddf4d46c89a1b6009f91bc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 08:55:22.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e97c41744d564c228ca78665cfcd82ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c56baca89ce447490119496150229f3 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"23e0819d688362ce0b6c564d469f57ea9cbe2731","revisions":{"76c9672674be6b1bde8439e2afba708ef1d04beb":{"kind":"REWORK","_number":1,"created":"2023-05-22 09:06:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/883716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/1"}}},"commit":{"parents":[{"commit":"64ab2e983de8bec2da394110334c88a7b35d5773","subject":"fluentd: pin fluent-plugin-elasticsearch to \u003c5.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64ab2e983de8bec2da394110334c88a7b35d5773"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-22 09:06:22.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76c9672674be6b1bde8439e2afba708ef1d04beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76c9672674be6b1bde8439e2afba708ef1d04beb"}]},"branch":"refs/heads/stable/yoga"},"64dc3b4816ff1ed2c235b9858dfc35e04c7c083d":{"kind":"REWORK","_number":2,"created":"2023-05-22 09:08:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/883716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/2"}}},"commit":{"parents":[{"commit":"64ab2e983de8bec2da394110334c88a7b35d5773","subject":"fluentd: pin fluent-plugin-elasticsearch to \u003c5.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64ab2e983de8bec2da394110334c88a7b35d5773"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-22 09:07:57.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64dc3b4816ff1ed2c235b9858dfc35e04c7c083d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64dc3b4816ff1ed2c235b9858dfc35e04c7c083d"}]},"branch":"refs/heads/stable/yoga"},"4fbc223905d9080b4ec2344cf9237d58e0edb4f3":{"kind":"REWORK","_number":3,"created":"2023-05-22 09:10:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/883716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/3"}}},"commit":{"parents":[{"commit":"64ab2e983de8bec2da394110334c88a7b35d5773","subject":"fluentd: pin fluent-plugin-elasticsearch to \u003c5.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64ab2e983de8bec2da394110334c88a7b35d5773"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-22 09:10:24.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fbc223905d9080b4ec2344cf9237d58e0edb4f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fbc223905d9080b4ec2344cf9237d58e0edb4f3"}]},"branch":"refs/heads/stable/yoga"},"033a82325307725fbabbee54493fb536153e0a64":{"kind":"REWORK","_number":4,"created":"2023-05-25 15:46:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/883716/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/4"}}},"commit":{"parents":[{"commit":"64ab2e983de8bec2da394110334c88a7b35d5773","subject":"fluentd: pin fluent-plugin-elasticsearch to \u003c5.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64ab2e983de8bec2da394110334c88a7b35d5773"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-25 15:46:00.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/033a82325307725fbabbee54493fb536153e0a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/033a82325307725fbabbee54493fb536153e0a64"}]},"branch":"refs/heads/stable/yoga"},"399227da70e7c22c1ddfddfa89e49736d45c294b":{"kind":"REWORK","_number":5,"created":"2023-05-26 12:18:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/5"}}},"commit":{"parents":[{"commit":"64ab2e983de8bec2da394110334c88a7b35d5773","subject":"fluentd: pin fluent-plugin-elasticsearch to \u003c5.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64ab2e983de8bec2da394110334c88a7b35d5773"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-26 12:18:18.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/399227da70e7c22c1ddfddfa89e49736d45c294b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/399227da70e7c22c1ddfddfa89e49736d45c294b"}]},"branch":"refs/heads/stable/yoga"},"f91460b9697580942eba071f0c119e64109e1567":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-05-26 13:31:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/6"}}},"commit":{"parents":[{"commit":"17fe689a8e055115e8c4f31b8bc2dd51d21818d6","subject":"CI: Fix centos9 builds and drop unnecessary zuul vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17fe689a8e055115e8c4f31b8bc2dd51d21818d6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-26 13:31:31.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f91460b9697580942eba071f0c119e64109e1567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f91460b9697580942eba071f0c119e64109e1567"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"1bbc63e4941792a93dbc5720f4d6130fb447db05":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-05-26 13:40:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/7"}}},"commit":{"parents":[{"commit":"44b339d9d968ef0054b7610a7fc7da1008389209","subject":"CI: Fix centos9 builds and drop unnecessary zuul vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44b339d9d968ef0054b7610a7fc7da1008389209"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-26 13:40:55.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1bbc63e4941792a93dbc5720f4d6130fb447db05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1bbc63e4941792a93dbc5720f4d6130fb447db05"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"8103ab13cc23548e553901647fcb803077a4f816":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-05-26 14:15:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/8"}}},"commit":{"parents":[{"commit":"23607a095980126bb430e8043e387596d1c0798c","subject":"CI: Fix centos9 builds and drop unnecessary zuul vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23607a095980126bb430e8043e387596d1c0798c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-26 14:15:19.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8103ab13cc23548e553901647fcb803077a4f816"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8103ab13cc23548e553901647fcb803077a4f816"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"3241c6dab7df9dd4e80c2ffc2a887d79c2c9ec73":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-05-29 08:52:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/9"}}},"commit":{"parents":[{"commit":"aa292da9b1ebf9091fffa63dc0f67fd86f38543e","subject":"CI: Fix centos9 builds and drop unnecessary zuul vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa292da9b1ebf9091fffa63dc0f67fd86f38543e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-29 08:52:32.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3241c6dab7df9dd4e80c2ffc2a887d79c2c9ec73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3241c6dab7df9dd4e80c2ffc2a887d79c2c9ec73"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"234739968ca9c0d6e1d697d0bfdd627eee5ed113":{"kind":"REWORK","_number":10,"created":"2023-05-29 11:10:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/10"}}},"commit":{"parents":[{"commit":"aa292da9b1ebf9091fffa63dc0f67fd86f38543e","subject":"CI: Fix centos9 builds and drop unnecessary zuul vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa292da9b1ebf9091fffa63dc0f67fd86f38543e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-29 11:10:58.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/234739968ca9c0d6e1d697d0bfdd627eee5ed113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/234739968ca9c0d6e1d697d0bfdd627eee5ed113"}]},"branch":"refs/heads/stable/yoga"},"9643efedf4689db9a319fded2584662641e7d7b8":{"kind":"REWORK","_number":11,"created":"2023-05-29 11:12:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/11"}}},"commit":{"parents":[{"commit":"f01f1a439364c91c1a6667f70c89bf0f5876088e","subject":"CI: Fix centos9 builds and drop unnecessary zuul vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f01f1a439364c91c1a6667f70c89bf0f5876088e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-29 11:12:01.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9643efedf4689db9a319fded2584662641e7d7b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9643efedf4689db9a319fded2584662641e7d7b8"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"54dce7d767b917c0e5a459f7c8451e4517d0c8d1":{"kind":"REWORK","_number":12,"created":"2023-05-29 11:12:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/12"}}},"commit":{"parents":[{"commit":"f01f1a439364c91c1a6667f70c89bf0f5876088e","subject":"CI: Fix centos9 builds and drop unnecessary zuul vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f01f1a439364c91c1a6667f70c89bf0f5876088e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-29 11:12:35.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54dce7d767b917c0e5a459f7c8451e4517d0c8d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54dce7d767b917c0e5a459f7c8451e4517d0c8d1"}]},"branch":"refs/heads/stable/yoga"},"13a19e75ca04becf0f88536c2630e17adeed54ba":{"kind":"REWORK","_number":13,"created":"2023-06-26 08:17:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/13"}}},"commit":{"parents":[{"commit":"75f9186288e6410935d19f430394410289ecb63b","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75f9186288e6410935d19f430394410289ecb63b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-26 08:17:42.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13a19e75ca04becf0f88536c2630e17adeed54ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13a19e75ca04becf0f88536c2630e17adeed54ba"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"6899de00b264e3e515b640d8069872e89ef3231f":{"kind":"REWORK","_number":14,"created":"2023-06-26 08:18:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/14"}}},"commit":{"parents":[{"commit":"75f9186288e6410935d19f430394410289ecb63b","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75f9186288e6410935d19f430394410289ecb63b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-26 08:18:11.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6899de00b264e3e515b640d8069872e89ef3231f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6899de00b264e3e515b640d8069872e89ef3231f"}]},"branch":"refs/heads/stable/yoga"},"eca7e99e781092856a3e104ee1c2f53ebe45b8e9":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2023-07-31 13:37:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/15"}}},"commit":{"parents":[{"commit":"d49cca54922507c040dd1003e628938b1aaf5cf1","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d49cca54922507c040dd1003e628938b1aaf5cf1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-07-31 13:37:49.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eca7e99e781092856a3e104ee1c2f53ebe45b8e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eca7e99e781092856a3e104ee1c2f53ebe45b8e9"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"1f3c38138090c58400f75235f7a2dd5d3a6819a1":{"kind":"REWORK","_number":16,"created":"2023-08-16 11:30:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/16"}}},"commit":{"parents":[{"commit":"d49cca54922507c040dd1003e628938b1aaf5cf1","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d49cca54922507c040dd1003e628938b1aaf5cf1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-16 11:30:07.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f3c38138090c58400f75235f7a2dd5d3a6819a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f3c38138090c58400f75235f7a2dd5d3a6819a1"}]},"branch":"refs/heads/stable/yoga"},"23e0819d688362ce0b6c564d469f57ea9cbe2731":{"kind":"REWORK","_number":17,"created":"2023-08-16 11:32:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/883716/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/883716/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/883716/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/883716/17"}}},"commit":{"parents":[{"commit":"d49cca54922507c040dd1003e628938b1aaf5cf1","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d49cca54922507c040dd1003e628938b1aaf5cf1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 10:25:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-16 11:32:12.000000000","tz":0},"subject":"opensearch: move to yum/apt repos","message":"opensearch: move to yum/apt repos\n\nsystemd-tmpfiles is needed otherwise opensearch\npostinst on Debian/Ubuntu fails\n\nChange-Id: I5aeb5796f8de7697a1c4a7404fe292ef9eb522ee\n(cherry picked from commit 6bd7898a1d95157c713bdff0fe5bdf5a59b65550)\n(cherry picked from commit aab455c5eb063b5e843bc0919bbae8f1687cf1d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23e0819d688362ce0b6c564d469f57ea9cbe2731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23e0819d688362ce0b6c564d469f57ea9cbe2731"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
