)]}'
{"id":"openstack%2Fkolla~883928","triplet_id":"openstack%2Fkolla~stable%2Fzed~I2a5a0d761223b4e99c8df59f80261108380d5038","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-24 15:25:37.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-24 14:19:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I2a5a0d761223b4e99c8df59f80261108380d5038","subject":"opensearch-dashboards: Fix permissions","status":"MERGED","created":"2023-05-24 13:48:36.000000000","updated":"2023-05-24 16:48:27.000000000","submitted":"2023-05-24 16:47:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883928","cherry_pick_of_change":883941,"cherry_pick_of_patch_set":3,"meta_rev_id":"cd648f2c11fb816ad1561fb776e14f1fad2b5e14","_number":883928,"virtual_id_number":883928,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-24 16:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-05-24 13:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-05-24 15:25:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-24 14:19:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-05-24 15:25:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-05-24 13:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 13:49:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-24 13:51:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-05-24 15:11:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae2382e7f1f99dd5f573112dc738b9b3c83f7293","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-24 13:48:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* kolla/common/users.py","accounts_in_message":[],"_revision_number":1},{"id":"32dba079ea69c3b829d5823f59aed43bdee189a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-24 13:49:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"17c1bafc11cc4e77f142e78e1c872624bf3ddbdd","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-24 13:49:08.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"0b04912c0c3bb530e9603a29e561c8adc115f0d8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-24 13:51:26.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"6f6911ffe99866c672705cd31815f83ebd7d74a8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-24 14:19:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8b294d427f7c35b47943b5196738b101d33d13c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 15:11:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80aaeeadaf8b4265b77b4297927ffc47\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1451648e069f4b30b14a53468faed48f : SUCCESS in 22m 55s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/7a9af96e485e45afb9f0a056b3bdd0a4 : SUCCESS in 1h 07m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6fa792bd464c464a9b5307ad4370227a : SUCCESS in 18m 04s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c2b9bfa8b4974d77a88c49ba689961b6 : SUCCESS in 1h 02m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/61b2bada47394af7943ab8bf3d77e1b2 : SUCCESS in 1h 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bff918188dca4eff829dfe26cadb7dcb : SUCCESS in 2m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9eba3d66927c4092a80e643779da7301 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b79d573c8194d94b180a13f379a5580 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3216172a93224a7fa47459835c980f5e : SUCCESS in 2m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7a9b33905c24b3895d8e2b74743a686 : SUCCESS in 6m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/aaeab456627a4236806a1c8018035f29 : SUCCESS in 19m 44s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/6fb89c36205f40cd985f79943941827d : SUCCESS in 1h 04m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/72d0fa49605b4ced9ae82d1a9dea6c01 : SUCCESS in 18m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7153d07f244a4d7cb85f71750ece36aa : SUCCESS in 46m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/baf6013b8dbe43b897add5ec00a6451b : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":2},{"id":"24aa80164776cc6c689de2ee03cc80d49ab145a6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-24 15:25:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9de1958ead8883c940aa661d643b8d06e72437ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 15:26:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"059c9457da2b31bf35b3a0bb2a58c1bcc468416d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 15:45:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/495785a72d8841ffbea2ba1383d0adbe\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fb2ca25d5b1d4b2eaec7e519d2019403 : SUCCESS in 24m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4483ace497ea42d38467158652971bbd : SUCCESS in 19m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/de486d305e8f40be8c6d3e2e90e425ad : SUCCESS in 1h 53m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b30a66f2112e4267bf8acff94204d7da : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4dae61909bac4bc280c83840f7ef48b8 : SUCCESS in 3m 04s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e69ff0fd89da44da902971d4a7714bc9 : SUCCESS in 18m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f916dc7e16e241e0a039ba5463017835 : SUCCESS in 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2073d2af7b2abf8c3aafabf6d82cad97a7dfbb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 16:47:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eb4d17dcf774455abe6e793bd004429\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a4dc48b444fc447cb5d480f60feceba5 : SUCCESS in 22m 32s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fa5aedcc9cd04342819cf71bcd94d7fc : SUCCESS in 1h 02m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8e098c2d26234e5da7deb32cf84a69c6 : SUCCESS in 1h 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4ae22f6bb7467d938995702703108c : SUCCESS in 2m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33eafe8e86894375916121a1dab72e0b : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16885b7efca142bc8e2c2f8d3685b8d3 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/470c4e4a5e804be1991efabea88a8e5f : SUCCESS in 5m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f13652b2708d40bba4b6850fee78c5cd : SUCCESS in 22m 05s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/17720926afe04c53a9f607ce1747b476 : SUCCESS in 53m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/979ac9f55c044b1988cdeb4750d91ec4 : SUCCESS in 21m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6a5b3bf61add4cdebb90d77f99e91061 : SUCCESS in 52m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aeae6e02080142cf8d39226105448b97 : SUCCESS in 1h 12m 42s","accounts_in_message":[],"_revision_number":2},{"id":"c898cbcb5cdf390ff504e7c4bc151331a4f06da8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 16:47:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cd648f2c11fb816ad1561fb776e14f1fad2b5e14","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 16:48:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/368d63dd26934fd8a987c0821adb7ccf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc71d60202be4ef3a42d196fde447d1a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d793de2a482ea21a3dec90828605f3fb8d9e4380","revisions":{"928a20abfd4ca47efd4725c47e475c3f25ec37b4":{"kind":"REWORK","_number":1,"created":"2023-05-24 13:48:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/883928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/883928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/883928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/883928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/883928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/883928/1"}}},"commit":{"parents":[{"commit":"37bc1ffa3abb2dfc2c159de0d4540da50264d646","subject":"Merge \"opensearch: move to yum/apt repos\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37bc1ffa3abb2dfc2c159de0d4540da50264d646"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-24 13:48:36.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n(cherry picked from commit 6786e4384c1a19e4658310e109b8ab9704f5eff9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/928a20abfd4ca47efd4725c47e475c3f25ec37b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/928a20abfd4ca47efd4725c47e475c3f25ec37b4"}]},"branch":"refs/heads/stable/zed"},"d793de2a482ea21a3dec90828605f3fb8d9e4380":{"kind":"REWORK","_number":2,"created":"2023-05-24 13:49:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/883928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/883928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/883928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/883928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/883928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/883928/2"}}},"commit":{"parents":[{"commit":"37bc1ffa3abb2dfc2c159de0d4540da50264d646","subject":"Merge \"opensearch: move to yum/apt repos\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37bc1ffa3abb2dfc2c159de0d4540da50264d646"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-24 13:49:05.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n(cherry picked from commit 6786e4384c1a19e4658310e109b8ab9704f5eff9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d793de2a482ea21a3dec90828605f3fb8d9e4380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d793de2a482ea21a3dec90828605f3fb8d9e4380"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
