)]}'
{"id":"openstack%2Fkolla~883941","triplet_id":"openstack%2Fkolla~master~I2a5a0d761223b4e99c8df59f80261108380d5038","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-24 12:52:33.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-05-23 14:48:50.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}},"hashtags":[],"change_id":"I2a5a0d761223b4e99c8df59f80261108380d5038","subject":"opensearch-dashboards: Fix permissions","status":"MERGED","created":"2023-05-23 14:15:55.000000000","updated":"2023-05-24 12:53:36.000000000","submitted":"2023-05-24 12:52:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883941","meta_rev_id":"f6ae4a76d792057aa385b118f7e3ed0c48fc017e","_number":883941,"virtual_id_number":883941,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-24 12:52:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-05-24 07:38:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-05-23 15:18:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-05-23 14:48:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-05-24 11:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-05-24 11:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-05-24 07:38:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-23 14:36:12.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-05-23 15:18:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-23 15:52:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-24 07:38:02.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-05-24 11:28:40.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"01ea267899057e5bc5df8092f7219aa04661da0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-23 14:15:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afa8cc88743ea89cff1ef2c1b500d66ae038c4a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-23 14:17:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e8886654f1bf2f2e164adf51fd7ddd9a379920ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-23 14:27:51.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"60882b6523284b4dabec15fb1d6355cac093b7ca","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-23 14:48:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b332c4ba6c8136f15fb573bbc07dc95fec0b3008","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-23 15:18:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7aa83f8cf14ba5b647adaf1afd011289628df7d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 15:52:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf61892316b54030920b78da1c816516\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c659f0eed2484e3881c7fadf444589ac : SUCCESS in 21m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/66f8890d02d5461bbc844381f1396051 : SUCCESS in 1h 09m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b40f28f2d07641fc973fae0e14d0e9d0 : SUCCESS in 21m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/485167edcd0c4a8184068d66388487e4 : SUCCESS in 1h 03m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/aaf770a8f85a4deab58ae74795e45db3 : SUCCESS in 1h 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a467e230a436493188680ef2f3b19d37 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9613ad7f75af4ce5b7f9616c9b957b6a : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1342de7782b4df0b2f31cd6be0e6257 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df74bf575fce47e98f8ee06d0f3f46a8 : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bc939b33aa041d1b8859b4c9b5f7bcb : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44df7c8bfb694121b1269dbb5962a61f : SUCCESS in 4m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1ad591dcfeda434486df95b007d0b78a : SUCCESS in 18m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d1c3d5c089694d42a7cf482448c285ee : SUCCESS in 1h 03m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8d771083b54244849edfdb6b6c8e3431 : SUCCESS in 1h 06m 13s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/408938dc54204fa4893613e16f151311 : FAILURE in 23m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/49f5f25d2895475281428e9252599190 : SUCCESS in 25m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ceb88a2ecab04ef6bb848e6663e5115a : SUCCESS in 57m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/df781b6f079843b1a16e9b3964492113 : SUCCESS in 57m 19s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bba05e0cd24b4f2a9bc9eff4e72d33b8 : FAILURE in 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fe1bb1db607cf89d97449a92e94015c7f0be9c07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 16:05:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60e08d568bd441028786a5063f2aa6f2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/38d19957f1af46d88a8b71a09cb3d3f8 : SUCCESS in 18m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e11147d4efd74c4d9b293d83fd93e9bd : SUCCESS in 59m 00s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5a05cf2eae394590ae61ca1a7950777e : SUCCESS in 1h 36m 31s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93f3afbca9b44a37a26911afb1de2ab4 : SUCCESS in 2m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d13d95134cab447894f91eb6a856bf95 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f6d222c74492439aae7bead69610cde4 : SUCCESS in 2m 25s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6d2e4dc1bb3b4cb0b8e871fa48445dd2 : SUCCESS in 22m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8cbb8aa8bb6d4d0f81bf0b77e741177b : SUCCESS in 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eb2c47fe6f32e62e5b72ca20e3f55ca8a6f3d3f7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-24 07:38:02.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"e5ec84dd3978412ce14d5d678a3d372f5b046658","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-24 07:41:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d757844c285f4f53bd19644e6351d48f98d1e524","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 08:42:32.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca7c122522c04c6dae24a5b06107c550\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/472b74af9bb843f5a0c0e46b63244464 : SUCCESS in 22m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1409d5ba31294419b64a58ed54f91554 : SUCCESS in 29m 40s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c244d8e7fbfb43518f66d5e7922758fe : FAILURE in 57m 47s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0822e1da86e9449a85cad57daaa2af9d : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/794f0425386b45a798702f0646807a2f : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a4c4f818839f4183bd204e915d1ddfa8 : SUCCESS in 4m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7faacac0e5244c0896bde988343b804c : SUCCESS in 21m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c32e601ecd3340a298f1acebdeedf1fe : SUCCESS in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"51baa88ad18ae9563d5992bb729d88c540582c2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 09:01:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bce5fa215d9442d5b945ebc20a5d7c12\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ff1d0f0412284d379c745a1b7d58c213 : SUCCESS in 21m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ed951de37e854c509ffc638b40d3cfce : SUCCESS in 1h 01m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4fafe70191c54ed38b9d3664597900b0 : SUCCESS in 15m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e8ec1d0cd10441c0b3d68da2adcd80c8 : SUCCESS in 1h 12m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b20105f24fa44a9fa7bd2e874916ddd2 : SUCCESS in 1h 07m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaabad24e25e4593a543a052cc190f81 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e87751c0bec04dd8b4152f7e489b2780 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd80989a322c40dda23db68b61b49277 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37735e89cb26467db2dd1b8141b85048 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fdb6729fb054584b26ae2f8908c6f4d : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6bfbec1400844b28d6c7d6c92cf5c6d : SUCCESS in 3m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/91c663da9533443f91f2586b421fa6f2 : SUCCESS in 19m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cad0393ac6334e22a059d66e3d77fb86 : SUCCESS in 1h 12m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e961a23c31d4496ab518919692e657bd : SUCCESS in 1h 09m 18s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6d12175ed8734f9db1f3732d9de5fbab : FAILURE in 26m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f545bc1bf8fd4d17896e299a06525a43 : SUCCESS in 27m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d8254c5ec08e407e90f78bf53156f90e : SUCCESS in 1h 02m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/197d9bda0a2043a7bae0ce8dd7d8335c : SUCCESS in 1h 00m 28s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/99622ca39807401db7211ba78f5456ad : FAILURE in 23m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e15b6c4f08c6f05dab9060991246ca24639b79e2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-24 11:28:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"32bbddc54c652ad1cd4ab274ef46991500a1c957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 11:29:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7c40ab9f10c234814f12a24a5c9e44dae8405783","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 12:52:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97e5e288d4c646878d696b70db171579\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/703b76b9441943cab5a2aee51735b91a : SUCCESS in 15m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d404fa32bc5d474a8765c0fc82583fda : SUCCESS in 1h 12m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/113358fe86594662921be4edf15c8c7f : SUCCESS in 58m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/641c045d30b94b41a30d15f33b0da08d : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12ba0b92cbbf4398bafb4c98681ef529 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e66eec366a64a249a0e6ccdd8b667e9 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e126821625444a72a2bc7088eb105251 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8dee313564645649cfe1743d243b191 : SUCCESS in 4m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d48a524138e347e9963af7656c9f7356 : SUCCESS in 18m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e55ed02518e2401184ec40086dbf361f : SUCCESS in 1h 02m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/124dbfb765f348b09bb34899bae638ac : SUCCESS in 1h 19m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/55e69b451300417a89339027a662ae2a : SUCCESS in 19m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/97252a29890740289bc8f6919fdbb630 : SUCCESS in 1h 07m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/08e17ea27610470bb49ef990865310de : SUCCESS in 1h 00m 51s","accounts_in_message":[],"_revision_number":3},{"id":"66c4a26ec949fb3566695a9bb19f78310f1c5d59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 12:52:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f6ae4a76d792057aa385b118f7e3ed0c48fc017e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 12:53:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6462e679a06341cb8d2f25bc5429cd20\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecbe411d2f7344bda2b4a81c351e98c8 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6786e4384c1a19e4658310e109b8ab9704f5eff9","revisions":{"3a5a3bdc664da00191fb4a4d5b3633d63772b413":{"kind":"REWORK","_number":1,"created":"2023-05-23 14:15:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/883941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/883941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/883941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/883941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/883941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/883941/1"}}},"commit":{"parents":[{"commit":"58030f8677547efa3757451097a75747fff4438f","subject":"Merge \"ubuntu: use Antelope UCA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58030f8677547efa3757451097a75747fff4438f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:15:53.000000000","tz":120},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a5a3bdc664da00191fb4a4d5b3633d63772b413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a5a3bdc664da00191fb4a4d5b3633d63772b413"}]},"branch":"refs/heads/master"},"1ad9da92312336cd653c0414b408b9f2c07e126a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-23 14:17:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/883941/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/883941/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/883941/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/883941/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/883941/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/883941/2"}}},"commit":{"parents":[{"commit":"239f9adff7c08f1fa24d9a71b54bff532467ec59","subject":"CI: Move useful rocky/ubuntu jobs out of experimental","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/239f9adff7c08f1fa24d9a71b54bff532467ec59"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:17:35.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ad9da92312336cd653c0414b408b9f2c07e126a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ad9da92312336cd653c0414b408b9f2c07e126a"}]},"branch":"refs/heads/master","description":"Rebase"},"6786e4384c1a19e4658310e109b8ab9704f5eff9":{"kind":"REWORK","_number":3,"created":"2023-05-23 14:27:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/883941/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/883941/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/883941/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/883941/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/883941/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/883941/3"}}},"commit":{"parents":[{"commit":"239f9adff7c08f1fa24d9a71b54bff532467ec59","subject":"CI: Move useful rocky/ubuntu jobs out of experimental","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/239f9adff7c08f1fa24d9a71b54bff532467ec59"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:27:51.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6786e4384c1a19e4658310e109b8ab9704f5eff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6786e4384c1a19e4658310e109b8ab9704f5eff9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
