)]}'
{"id":"openstack%2Fkolla~884001","triplet_id":"openstack%2Fkolla~master~Ibf3983f49795eea1f75aa7efb120c812adab4c86","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-18 15:30:14.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-18 14:09:37.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-08-18 15:30:14.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-08-18 15:30:14.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-08-18 15:30:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf3983f49795eea1f75aa7efb120c812adab4c86","subject":"ubuntu: mark collectd and telegraf as buildable","status":"MERGED","created":"2023-05-24 07:56:25.000000000","updated":"2023-08-18 15:31:12.000000000","submitted":"2023-08-18 15:30:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884001","meta_rev_id":"3e2f1b67056ebe8dddd3ce396543f59a880e48d7","_number":884001,"virtual_id_number":884001,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-18 15:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-18 12:47:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-18 14:09:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-08-18 12:47:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 09:17:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-24 10:24:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-05 19:05:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-06-26 07:53:14.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-07-15 06:56:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-08-18 12:37:38.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f72ef5c7b35ca5d323509a81284285063810d85a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-24 07:56:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4251f26d03e11295c7534896dc18e2bd5008d70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 09:17:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9eaafbce80ab4bea91a01c2666e70e61\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c5c5f771c8ae4316a8d64da129761d3d : SUCCESS in 20m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/55117a3b54db4c76af05fb0248d78648 : SUCCESS in 1h 06m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d06f7fb9df25441288a156bb88a0340f : SUCCESS in 16m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e36b2a219edf42cb86908cce9771a755 : SUCCESS in 51m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e36e1814454047d68b9966fc08642249 : SUCCESS in 1h 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21c1a57e08234e6794a8c741758c307d : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad96ecee58954465a5dd5081a656590b : SUCCESS in 2m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40b9ac56c3e145b6bf55d9b95f24f85b : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fc4ef18e13547bb9344bc3d2006b1f9 : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13c2278bab1044a1b9e50f152e3eee0e : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f9693f55e7d41c3a4bd2cbcbb61048b : SUCCESS in 4m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cb901b157b9e497ab877ee2a9112d865 : SUCCESS in 18m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f372c0f9692244438857162ae39a79b6 : SUCCESS in 1h 15m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dc65b70310e847e9a7cf491a98dc1cae : SUCCESS in 1h 15m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b81a2d4258bd44948602587e54415dec : SUCCESS in 15m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0351e378cdee4d468a87fe6726424f89 : SUCCESS in 58m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/922f245ce94a414e8a360e2169e393d5 : FAILURE in 47m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6cfccfa0c6b7598a942931ba29cb4d3026d38fc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 10:04:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cf0443a208643d7a30f6f0cb3ddf766\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c02b04b6fff44099a97c13f435dcb809 : SUCCESS in 26m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/85eb9ca955ad4fc29020ef0b1f828272 : SUCCESS in 27m 36s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/92e8935e8ccd43668d21d9ba9fc21d80 : SUCCESS in 2h 06m 33s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb0780a994bb4aaba449a20726b67ab6 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/afa7fa82aff84a15b0266cf33c457bf8 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ba10e2fa9ea4127bbef022e825b9fd9 : SUCCESS in 3m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/870a1a2f29044d7cb13e4e0ee7fbad7b : SUCCESS in 26m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/67bb0eebb9954a80b3db6ec749111731 : SUCCESS in 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82b9234ed2d48048f47e67548f0387fde7eb07de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-05 19:05:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e2a1a9cb2f0f3c60a499b09da3cef38b26b831ca","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-06 10:35:20.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3122cf736eefeecbf68f2f7114ac4cb900643b0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 11:56:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b9d685f75e24bbaa6c22705f84ef6cc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/94cf434d5f4f40d89bef8dea3c54794c : SUCCESS in 34m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8cce3e4eeac7431bb2f43b3d6a8c1c1e : SUCCESS in 1h 10m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cb970098a06f4f999367ff487ae89400 : SUCCESS in 20m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c72c1e288ed9429ba0b2f99368220f17 : SUCCESS in 56m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/40a4055a1b1d43ae941ab157d2a20c2d : SUCCESS in 1h 14m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f35eb3e59514de1a20c57f749e6b794 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6458a654c6740d890e18d2c6a1aee9d : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0475e46486446978a0a50a3d179a4e4 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ece9af763506425e9ff3e386281f2b3a : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea0dbe4a1963434395b64e8d9497e9c0 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e435f1b15c4cbe95e793ff8c5c24f3 : SUCCESS in 5m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/71c38e71aa224b95b9706e03e7189d52 : SUCCESS in 25m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/74ff1ea813494edda5f4f1f4747d2e49 : SUCCESS in 1h 02m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2e1df3c8c56549f08ec375d7e6f69845 : SUCCESS in 1h 14m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/643a2c15bc61416daa6dc79bbd6644a6 : SUCCESS in 17m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/303894d282e94f1b82a188db1717c21e : SUCCESS in 58m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b3c072f0d9724f4e9eae27916b44d799 : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e6786086e096880fffaf974f25478fd3e59d12e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 13:03:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d38a8d082b49422d89afbd080696b359\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8820a314dc074944b88724619f1fcb27 : SUCCESS in 20m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3ab839873a7248cfbb283545bc8f5616 : SUCCESS in 1h 28m 17s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/129c82e892f940a7b02f33f9b82fedfa : SUCCESS in 2h 25m 52s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae7bee008c2b4d308c226b30ee3f2785 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80fdcdb4d5d24c4a85287a9b812c3806 : SUCCESS in 2m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7aeb6d291c7a4fa9a386ae62e2688c81 : SUCCESS in 2m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cec7114d22444a09a4c7911e4e9a4cc3 : SUCCESS in 16m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7ad023c459dd42a1a571339defc79917 : SUCCESS in 15m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cf6b1fb9c8cf85f961b84066a99823eeaae239d","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-06-26 07:53:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3919f7908fbc5909f1ffaee4759dc5cd602f6e1c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 08:15:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* kolla/image/unbuildable.py\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"279ea968598e893a894298aee3effb7144706fce","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 08:16:17.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d4601f5a265572ff1b4a1f09059f66ddfe49d2aa","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 08:16:23.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"25ca75c08961d71049dd10fc04e0bede0819437f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:57:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/162f1d7b0e6d45a9b925d78af4666970\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fef164d7544c4019ac21772dc5ba1d2d : FAILURE in 30m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7e71b967f7f946abbe5ea95c2042418e : NODE_FAILURE Node request 300-0021541972 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/614c866aac4841d1b61cba5e26edfcd9 : NODE_FAILURE Node request 300-0021541973 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43fafcf6d76a41dd8a9f10c70374a80d : NODE_FAILURE Node request 300-0021541974 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65c31a2f54b747b39b85324c18689953 : NODE_FAILURE Node request 300-0021541975 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f226791d4f954296ba2086af9b2f8a6d : SUCCESS in 2m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/814dfb3959e1493089b50c7b44b23952 : FAILURE in 34m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/84d8c7dc590c408c913f1141684247d1 : FAILURE in 32m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f2282ba40772fb3ed8494ffb8edcba2b3148f732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 09:41:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/874cf39b85d443d38e4a65406472432c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/977445efd1804456b4ffd75524554cf4 : FAILURE in 24m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c8b26be9946b47eeb105d98b02551f6b : FAILURE in 27m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7ed41ad8cba04a46a33667b126a8fdcf : FAILURE in 23m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3fbbe668aa2b4be69ddacd9725b13fc6 : FAILURE in 22m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bd2ddf5e0e944920825dc777d8411c3a : FAILURE in 23m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a740f384a6c44d6996f2b0e314d9fed : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4d21a579fba42978c32f4b3f1fb95c4 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/594b49c9e86a42d2b295d5cfbcf4dd12 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9138c59c39654cd3ab11eccba5f81da0 : SUCCESS in 3m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dd81b12247b4f66a67b2c4ebab7b173 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f803dd62d1d4459d91f632657a1d9570 : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6aa202f474bb4cb28a5fcb1ae59817d9 : FAILURE in 19m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/08574a205c0243219a3183ba8ae6f8cb : FAILURE in 17m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2e0c465b9bc4467bbc6de5973ac81813 : FAILURE in 19m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b6ae77e8cdf74f0cb61a4f82f3068224 : FAILURE in 1h 11m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1c76d910746b4f1ca36321f9c51db62e : SUCCESS in 1h 05m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/56da01f6851a41528be0bb4ce2691476 : FAILURE in 1h 16m 42s","accounts_in_message":[],"_revision_number":3},{"id":"c9faa50f804df3cad5e5be06ab7e878ef5d961e9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-07-15 06:56:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9f209277181cd7a5aa23ebcfeddfbb884cd5f4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-15 08:20:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb3741a4f4874ea9aa84d49fa7e955c4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2bf060799e8e4f44bb5c46ba6303d871 : SUCCESS in 31m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0937e48a81a349918a30bd5a51b25709 : SUCCESS in 28m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/607286a379cd458db86ca33221162a84 : SUCCESS in 1h 21m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/239afdaa22ca42219097adb9309b6438 : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e0d0571622a43289afcae8d234a77f6 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3267b5319c9c433d91bb70c564267e27 : SUCCESS in 5m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cd1891be056b46a78a46c33eda5161cb : SUCCESS in 34m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/563491a944014af2be4f82ff4cb38920 : SUCCESS in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f5634cc0d848f9af1d9942a1bdd1164c15b0ae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-15 08:23:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cda0aa5235794b529c6c30b6e30c68c3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/47046678a6d74a66961c57807efa6262 : SUCCESS in 16m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/31dff0d44c5c462ea6650b001cd13a1e : SUCCESS in 45m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d7cdfab8b747479d9d35897de6ca8cd8 : SUCCESS in 13m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/831146242f0847ddbcd29f7ba690c1b7 : SUCCESS in 53m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/22c6a9f8f0584e429e8b7d2d41bb3a0d : SUCCESS in 51m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7f9fd50778f4742b0be44c9f444ac66 : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/481317fb7e894931849ca9f5b316f873 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f997db44487b4e248dc3ae04ffb211ae : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/071e2dfce8e142519eb8e9103e8ff6e4 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4448c17eeacf4cd78f5173a186269d0a : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1dddaa960444e2fbeefcf80f4d1b3d9 : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/716aeef8a0f3475b81c76e8683908db2 : SUCCESS in 18m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/94dd92c929564185b0c846a76758021d : SUCCESS in 1h 07m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3c62755d230b42eabb3864c88c656b95 : SUCCESS in 1h 20m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/57cf9cb6657a4b529be165c937bb9f88 : SUCCESS in 19m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5d110772df134f0380ba5801854a8d32 : SUCCESS in 44m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/36fed8bdabc74b2985af343c9789f601 : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":3},{"id":"78ddce6e50f189c9f19c3de4f09e747a8257a7ee","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-07-17 19:55:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c9e557b061acfdcdf243e08f890ea0b5291dff65","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-18 09:47:03.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nThe following files contain Git conflicts:\n* kolla/image/unbuildable.py\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e2c237baabda4d13cb5d975dbef91b1449e6529f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-18 09:47:30.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"a76c4ecdbaa3b8f8a91e152206d6d2bd8ce6efbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 11:05:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e95d48105bb4ba392c47d3413dac7e6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/733416ebe0b8443dbfe90d468f3717b9 : SUCCESS in 25m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b2c2d83287324812b19ae2acf821c675 : SUCCESS in 1h 11m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fbf3deeaefb846788df5ca17a0f88507 : SUCCESS in 24m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f0dbe5a4e3354b4485f73ef93ffc6ba5 : SUCCESS in 52m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9be4386d46204b8e93300dfd33b1920a : SUCCESS in 1h 02m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ce1996553b40f1b8795e79f603d79c : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4239923cb5e0402dbf40c34ee3046891 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e6f7a99eddd4da4bd20ff28b990c70f : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e03535578bf840cab2f262b2354c4bd9 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70dc918ebf114c59a57a6a1839e7b40e : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3257628d22f84e0c8875435fa16a950c : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f3ea79b4204335a6b8dcca70f8b19a : SUCCESS in 4m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/85a94b12f74548778f4b02d1fbf405ea : SUCCESS in 18m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d9e0bf74598f47c18828828c8a2d8199 : SUCCESS in 1h 03m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ffd1b7f5d8764e0d9869c19ca7b4a6a8 : SUCCESS in 1h 06m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c8038b2a951a45a7a31947664eaa16a6 : SUCCESS in 18m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/89be59485fcb4bc4b7d3529f6b55a739 : SUCCESS in 1h 01m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f2705c9993794d4f85811b46bb577648 : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":5},{"id":"8ca0d0e3177095cb35a11f863cfe57c9d3919aab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 12:06:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f4070cbcf7e45f387c4fb258a336d94\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/787bb167b3d74b818f9693e239d426bc : SUCCESS in 28m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/01b48727d2a94ec0bab8a98c130bf4e7 : SUCCESS in 1h 25m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c0a17602568d4cdc9934e94c46b9fb4f : SUCCESS in 2h 16m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9ea069dadb949ee9d8eb19a16ca58fd : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7406db2fce249d19d10a98cfddf7f30 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f520481a27d14226b9630c5a22b343fa : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/06275055347643a8a66adfcf39172c6c : SUCCESS in 5m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b2aa0897522c44ffbac9f27d4fdebe3e : SUCCESS in 1h 19m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0f79442ba1284c6fa8583786f8eed7ed : SUCCESS in 1h 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a68cac676c542f40c8296529a0a1dfcca4eb8efb","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-18 12:37:00.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"ca6cbd69d593c7f58faad9c46908c884a54ed57e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-18 12:37:37.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"932bac3b591abcd3d240afe168e6581a532a81f7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-18 12:37:38.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f258cfaca814bf4f3ddb51bdd5f6fe5fee76d999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-18 12:47:11.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"a36384eeccce8bdd62d0c77387df25f15fc76d84","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-18 12:47:34.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d29a4e248cd55ab2fd6fa0f751793cea4738e0fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 14:06:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a857834d971481fb1b8e6432c29118e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/eaaf3a2aadaf40e7bf232a9ffd54e275 : SUCCESS in 19m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9976704059084963871e2796152e375d : SUCCESS in 55m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dd4a7599acc14dc2b4c7a134c901b737 : SUCCESS in 14m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e9f9c6b8f7c84806bc910bc1d1c279d1 : SUCCESS in 58m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/620bef1e38214ed39b1902ab201b0b6b : SUCCESS in 58m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e7c6ea1e2374c8aa64be920b11eb633 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38b47d8f5bfd44b9b1569aa7803e0c12 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a55869fb8394d76b10ad399e0b90e71 : SUCCESS in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9221abef8edf492b89951742171ed9fb : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed036da891984f128eff090f94d62065 : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f497c7c8dad4ae7b3a4e5130a1f86f1 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6684fc64921499fb168c6f4b2180c40 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/13ee6a5ef4e345b69af70ae2c20083a6 : SUCCESS in 20m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/222634fa819d4fd29896c08e663cf053 : SUCCESS in 1h 10m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/126fcb4f0da84e2690b92d3deb1955f3 : SUCCESS in 1h 13m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/225ac51dd7b443828b15f7f963037019 : SUCCESS in 24m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6c8670876b7e46708ae2a2b2e9ed38be : SUCCESS in 59m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/128f6f1d13474f83ad412efa1c9f32b7 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":7},{"id":"03446ae53947f189183f7616aaf81da2d7f615d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 14:06:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c0d293e125de53fc111680f2e5cc52b49ed78caf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-18 14:09:37.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"25f90c5c46839fcbe7cd421749115a639815961f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:09:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/135b8a2d6e284ce1b1b456cead59e3c5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e8c53257a5944e00a8816726109d9976 : SUCCESS in 18m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e43fad7233a342ef9bebbb8b58b351b8 : SUCCESS in 17m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/27ee9b412a2e4878a4d30301d128eae1 : SUCCESS in 2h 12m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f75e3ed763b474f9a7031a7b246fef0 : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68e259284329486d93277573cad0a26a : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d20e0fc8022447d8812998e64cee90dd : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/82d0630e99b74f6eaf6da0bb12dfa46c : SUCCESS in 5m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0379445e814a493da90d666c7d113321 : SUCCESS in 1h 17m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9a347ef301d74382b014920a1ce76bc3 : SUCCESS in 1h 11m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9bf758531cd415954aa5008427c09c60670cf9fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:30:14.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22488dd53a16485aba7261077e49cd0e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c99c1473e728415cbb42675e43d3c647 : SUCCESS in 20m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d55acae5be2648fdacc445131162e09d : SUCCESS in 57m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ce312719c2aa49bfa39415f4b877bc98 : SUCCESS in 1h 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b8d8d263c564847b8e9fb2130edb8ea : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cc660cea19a44978ef72cc873f2d5e1 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2c22f185396483b82d2c49dee75cca9 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c0b5d843832484884e965efcca349b2 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44516b6ef6fc4750836f4b77f8c37c8d : SUCCESS in 7m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d2ce97d5adee401face18a6e6af24be2 : SUCCESS in 18m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/18dfff6bb09d4abfac29f0f1fa63ad99 : SUCCESS in 1h 14m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/999df8e13cd543dd9b60374043cc9808 : SUCCESS in 1h 18m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/52cc412b08874f0da05116342603be80 : SUCCESS in 17m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e92e86ead3e04b1e800af804ae827dd8 : SUCCESS in 54m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c3fd6a121312435a94b2f96efebe34f4 : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":7},{"id":"4739724235b4de927ec9d08bc2c9ff73d87658cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:30:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"3e2f1b67056ebe8dddd3ce396543f59a880e48d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:31:12.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5abec2690a149039ec05bf2e71cde31\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b6f68f5aa042a389f0fe6d739c9bc6 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f000360c2d699e5b2bd9bee00e78055deb1540b4","revisions":{"9f233306d7f7d3f69716a0f5088f9f9bf21ac3df":{"kind":"REWORK","_number":1,"created":"2023-05-24 07:56:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/884001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/884001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/884001/1"}}},"commit":{"parents":[{"commit":"58030f8677547efa3757451097a75747fff4438f","subject":"Merge \"ubuntu: use Antelope UCA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58030f8677547efa3757451097a75747fff4438f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-24 07:55:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-24 07:56:23.000000000","tz":120},"subject":"ubuntu: mark collectd and telegraf as buildable","message":"ubuntu: mark collectd and telegraf as buildable\n\nCloses-Bug: #2019756\nChange-Id: Ibf3983f49795eea1f75aa7efb120c812adab4c86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f233306d7f7d3f69716a0f5088f9f9bf21ac3df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f233306d7f7d3f69716a0f5088f9f9bf21ac3df"}]},"branch":"refs/heads/master"},"89e3981125216c541ff329ec0e7799f45d6b8ca0":{"kind":"REWORK","_number":2,"created":"2023-06-26 08:15:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/884001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/884001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/884001/2"}}},"commit":{"parents":[{"commit":"b33167972d6e17b2eaab85e72043e3b4efa4b789","subject":"opensearch: drop unnecessary workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b33167972d6e17b2eaab85e72043e3b4efa4b789"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-24 07:55:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-26 08:15:44.000000000","tz":0},"subject":"ubuntu: mark collectd and telegraf as buildable","message":"ubuntu: mark collectd and telegraf as buildable\n\nCloses-Bug: #2019756\nChange-Id: Ibf3983f49795eea1f75aa7efb120c812adab4c86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89e3981125216c541ff329ec0e7799f45d6b8ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89e3981125216c541ff329ec0e7799f45d6b8ca0"}]},"branch":"refs/heads/master","description":"Rebase"},"0ae8684bf5965be9c81a09e0d89f9731b3f5c1e6":{"kind":"REWORK","_number":3,"created":"2023-06-26 08:16:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/884001/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/884001/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/884001/3"}}},"commit":{"parents":[{"commit":"b33167972d6e17b2eaab85e72043e3b4efa4b789","subject":"opensearch: drop unnecessary workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b33167972d6e17b2eaab85e72043e3b4efa4b789"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-24 07:55:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-26 08:16:17.000000000","tz":0},"subject":"ubuntu: mark collectd and telegraf as buildable","message":"ubuntu: mark collectd and telegraf as buildable\n\nCloses-Bug: #2019756\nChange-Id: Ibf3983f49795eea1f75aa7efb120c812adab4c86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ae8684bf5965be9c81a09e0d89f9731b3f5c1e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ae8684bf5965be9c81a09e0d89f9731b3f5c1e6"}]},"branch":"refs/heads/master"},"a8a3d3aed0c271ca7f29e38724eb2efd9b0a9742":{"kind":"REWORK","_number":4,"created":"2023-08-18 09:47:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/884001/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/884001/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/884001/4"}}},"commit":{"parents":[{"commit":"a004abbdc36b425576c7a9a8c480df938f5c0a1f","subject":"Merge \"rabbitmq: bump version to 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a004abbdc36b425576c7a9a8c480df938f5c0a1f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-24 07:55:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-18 09:47:03.000000000","tz":0},"subject":"ubuntu: mark collectd and telegraf as buildable","message":"ubuntu: mark collectd and telegraf as buildable\n\nCloses-Bug: #2019756\nChange-Id: Ibf3983f49795eea1f75aa7efb120c812adab4c86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8a3d3aed0c271ca7f29e38724eb2efd9b0a9742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8a3d3aed0c271ca7f29e38724eb2efd9b0a9742"}]},"branch":"refs/heads/master","description":"Rebase"},"02f207d5462f1c90f7df00f26c51b02db41cb87f":{"kind":"REWORK","_number":5,"created":"2023-08-18 09:47:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/884001/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/884001/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/884001/5"}}},"commit":{"parents":[{"commit":"a004abbdc36b425576c7a9a8c480df938f5c0a1f","subject":"Merge \"rabbitmq: bump version to 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a004abbdc36b425576c7a9a8c480df938f5c0a1f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-24 07:55:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-18 09:47:30.000000000","tz":0},"subject":"ubuntu: mark collectd and telegraf as buildable","message":"ubuntu: mark collectd and telegraf as buildable\n\nCloses-Bug: #2019756\nChange-Id: Ibf3983f49795eea1f75aa7efb120c812adab4c86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02f207d5462f1c90f7df00f26c51b02db41cb87f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02f207d5462f1c90f7df00f26c51b02db41cb87f"}]},"branch":"refs/heads/master"},"faf444f8907cb1894a8f2095ca0385812c2d9c43":{"kind":"REWORK","_number":6,"created":"2023-08-18 12:37:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/884001/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/884001/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/884001/6"}}},"commit":{"parents":[{"commit":"a004abbdc36b425576c7a9a8c480df938f5c0a1f","subject":"Merge \"rabbitmq: bump version to 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a004abbdc36b425576c7a9a8c480df938f5c0a1f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-24 07:55:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-18 12:37:37.000000000","tz":0},"subject":"ubuntu: mark collectd and telegraf as buildable","message":"ubuntu: mark collectd and telegraf as buildable\n\nCloses-Bug: #2019756\nChange-Id: Ibf3983f49795eea1f75aa7efb120c812adab4c86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/faf444f8907cb1894a8f2095ca0385812c2d9c43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/faf444f8907cb1894a8f2095ca0385812c2d9c43"}]},"branch":"refs/heads/master"},"f000360c2d699e5b2bd9bee00e78055deb1540b4":{"kind":"REWORK","_number":7,"created":"2023-08-18 12:47:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/884001/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/884001/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/884001/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/884001/7"}}},"commit":{"parents":[{"commit":"a004abbdc36b425576c7a9a8c480df938f5c0a1f","subject":"Merge \"rabbitmq: bump version to 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a004abbdc36b425576c7a9a8c480df938f5c0a1f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-24 07:55:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-18 12:47:11.000000000","tz":0},"subject":"ubuntu: mark collectd and telegraf as buildable","message":"ubuntu: mark collectd and telegraf as buildable\n\nCloses-Bug: #2019756\nChange-Id: Ibf3983f49795eea1f75aa7efb120c812adab4c86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f000360c2d699e5b2bd9bee00e78055deb1540b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f000360c2d699e5b2bd9bee00e78055deb1540b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
