)]}'
{"id":"openstack%2Fkolla~884208","triplet_id":"openstack%2Fkolla~master~Ie1b36eb9a403611c2f86aa4c1d500d324e5291f3","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-25 14:13:52.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-05-31 09:33:09.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-25 11:45:11.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-05-31 09:33:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1b36eb9a403611c2f86aa4c1d500d324e5291f3","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","status":"MERGED","created":"2023-05-24 14:19:50.000000000","updated":"2023-05-31 09:34:12.000000000","submitted":"2023-05-31 09:33:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884208","meta_rev_id":"e848fc37c89cb076f92daaae3ad6bccaacef7afd","_number":884208,"virtual_id_number":884208,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-31 09:33:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-31 07:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-30 15:50:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-05-30 15:50:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-30 15:50:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-05-31 07:43:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-05-31 07:43:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 14:28:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-24 15:32:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-24 16:29:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-25 05:09:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-05-25 10:38:52.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-05-30 08:57:16.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"78f0c3db7fba3a365e2aed081e7707f1d455ec2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-24 14:19:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62e06fea30a467f7e50262ded3c4117adc8ae257","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-24 14:28:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e614193a2d92bc3a06d67f5b69410533958c0893","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 15:32:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/787c62f235874987a5106531ba695444\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7077e00f85cf4a1a8f3525790e6a93ec : SUCCESS in 31m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c961d028609e4917ae02424dfb46b839 : SUCCESS in 30m 44s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d6b867022ab2446f8cae159342c3da1c : NODE_FAILURE Node request 300-0021298942 failed in 0s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac99b3567c624b1da9f16de7ced79916 : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3041b8a6f53418794f7edef0b54fb5f : NODE_FAILURE Node request 300-0021298944 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/051f2593b2574f539685b4fc825e3ba7 : NODE_FAILURE Node request 300-0021298945 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7b96c174fbdb42c6a6859c2bb4b72887 : SUCCESS in 1h 11m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d6c0e5d0244e45fd8ec1dfc0c1952207 : SUCCESS in 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c2768990cd5634ec09a95b2e752e6c40e52d797","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 16:29:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/313a7fc6f6244813a1a62a1ee41bf00a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e8d4b8d189734613a4f805091d03565c : SUCCESS in 16m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/01955f1bd7dd4dd69ef4aa9eaa665587 : SUCCESS in 1h 11m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8f9734b43a484842be6ede36d5d776ac : SUCCESS in 16m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/98f9023e30b240ea9b9e8e7d018a5215 : SUCCESS in 1h 12m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/005bc85432f0451c9a4dd571cd277342 : SUCCESS in 1h 23m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4242a462d8a84c1ba6d8b1ef9ede5682 : SUCCESS in 2m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e722b62740ec4aa8ac16c23012268d0c : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b7dd3cbed994ce1996fcee883e9ebbc : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2d70f65eb5c41d78d622e0db944d997 : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/667923748e58419396ff6ed7cda0ae08 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ca809626d524e37b090dfa4e4c4f95f : SUCCESS in 5m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/606e608970fb482fb48e87b3088c3efa : SUCCESS in 22m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6ab4161a1fbd4314a6f9309b673924b4 : SUCCESS in 1h 11m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b2cc4aa16cd041bf96188490f2565830 : SUCCESS in 1h 10m 34s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a3f64ac82c144b4e9bf02156894d7f05 : SUCCESS in 1h 37m 14s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5d14cb7da9f84321940841dff3c11fb1 : FAILURE in 1h 28m 36s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/1451be6fe8674b90998da051c2383113 : SUCCESS in 24m 40s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/416bf9995e684927aeb5a0e4e1622328 : SUCCESS in 18m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/abce8bb8c61a401fa0f0f256e4ecfc9a : SUCCESS in 1h 07m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f4f373a83a2d4f25ab1c1e88783b43b0 : SUCCESS in 1h 00m 41s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/fc50d2ac90b24a8aa7b917510ec82bc7 : SUCCESS in 1h 26m 01s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bed40f59fc0a4a398272821740fa773e : FAILURE in 2h 01m 00s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ca9eaf2ce00740f09af4a2d01cfe15b7 : SUCCESS in 25m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f675fb60fa0859df331ea29dc9d5d51e66a41f25","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-25 05:09:59.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"161bf68f5a087eab621a604e04d31d0534eebbc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-25 08:58:56.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ea7535cab351b835023ad26514ff21608fc493bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 10:10:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b95b10050b14b458af3b3e18fbab732\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c05262b6a337403688ec72baf6c90139 : SUCCESS in 16m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a1d7c21961da4b38855457fdda0a715f : SUCCESS in 14m 57s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cff3ac4692ff45069983d41c8c2a6ad9 : SUCCESS in 1h 09m 06s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab99c126c8f14117a256dfde3c3a96b0 : SUCCESS in 2m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68f5b37737834f9580ee42744c7e49a4 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/339dd05850834902bb5ad72ec9d63532 : SUCCESS in 2m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2fa0fc8eca74402ea3e853358dc8dd45 : SUCCESS in 37m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f3775922b9254ea58b070f343991e350 : SUCCESS in 16m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"29dcc6de600ab36b44c035cedfeadb686b353d4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 10:36:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a6ffa362d344634afc53ff6c6c3a2a4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8dbb639e554b42b1a0e366459dc1c574 : SUCCESS in 20m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7cba6d19b1bc45b6a16c3c16449319c5 : SUCCESS in 56m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a1d59e3b4113498faaf605aecbde316c : SUCCESS in 26m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2e2f475b897944f3a34b55106fb6a5d8 : SUCCESS in 52m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5fe3ace9af0240c2beb8544a9467f156 : SUCCESS in 1h 08m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/824f6a60de39453fa0e13af37b7bdf90 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce13e0181a28487e8a7660353a71e219 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/963ff3ffa93c432fad6bd976186e6567 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0fddb42eb484c2889430cfff69249ca : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0313e4a4d5b144e0a23a72d387e20ecd : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6731318e6bb049a894d9c16d03d66d7b : SUCCESS in 4m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e985912d67b543e5968d2a8a9bfec93a : SUCCESS in 23m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/30b28711315e414c868f7676fc272637 : SUCCESS in 1h 00m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3d3de13606654edab97536e162e003d2 : SUCCESS in 1h 14m 19s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/39fc36f9ec054aa5843ec16c366a2cc1 : SUCCESS in 1h 32m 00s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a9ccbd0a5c2b4256b3c0f88326cf6bc7 : FAILURE in 1h 07m 43s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/636f75be5da2494aabf5a524b1bf5bcc : SUCCESS in 24m 31s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a839f2dc92864e728509e141b2466c4d : SUCCESS in 15m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b52b1ce80b214a929dae663ed08a882c : SUCCESS in 50m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/481a0cb9b88146a1a09c27a6db7ab87f : SUCCESS in 1h 04m 39s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7e04aef64fa841bd8a00837a653f5925 : SUCCESS in 1h 14m 14s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/054390712c324974af6f1fc22c98b8c5 : FAILURE in 54m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/4e96d5e0672e4c1ebc5ac45f61302529 : SUCCESS in 21m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6327172e8ea4d6610cc318ea04d15ede3fd54c5e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-05-25 10:38:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6a3e62dc145e9e0397cd408739ea2351acb757a9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-25 11:44:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"205cd5ad2b057f0b95f96d43506fc4a128d5ad14","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-25 11:45:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b614c7be34a4b74e277f0f8d5c6d3a1280c7a8fa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-25 14:13:52.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"fa8608cbacb1c3225810a87d5e6e43b2e9a32036","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-30 08:57:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81468a14089443aceff268a7bd0415f0032b32fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 08:57:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bbdfde5a6787e59d9f8366b5e8c43aadb5a69664","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 10:14:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b9ed00ed0f540ce9074efbe772f5103\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/519719833fa84237a8141c803d4195b0 : SUCCESS in 18m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2085c5968db840838d748ffd37b97cc7 : FAILURE in 58m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f9793869baf1402e9f678ce521bee7c4 : SUCCESS in 1h 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283bb0637a6844a5a87d4a46c69aa978 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6791e917258b42cab1f819651be4ba7a : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0fd253fc8304ae6bc1ee10bd67fd81f : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eeb91ff1efee45d89575e7f1dbd19baf : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3c74311706148b3a9892a6293afe9c7 : SUCCESS in 4m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/76743e7faf4f48349e933561dbd7b36e : SUCCESS in 23m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0c38c86d44e944848928f07dc050a36b : FAILURE in 51m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2156a34a79494f2fad1e61591f316ac4 : SUCCESS in 1h 02m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c753bbdc6c6c4d358d977d962dc47b5a : SUCCESS in 22m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9674a0f0f35844738ceb497f9cf5b9ee : FAILURE in 44m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/eb18f81df088430f8810fbe3c1a07fe2 : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":2},{"id":"26023ab4d9648b51da59c1e029fc3d689dbe5682","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-30 15:50:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a703181abaeb8961f976e8af3702a76dfe2011cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 16:56:58.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d369b42330ce48ebbbaa75a5a91c1138\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/720fc275e2424600ba7824869196735c : SUCCESS in 22m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/06b8471c9c884798a00edf0fe08f4b1f : NODE_FAILURE Node request 300-0021348479 failed in 0s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/21b23539b371474184ab46dff8537cc8 : SUCCESS in 1h 02m 21s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c34289c20b53434a98beed3ab207ca74 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5a7bf92ef554462b0e03d7cb6300ce7 : NODE_FAILURE Node request 300-0021348482 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/178d3e1c89d643e29a22a0eba913f169 : SUCCESS in 3m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/24a0b8eca4a0470da18cea869c0d7032 : SUCCESS in 23m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/64ed2e8447424764be7b6897ef0b7c0a : SUCCESS in 20m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a7c72e9e4e9b49d45643ae2fd7dcbe26408fef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 17:42:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f812c7a672d44f99b037a70a87424a8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4cee80bce85d49c2bbe389a73da9da49 : SUCCESS in 24m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8e073b023901444f8fc0c18e85d19830 : SUCCESS in 55m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3b11021cfd9045c3b317744f3fd24a1b : SUCCESS in 16m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b86a38cb21964ae6a02ef4309c3e6075 : SUCCESS in 1h 10m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e85a1b2c8fc0400b8b1f4842522b2f77 : SUCCESS in 1h 23m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/594d83583bf94a11b349329ecce9431c : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed8d14ef8c6241c38d2c6aa0a0142c95 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13dce3a09bea4581bd0438fac3ccaebf : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1461d61399224538a70cd0210c9c440c : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4363bcf0d445498596eaed8a45732bee : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016db69b39064295a7645f30347d8349 : SUCCESS in 4m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5ad650abdaa1433bb55a4fe9d011afbe : SUCCESS in 18m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4284364ac5ad4b38bca92b2fa870f934 : SUCCESS in 56m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/256fd8910b36476c9b50fa65b5345fd5 : SUCCESS in 1h 11m 39s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f20416ca61e54f49b3555805c0de466f : SUCCESS in 1h 33m 12s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/361ddd4799f44d9f8206411407b6d16b : FAILURE in 1h 00m 55s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/601d5abae8a742beae90d9dd42460052 : SUCCESS in 27m 49s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e3c71416fec244148ae1df85dd80e0af : SUCCESS in 17m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1372a45e738a4d90a0eb285b4a2cf039 : SUCCESS in 55m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/463f3fcfad614123bb1d5ad9f65be0c3 : SUCCESS in 1h 02m 19s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f3b52d89442d4d90a6ac37f9114fa7fe : SUCCESS in 1h 28m 10s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/629cc0f4e24046c3a5dd751f91a8134b : SUCCESS in 1h 45m 18s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ebc2093a9d8d44409658e1ea5e4b95d1 : SUCCESS in 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"95bcff5abfd5fb64d7c01cfe67db6d99740f3ea0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-31 07:43:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"1d6aa2a125b1c9878d4acb3ff83a458669230469","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 07:44:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a020026e771beb43f177ea140ee8cc631213601","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 09:33:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e7bc6a6e60044e489247a7a148871eb\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4944059cf539466b912a9ee35d979c70 : SUCCESS in 18m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/13634990842d45bcab711e212a194bf5 : SUCCESS in 1h 02m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/12457e96c76244dbbc9837a0d8a1eed7 : SUCCESS in 1h 09m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0044e62a31c4b87a0aa33cc4e6fcc67 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a4361e15b994c56aaa0de33823b5a06 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc4c0eb619714064828bde66e0fb02f4 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d6162f6296246058203d0d14dc0e93f : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd0f591b95af4ebabe27a141cbb6f1c0 : SUCCESS in 4m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/46e3629a3b2e4c72844fd8f7e94ed74c : SUCCESS in 17m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fd35c559c0e9412d8227f25aa968f1ce : SUCCESS in 1h 03m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/02fa48d0cd5f4e28ad9522f7116fc873 : SUCCESS in 1h 12m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/04a947946be14b04ac6ea5cfbef018e3 : SUCCESS in 17m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f9c35177f6a7400cba63adddbfe7617d : SUCCESS in 53m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b6180613604c417cb449e59039626473 : SUCCESS in 1h 09m 10s","accounts_in_message":[],"_revision_number":3},{"id":"1b6e382970cc34c486b5b323e3c8670e034b0c0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 09:33:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e848fc37c89cb076f92daaae3ad6bccaacef7afd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 09:34:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00365ff892bf49568f2ce2305619244b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a9eec6545a44a8a90a50e1b741c13d7 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9035251443ed6cbb110e9dc5efdbc9d4be797ef5","revisions":{"883bab10d151e956abedbe6ae2688557d61e8170":{"kind":"REWORK","_number":1,"created":"2023-05-24 14:19:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/884208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/884208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/884208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/884208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/884208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/884208/1"}}},"commit":{"parents":[{"commit":"6786e4384c1a19e4658310e109b8ab9704f5eff9","subject":"opensearch-dashboards: Fix permissions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6786e4384c1a19e4658310e109b8ab9704f5eff9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-24 14:19:16.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-24 14:19:30.000000000","tz":60},"subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","message":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists\n\nIf mariadb-clustercheck starts before mariadb, the directory may not\nexist, and the command fails.\n\nThis is necessary for the refactored MariaDB restart in\nI5a12670d07077d24047aaff57ce8d33ccf7156ff.\n\nChange-Id: Ie1b36eb9a403611c2f86aa4c1d500d324e5291f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/883bab10d151e956abedbe6ae2688557d61e8170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/883bab10d151e956abedbe6ae2688557d61e8170"}]},"branch":"refs/heads/master"},"ff46f929dfbe530b47c3bec041fc46c793af30de":{"kind":"REWORK","_number":2,"created":"2023-05-25 08:58:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/884208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/884208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/884208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/884208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/884208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/884208/2"}}},"commit":{"parents":[{"commit":"6786e4384c1a19e4658310e109b8ab9704f5eff9","subject":"opensearch-dashboards: Fix permissions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6786e4384c1a19e4658310e109b8ab9704f5eff9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-24 14:19:16.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-25 08:58:48.000000000","tz":60},"subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","message":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists\n\nIf mariadb-clustercheck starts before mariadb, the directory may not\nexist, and the command fails.\n\nThis is necessary for the refactored MariaDB restart in\nI5a12670d07077d24047aaff57ce8d33ccf7156ff.\n\nChange-Id: Ie1b36eb9a403611c2f86aa4c1d500d324e5291f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff46f929dfbe530b47c3bec041fc46c793af30de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff46f929dfbe530b47c3bec041fc46c793af30de"}]},"branch":"refs/heads/master"},"9035251443ed6cbb110e9dc5efdbc9d4be797ef5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-05-30 15:50:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/884208/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/884208/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/884208/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/884208/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/884208/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/884208/3"}}},"commit":{"parents":[{"commit":"600ed4f149a1cf542f54d7fc5005030f53c69a83","subject":"rabbitmq: pin to 3.11.16","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/600ed4f149a1cf542f54d7fc5005030f53c69a83"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-24 14:19:16.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-30 15:50:01.000000000","tz":0},"subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","message":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists\n\nIf mariadb-clustercheck starts before mariadb, the directory may not\nexist, and the command fails.\n\nThis is necessary for the refactored MariaDB restart in\nI5a12670d07077d24047aaff57ce8d33ccf7156ff.\n\nChange-Id: Ie1b36eb9a403611c2f86aa4c1d500d324e5291f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9035251443ed6cbb110e9dc5efdbc9d4be797ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9035251443ed6cbb110e9dc5efdbc9d4be797ef5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
