)]}'
{"id":"openstack%2Fkolla~884368","triplet_id":"openstack%2Fkolla~master~I169e51c0f5c691518ada1837990b5bdd2a3d1481","project":"openstack/kolla","branch":"master","topic":"bug/2020761","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-28 13:40:28.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-01 18:58:36.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-08-07 07:36:52.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2023-09-06 07:21:45.000000000","reason":"Change was submitted"},"32398":{"account":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"last_update":"2023-09-06 07:21:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I169e51c0f5c691518ada1837990b5bdd2a3d1481","subject":"Improve kolla-toolbox offline support.","status":"MERGED","created":"2023-05-25 11:55:42.000000000","updated":"2023-09-06 07:22:53.000000000","submitted":"2023-09-06 07:21:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884368-bug/2020761","meta_rev_id":"f98a491474eaf55bd1860c82e0d791e3eb55beb8","_number":884368,"virtual_id_number":884368,"owner":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 07:21:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-28 13:40:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-01 18:58:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-08-07 07:36:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-08-30 05:36:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-09-06 05:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-08-01 22:39:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-09-06 05:44:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-08-01 11:38:03.000000000","_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-25 11:58:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-05-25 12:08:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-25 12:19:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-25 12:31:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-20 14:40:23.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-06-20 16:57:09.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-07-13 01:26:39.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-08-30 05:36:48.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-09-06 05:59:24.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"96986e677dfd1f462dbc69bd1500d9c29e5dc21a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-05-25 11:55:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66a71a77c47f4455d76bfe985aa909dc9872e916","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-25 11:58:35.000000000","message":"Patch Set 1: Workflow-1 Review-Priority-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ff7ea19ebfa7eb13489628da8529112fd9c55ea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-05-25 12:06:30.000000000","message":"Patch Set 2: Commit message was updated.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ff4ebce9141fdfc80ddd7e05d72a48d290ed8a79","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-25 12:08:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2698e735f50d80948ebd03aca1cabb0415ec7b7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 12:19:34.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/184a2414c0364c59b3cc7af5d3abd5c3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2ff477c151374a7b87f285b29fb46ae2 : FAILURE in 4m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/926c945b16f14f37a8618f870317b589 : FAILURE in 9m 21s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f8263a1f67394b8e9afc9a53b6544864 : FAILURE in 10m 15s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/efcad84e4ffe47e59d201a0b4e842335 : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3455957d456a486ab1427e6aabb387aa : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/abcb7f4eea6244f6a456d2dd5f156f5c : SUCCESS in 5m 47s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8e13c0241f0f425da1a2b74d9e09d597 : FAILURE in 5m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d4e5d3c67ed64516b442331e31a59eb3 : FAILURE in 7m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"171f2a190bc1cc94399c0e23cb67a315e1de06ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 12:31:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c424cef7b80a4015a1699df38f09201e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b113ca7549f8415697c3e5f94abb9f88 : FAILURE in 5m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b5216fe7917546549d64257d8df07014 : FAILURE in 9m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e6af133a4e804162851fcf0f845f1fb0 : FAILURE in 3m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d7ed4de267a2418dab00d783c71eb509 : FAILURE in 14m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2306e5720bab40fcaf952d8499012242 : FAILURE in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99a48cc524c64f36b884741e675a2ba6 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6001fd0b74264398b1d53fb5655343b2 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d6a600b766f477b968af5b74da20f49 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/474d3b3df16e46789c8ecbda0644029e : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3c308b276a94609a26b73266cf53a65 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea3acda689d4abdb3c5131cc44d03ec : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19b10ef7320b4f82b250ac78dc7ea0dc : SUCCESS in 3m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5f4e4faa5d7540ca9a14e5755586008e : FAILURE in 3m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e6d049d90b8640bab72a7e360ad056c9 : FAILURE in 8m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5a9f1d327bcc4fb5a75608107e75f01e : FAILURE in 8m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fc17e225eaaa4e65b35326112fae67ac : FAILURE in 4m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a77ced59e7064e72925b882694d83009 : FAILURE in 8m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/476b50c8365a4756a282034a3f269367 : FAILURE in 8m 52s","accounts_in_message":[],"_revision_number":2},{"id":"31355437ddf1100e89c70ab35aa2a80425bf09d5","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-05-25 12:38:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ef18ebd1a340f3f49ebc47713a48522ab8f5af4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-05-25 13:09:20.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f59aa4922ce358590f8ce35ad48ab9807e0ea124","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-25 13:09:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"846a4de214e457d824e78470dc9a865edb33a26f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-25 13:13:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"20de07ee9ab80df59c76fd06ffcbf32ed4a08bd3","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-05-25 13:13:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e30f961d0cb412f1ab5b5d4a70d26364411fb398","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-05-25 13:14:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b2d0dd3779b6a1e6364e087db9c08414016ca207","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-25 13:15:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b836a483e0962babfc05cf6a0465b9476378e184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 13:44:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/211fce66ea074498879beb5236186078\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/28db6c70688e45a5a531491fde937abc : FAILURE in 21m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/537a0e173e694dd3aedeffb233770d99 : FAILURE in 22m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cf1ac3f1465d425f834e069478ef7b26 : FAILURE in 23m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3bb841d825614b13aa3970603e10ff7d : FAILURE in 23m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f906cba37aad4d938ef6a8b243ac9961 : FAILURE in 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2efa6cacabd473cb803c28f724a6fab : SUCCESS in 2m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/122ad94cd7bc4d5aacbc8e8c7b326748 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd4a695667d841a38944cd4e8edc1896 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f12a12268c3440e88b95f8a5187138aa : SUCCESS in 2m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01fda4e185f94c96b597afa219329e72 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf84a64238a4978825b4888ad892d62 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7627687797544e499262dddde23d5828 : SUCCESS in 3m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/983c593d60914579b4c684a481524117 : FAILURE in 15m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fb0991bfbe364833952e623b47e2669a : FAILURE in 19m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/71578809fdf044e0adb5d654af0687ab : FAILURE in 20m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2d466e6de26042e9b1b1bd6bb59650d4 : FAILURE in 21m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/798741c7f4d84b05868bccb1f50750a5 : FAILURE in 16m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/17f70e9011da4ff4aba007b46904eb89 : FAILURE in 15m 20s","accounts_in_message":[],"_revision_number":3},{"id":"70f9ae8ace48648349aec0e7d2de198fb7c290c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 14:13:22.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3295070ec8964b0687e30adce9be96e1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/693ce41ab0384129aa1cf72e3ce75d45 : FAILURE in 1h 01m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2f110865d6054f5ba57ed98a4bd5a069 : FAILURE in 13m 50s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eefa7ff4aca84286a9b4086815e26dd3 : FAILURE in 1h 00m 35s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bc052e3551b46cabf3ae7517da091b9 : SUCCESS in 2m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e556df7631e24558baf16a860d5dfbe6 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/43741a5ec27d4c2ab4fcaf4c6db74c7e : SUCCESS in 2m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4c46b7c3901a4e9d8df3dde6d334a4a6 : FAILURE in 56m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/06998790c3fa41cc98497e0b382b3a23 : FAILURE in 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"78f8238e31fd98418eac799d40924f5537ba637a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-05-25 14:39:39.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority-1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"919561251d1c2a66b800ffd518ae79c034b8b785","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-25 14:43:33.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a92cf43aaf1f2280f985809bd8bf462a6cfb8ba8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 16:00:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2c8a5506c624a54afb23aa747a135fb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b08b3f3225424a819f50dedb71db29a1 : SUCCESS in 20m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/251ad3e69a234b98b9f522c7ac4c8182 : SUCCESS in 1h 18m 34s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8d77ba0f246149a39ea6daeed3f85385 : SUCCESS in 1h 01m 22s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ba0232495dd45adb2d6d0f7681d02bb : SUCCESS in 2m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/799063931a80419984353444f49328d5 : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d1e3301e9e9443f90098f3623ef76db : SUCCESS in 2m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/69aaf295cbe24416bedd2317c6a60119 : SUCCESS in 15m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ae02d5ad70b343f99d459f545eeacc8f : SUCCESS in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7eab68d8b7d1e0eb162165fa2023a129f9d48d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 16:04:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f29ea44b408b4d738a9c7bd7182fd7b3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5b56b6b0100d43bfafa8efffdd723f8b : SUCCESS in 18m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/81ce2e8bdaa8413a933d46084097feb1 : SUCCESS in 1h 02m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d96b3b080dcc47dc9ca506a91f59286c : SUCCESS in 18m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1ab8ae6bd3ad49eab7fde05c096b0e73 : SUCCESS in 48m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e7a64cd8eed7410f9295271e88809814 : SUCCESS in 1h 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/075860dfef554709958be26673727ee3 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ad7d8ac40114bc6b91ebc9e8ea5f3e3 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c1ad37b2ae04b1ab5e037b5c8c326d8 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebf2110f59674a37bdffe6666c812aad : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/577bd2cb46fc4d04851b22ee17ca5ebf : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37464132e8fb4ca0932bae642a055233 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60df2426f2d142be8e474d61e400ba60 : SUCCESS in 2m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cb92dea6d5d545c493a9628033cd6f1d : SUCCESS in 16m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6ec330b1f9164c39b73dcaede03334e6 : SUCCESS in 1h 10m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f399648c02fa4bfc85de318e66d46fb9 : SUCCESS in 1h 09m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/915ae461ce2446e5b0d43808679d62c5 : SUCCESS in 24m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/edb2fbdcd4914e89b8d256e3e66202b1 : SUCCESS in 55m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/02ff72d51cf64825a71f43aa5a7df21c : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":4},{"id":"66e89c4ccfd2691b870a11ea265fc96c681f2c5e","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-20 14:40:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c465f06e1c3fdba5842123f2a5bd0f0d39550911","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 14:46:08.000000000","message":"Patch Set 4: -Review-Priority\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dd977501e7aff21a9730c2a0793af12ed75b24ec","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-20 14:51:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5cb1e0f2bcf3ad954da839e630c13a267df16121","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 14:57:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"70f776d54cf3de2587d6f9e0b6bf766cd75b1683","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-06-20 15:44:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"55f8db9ac3332729de233f9f5d527e7c71401254","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-20 16:57:09.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f5181b65aace58b66d1dd64a5bd4dc9430d04998","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-30 05:35:21.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ab655939ac4bf8fdd2bd7e3cb198675ecbce310","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-06-30 08:36:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cac801e3471053e34a618a06d79f015a06a9e27d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-30 09:07:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a5e9f227eb890731f145b6604e2c6c932e25656c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-30 09:13:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2c3eadd05937eededb34939adc383dc514972131","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-30 09:18:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2065588e7d4b666525235ec62c9abce82681f9bd","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-06-30 09:19:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1500a50935e22b3e7fb53b4744dcd4dc35521d92","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-30 13:00:52.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc9c307fdee4549f14494778505593dea231366a","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-07-13 01:26:39.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"abd687db97474326c0ebbce7c220817aa9913d08","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-08-01 10:50:26.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"48ef97ed0ed2e12e6ee12a892dbd809f0ac44ef1","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-08-01 10:50:31.000000000","message":"Patch Set 4: Backport-Candidate+1","accounts_in_message":[],"_revision_number":4},{"id":"33e52481dc349f38b8cfedd4727ec04e01c05c72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-08-01 11:37:59.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8198129a26cba5d1dd099b75f83ab0d16fd6cc09","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-08-01 11:38:03.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f909340b47e2edd468a216d3cef53322c7cde93d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 13:33:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/980083235d9744128fb6f5f15c3f5116\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/38ce6908b5924d7f82d95e420006fe07 : SUCCESS in 15m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8a9d719576404f7990eb03e8745b184b : SUCCESS in 58m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9b57bfe72f7a433bbfb61d9685bfec73 : SUCCESS in 16m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d67e5a1d4c8146da8ea47d59a2ceca10 : SUCCESS in 1h 02m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b729d78936cf4b678879523dff38c77b : SUCCESS in 1h 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6176a4dc20a49b7a3622c0cdea01f5a : SUCCESS in 2m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d14e87e410444daa2a8ae5220089d74 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f898b88e5e449f9820dd8517ce4a0c8 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c3741e7966b4213a5b588b751700f64 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebcc4a41b7af43fb9ddc432eb826d4a4 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af4d592dba94450913b719e22899c91 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63f57bd717a2476ebd08491ddd5d7414 : SUCCESS in 3m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bef9c77c81ee41e5b4cce11d8ec79e9d : SUCCESS in 13m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/35a30c70aa3b4578882cd308334bf0fb : SUCCESS in 52m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ae164d2116154ef6a5d97c467ddbc580 : SUCCESS in 1h 53m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/57e6aef498d0492aa69ab2e894a9b7e2 : SUCCESS in 25m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/de88e73cfbe54e6c8fd1c4e639c3e6cd : SUCCESS in 59m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7808d277b4824661a4b257a8ed6955f5 : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":5},{"id":"10fcb90113856a9978b30658096b7b4224fde05a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 13:37:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6220d6f3f5dc4fe492c63597b4950533\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3aea878944624c85a04e4d86d642f2ce : SUCCESS in 58m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d688dae9810b451ca912ad57f28bfb5e : SUCCESS in 19m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ca48a96d1f8445dab34194773bc0ea08 : SUCCESS in 1h 57m 41s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3235805c46e94015bd04a46612f80fcc : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3fc163fa7b1946639fcc0a856f766099 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1139d41c4c8b4de2a8581d072f909a5e : SUCCESS in 2m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/860b089068c24f30a51d7a7b60e46f05 : SUCCESS in 48m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/be5bfa9fd0f642d39d2833195e307c3b : SUCCESS in 19m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"da9999fc02375622a829f97bac8ee6ff9441ea14","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-01 18:58:36.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d05a6db0b2f1772171b226b11dae01ceb217b980","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-08-01 22:39:14.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cc9d967d8e55d85994ea58e185fc4826f8b2521b","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-07 07:36:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c9589fe863d495570704bc6193cfd7bc2184bf60","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-28 13:40:28.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8be2e8b0894ffeef84309bc7bf136cd68dd1444e","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-08-30 05:36:48.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d92230e985e320fc4e922a1fdbc499f693dfacb0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-06 05:44:14.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"845ec9c672937ce7be8c4d3d4ad741df751454a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 05:44:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fbeb51cfef2d0da15610378154a5eb5167acd4b1","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-09-06 05:59:24.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ecd409c8e5cd229351c2ad96a11139f078ba7417","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 07:21:44.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f87463108bfd4ea78f24cee64b6e1f1c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/856079acdd974842965ddbba14485cdc : SUCCESS in 20m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/961e0287c2354045ae8307f25dcec35f : SUCCESS in 47m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4a953d6cdea24215a903ca8cd4dc9190 : SUCCESS in 1h 06m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a04a6db8f324546a3fd2c8a0ba567ba : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa401b4cb6584e28b6898e568da61d7a : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc335ff6a58d45d5ba94c50981452afd : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fa17580c1b643a3b26695aaaae28a37 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52450d857322416db5e096dea0aec865 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b12cb67f52e04df9a54031479775d7d2 : SUCCESS in 3m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/12e25bbbb44144ae90809d796647f9c8 : SUCCESS in 19m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ef976f8a20ef46898c87566b875add48 : SUCCESS in 1h 12m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7cd155fc83a44ee6a6d0de5456f09885 : SUCCESS in 1h 31m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fdc83ca31f2b44ba863cf2c636a95179 : SUCCESS in 18m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c1c77d5ec2bb4fa2a61e68b350014d48 : SUCCESS in 1h 09m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2aa3823bc90e494e96702d65476c00a2 : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":5},{"id":"bba0d0c6571e42292e823ab44487129e7218f8ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 07:21:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"f98a491474eaf55bd1860c82e0d791e3eb55beb8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 07:22:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/092c6133e5bb4580909a011f94713a86\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce9bd8e306349aeb8c0f84f83183813 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01de573cc8324e22b798ef872d83683d : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7f38e1df71eafb250b14c4e125c7313bd9218f7d","revisions":{"d3bc88d50bb7cae873720be91c8f80083f0d4bb8":{"kind":"REWORK","_number":1,"created":"2023-05-25 11:55:42.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/68/884368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/884368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/884368/1"}}},"commit":{"parents":[{"commit":"6786e4384c1a19e4658310e109b8ab9704f5eff9","subject":"opensearch-dashboards: Fix permissions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6786e4384c1a19e4658310e109b8ab9704f5eff9"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2023-05-25 11:55:22.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2023-05-25 11:55:22.000000000","tz":120},"subject":"Improve kolla-toolbox offline support.   * Use structure that allow for custom package version and base URL.","message":"Improve kolla-toolbox offline support.\n  * Use structure that allow for custom package version and base URL.\n\nChange-Id: I169e51c0f5c691518ada1837990b5bdd2a3d1481\nCloses-bug: #2020761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3bc88d50bb7cae873720be91c8f80083f0d4bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3bc88d50bb7cae873720be91c8f80083f0d4bb8"}]},"branch":"refs/heads/master"},"ddca9b8230c3d2392c9b5d9630245831d9ae77ac":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-25 12:06:30.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/68/884368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/884368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/884368/2"}}},"commit":{"parents":[{"commit":"6786e4384c1a19e4658310e109b8ab9704f5eff9","subject":"opensearch-dashboards: Fix permissions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6786e4384c1a19e4658310e109b8ab9704f5eff9"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2023-05-25 11:55:22.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2023-05-25 12:06:30.000000000","tz":0},"subject":"Improve kolla-toolbox offline support.","message":"Improve kolla-toolbox offline support.\n\n  * Use structure that allow for custom package version and base URL.\n\nChange-Id: I169e51c0f5c691518ada1837990b5bdd2a3d1481\nCloses-bug: #2020761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddca9b8230c3d2392c9b5d9630245831d9ae77ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddca9b8230c3d2392c9b5d9630245831d9ae77ac"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9c617931d095cb61571665e5cd901642ebbd4073":{"kind":"REWORK","_number":3,"created":"2023-05-25 13:09:20.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/68/884368/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/884368/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/884368/3"}}},"commit":{"parents":[{"commit":"6786e4384c1a19e4658310e109b8ab9704f5eff9","subject":"opensearch-dashboards: Fix permissions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6786e4384c1a19e4658310e109b8ab9704f5eff9"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2023-05-25 13:09:08.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2023-05-25 13:09:08.000000000","tz":120},"subject":"Improve kolla-toolbox offline support.","message":"Improve kolla-toolbox offline support.\n\n  * Use structure that allow for custom package version and base URL.\n\nChange-Id: I169e51c0f5c691518ada1837990b5bdd2a3d1481\nCloses-bug: #2020761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c617931d095cb61571665e5cd901642ebbd4073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c617931d095cb61571665e5cd901642ebbd4073"}]},"branch":"refs/heads/master"},"ec58ec1b946c65efc76caafa8380740c7373083c":{"kind":"REWORK","_number":4,"created":"2023-05-25 14:39:39.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/68/884368/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/884368/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/884368/4"}}},"commit":{"parents":[{"commit":"6786e4384c1a19e4658310e109b8ab9704f5eff9","subject":"opensearch-dashboards: Fix permissions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6786e4384c1a19e4658310e109b8ab9704f5eff9"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2023-05-25 14:39:25.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2023-05-25 14:39:25.000000000","tz":120},"subject":"Improve kolla-toolbox offline support.","message":"Improve kolla-toolbox offline support.\n\n  * Use structure that allow for custom package version and base URL.\n\nChange-Id: I169e51c0f5c691518ada1837990b5bdd2a3d1481\nCloses-bug: #2020761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec58ec1b946c65efc76caafa8380740c7373083c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec58ec1b946c65efc76caafa8380740c7373083c"}]},"branch":"refs/heads/master"},"7f38e1df71eafb250b14c4e125c7313bd9218f7d":{"kind":"REWORK","_number":5,"created":"2023-08-01 11:37:59.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/68/884368/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/884368/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/884368/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/884368/5"}}},"commit":{"parents":[{"commit":"6786e4384c1a19e4658310e109b8ab9704f5eff9","subject":"opensearch-dashboards: Fix permissions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6786e4384c1a19e4658310e109b8ab9704f5eff9"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2023-05-25 14:39:25.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2023-08-01 11:37:59.000000000","tz":0},"subject":"Improve kolla-toolbox offline support.","message":"Improve kolla-toolbox offline support.\n\n  * Use structure that allow for custom package version and base URL.\n\nChange-Id: I169e51c0f5c691518ada1837990b5bdd2a3d1481\nCloses-bug: #2020761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f38e1df71eafb250b14c4e125c7313bd9218f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f38e1df71eafb250b14c4e125c7313bd9218f7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
