)]}'
{"id":"openstack%2Fkolla~884375","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I2a5a0d761223b4e99c8df59f80261108380d5038","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-17 08:58:36.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-08-17 08:58:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a5a0d761223b4e99c8df59f80261108380d5038","subject":"opensearch-dashboards: Fix permissions","status":"MERGED","created":"2023-05-25 15:46:34.000000000","updated":"2023-08-17 08:59:30.000000000","submitted":"2023-08-17 08:58:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884375","cherry_pick_of_change":883941,"cherry_pick_of_patch_set":3,"meta_rev_id":"9bf048958b2167b7de24f216de8cd52e7dab1c7c","_number":884375,"virtual_id_number":884375,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-17 08:58:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-16 12:54:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-16 11:32:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-17 07:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-17 07:10:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-25 15:46:34.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-05-25 17:15:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-25 17:19:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-29 09:40:08.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-16 12:54:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-17 07:10:19.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"bd1b7965ad3773ad116af6f0e91ada1994becff1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-25 15:46:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* kolla/common/users.py","accounts_in_message":[],"_revision_number":1},{"id":"40d22d65f973970e2e9abd59f383f38930dc4947","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-25 15:47:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"45167a099a2312416c4ca9b2bff4a5e514c0f9e1","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-25 15:47:59.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"790df1bb1a5d2c6f6e054fc5f25f0a96849ba69d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 17:15:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9064fc4ab68c4c78a68570d298100c38\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e36ffc7d584248d88130e5f451403849 : FAILURE in 44m 22s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7a01b67510f04663beb6efe1906260cc : FAILURE in 42m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/892c4dd92f5442e8bc2724145ba7b3b3 : SUCCESS in 41m 28s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/77613cd1321f44cb8fdba00c4882df47 : SUCCESS in 1h 26m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1601332db6e7477aa371d0ba1215d3a5 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a08d804e8e4343adb695ec7c69fceaac : SUCCESS in 4m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1bd8f006d0504043b2451beb59e95d9d : SUCCESS in 39m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1c5a774a424fa07e2788ccc8d429ed31fa03ecce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 17:19:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6479eee49c5d409c978638b4468c0699\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d4dc177e59d14b57a062f54a6610ac15 : FAILURE in 29m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f1ebd8681af34d229173f0c297567141 : FAILURE in 36m 54s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/a516377251cf4e4aab0168e08c0fc6c4 : FAILURE in 35m 03s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/adb3159b4ee240d692072a50642afb18 : SUCCESS in 1h 23m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4c55f985282d45f19f7bd26edf285458 : SUCCESS in 1h 15m 42s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/2440db5626204dd7b853acf2b43dd35d : SUCCESS in 1h 01m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3a9ba8506f7640c48661b41de38eb4e4 : SUCCESS in 1h 20m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ff505feb81484cc29e0fd768c9c7a4c7 : SUCCESS in 26m 36s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cb3c4fcc67754d868b3a025d00cfb8cc : SUCCESS in 23m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f77fb1644df240da840b0313f20b0b8a : SUCCESS in 1h 00m 03s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ce244dc8d54a4ebfa00043b2f3767400 : SUCCESS in 1h 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccaf4ae6b76b487eb16c9c54098f3b2a : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e806d55112e4017a9e0304b7fd82898 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47cf07ba683441a583076063948bc2e7 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de479973e0bf4d95aa447377f86ac86c : SUCCESS in 6m 19s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/ad6c98e15b2948e69d6f53e2c2efc892 : SUCCESS in 28m 05s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/70e1d2b3bd59473283ceceb460837251 : SUCCESS in 57m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/65477a9c0e1f40e4802ea985ed35ff5d : SUCCESS in 22m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d68ed35ad7c42f2bec243017be276c8 : SUCCESS in 34m 12s","accounts_in_message":[],"_revision_number":2},{"id":"59774db914b5ff6a60316c9666ab7bc772e9cc6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 08:52:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8191d9f9c702cb458e5c42e4489e03a14113d78e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 09:27:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/118695fac20340bdaae7933a87c73763\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f6fae3a907b49aa9f2a23c32cb4623c : FAILURE in 32m 45s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bd7e58b1d4c6497398b9bc79a9d341c0 : NODE_FAILURE Node request 300-0021334662 failed in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c94e5f99499b48439e337ec64854daf3 : SUCCESS in 29m 56s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8053450ef8be41c6ae5de29b412e07a4 : NODE_FAILURE Node request 300-0021334664 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82f83cdd484e4fee8d1fd602032d6d0e : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b15cb2c8a6a84e6183f59f289736d91d : SUCCESS in 5m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e49c016d722e4a8cb7f4341cd0c6dcc7 : SUCCESS in 26m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5929abdea982d8eb3f2491952a6f19fecf5dca4a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-29 09:40:08.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e690ff512e6cc63273059063edd77999720275ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 10:26:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d73da431dfa40248c8034edb2d907cd\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/829dd80495974d26bb101f8c6d137fac : FAILURE in 35m 14s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a40e514fa3dc4f359695fa4caf90b5d7 : FAILURE in 40m 35s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/22e1728e2d5f4d16b2483f4c84dc4af1 : SUCCESS in 26m 35s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b212934836ea43339a36fbc9c6862213 : SUCCESS in 1h 11m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6ff00d9d88fa46e2b67485d5e4c51ede : SUCCESS in 1h 07m 09s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/3c337bfd53084546877ef2a650c6bf95 : SUCCESS in 1h 02m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/69ee39d266124a6da224d099873499c9 : SUCCESS in 1h 28m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c3aa2bf6bcb84c77afd38898d088d90e : SUCCESS in 22m 55s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5cb79f434d354a35bfdf827559b2aa55 : SUCCESS in 15m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/92d770fb5bfe43f0a7849fd6f31b259b : SUCCESS in 1h 00m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1844064ba5584fc28d7f7d2adbab4149 : SUCCESS in 1h 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10964cf5f92148c984d0ce5956b17566 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/885d0a65fc7345019efadf640befeadf : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16ebc9b196614230a00b30637e6fe788 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd10e68ab4146bc80e62e1302d79c6b : SUCCESS in 6m 10s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/0e8e614832954866b1bee259f8bceba8 : SUCCESS in 19m 54s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/9fbb82d7dd7b4a05ab805e68b90ac3b9 : SUCCESS in 1h 00m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/549157d12f3741bfa3b20b571ca95b44 : SUCCESS in 18m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2e9dd7cbd6540c18db67c70cf816358 : SUCCESS in 22m 36s","accounts_in_message":[],"_revision_number":3},{"id":"024e858a532425438c49d08912f63c9f4d00b28d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 11:12:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b52d15c11aabc19b4987554304bc03b88a6c5202","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 12:42:10.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/883716 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"1c2097f8c9ee4c955ca98d36afcd65e2e77ccf72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 12:42:42.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 884487,10","accounts_in_message":[],"_revision_number":4},{"id":"9ac33977b6bdccf19f63e70061bb82aeaaf81530","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 08:18:29.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"bc8b0351240d3207c8da3b807c90ed44bb6e1a84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 09:21:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ac893bbdfa436cbaca3ac1275ee7e3\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6997599d82c54e22bc90364381cd908c : NODE_FAILURE Node request 300-0021542077 failed in 0s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a4e7334f4f004cdab1cd22b31763f8cc : NODE_FAILURE Node request 300-0021542078 failed in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2d535e6b531042d9a23e451b852271dc : SUCCESS in 1h 00m 06s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/38532d5e274342838893f8c5dddb6386 : NODE_FAILURE Node request 300-0021542080 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11092a0dcf9b4d7ab5d805001dfefb1f : NODE_FAILURE Node request 300-0021542081 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff0265ebad8f4dc8ab681f42e2671503 : SUCCESS in 5m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/38f668eb59aa45f2913d4f2658918273 : SUCCESS in 34m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9e16e19c63a170282324e57f0619d4a783997162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 10:52:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02723bb9766d4c0d90c0cc7439390ddb\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a0a2c32c70004134800c1510828f943a : FAILURE in 39m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1367e57a646746ef9cf39a7faa585eb4 : FAILURE in 32m 50s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/0b5d13e1ce074bc3936bf6194bf4fadf : SUCCESS in 24m 51s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1db29d2f59eb4cf4908d08bb0c7780e0 : TIMED_OUT in 2h 01m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/09df6097fb3549b0bf38b3e044e9415d : FAILURE in 34m 48s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/4e9d772e6b464947925afb3b82caf0a6 : FAILURE in 1h 20m 30s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d7574108aa794975b9f9039e40a54156 : SUCCESS in 2h 23m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d4ad122f424541c2b217e2b28921a858 : SUCCESS in 21m 04s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/956bac245029477eb33b015aaafa067d : SUCCESS in 20m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d2e06f9ae75c446abfe5ff6972810dcf : FAILURE in 29m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b2d87439021644e0bad3f0f8e3054b46 : SUCCESS in 1h 14m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0a29ad1a82a4474aec0fd6565a6dd1c : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/effd4ca2736649a592ee598c6924a100 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/016aa3a51c2842a880490ac77fe75082 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4876d5d77b0c40508a0a7606fb7a0f03 : SUCCESS in 6m 05s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/69acafda2e7d49b1b4a46e78e20aca05 : FAILURE in 14m 23s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/92e845d634834a2093a875ccac94e4eb : FAILURE in 15m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dcf7ba1a9a4a43818738e51db9f10cfb : FAILURE in 28m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd7805d12bc74bb5b9a74276bf77a336 : FAILURE in 17m 45s","accounts_in_message":[],"_revision_number":5},{"id":"69dfb473fc87e9a6d24c4b123e708e99445d17dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-07-31 13:37:55.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c1f3bd0c6a057ee2ad9257fbe3678252dcb832ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 15:15:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a58be23aedaa476cae503159ee8bbe85\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/94549b557c4d4614b91b3751f3016675 : FAILURE in 41m 56s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7384e4d2746f4e09965a38f3be230cfd : FAILURE in 37m 17s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/24a723f68fae4e9192190248768054ee : SUCCESS in 32m 46s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/98762e5a169640d49628584c7a2fa4cf : SUCCESS in 1h 12m 57s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cf06e1832f1c4bc4a6640ebb72ec744b : SUCCESS in 1h 11m 29s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/7c4ba8972b5d4251a2fe3af660f0725d : SUCCESS in 1h 17m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4db02a1a766542a8b4f7688c4f647195 : SUCCESS in 1h 16m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cc170966f5a04646810b70048f8330bf : SUCCESS in 21m 03s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1705314553224d27a5b25b68812e56e6 : SUCCESS in 19m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bd2aa98b274b4966869e158d5aef8c0d : SUCCESS in 1h 10m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e0736d3333d34db281d645f89f887762 : SUCCESS in 1h 00m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fbe557e1ebd4b3ba9cbcdf2b6781f6f : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/feed3f59b97c4ad6b8a952524b0198d4 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be823d1593d04789a04ea02131525b50 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980334cea0cc418a8ce3cccf3ed3f622 : SUCCESS in 7m 02s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/2b511de0929248e8a6e2ddc5c92a5932 : SUCCESS in 19m 22s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ee2d863450d34659bfcf385324d03076 : SUCCESS in 56m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/804b9bd240184bb382d8547745a2cf49 : SUCCESS in 18m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2aa80b4e66f54f6c80eeb635fdf2edf6 : SUCCESS in 25m 34s","accounts_in_message":[],"_revision_number":6},{"id":"e0945f720085160a49d6a594c4e669eb96061c87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 15:46:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3319484baf74093be0517998186fae3\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c2abdde182446e3af9bcc7fba86b691 : FAILURE in 2h 05m 33s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7994d55550a146d2a8b68dd7e589b60c : FAILURE in 33m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/10303834fb194d6c9077b0deae35f399 : SUCCESS in 1h 43m 42s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ba76444e7cc45d38be3d827fa03df6f : SUCCESS in 1h 18m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7c737c624ae46569bcc870a182fa74f : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92d0804e983c4db18d962ff974d61fcb : SUCCESS in 6m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/86809f92d33d4374ac0d09154c4a0344 : FAILURE in 1h 33m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"63036b47df74870121b2e2e62862dc7dfe0897f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-16 11:32:55.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"c57c45d24c3471b2955769aee1d63b4ee598be45","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-16 12:54:24.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"f977ae1d520fc1843fcc4bafab0046808daa17ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 13:05:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3cf1c75f35945eba608421e87de7a18\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/deaebee519cb4cf9b5f456be1214ba29 : SUCCESS in 33m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/911ba0a01e2e498797367416723d3aaf : SUCCESS in 40m 30s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/bc993e088dc044e6bbcdae409ef4349a : SUCCESS in 21m 03s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d6b91f96e6704cac8aa1955c0c6a11f7 : SUCCESS in 1h 20m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f1e2068579e047aeb4597e1368dc53b4 : SUCCESS in 1h 08m 53s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/e6bf0d2130974a34adad4df1e266401d : SUCCESS in 59m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/eaae0d9b571449bb9cd6754d394272d9 : SUCCESS in 1h 25m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9d21e6ab02ec4bb68a3e78238a2cece0 : SUCCESS in 19m 45s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9d92ebd855384cdb9509eee0e34f893d : SUCCESS in 17m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/74ee77188eff4a72b1282b10103fa232 : SUCCESS in 1h 06m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c38a2ed1b1c4467cb73dd6ee035dd6ec : SUCCESS in 1h 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec7b8087429a4d2294ecbc326f5500a3 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/961ac10ea78d457a9e03ff29da2f355d : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba2bc0892a094f98957214b9dafce342 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ae810b0d2a46debaa281b1e5bfcabb : SUCCESS in 4m 54s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/ba61ba3db9bd4fbeb45bad1d9185ca7e : SUCCESS in 20m 55s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/32aaf86588aa410cb25077a90454700a : SUCCESS in 55m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/988e7350cb7b47b4a29058ff3125c0c3 : SUCCESS in 30m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c30e28c9cf9481e986c27cd667462c2 : SUCCESS in 36m 50s","accounts_in_message":[],"_revision_number":7},{"id":"f54ece3faf1d713f020bf3d6d7b759d5cee41357","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 13:37:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7a79d9c96454904ba0a003489aa54aa\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4fe46fdc757f47f588ace0f61a4e648e : SUCCESS in 2h 01m 50s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/52a18877c08540109d967c1f28b0b191 : SUCCESS in 21m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/04e83e6b4f194c4ab9b645df774f439f : SUCCESS in 1h 28m 12s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c714383153fc4f789d5f5ae1a1782a16 : FAILURE in 1h 06m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36f72351edad41cd820a82c228496c8e : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2dc86e685a0844d289a469d58f6df339 : SUCCESS in 4m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/73a0020fb9754b5c81bde3887a42834b : FAILURE in 1h 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"16f6b56f435f2815514607a864931392141641dc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-17 07:10:19.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b3761cd999eff142d88d424f856568d2bb4e7edc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 07:10:49.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"8988216790b40dca5ee09e2eaf5b32b845429b53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 08:58:36.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c97ef9efc6d44e11878d74d7969238cc\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ac70f6b120344d5d845ccfdbf64ddeb7 : SUCCESS in 37m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2a447a4e7ea740e8b72cdca06504cc62 : SUCCESS in 1h 31m 26s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/812779f5361d48d58b982cca247c91ee : SUCCESS in 1h 28m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/32778148b19e478da665bf684c1daa11 : SUCCESS in 21m 06s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1126d01893cc43d3bdcfd28165fac732 : SUCCESS in 1h 02m 54s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b1369699ea72418b99aae5257b35d5e2 : SUCCESS in 1h 08m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f221d96844d416ca3d5d86cfd8b97d8 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9266a57aafd4deda99069cd0b9e3aed : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee67e7e6db224009a0254dc8d4c8d0b2 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5caa2fbe5eb947c5ad1f618d664b6ff5 : SUCCESS in 4m 58s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8ba98b01c1d643798391c32e94033dba : SUCCESS in 1h 10m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7907dad046049d884dba98375996a8b : SUCCESS in 17m 24s","accounts_in_message":[],"_revision_number":7},{"id":"9afc9c00a0a8f52299fe14bde7decd67ef7c8b47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 08:58:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"9bf048958b2167b7de24f216de8cd52e7dab1c7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 08:59:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4288b1ab0f654c5dafd65aa91fd65a2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33f1e73b1fb474dae5e722b4ecad2e2 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7a30d4d7689299ed1f0007326ba782b5047c1188","revisions":{"3ae297492d0cc38576eb2214f3d67c468f89972f":{"kind":"REWORK","_number":1,"created":"2023-05-25 15:46:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/884375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/884375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/884375/1"}}},"commit":{"parents":[{"commit":"033a82325307725fbabbee54493fb536153e0a64","subject":"opensearch: move to yum/apt repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/033a82325307725fbabbee54493fb536153e0a64"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-25 15:46:34.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n(cherry picked from commit 6786e4384c1a19e4658310e109b8ab9704f5eff9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ae297492d0cc38576eb2214f3d67c468f89972f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ae297492d0cc38576eb2214f3d67c468f89972f"}]},"branch":"refs/heads/stable/yoga"},"fb967b1e055c0264366ef0198643307ed371a014":{"kind":"REWORK","_number":2,"created":"2023-05-25 15:47:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/884375/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/884375/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/884375/2"}}},"commit":{"parents":[{"commit":"033a82325307725fbabbee54493fb536153e0a64","subject":"opensearch: move to yum/apt repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/033a82325307725fbabbee54493fb536153e0a64"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-25 15:47:51.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n(cherry picked from commit 6786e4384c1a19e4658310e109b8ab9704f5eff9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb967b1e055c0264366ef0198643307ed371a014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb967b1e055c0264366ef0198643307ed371a014"}]},"branch":"refs/heads/stable/yoga"},"3225160eb3eeb3b88912450d6a4d59171826e8bd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-05-29 08:52:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/884375/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/884375/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/884375/3"}}},"commit":{"parents":[{"commit":"3241c6dab7df9dd4e80c2ffc2a887d79c2c9ec73","subject":"opensearch: move to yum/apt repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3241c6dab7df9dd4e80c2ffc2a887d79c2c9ec73"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-29 08:52:40.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n(cherry picked from commit 6786e4384c1a19e4658310e109b8ab9704f5eff9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3225160eb3eeb3b88912450d6a4d59171826e8bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3225160eb3eeb3b88912450d6a4d59171826e8bd"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"832b4aa9b34f90bcb39595dd484b9005740a6315":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-05-29 11:12:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/884375/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/884375/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/884375/4"}}},"commit":{"parents":[{"commit":"54dce7d767b917c0e5a459f7c8451e4517d0c8d1","subject":"opensearch: move to yum/apt repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54dce7d767b917c0e5a459f7c8451e4517d0c8d1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-29 11:12:42.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n(cherry picked from commit 6786e4384c1a19e4658310e109b8ab9704f5eff9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/832b4aa9b34f90bcb39595dd484b9005740a6315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/832b4aa9b34f90bcb39595dd484b9005740a6315"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"2dadbb756a080c0d95e06e5c4c48bdb6f7464942":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-06-26 08:18:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/884375/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/884375/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/884375/5"}}},"commit":{"parents":[{"commit":"6899de00b264e3e515b640d8069872e89ef3231f","subject":"opensearch: move to yum/apt repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6899de00b264e3e515b640d8069872e89ef3231f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-26 08:18:29.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n(cherry picked from commit 6786e4384c1a19e4658310e109b8ab9704f5eff9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2dadbb756a080c0d95e06e5c4c48bdb6f7464942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2dadbb756a080c0d95e06e5c4c48bdb6f7464942"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"004a5594e5ff6c0ed47d2b3c5a8a9b956c4ef258":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-07-31 13:37:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/884375/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/884375/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/884375/6"}}},"commit":{"parents":[{"commit":"eca7e99e781092856a3e104ee1c2f53ebe45b8e9","subject":"opensearch: move to yum/apt repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eca7e99e781092856a3e104ee1c2f53ebe45b8e9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-07-31 13:37:55.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n(cherry picked from commit 6786e4384c1a19e4658310e109b8ab9704f5eff9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/004a5594e5ff6c0ed47d2b3c5a8a9b956c4ef258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/004a5594e5ff6c0ed47d2b3c5a8a9b956c4ef258"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"7a30d4d7689299ed1f0007326ba782b5047c1188":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-08-16 11:32:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/884375/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/884375/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/884375/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/884375/7"}}},"commit":{"parents":[{"commit":"23e0819d688362ce0b6c564d469f57ea9cbe2731","subject":"opensearch: move to yum/apt repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23e0819d688362ce0b6c564d469f57ea9cbe2731"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-16 11:32:55.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n(cherry picked from commit 6786e4384c1a19e4658310e109b8ab9704f5eff9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a30d4d7689299ed1f0007326ba782b5047c1188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a30d4d7689299ed1f0007326ba782b5047c1188"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
