)]}'
{"id":"openstack%2Fkolla~884614","triplet_id":"openstack%2Fkolla~master~I18d3ca8bd83a72dba15fc7e543270e260ae4dab0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-29 15:17:09.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-29 13:42:18.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I18d3ca8bd83a72dba15fc7e543270e260ae4dab0","subject":"rocky/centos: mark bifrost buildable","status":"MERGED","created":"2023-05-29 12:41:08.000000000","updated":"2023-08-18 12:41:23.000000000","submitted":"2023-05-29 15:17:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884614","meta_rev_id":"c84a878486191d8347ac2e638fba1b34a59397bf","_number":884614,"virtual_id_number":884614,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-29 15:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-29 13:36:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-29 13:42:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-05-29 13:33:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-05-29 13:36:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-29 13:13:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-29 13:33:13.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-05-29 13:57:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-29 14:00:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76fd7730443f7754f24a5fcf3d606ca30bb757f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 12:41:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"581eb6a15ad44c47e3bbe710d5c372da29b0425d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-29 13:33:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2defca3657922eb5a79b56e1fec20989132ef8d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 13:36:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5380a8579cb989283d26168accf6a61b4e742a7f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-29 13:42:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d5211d3b155d0f9f8caf947d7d375e627eab41c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 13:57:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f953dfe7a04418f8de61111f00b8cbb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/438dc2ff336c40058ead845d666f7c51 : SUCCESS in 29m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1275cda4b5b24c619ca7321a41920b3c : SUCCESS in 26m 38s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/68d358b1420f40a2a84b311ed96bebe8 : SUCCESS in 1h 14m 41s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/561abc2e849f4db8a6b798b0e83c36bb : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a44d92bf1a854796a23439e4c89a56ad : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/75d47db7ab734641943a137354ebb9ee : SUCCESS in 3m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f2b258f43e1e40eba4382247981dc2d7 : SUCCESS in 26m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f1dc19f1c26f43469ef7997153dee893 : SUCCESS in 23m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64acc80cdf44393a9918a2e06ba45c0e8108d205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 14:00:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/676b165a15db4f50b8f9a9a0a8eec69b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/94b8fd15731748f1a68e4e3e2bb2c352 : SUCCESS in 26m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bc6853b513734093a5b7e226afe65559 : SUCCESS in 1h 07m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a72bcba2882849dc9611bff51f672ad2 : SUCCESS in 18m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2339c83f59e347f189521080530dd8f8 : SUCCESS in 53m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c9978e979c224314aa45398dc0f6d6bd : SUCCESS in 1h 01m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907e1451fe45494bb79019bbd0d3b2ab : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bd8515ec50843ffbb4f5f19cb80b715 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48c47cfb7b4248ddb5493d8b9e72fabd : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34b69dcfaa324981bd078ca4546e228e : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ae0a065a9f24659bd6dda4b2b03d06d : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/444b3b87d71d41acbcac18ecbb99abae : SUCCESS in 4m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9f27f7733a794562bfa91cfda8d58548 : SUCCESS in 22m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5d148e026e6d4bc498d8bd4123dd9ebb : SUCCESS in 57m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/71125b8e9b084d6d981b092a13330c19 : SUCCESS in 1h 03m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/44ff3591fdd941888e2f4333e68ae2f5 : SUCCESS in 25m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3b9abc518036447091b4c6ae0029e8f4 : SUCCESS in 58m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/99b3c3bb6ee347ba811b5b271b36ebd9 : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"da20b054164386d60a40599d6c6b7db67b56de23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 14:01:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac294e2c18d936b9a244f2a373bc0f0dcbad4912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 15:17:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfb7d57eed3c4ce79407c53d5659234d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/46784f1f77f64a12b91d90ca08102855 : SUCCESS in 16m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9f7ac150b233487ab0203e905aa7d744 : SUCCESS in 59m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/82a24506a9314fc294e461301842b8e9 : SUCCESS in 1h 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5805e9e9b14a4be2b907c2d64f216a7b : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fcd5734278c45b4a2f420a86ff4e71a : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba0119603dcf4d7aad65a3fae9228ed3 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ccaa2ecc39e4999bc36ca1e64147932 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83509cb389d149bc89fce7e8ac044240 : SUCCESS in 4m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/806c2181b7e948fdac80a9900f3dcf64 : SUCCESS in 24m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ed0629fc21c348af8c6bf7e57bc32b7b : SUCCESS in 54m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4995c8a03be4471cb8cba19c60689c5d : SUCCESS in 1h 11m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b21a0db57f4a4f74a8989400584d8e51 : SUCCESS in 20m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6277a67b09cd45b683aed4313fce6067 : SUCCESS in 1h 06m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2cdf6607ee8e4da587383664288937a5 : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"dbcbd35a6bf5752f64ef4cbcf94fb40d1322213a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 15:17:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3439369d3911bf362733863e990dad355a8a8fb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 15:18:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe96d27837e04a0ebf902c82260fa463\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2299ffe798d34c85a3d09a4c790160d5 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1},{"id":"c84a878486191d8347ac2e638fba1b34a59397bf","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-18 12:41:23.000000000","message":"Created a revert of this change as I5cbaf2017b424d915049160ec576f235ef01c020","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"878e9c0524e97dda7d61c6161d33e26052d6285f","revisions":{"878e9c0524e97dda7d61c6161d33e26052d6285f":{"kind":"REWORK","_number":1,"created":"2023-05-29 12:41:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/14/884614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/884614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/884614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/884614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/884614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/884614/1"}}},"commit":{"parents":[{"commit":"9e1b28022f2838d689312db32efe1475dca5be89","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e1b28022f2838d689312db32efe1475dca5be89"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-29 12:40:51.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-29 12:41:06.000000000","tz":120},"subject":"rocky/centos: mark bifrost buildable","message":"rocky/centos: mark bifrost buildable\n\nChange-Id: I18d3ca8bd83a72dba15fc7e543270e260ae4dab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878e9c0524e97dda7d61c6161d33e26052d6285f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878e9c0524e97dda7d61c6161d33e26052d6285f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
