)]}'
{"id":"openstack%2Fkolla~884677","triplet_id":"openstack%2Fkolla~master~I3a23670f85336344d8aebaded74dd65871837b82","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-31 09:31:39.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-30 15:51:09.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-05-31 08:20:52.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2023-05-31 09:31:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a23670f85336344d8aebaded74dd65871837b82","subject":"rabbitmq: pin to 3.11.16","status":"MERGED","created":"2023-05-30 10:45:12.000000000","updated":"2023-06-21 17:20:03.000000000","submitted":"2023-05-31 09:31:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"884677","meta_rev_id":"42688eca6eecf310869d1824a481efba998700e8","_number":884677,"virtual_id_number":884677,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-31 09:31:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-05-30 15:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-30 15:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-05-30 16:19:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":2,"date":"2023-05-30 15:49:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-05-31 07:42:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":1,"date":"2023-05-30 15:49:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-05-30 15:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 11:42:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-30 12:35:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 13:20:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-30 15:01:14.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-05-30 15:47:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-05-30 15:49:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-05-30 16:19:30.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"},{"updated":"2023-05-30 17:30:24.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"}],"messages":[{"id":"cf71675c717ddea6bdfe0c6124272fdd5d3768d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-30 10:45:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f9be2ca3f40154a9e98b04dae54e42f30ff43ec","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-30 10:48:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01929ca0082582786de76a601be8daa779573a02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 11:42:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc87a67ff25e43d3905e0928e430a5d1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6d97dbb54bb74df58995d40d556a43d4 : SUCCESS in 19m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bdcefd9e874142f8b3baf4e0e17cccd8 : SUCCESS in 18m 54s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9e4056efd3b943d4829c156b7722c3c9 : FAILURE in 56m 22s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a8736fabe9a4847b14c6eedc519e37e : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b92e850e12f473aac9e0ada68417608 : SUCCESS in 2m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5d86a0cc0b448e98130e1f153d1d6b3 : SUCCESS in 2m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/66ffe03e898a43cb85c6c9df93fd7882 : SUCCESS in 19m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5907bba1014e4ffdbd1ed4295b958c8c : SUCCESS in 52m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8719fced0fca81afd9cc0ada6c918f819d22b853","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:35:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2283550a68d047fc877e1828b69c48a3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/19ae0ef8943c43b2a6b24d95d83cbd90 : SUCCESS in 29m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3c63849ba1e2439ebe22b4765a9147b0 : SUCCESS in 56m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9d75c26c12ce44bb91ba57c032bd7404 : SUCCESS in 16m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a669993b3cd34ccf81b151bcd9eeaf8d : FAILURE in 52m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a9ceded421b04794939744f60d794541 : SUCCESS in 59m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f19ea9adc42b43ca942c9b725e432012 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e5c220e81334aecbf4cab24e1acf8c3 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7722250c1b0473a9cc397bde990c7f6 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1abb6f12180e4741a6ac3d9c00450f58 : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13067945dca44de59bf237a016a49f0f : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8781d08888b345ea8d2a04697b1ca885 : SUCCESS in 5m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d2b05abf762a478a8809b09b05397286 : SUCCESS in 19m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/795c7711f5714b118f9e734723656961 : SUCCESS in 59m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/75d0ba98b41741bf842435e25790dfc6 : SUCCESS in 1h 03m 07s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ab1998216f8d4591aa65ba375957bdcf : SUCCESS in 1h 32m 24s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/691cd1e69cef4f31a38edf58f42c0167 : FAILURE in 1h 06m 38s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/dd12bda438b44ae8990bd5c2e153ab49 : SUCCESS in 1h 42m 53s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f84d7bcb7bef463baae6529b2e64bd6b : FAILURE in 48m 46s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/dc5206d1bcd14572b5e4395c62404c84 : SUCCESS in 1h 09m 08s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/f059981621e54b54ad92cdfd7030064c : SUCCESS in 25m 03s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/203b3a37e86e4b4e8b2845eda55ce5ca : SUCCESS in 58m 50s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/6fc9c19ed2bc45c4a85d6b201e297cba : SUCCESS in 1h 18m 00s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/e20b1d6e005b42c8860b26beff5b7401 : SUCCESS in 1h 06m 02s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e12e9cedf7724539b829529ce50ae1c5 : SUCCESS in 33m 36s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/25819020d8674a8a99303683778ceaa7 : SUCCESS in 28m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/030ad21a8a2240fc8f7bd5aae6fc159b : FAILURE in 43m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cfa51e7e26cc46d69b4c053f5f44c40d : SUCCESS in 1h 01m 54s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6ef0f28cc5004e31b09f50b97d04b122 : FAILURE in 1h 07m 57s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c3abd4c485744df3ae4ddbca6dcc0f00 : FAILURE in 1h 45m 42s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/c1a85e2e5ed54c1f9c36e81c7594d8d4 : FAILURE in 1h 08m 18s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/95e608d6545941709cd3460aff282b59 : FAILURE in 54m 18s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/8aefec9551ff4e1facf6fcc2a9f28a28 : FAILURE in 1h 08m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/4757fe9c7d7c4ce88e2bb7fd61892f41 : SUCCESS in 26m 38s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b51a45791f984162b740928eaccf3083 : FAILURE in 40m 28s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e1af2e3dc2da4d03aa5e2d698a380173 : FAILURE in 48m 26s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0247d3dbca6a4d79b5c7e78e56824f97 : FAILURE in 1h 00m 33s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d604feadb14a470fbd6ac8ad1abd1eec : FAILURE in 26m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb67b752a3941ae3a0e8e80a1b70d4fb11b2215d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-30 13:10:24.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e8c572f35c2dc78cf99e42d6331bc2d1f8bd9495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 14:54:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f55ad46ad2754652a41cbc3983937e78\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4915d455cc7c4cc5971f079e49af1830 : SUCCESS in 44m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4be14bbac5cc4df59c1acf24bbac1938 : SUCCESS in 1h 02m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4e5cc35e72774295881d6e6702511939 : SUCCESS in 18m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/46414f00a0974285856dcb8b3696e191 : SUCCESS in 59m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1f57fc77783f4cac8dc1001744cfb476 : SUCCESS in 1h 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ffd5f23d67641a699305e519f52bc18 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f57c82a40e5340ac97347e63bc92b8e3 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48c1a05a94674e309e5da735d00e717f : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4810e18add814af2ab7a637d09fb724b : SUCCESS in 3m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9df1362b5de340d9af597880d1fafa6f : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/296eb202e3c841c5a6514a0c1e7e834a : SUCCESS in 6m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cafd1fa5a1944736ae179c8db1bafd3d : SUCCESS in 25m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7918e0cc613046adabbccdb4cc4c2e47 : SUCCESS in 1h 01m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/abc9d98a3c494cfbb67b6c335fd736a6 : SUCCESS in 1h 14m 24s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0fca860871234e5781720ad0d6c45d38 : SUCCESS in 1h 34m 23s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e7ebc2e082e244e4b6854ceb48baaaeb : FAILURE in 48m 09s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/e5490863a4d44751a95f1d5fd227224e : SUCCESS in 1h 19m 34s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/9174ffbf7a2445cb9a9adf65d7a2ad1e : FAILURE in 47m 53s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/7e433295138c45f192924439109bf52d : SUCCESS in 1h 19m 06s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/ee43d28adac9467db0b6e30a16fa6593 : SUCCESS in 30m 46s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/cc0ea4c6f8cc467689e9a5da3aa886e8 : SUCCESS in 58m 18s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/13a0b35b4036414cb6f46b56a4e95831 : SUCCESS in 59m 20s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/36b84e63b6d84db395c5a64043c6aaff : SUCCESS in 1h 10m 28s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0f8a3ccdb6014a10bc12e7e1243c60b5 : SUCCESS in 28m 38s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/98af3ec92a394dfdab752ba10e57beed : SUCCESS in 16m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/02022b3b07484d9382c2637a3db84e69 : SUCCESS in 58m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/696b7ad7f65c44249b38ac0fd4cfb62a : SUCCESS in 58m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5bb7926801b14ae7a415565e7dc897e5 : SUCCESS in 1h 28m 20s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/98eefb62e41d42fdb89633eeefeb09da : SUCCESS in 1h 38m 33s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/6a4616046b7f4e89bfe949b6c753c85b : SUCCESS in 1h 30m 22s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/586e21ab2a32440c9352cbdb83655073 : SUCCESS in 1h 04m 50s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/4d15b723ff4b4b95a9b419c96a8dbc3a : SUCCESS in 1h 12m 12s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b48f2e33a316463984d0567078c5258f : SUCCESS in 21m 42s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ce6cc35a443140c8b8094f6334d3dced : SUCCESS in 57m 16s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8fc3a654943c4f6ea169cbae787e39ad : SUCCESS in 54m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b2fdabb7aea0422ca80c89dc737f8088 : SUCCESS in 1h 02m 03s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3b1a7a9553ce4ea9a475a8c46d3de8d5 : SUCCESS in 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d26bc80fc8b44ec6de4bfbaf1157f72341b2b4ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 14:58:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5cdfe2790e642028755fa06dfa4b019\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0077c2d8e18d4e30880aa816c956170d : SUCCESS in 55m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/de5babaaaa1a4e65b610102af2f8936c : SUCCESS in 17m 38s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b46388b3baa8421fab3c7e5f1d82e331 : SUCCESS in 1h 46m 20s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05d960e549644b36b367712ddaba6a16 : SUCCESS in 2m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5dfec5eb74264c2c99dad320ff10fa5b : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f18d06096a8433d85a88f3d2ad9f1c0 : SUCCESS in 2m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bfeb581c8673421384d417752aa45a9f : SUCCESS in 20m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/74d5bbfa50674e29903b06fe0fd3375a : SUCCESS in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5266e7369b54d2696a21216faf6fb261dc782d34","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-05-30 15:01:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"eea110cc03a3bdb7975c19b6bae1ec7977c61394","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-30 15:35:18.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"41c2d280cf367f93462821afcb66ad16e8b9b32f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-30 15:47:43.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"488dcf0272b560b33f5eafa5db497c751b002b07","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-30 15:49:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"263dded0f5132463fd13ab8a4a1c291b533c8771","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-30 15:51:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8fc561ae39e2e4169fb9691c8826b4170d285462","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-05-30 16:19:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"74103038b8e91d51d7e2a289ce9f6b7c85db61fa","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-30 17:30:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fcf458c7c4ce21931f67a8cba4ab9c8d10525d98","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-30 17:33:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"72c1e7554d867ebcfcda11d155af4abc647a4ad6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-30 17:53:44.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6bde2779c5189c01ce555dd5ff3032ba63dfabd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 18:00:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c271f7e19a5460c9dc5b9ade438b781\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dcdf3d150ab84c3680ab85949b046e74 : SUCCESS in 22m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/60748ffe0fad487db7623048a46be0da : SUCCESS in 56m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0dd27ccfd838403ba85db4cc0dcdb0b4 : SUCCESS in 19m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c3749a057ae24747bbb60d870c26e9b3 : SUCCESS in 51m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a1b976d591a1400e90ab334949db0fb9 : SUCCESS in 1h 04m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc79d3bf0e44a58b774534211f3196c : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe71de6001f54528a2454a97b1011cfe : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/faa67fb435d2479f8ff89ad868fa6996 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01529f14f3ca42c9961725436d89f5c8 : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22d67a46d61b433ea1ed34f792644e49 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab7806116fdf4ae3882c4d6c4866245d : SUCCESS in 4m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a057ceec71884a3083859c86245e91ea : SUCCESS in 24m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/89236c9d347d420194428f83c0fd8bc2 : SUCCESS in 1h 00m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cb70c0c47d0f48ed8769a640a08c78f7 : SUCCESS in 1h 07m 37s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/816365316f124181a274fb3d999c2fbe : SUCCESS in 1h 43m 13s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/11fabfc59eea41169f92ea3c5a46413b : FAILURE in 1h 05m 54s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/aab1871a29a641aa8e145f8a72930387 : SUCCESS in 1h 39m 05s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/821be21fa264498dbc7c572c566a8d90 : FAILURE in 1h 14m 23s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/de1378b1102b41a992e5ba773f3fc899 : SUCCESS in 1h 23m 40s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/1c0c399cc999474ba2164672d1d8107f : SUCCESS in 31m 09s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/1f571965bf72456ea5b75264c680df2b : SUCCESS in 1h 02m 32s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/aee3cf07ed5b4409afc86492e40ba2d4 : SUCCESS in 59m 44s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/64e64c0ed767422c94e86fea39d0603f : SUCCESS in 1h 12m 54s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0f78f855b5b64b1691dd9b7c8b98ea8f : SUCCESS in 29m 10s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b89ff1ed4737477498027c2cdf5ed50e : SUCCESS in 16m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d7b60cb4302b49b2bee8d5fc05703ea9 : SUCCESS in 1h 10m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d3271a62a33e46839aea30eb907d963b : SUCCESS in 59m 57s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/cd7eb64a776145d2beba72b5186458d2 : SUCCESS in 1h 28m 27s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6163ac72b2d24053af7469d36c4e6eba : SUCCESS in 2h 14m 48s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/4bfa0089a99f447b874e00d3ae502552 : SUCCESS in 1h 27m 44s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/54eacc8cb8214a18be8a1b897250b77f : SUCCESS in 1h 25m 20s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/b041d21e18b447b9b3d6ab6ade0e83cf : SUCCESS in 1h 12m 17s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/dadcdf0d4a8b40d09a3c09a3fd28258a : SUCCESS in 29m 57s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/68ea4e11ff204bfb9161d9d67cd23aca : SUCCESS in 53m 59s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/70ebd81191de414aa6efc30234c698d9 : SUCCESS in 1h 26m 39s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/6c25a128bf8445968869d9a19afdf9c8 : SUCCESS in 1h 35m 09s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/69388e47d0b04a3baaf59b6525cb7aa6 : SUCCESS in 43m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"14846746d43e3e933bf6e0ea6f585fcc9269bfa9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 18:12:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f97029704a144fbe97c14f337d92d7d7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/98fee71b5c3e48439aabcb0aef1aabb8 : SUCCESS in 1h 45m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8cf4de4ece1a4dab959ae9b70716aee8 : SUCCESS in 18m 49s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cce793b379744d019c769f733b145a7b : SUCCESS in 2h 34m 34s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2598f5d99d546428641f4dda3feeef2 : SUCCESS in 2m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/015888f1059b4eab9d7737908b1f9d85 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/520f97efbd6747f39c6bcc455dca64c2 : SUCCESS in 2m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9f8a182171554885b0e349fc04547a5e : SUCCESS in 1h 37m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/01ccefffb38a442a9417d47d154ef455 : SUCCESS in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dfd3d897a1eaf8732f1f9b6b8f4a3979ae6764ed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-31 07:42:22.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a34b1f44af720d7e50e95f3027d211874572a5fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 07:42:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea07f96fcb115521951a0742a2e49207594e6ff4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-31 07:43:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"74b84377a3fa48343fa33abd2789897cc7032dd5","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-31 08:20:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dc8d703ea776ddc312d27945a58b5cc13c01b038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 09:31:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47115754bf644421a256e42c6f721c78\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/78096d1f1971488baf13fe5e14bd34d6 : SUCCESS in 17m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aa294d0b6f074b91a95e48f14951d3bd : SUCCESS in 1h 02m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e22c731e9776429a80fff19e4e4b1bde : SUCCESS in 57m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/045e3eecc9384057984642139b3c0453 : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c69cd88ab97b4c1e800374511cb48c44 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c2916a71fb448318d0531c31780d258 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4615c0916c244a8a12c0b909aeb4809 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3bfe3fb8bf42969df74b67bf41437f : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/adda66cdb91240eeaee5893f618feab1 : SUCCESS in 23m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/94dc95d596404a36baa26e04bc67dc6e : SUCCESS in 1h 07m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/52d9c8f08171424f99ec3c8be0d2218a : SUCCESS in 1h 15m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/578a946d7449414db8c02458d3691120 : SUCCESS in 20m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/036fdd3773f848de9f582a4c9747c71a : SUCCESS in 1h 06m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e2b15c595bad4a029d4ed3d98bf7c9c1 : SUCCESS in 1h 06m 28s","accounts_in_message":[],"_revision_number":3},{"id":"58567e4b4c41e52fb10a34eaaaef2f66f3a95bba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 09:31:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1b6ddd163a77ebcf48d8ea24c798a80fc57c22f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 09:32:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a68b753b1f164b4883fcd14094a9cd7a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78d33613c97f4708ab2590b634c69819 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3},{"id":"42688eca6eecf310869d1824a481efba998700e8","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 17:20:03.000000000","message":"Created a revert of this change as I4659bcd48ea32294b3ee84af9af94fae32b38bf3","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"600ed4f149a1cf542f54d7fc5005030f53c69a83","revisions":{"c5319f39e5eb14073bd0d0f6e86c27cc9a34af95":{"kind":"REWORK","_number":1,"created":"2023-05-30 10:45:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/884677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/884677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/884677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/884677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/884677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/884677/1"}}},"commit":{"parents":[{"commit":"878e9c0524e97dda7d61c6161d33e26052d6285f","subject":"rocky/centos: mark bifrost buildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878e9c0524e97dda7d61c6161d33e26052d6285f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-30 10:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-30 10:45:10.000000000","tz":120},"subject":"rabbitmq: pin to 3.11.16","message":"rabbitmq: pin to 3.11.16\n\nSeems rabbitmq node drain got broken in 3.11.17\n\nChange-Id: I3a23670f85336344d8aebaded74dd65871837b82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5319f39e5eb14073bd0d0f6e86c27cc9a34af95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5319f39e5eb14073bd0d0f6e86c27cc9a34af95"}]},"branch":"refs/heads/master"},"5e9b1bc8ddb0c55c98f84e5cc581732be0ad497b":{"kind":"REWORK","_number":2,"created":"2023-05-30 13:10:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/884677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/884677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/884677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/884677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/884677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/884677/2"}}},"commit":{"parents":[{"commit":"878e9c0524e97dda7d61c6161d33e26052d6285f","subject":"rocky/centos: mark bifrost buildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878e9c0524e97dda7d61c6161d33e26052d6285f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-30 10:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-30 13:10:24.000000000","tz":0},"subject":"rabbitmq: pin to 3.11.16","message":"rabbitmq: pin to 3.11.16\n\nSeems rabbitmq node drain got broken in 3.11.17\n\nChange-Id: I3a23670f85336344d8aebaded74dd65871837b82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e9b1bc8ddb0c55c98f84e5cc581732be0ad497b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e9b1bc8ddb0c55c98f84e5cc581732be0ad497b"}]},"branch":"refs/heads/master"},"600ed4f149a1cf542f54d7fc5005030f53c69a83":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-05-30 15:35:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/884677/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/884677/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/884677/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/884677/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/884677/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/884677/3"}}},"commit":{"parents":[{"commit":"878e9c0524e97dda7d61c6161d33e26052d6285f","subject":"rocky/centos: mark bifrost buildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878e9c0524e97dda7d61c6161d33e26052d6285f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-30 10:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-30 15:35:18.000000000","tz":0},"subject":"rabbitmq: pin to 3.11.16","message":"rabbitmq: pin to 3.11.16\n\nSeems rabbitmq node drain got broken in 3.11.17\n\nRelated-Bug: https://github.com/rabbitmq/rabbitmq-server/issues/8414\n\nChange-Id: I3a23670f85336344d8aebaded74dd65871837b82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/600ed4f149a1cf542f54d7fc5005030f53c69a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/600ed4f149a1cf542f54d7fc5005030f53c69a83"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
