)]}'
{"id":"openstack%2Fkolla~884815","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I3a23670f85336344d8aebaded74dd65871837b82","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-31 13:22:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a23670f85336344d8aebaded74dd65871837b82","subject":"rabbitmq: pin to 3.11.16","status":"MERGED","created":"2023-05-31 10:02:41.000000000","updated":"2023-05-31 13:23:17.000000000","submitted":"2023-05-31 13:22:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884815","cherry_pick_of_change":884677,"cherry_pick_of_patch_set":3,"meta_rev_id":"275ad961be20994d7d7430d93f253241dae059c7","_number":884815,"virtual_id_number":884815,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-31 13:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-31 12:05:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-31 11:30:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-05-31 12:05:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-05-31 12:05:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-31 11:30:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-31 11:32:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-31 11:50:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-31 12:05:56.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"6cb7731d7d150ac4fa87b93a54422e2a5576b2b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-31 10:02:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"15f39ac75829d1c560c7c74d2fed2a7881a46a60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-31 10:12:36.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1dae4bc53004dcdd72fa3beffc3e52792e5e4724","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 11:30:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"40b77d9e5cf78389ee097854efb0b57e8e8bc66b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 11:32:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b09a8fbb1304b498610c81909522075\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/124ad130d9854d9db4d0a2173e82154a : SUCCESS in 23m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3a0bcb5c526e49bfb94471c10f2b129a : SUCCESS in 24m 17s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/26810797af194f60ab8a2d567d6d763b : SUCCESS in 1h 10m 44s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75cdc21518d647f3bb72705fad7ca46b : SUCCESS in 2m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e50b268e5854426b6035f4d0f6224dd : SUCCESS in 4m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8f6089f858494f129d0ceba4ac840463 : SUCCESS in 1h 16m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/654a19da16584ae995530db3a57d3ac2 : SUCCESS in 20m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b718e23bb01f3eb505e392fab6985380bbe0a51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 11:50:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be40db8c9b874347b2ce3eea283f798f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5615ebdf30934447a1e6099b82f6be50 : SUCCESS in 21m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4517c1f587134b8f935cbd976eb6c274 : SUCCESS in 1h 08m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1c338c8b6ea3493bbc8b73576e8f4851 : SUCCESS in 25m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8dfce9c0c0a8408b9e13a28d3878917c : SUCCESS in 58m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7c122c3b1bdb405e8f04e00c1c01c683 : SUCCESS in 1h 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62dc2335bfab4c73a4d0cb7ccb7cd041 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d3fc0cefcbd49bcbb6c66dd71bbb353 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4470fab5f1c540409aa4af6b05186b34 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/defa72762a794e3da3e37167415360de : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc5eef28800421b87287c106bb1c508 : SUCCESS in 4m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6acccdee219d469f8dc752bb1b430016 : SUCCESS in 22m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c28742bb50414f15b43b7f4fcc57672d : SUCCESS in 1h 05m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bff44fcf739541e08d945b74a7a26f82 : SUCCESS in 1h 12m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/546fc70bda5c43c1a29fe52b3d2455a7 : SUCCESS in 26m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/efd5636c526240d29907cbff1473b10e : SUCCESS in 55m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/00f20aa1c6b844ea9cd35fc3d3d400af : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":2},{"id":"32fe9e31d57de05a9dd0771ffbaafd6f43f027e4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-31 12:05:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f8f994b25bd472d59ff6a3e875e21573d71e068f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-31 12:05:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"de05515c7da65db1b09e02c89a55ac340046ec09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 12:06:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8fa476d97dccd1a6ff505659209002617c9f9311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 13:22:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1863c0f8103a4009bf0cdda0bc2120ae\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fe42ffee4a9a4c3689dc9db5197f7bcf : SUCCESS in 17m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aa704244bb554803a2ac12c4f7144b32 : SUCCESS in 52m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c7ff54bc6f6a4bb6ba1d92f508361303 : SUCCESS in 1h 01m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5645fd371bfa494c89eea27cf47d8429 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db33c62f700f478481c59dc953d33a3a : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f374efc2da546949a5ad50e1f706e4f : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b876af619974bd39e75228b52f717a2 : SUCCESS in 4m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ec747e817eb94a8fa948b481d5764d67 : SUCCESS in 18m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/be78f778c7bb4994b9f9e4df8414d0db : SUCCESS in 53m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dcef5f7c21664628960cc850d49bfd90 : SUCCESS in 59m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c6ab70cb08b94d99a49ab0769ae3ed5a : SUCCESS in 18m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ed290d06df2142a3953fc94a27c72287 : SUCCESS in 57m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b410c45b5f5640dcb29ea2aa0c77c171 : SUCCESS in 1h 09m 17s","accounts_in_message":[],"_revision_number":2},{"id":"5af68fade82a8b1bdef35658733c39f1107d1b2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 13:22:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"275ad961be20994d7d7430d93f253241dae059c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 13:23:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d716c64691d942e38dc3c5e1bbed3e63\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/339c7d925156477ba8014a7fd9f8cfc3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"10be550bc562a6e4f11d4740a18cb89ad88dbcf2","revisions":{"844408ba76fc2a2f623e4a7b3413bc88fd555dc8":{"kind":"REWORK","_number":1,"created":"2023-05-31 10:02:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/884815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/884815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/884815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/884815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/884815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/884815/1"}}},"commit":{"parents":[{"commit":"3f911485bbef1e8d1f61daf938da1230ea8f927e","subject":"rocky/centos: mark bifrost buildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f911485bbef1e8d1f61daf938da1230ea8f927e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-30 10:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-31 10:02:41.000000000","tz":0},"subject":"rabbitmq: pin to 3.11.16","message":"rabbitmq: pin to 3.11.16\n\nSeems rabbitmq node drain got broken in 3.11.17\n\nRelated-Bug: https://github.com/rabbitmq/rabbitmq-server/issues/8414\n\nChange-Id: I3a23670f85336344d8aebaded74dd65871837b82\n(cherry picked from commit 600ed4f149a1cf542f54d7fc5005030f53c69a83)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/844408ba76fc2a2f623e4a7b3413bc88fd555dc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/844408ba76fc2a2f623e4a7b3413bc88fd555dc8"}]},"branch":"refs/heads/stable/2023.1"},"10be550bc562a6e4f11d4740a18cb89ad88dbcf2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-31 10:12:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/884815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/884815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/884815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/884815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/884815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/884815/2"}}},"commit":{"parents":[{"commit":"3f911485bbef1e8d1f61daf938da1230ea8f927e","subject":"rocky/centos: mark bifrost buildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f911485bbef1e8d1f61daf938da1230ea8f927e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-30 10:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-31 10:12:36.000000000","tz":0},"subject":"rabbitmq: pin to 3.11.16","message":"rabbitmq: pin to 3.11.16\n\nSeems rabbitmq node drain got broken in 3.11.17\n\nRelated-Bug: https://github.com/rabbitmq/rabbitmq-server/discussions/8415\n\nChange-Id: I3a23670f85336344d8aebaded74dd65871837b82\n(cherry picked from commit 600ed4f149a1cf542f54d7fc5005030f53c69a83)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10be550bc562a6e4f11d4740a18cb89ad88dbcf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10be550bc562a6e4f11d4740a18cb89ad88dbcf2"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
