)]}'
{"id":"openstack%2Fkolla~884823","triplet_id":"openstack%2Fkolla~stable%2F2023.1~Ie1b36eb9a403611c2f86aa4c1d500d324e5291f3","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-06-05 11:54:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1b36eb9a403611c2f86aa4c1d500d324e5291f3","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","status":"MERGED","created":"2023-06-01 08:26:44.000000000","updated":"2023-06-05 11:55:54.000000000","submitted":"2023-06-05 11:54:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884823","cherry_pick_of_change":884208,"cherry_pick_of_patch_set":3,"meta_rev_id":"a6944a958b488d4e7a4443d512e7c7598c206b15","_number":884823,"virtual_id_number":884823,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-05 11:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-06-01 09:34:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-01 11:03:13.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-06-05 09:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-06-05 09:57:36.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-01 09:34:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-01 09:47:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-05 09:57:36.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"e208c65198ce30981f72ddb04dcc0f3e916fbb66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-01 08:26:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"adb0d2fdf960949c7034e22206647190044fa905","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-01 09:34:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e4e37628dac0968734eb85615412619243fa6fa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 09:47:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/776a582296c14e9dba7fa53a6a344547\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d0918de8dcb24d75a7297486b5359c33 : SUCCESS in 29m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a6041583266642ceba6217491f3c38f3 : SUCCESS in 59m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5f4080d87def477ebb5fcc293f9ed7be : SUCCESS in 19m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/318a1c9d72834c4ea2b919b0d9a26b8d : SUCCESS in 58m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e428b6163e3743eb8f5f4a0c2d58faff : SUCCESS in 1h 02m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/437b70e864384483a829103ff094364d : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/486e1bbb01c741d881adcdaaaac45789 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c293c5c825034ab69373527f24fbb515 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d0fc3a78999464abdcdb400ceb0df11 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8ffa65476a24481a5db5e1a3203c7f6 : SUCCESS in 4m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7acfb8b808b64d349b89977ba5769ea5 : SUCCESS in 24m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b4761b8aca6f475497ae2d84c14455dc : SUCCESS in 1h 10m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e079eda847e045cfb360c3bd46b598ba : SUCCESS in 1h 15m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0011a10d3a614db9a4136b218f6db95d : SUCCESS in 27m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b602dc0d67cc4549a9159a01abf01709 : SUCCESS in 57m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dadb2452c1c14375b5567a89b2c0272a : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"70d954e2f55c787ad87a921d8faa807ee5067992","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 10:53:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e21639bcfb7463d8d01cd18646da125\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/94cc50f023ed4c6d8896671137f3a3d3 : SUCCESS in 1h 13m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2be7422a588047329fba92a8341580ef : SUCCESS in 1h 13m 41s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/654f4c8f96f742728ce2e82282dfd0fb : SUCCESS in 2h 25m 17s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecaccccdf9ac405f8464a551d89c5c9b : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eab4fb34cceb468987e5dfe66f8e1683 : SUCCESS in 3m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cc5fd7f3c92943918007c4a84bce55e5 : SUCCESS in 25m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f1f3608f854a45cd8b165b78e9bd8f5a : SUCCESS in 25m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2af2bc2f97a7caed4215e1edcf34a1bf6a66f96","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-01 11:03:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"363c53dc34017ddfda5338611161184ce10736d4","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-05 09:57:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c599787b9a004b6518a99ad58ecbcf65bf500993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 09:58:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5eeff7fb76925e9a2d11c29f427fd22fd5282c1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 11:54:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b7eab8ada6043e6b974d27077986910\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c867e87287fb477886496703691d1d62 : SUCCESS in 21m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/66744a0c52a143929426d0b15ee74fb8 : SUCCESS in 57m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d89e712b77dd475ba040473dba768810 : SUCCESS in 1h 17m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9d835d440154323af2efb939834eb02 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5544e0aa3bb940b798112582d1870457 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/deae45da20904be1a8d86482fa37f41c : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e39fb566b2234df3bb85c0335ae7666a : SUCCESS in 4m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/019950992c814ecbb58d46ee5eff1168 : SUCCESS in 22m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1045e0f534a14449b2692e7aedcfa9d0 : SUCCESS in 1h 00m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5d42acec0292410b8c5b1b9823683162 : SUCCESS in 1h 12m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/93888bc1c8944008b62c34a4181687ec : SUCCESS in 35m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6da191e3a91246f9bd3dc992771a3416 : SUCCESS in 1h 12m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3b1f6ff6dfa8420da7a4d664632bb71e : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"63e5167a81b2ab44af36d227656666aeb1798e38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 11:54:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a6944a958b488d4e7a4443d512e7c7598c206b15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 11:55:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5328b12744a4db0a752ee7f06869d06\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5cabe8e57344d6c8504a1919feb3b1e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ca7ec16135bf62637ca1115e122f66cb1eefb27","revisions":{"2ca7ec16135bf62637ca1115e122f66cb1eefb27":{"kind":"REWORK","_number":1,"created":"2023-06-01 08:26:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/884823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/884823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/884823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/884823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/884823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/884823/1"}}},"commit":{"parents":[{"commit":"10be550bc562a6e4f11d4740a18cb89ad88dbcf2","subject":"rabbitmq: pin to 3.11.16","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10be550bc562a6e4f11d4740a18cb89ad88dbcf2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-24 14:19:16.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-06-01 08:26:44.000000000","tz":0},"subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","message":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists\n\nIf mariadb-clustercheck starts before mariadb, the directory may not\nexist, and the command fails.\n\nThis is necessary for the refactored MariaDB restart in\nI5a12670d07077d24047aaff57ce8d33ccf7156ff.\n\nChange-Id: Ie1b36eb9a403611c2f86aa4c1d500d324e5291f3\n(cherry picked from commit 9035251443ed6cbb110e9dc5efdbc9d4be797ef5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ca7ec16135bf62637ca1115e122f66cb1eefb27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ca7ec16135bf62637ca1115e122f66cb1eefb27"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
