)]}'
{"id":"openstack%2Fkolla~884856","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I3723e418c061c387c7f18cd33556ebb74ba1bd95","project":"openstack/kolla","branch":"stable/yoga","topic":"fix/fluentd_gems","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-07 12:54:39.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-01 11:13:23.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-01 12:34:10.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-06-12 14:24:23.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-06-12 14:24:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3723e418c061c387c7f18cd33556ebb74ba1bd95","subject":"fluentd: pin td-agent to 4.4.2","status":"MERGED","created":"2023-05-31 10:04:45.000000000","updated":"2023-06-12 14:25:27.000000000","submitted":"2023-06-12 14:24:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884856-fix/fluentd_gems","meta_rev_id":"315a683911616478ef7585bab9dd2a8783060610","_number":884856,"virtual_id_number":884856,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-12 14:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-07 12:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-06-03 20:00:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2023-06-01 12:34:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-06-12 12:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-06-12 12:55:36.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-06-07 12:54:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-31 10:52:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-05-31 11:19:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-31 12:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-31 12:46:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-01 11:06:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-06-01 12:13:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-03 20:00:17.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2023-06-12 12:55:36.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"ee47ed3c2ef6b37c43d7b6d1110cd40036d2478d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-31 10:04:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08820baed75f2083ed60249dfe01b4aeb606a4bb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-31 10:52:56.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d97a05c8d91263651cbd7efb056651b6d31b07e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 11:19:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4faa84198974bd8b04e82841bef5ece\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9013c8a67319445fb858d1ee13869513 : SUCCESS in 34m 11s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/30ef1f1fa84c46edb908675a2f08afdc : SUCCESS in 23m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d804717af9bb4779bbd43b63549d52e9 : SUCCESS in 25m 27s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b25a56a65ed547ffbbf5f81b63e153cc : SUCCESS in 1h 09m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f0051d392c3452b91fb329fc1bb7b6b : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65a794c883cb4a828abe1e9c5aa0ca37 : SUCCESS in 3m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd4903cf3fbd4ac8adf0cf2bc21070f0 : FAILURE in 40m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc913fe2b0168252a9fa5d5441779cb7a09aa156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 12:08:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fbe24b3af9b48919367478f64e74cbf\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d3a8bdcd46614da8834623072a0bd6f3 : SUCCESS in 30m 49s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/10be6b6cbb6d46c99e7f269a72b335d9 : SUCCESS in 40m 09s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/96945f14db544be8b5bf1086ea26199f : SUCCESS in 24m 48s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b9105cd4043448f697bed37ac34930e1 : SUCCESS in 1h 24m 05s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/601c758f6c9d4cdb9648a96a51e66643 : SUCCESS in 1h 09m 03s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/4e06f8f1b7cc4b65be2549551973ce0c : SUCCESS in 1h 16m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/05a89d8dd0c948a99603cb3dc4c7f142 : SUCCESS in 1h 27m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ed98310eacf841cea7809ac304b53ff3 : SUCCESS in 25m 11s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/65c559f083454e129244a63df1ed0aa8 : SUCCESS in 23m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e6e85783483b46508741384b3eab0c83 : SUCCESS in 53m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/748ad72f9be34b23a5bad6866aa58008 : SUCCESS in 1h 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf6092d996054d77b70ec2ef47ed7c46 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ea83242e30e4509b87287685adb385a : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f0bd22a16244fdcb027105cbbec1695 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2660d008018f40cf9675dd149b42b6bf : SUCCESS in 5m 27s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/2808eb8c8cf440f8890743b3be0e0730 : SUCCESS in 28m 28s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/4dd50f2b232c42f38cb8c7ff751c0b04 : SUCCESS in 56m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d96da4da2f1a43178272899f5f343783 : SUCCESS in 22m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/73ec184bbdf5412c8c8e6eff91bea030 : SUCCESS in 23m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9fe4b1f7069197e02543224c4574b2583bcf4d7f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-31 12:34:29.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a39a504af28b9ef4cf55ef9e00d9c56fe5bce249","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-31 12:39:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebdda006c22955e99a062aafb83b366ba0a52362","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-31 14:15:32.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"e4b625e269c03fb6c9c3ae371c9150fb955b75e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-31 14:24:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f59be7b8677d308a12b0725fb825d302953b8b34","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-31 14:25:37.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"388adf9f3e14bdd7dfd1a763c8aa13a09ef61042","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-31 14:34:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbfdf012e4f24c84221c033702a19372b76bde9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 14:41:37.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7ec9863f39b4e0cbb31456a7da2d01e\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fcb68acfb07f460a90c6bb2344415e75 : NODE_FAILURE Node request 300-0021361147 failed in 0s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6471cfe73dbd48e08780a2aa507c8803 : NODE_FAILURE Node request 300-0021361148 failed in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b4d863e7478412490697a85fc105e2b : FAILURE in 9m 50s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/06cc864519b5431fbaccf7b948251411 : FAILURE in 15m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cde11924d87347099f1e9c252a2e6ed7 : NODE_FAILURE Node request 300-0021361151 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1206c78e020f491699ea5c4babc4b982 : NODE_FAILURE Node request 300-0021361152 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/77bd727da45245f694bc8e20511f1ef0 : NODE_FAILURE Node request 300-0021361153 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8dc357bcda6b942721555aab8702bb72de54eb90","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 14:49:18.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8dcd8de5c93d5cece7869d1006c7f07e5d00bfc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-31 14:56:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"384cbc74b8c06e50ef683ef2f0de3661b185b156","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-31 14:58:13.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"72437f31190f6c896cef725fa5a2df4d2e0b74c6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 15:02:07.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"92178699b44a605df93e18e2aa720820c0a34ef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 16:32:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b43e1ed0c45a4d6481d8651a03e04bd4\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0d15a3b1f7a94c0f8f73ad2166f7880a : SUCCESS in 34m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/da32863603224af088a1d936a45a9698 : SUCCESS in 37m 52s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/b44b6b09a882462d992da5e19e91c022 : SUCCESS in 30m 36s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/129d6a300eb4414bb0893af0846eda51 : SUCCESS in 1h 16m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b8f02cc1f5644050928a6754eba66721 : SUCCESS in 1h 12m 23s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/baa91832ffba4143a6b9977547bd06df : SUCCESS in 1h 03m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d325c0fab6a2422b89283da8fb49932b : SUCCESS in 1h 26m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6d5e6212b0b34c418aa1799ebd376bbe : SUCCESS in 27m 06s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c46e91f59fc646e782f82bea4d32d909 : SUCCESS in 18m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2e0c96e0a9984c3faab205e20ebd9841 : SUCCESS in 1h 04m 13s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f0dffba921db4a25ba324db7f0f2f40d : SUCCESS in 1h 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d75b39711c64005946f3fc517904e7b : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04831c3ce4124a38a99cb0f2f94da8ac : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0710ae32e09a43b68fb1d1b0c5c65747 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a19435d2fa4bf8b5b6ac19fd1c18cb : SUCCESS in 5m 27s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/f73d84915ea74369a8a45e04664970ae : SUCCESS in 24m 06s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f4a16230bcb5496fbed432dad4046219 : SUCCESS in 1h 05m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e07f92bd252147e4806722feef2175fe : SUCCESS in 25m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/904cd070e6454a709c989d0ab6188f8f : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":3},{"id":"81e35d19afad957f7ea2fae358aae4ddd08e533e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 17:04:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb9aa59d8e9b4505a2d3bc0601560f1a\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1902c0dcf49e4a5d843f5d7385d9e961 : SUCCESS in 2h 05m 32s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5e3f9ebc8e4432687bf3c92457242d7 : SUCCESS in 1h 36m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f54a4d29b894d92ae821a4b49ff806e : SUCCESS in 1h 32m 15s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e9e8ed8367d424ca7e889aa8314b454 : SUCCESS in 1h 10m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ea86ff65fd042fd89f9f7860d83bc91 : SUCCESS in 2m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e852bd4aafda46adac1faf5aa82f6aaf : SUCCESS in 3m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b294b5777ac43dd81594c72a4dfca7b : FAILURE in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"61cd3ada7fe2bcf3b8f8c39cbe2f41843bea4e3a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-01 11:06:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5ccda76966a5024ce5a8f3b9b382656fb1236f6a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-01 11:13:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c18048228fcd9387667983271f9f28137a10d496","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-01 12:13:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1af707deaf64592b33bebb0bbc66a3f50fcb3ead","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-01 12:21:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"019448ec5e77fc8cd60c8fd5796a3e7367b8bee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-01 12:26:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d3275718d46d362cda919f7784909d6e4d9349b8","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-01 12:27:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4d80110f46578a4c990bfa64bbe2e00cd169135a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-01 12:34:10.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5573a92acf76e8a3c2cf88b7a2e6754d40c08ee7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 13:30:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6241715fe485477aa4d50705e53f9193\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/233bac5b35c140cdb02aad911182edcc : SUCCESS in 23m 22s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/71ed39f107ae426a874dca734b6c0ef0 : SUCCESS in 20m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c9fc77cb41a4413b74833daf25a6b4c : SUCCESS in 18m 41s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/56ce0a9226ea49768b37b888b3f63876 : SUCCESS in 1h 03m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/713cfd50f6814fe4ae9c42edbf7173f1 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/86d1e40982e74088ab6f302bdf07a353 : SUCCESS in 3m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9314e1c61f9c4d7cb590ce417792105a : FAILURE in 36m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e7801b07728fb54ae58c3aa972da163c399a4004","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 14:24:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48e9c5257c4649c69975648c7b7d3063\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/83cf0b1eead44047999bd68c430fe9b4 : SUCCESS in 40m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1a8decc0c03c4a1081e5b65db7940e50 : SUCCESS in 40m 28s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/35c5d199d23a49bba2063d900675001c : SUCCESS in 31m 01s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2a34736e549044fb864b405c7f3c9988 : SUCCESS in 1h 10m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bbec003eb22e4d4daccb77d9e675bfb8 : SUCCESS in 1h 12m 46s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/e8d09ec5ce044c0e8137e3f194908a64 : SUCCESS in 1h 02m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/50cf3733364a442487f0c4dfed794a67 : SUCCESS in 1h 38m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b47598d9892f4760abaee0c2f9865ca3 : SUCCESS in 29m 02s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4939a4b45e8d48cda6f975c85e4028f6 : SUCCESS in 25m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/efe46e83618d4566a7cb0bee160c9c26 : SUCCESS in 1h 12m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/abf90fc3091b4b53bc7277220fabb40d : SUCCESS in 1h 24m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12d98c1b617d4a6ca8286963bc774716 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cc7ed8395704354a0fd5cc26d6e78aa : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8e9efd558ae42a69de5338b8f7e33b4 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3debf32b89ac443dbfed15ae73b4207b : SUCCESS in 7m 19s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/13665e9433ed4d6b8d83ec152c8b266b : SUCCESS in 19m 47s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ca847caff1724e9993c3988b58dbc9c1 : SUCCESS in 1h 17m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/acb68a1b840c41409b1515711baa367f : SUCCESS in 28m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d332a51bdc224612a8d8d87cbc5299fb : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":4},{"id":"67b26c83623fc8c20aa140ce776d291bb7a659e9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-01 19:25:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cf87b51d0072d7192fd5bd58949e733a2589cabf","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 21:52:41.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/893461ec509b4f8abae872dfdee45aab\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/55e844a6cd6b4c9994ea01b0c982db35 : SUCCESS in 41m 03s\n- kolla-build-no-infra-wheels-centos9s-source https://zuul.opendev.org/t/openstack/build/31262fae18a94ce69e30b4b00d529c47 : SUCCESS in 26m 18s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/15e3e40eaf6e4527b23a89cfd50f758a : SUCCESS in 41m 27s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/bb0b60fc6a0a43e5b7fafe32a80f29c5 : SUCCESS in 1h 40m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d345d0d63a1e44ac852fba8e6ff58313 : SUCCESS in 1h 51m 16s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/f1348bea52644cb9922c70119b00c59b : SUCCESS in 1h 25m 36s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/31ef2722877b458ab99ed897616ca0ee : SUCCESS in 1h 47m 16s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/6872bef956c54bb6bbdca00063ca1994 : FAILURE in 1h 17m 26s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/0722796e3420461c958720dd4105ac8e : SUCCESS in 29m 30s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/d524daf88b484378b0669efb2ceb11a5 : SUCCESS in 1h 11m 01s (non-voting)\n- kolla-ansible-centos8s-source-octavia https://zuul.opendev.org/t/openstack/build/68206e1ad5204c9a95fd68e92fd9ac61 : SUCCESS in 1h 17m 43s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/ad80ca61495c47cb80be47fbfb3234a6 : SUCCESS in 1h 18m 54s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/2d5c602fd5fb4f639025d7feb9d6e7f4 : FAILURE in 42m 00s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/d0d38da90e684efc8879dd9e33d4809e : FAILURE in 22m 27s\n- kolla-ansible-debian-source-ironic https://zuul.opendev.org/t/openstack/build/68a45ce0aa3447f7add865d45e60c541 : SUCCESS in 1h 14m 29s (non-voting)\n- kolla-build-no-infra-wheels-rocky9 https://zuul.opendev.org/t/openstack/build/bfa501499fa040f78085af6c725810f0 : SUCCESS in 19m 27s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf069f708e844ef98ab6bee7d36265b8 : SUCCESS in 22m 08s","accounts_in_message":[],"_revision_number":4},{"id":"282ad46c51099e6cd5f34537d46423fe21f1ac24","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-02 09:42:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"797c540ca6b0f3f7f5ce461c87ff186780e8fe1b","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2023-06-03 20:00:17.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"59c1f1d7d68d20c422a4fd4fdc5e6adcdaa760ea","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-07 12:54:39.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"99e9bf0ba6b18191d5d3ef05bc47aa65f96533c3","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-12 12:55:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ce0aa88c8f2c6ae950768e9dfe7372f67b1ab9a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 12:55:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c486863c63117a4279aff9161725e0cc17c2ded5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 14:24:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7289f8f4d2404db5967ce68e1b023f34\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e7451ee7dbd7475c8d64a3e0dc235070 : SUCCESS in 48m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9470a168cb344fa5870992ccbf60f777 : SUCCESS in 1h 10m 24s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/03e0ac8e4fc04323a2d7dd3584102151 : SUCCESS in 1h 28m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d023228fc03b42f5a86eb40bb2b502ed : SUCCESS in 20m 24s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d5d6709355b74656a5044d1dc6ff7488 : SUCCESS in 54m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c412e894363a47fba87137a387a56a31 : SUCCESS in 1h 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f3d4dbd880349d5af2d77565cac2b0c : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5482f79ad9440558c9d1e497129d62f : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4998274c6f2c4e7e93b4ae0259a52ddc : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fce1b0f78c742658c092d74d2a5237c : SUCCESS in 8m 30s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a81146befef74591be95e31419467df2 : SUCCESS in 1h 02m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/722894d5dd7d4343869dacf3eae9703a : SUCCESS in 23m 23s","accounts_in_message":[],"_revision_number":4},{"id":"0da6037e59fb1f2c75b6de13c1fb01eb5091ffb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 14:24:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"315a683911616478ef7585bab9dd2a8783060610","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 14:25:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8731fa62712d491ba6279a6f13207c71\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d31190e2de34c10998fcf116140beef : SUCCESS in 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"59a5de343ac3adf9463d7231fa0f5ff5f1f0a433","revisions":{"a071ccc6780bb0333c6cc6677929c2f895f237e4":{"kind":"REWORK","_number":1,"created":"2023-05-31 10:04:45.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/56/884856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/884856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/884856/1"}}},"commit":{"parents":[{"commit":"9c0a62a9526a366651ae7ce0a96676b18221b97f","subject":"CI: Fix centos9 builds and drop unnecessary zuul vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c0a62a9526a366651ae7ce0a96676b18221b97f"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-05-31 10:02:53.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-05-31 10:04:44.000000000","tz":120},"subject":"fluentd: pin fluent-plugin-elasticsearch to \u003c5.3.0","message":"fluentd: pin fluent-plugin-elasticsearch to \u003c5.3.0\n\n5.3.0 requires faraday\u003e2.0, let\u0027s pin stable branches requiring\nElasticSearch support.\n\nadded: remove newer plugin actively\n\nCloses-Bug: #2019557\n\nChange-Id: I3723e418c061c387c7f18cd33556ebb74ba1bd95\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a071ccc6780bb0333c6cc6677929c2f895f237e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a071ccc6780bb0333c6cc6677929c2f895f237e4"}]},"branch":"refs/heads/stable/yoga"},"27c5198ea1b51d5ac150580028b717c6da642dce":{"kind":"REWORK","_number":2,"created":"2023-05-31 14:24:49.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/56/884856/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/884856/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/884856/2"}}},"commit":{"parents":[{"commit":"9c0a62a9526a366651ae7ce0a96676b18221b97f","subject":"CI: Fix centos9 builds and drop unnecessary zuul vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c0a62a9526a366651ae7ce0a96676b18221b97f"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-05-31 10:02:53.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-05-31 14:23:50.000000000","tz":120},"subject":"fluentd: pin td-agent to 4.4.2","message":"fluentd: pin td-agent to 4.4.2\n\nCloses-Bug: #2019557\n\nChange-Id: I3723e418c061c387c7f18cd33556ebb74ba1bd95\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27c5198ea1b51d5ac150580028b717c6da642dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27c5198ea1b51d5ac150580028b717c6da642dce"}]},"branch":"refs/heads/stable/yoga"},"e1cf6ee2f82103cfa7dae42983b4ca3512d3bb83":{"kind":"REWORK","_number":3,"created":"2023-05-31 14:56:58.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/56/884856/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/884856/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/884856/3"}}},"commit":{"parents":[{"commit":"9c0a62a9526a366651ae7ce0a96676b18221b97f","subject":"CI: Fix centos9 builds and drop unnecessary zuul vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c0a62a9526a366651ae7ce0a96676b18221b97f"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-05-31 10:02:53.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-05-31 14:56:44.000000000","tz":120},"subject":"fluentd: pin td-agent to 4.4.2","message":"fluentd: pin td-agent to 4.4.2\n\nCloses-Bug: #2019557\n\nChange-Id: I3723e418c061c387c7f18cd33556ebb74ba1bd95\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1cf6ee2f82103cfa7dae42983b4ca3512d3bb83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1cf6ee2f82103cfa7dae42983b4ca3512d3bb83"}]},"branch":"refs/heads/stable/yoga"},"59a5de343ac3adf9463d7231fa0f5ff5f1f0a433":{"kind":"REWORK","_number":4,"created":"2023-06-01 12:26:35.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/56/884856/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/884856/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/884856/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/884856/4"}}},"commit":{"parents":[{"commit":"9c0a62a9526a366651ae7ce0a96676b18221b97f","subject":"CI: Fix centos9 builds and drop unnecessary zuul vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c0a62a9526a366651ae7ce0a96676b18221b97f"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-05-31 10:02:53.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-06-01 12:26:25.000000000","tz":120},"subject":"fluentd: pin td-agent to 4.4.2","message":"fluentd: pin td-agent to 4.4.2\n\nCloses-Bug: #2019557\n\nChange-Id: I3723e418c061c387c7f18cd33556ebb74ba1bd95\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59a5de343ac3adf9463d7231fa0f5ff5f1f0a433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59a5de343ac3adf9463d7231fa0f5ff5f1f0a433"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
