)]}'
{"id":"openstack%2Fkolla~885857","triplet_id":"openstack%2Fkolla~master~Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8","project":"openstack/kolla","branch":"master","topic":"rabbitmq-ppa-novemberrain","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-15 04:21:32.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-16 15:33:38.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-06-15 19:22:36.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-15 07:31:08.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-06-16 19:58:04.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-06-16 19:58:04.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-15 08:58:37.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8","subject":"Fix rabbitmq image build","status":"MERGED","created":"2023-06-12 10:15:01.000000000","updated":"2023-06-16 19:59:59.000000000","submitted":"2023-06-16 19:58:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"885857-rabbitmq-ppa-novemberrain","meta_rev_id":"406240d0824a43401c4d335a02cd6cb5e0ef83e7","_number":885857,"virtual_id_number":885857,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-16 19:58:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"date":"2023-06-16 07:52:06.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-06-16 18:04:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-15 07:31:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-06-16 18:04:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-15 04:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-12 10:16:44.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-06-12 10:16:44.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-12 10:16:44.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-06-12 10:16:44.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-12 12:47:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-12 12:52:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-14 16:50:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-15 19:22:36.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REMOVED"}],"messages":[{"id":"d72d112f94cf824ed44357d81b8249ca5a62f3a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-12 10:15:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1228d5f14bcacb555d3a851f9fdb4046ecd7012","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-12 11:05:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5a04988221ebe938568bcb9841e0d49d7b87b092","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-12 12:47:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9cc1d12c24928ac991623b86e4bc5fe28bfc10d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 12:52:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/561716f29c2d4aeea48fb9dbb38a3935\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6d08603bfe6849b0a9c5ca8b14a49e17 : SUCCESS in 19m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c7bd10ed484e4a0c89a2c6985e17a9d6 : SUCCESS in 58m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4a24c7acbcf54a70a6ce355b59d5ea81 : SUCCESS in 19m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/02856913416b43e1b1397e07c9dd6eed : SUCCESS in 51m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e7672a192f5c41caae030d1605704b2a : FAILURE in 45m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51722e2ef1fc45748bb4d7bd1d9ed479 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9adcc8a43a214cdebd18a908e476e825 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/526c916312b8486ca9890755befa88c8 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcd6d050280c4cf58fe431d796955ee3 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0fce835ba9e430f811933d888eac841 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c48a3cbebf4491a8968cd8909170b3a : SUCCESS in 4m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f1234ba321384ac4ab0cf47b095f8a10 : SUCCESS in 20m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/48ebb283e4cb4c3da1c993dbcdc59760 : SUCCESS in 1h 00m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/71b1d19451f2479fb7c00950089f900a : FAILURE in 44m 16s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/9e360ab40c134354a12a4e51c96b5d3b : SUCCESS in 1h 26m 38s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/01a321b3be7f44728b2b78c46947cc7b : FAILURE in 59m 11s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/80c332df6df14aa69f0c207b8b51355d : FAILURE in 1h 04m 40s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/17421f9869854dc48cda571d88ba541c : FAILURE in 42m 52s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/bc8aadb6ebe74aed9c1bd1a96844399b : SUCCESS in 1h 07m 48s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/4e8d4899b133474295d23a1592e05050 : SUCCESS in 28m 44s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/c4ef0987024f4b02bf46654b0eb3d6e7 : SUCCESS in 1h 01m 12s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b9cd1da74dc2419f9b52be510f480f9b : SUCCESS in 1h 12m 47s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/f96e69fd8aa04b5497e5af063d64cc02 : SUCCESS in 1h 06m 53s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a83303a6b3174fdab4d4df14c0e96da4 : SUCCESS in 31m 14s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5dbcd030400243748c92209039ff0631 : SUCCESS in 34m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/40a4b588c21449bc8b4fa79a6c78321e : SUCCESS in 53m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b586c03ba9284848976757b99e41c0d2 : FAILURE in 49m 57s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ea18654fdfdd43dd81152e2b23521efa : SUCCESS in 1h 23m 49s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b36e27c4e15547a28799765242ecf43b : FAILURE in 1h 20m 07s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/1198a54defb84b68a61911c31a160121 : SUCCESS in 1h 18m 04s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/76e35f76edc34076aca450391e36c288 : SUCCESS in 1h 10m 35s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/19f9fa3bb7bd471abb7202bfd3c1fda2 : SUCCESS in 1h 16m 19s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/8fbaec85b6f74103a7c54e8bc9621876 : SUCCESS in 27m 00s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2d85f1fc59514887b8d88caea2dbeeb8 : SUCCESS in 57m 28s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/725c2309389e43ce978fb3dca6ad8a0c : SUCCESS in 55m 36s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/fbab0c46438e4455bc593ebd4b496a5a : SUCCESS in 1h 09m 58s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8606ff0bb7714870a0c763ce7a0c69ff : SUCCESS in 26m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e48d42e30476d9a7c9eb30e7e0b092d0f762e804","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 13:08:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afb89fdf7b604f48b236da9b5fb8407a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fdefb281d8744c12bd873f377c696e55 : SUCCESS in 18m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1cb8ff852b474273ab4f3180083858a1 : SUCCESS in 15m 48s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b59bf47b4051430da53db983d4c4f6e7 : SUCCESS in 2h 00m 32s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3b19d3415ef454b88416e7ed87b0046 : SUCCESS in 2m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f536286a6706432ba004e6d7abc37602 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b1039b3b2de4ad2885ab6bffeb93fc8 : SUCCESS in 2m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/aa0cc2513c9e4c5290f60e155f8e9274 : SUCCESS in 17m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7ebc685f4afc4097940734ce8cbab2f4 : SUCCESS in 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"33a9aad2e5cf2df680b2f651b5283b54b34c6253","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-13 12:30:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48cfbc80339fc93d1606912cc848b97f8433fb50","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-13 13:32:41.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ccb6b250149bc785aa166bd392fdac18236930c6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-13 14:03:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6402900b167026e4a422d97edf0d193e67fc899f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-13 14:37:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0521bbe44214e9da004047a8b5ad478183ffd132","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-13 14:41:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e62b2c9c8e65320cd46e119e6f78578803fcbf1f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-13 15:06:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"70d38847690d1762455f4b543e96d5e633752a07","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-13 15:14:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"780f15a4ecd649c7bcc36bf7e82eab0c36c89a1e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-14 08:48:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4d2082fac9a3f7d04d775973f68af560b78455a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-14 09:04:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5dd8abc097205c61b28023467e622c7df66f043d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-14 10:17:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"62c7c9cace5701de1de6b084fb647b25f7e0f043","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-14 10:22:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f006a3685980f6037493787d98d2430a86375b3b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-14 10:23:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"66ed2a92e91e6d7b3e10792d6e560ba293f76033","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-14 10:27:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1d1f3f75a7b0d15a906e2f02f2534baf824e3096","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-14 10:31:05.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"477efa7b995b4c2d764288e9168bc5abcdd4b66f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 10:38:59.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e91ab6f2d994e07a6b92b198f65f91c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3c75751b23a546cab078652640b92dd9 : FAILURE in 18m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4a7f9340527844ebb93ac7d6d2bb3605 : FAILURE in 15m 33s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/164d881ae46941a2a82ed7cc8ac6b1ad : FAILURE in 20m 12s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77b15f75e1ff4ab58d7628d10df14d92 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8f5470d6d0c4dc1919f8c33141e5907 : SUCCESS in 2m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69eeba53244b436eaacfa926ef277707 : SUCCESS in 4m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ed02088b2fb34f689b64f9a5d30b42cf : FAILURE in 17m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e59c3f7e6510479d8262a55f9fabc448 : FAILURE in 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"65dbdd27d8a903a105c4de8042cbbdd2e9c44511","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 12:30:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe538527a1134edcb8a6962d7717d694\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/20396b78fe4e4a89a1f8fa1d1e13c42e : SUCCESS in 22m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/11e11205511144ad8bd9e7c697bc3a7b : SUCCESS in 1h 03m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d1fd9d6b02d849c7aed622cddf1a10d5 : SUCCESS in 22m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9b77b1bf6ab24ac6ac9d2e6cdde51436 : SUCCESS in 54m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9a64469d38ec45bd8efad93e161adaa9 : SUCCESS in 1h 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aad3f9cb46f4c07afdd0890af26b1ee : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b43ca538d0149439f88f33dcedc1a0c : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1419819ad4244a19a9a788345388fca : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee7dc8d68e5f4dffafcaa5aae1e12795 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfcc6f15818c4bf390f699049c140041 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8698f04b2942e0afdba43359849bbb : SUCCESS in 4m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2fc84287c3644acaa6cf91fe14595563 : SUCCESS in 15m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/29d941d27cf0414dade74897faa8bb90 : SUCCESS in 1h 04m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0bb5a8a3e55a459180b0f65b11db9e42 : SUCCESS in 1h 13m 11s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f12363c958894ae19db9bbf16cfbefde : SUCCESS in 1h 08m 28s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b58b00c3247046e9a63a3a0a3db6bfed : FAILURE in 1h 01m 48s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/b293b9f7eeda4488819d4404f3a184b9 : FAILURE in 1h 07m 04s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c8a9aa0b237a4c7ab8556ec348d5aa7c : FAILURE in 41m 24s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/a134e241c2564baeaeddb2e38a073f41 : SUCCESS in 1h 17m 17s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/54706ce1aa814779a971e08e36b5ae08 : SUCCESS in 24m 09s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/77639ba32cde4ddb8100aac5aa7628db : SUCCESS in 56m 40s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/16597ee52c2943e59744f9baecb7c0f8 : SUCCESS in 50m 02s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/1ea60d2003744d1ab71fe49b813b1bdc : SUCCESS in 1h 10m 27s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/70d2442053ea43f7be59533a9abe3025 : SUCCESS in 25m 18s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3f26fef3bd804d4483982e6ab4b2daf2 : SUCCESS in 25m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ec5e5aca42dc4c12b1f5465b05d603dd : SUCCESS in 53m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c44a4cf819d94af5b35505273fade720 : SUCCESS in 58m 06s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/436f0466946e4f149c7137c2f1775454 : SUCCESS in 1h 14m 45s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e890994fb47842089b20ce68c751341f : FAILURE in 1h 33m 06s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/7ea6600d364b4546907e7abd16c3989e : SUCCESS in 1h 33m 50s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/6bc846fecdf94200bf51d5fd23dcda4d : SUCCESS in 1h 20m 47s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/1413d2d171db4b5b87ad4df1cfcf8501 : FAILURE in 1h 58m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/1c37c37ed642427fb34519999c79e37e : SUCCESS in 23m 54s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f33acb36cc2a40d8bc9512d175a7b4a2 : SUCCESS in 57m 00s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2702d7c2e0734fa281e3f8434f0043de : SUCCESS in 56m 16s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e2011014e9f34fb49e71dcaca505fbd3 : TIMED_OUT in 2h 04m 24s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a6f95c886b5c4b2d875382212d6c4838 : SUCCESS in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0b49936bf9b12c79a0ecc18e5268b9b03e18af0f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-14 13:28:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a08b41fcdc5122c0e832222704247d48b32ef257","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-14 14:43:56.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b72ae70c0e2073884102aacbf6dc3b8406e22282","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-14 17:43:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"51192e46a0e117a5a6022efcf7ee70b6ede5b379","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-14 17:54:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"456422047db455bf6a7f4869cedef82ca0517205","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-14 17:55:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a2a9a8343aee54ae118bfb3e0461133b976d0a68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-14 21:28:05.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"beb5e8bfd19f6c45840966bfa200a9e314fdab04","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-14 21:29:35.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"495f77ce065f7fc28f7413bbb14db7f70ad7545b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 22:18:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b314fbc085eb475c92e680bcfd17369a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/aa85685d26344dc8a3f7e04154d17741 : FAILURE in 16m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0a4a826b6dea4c1a8445ae5dcf042496 : FAILURE in 14m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/df3c0270ced24b0abcdb51d47716e2bf : FAILURE in 17m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9573a176ab641d19e1ee76715588db2 : SUCCESS in 2m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f99f168d9d6e43c4a1e0dc6522d58de9 : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5b7aa01a37e8425db58ccc8932e28eba : SUCCESS in 4m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/aac85f7c439d44d8a5737d06c512a1e8 : FAILURE in 18m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/77dedc63d4674f44a8702661c0716915 : FAILURE in 48m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6c05ee3487bd6a00abe22d614144f3a04bee7d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 23:08:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab60c99099954c579ad590822a50482a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/16e60107ef3f4e0098b075442daadeca : SUCCESS in 23m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9ac967c6e5b448b6a1f1bab7239d40e6 : SUCCESS in 53m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/672447d69cc1449da4c9bf5cc65cafca : SUCCESS in 13m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0cef382286804c839500d251cf10ac8e : SUCCESS in 1h 00m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/541f5553b0fc4b25821ee4869dc10ee7 : FAILURE in 51m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98ae145e07b64b5ebbe6983dfd7b7c82 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6276d2c43c874baa8002f8a7aef7e876 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2155f070c35c4a438352115d97a596e4 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f791c3b8f20c4f138b76f54ea0b58f06 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4eaa31d758b34bab9683aadd4cabc252 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759e0f18be314ad3bf88884e23f5075e : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9d14b2d5b7aa4c82bd26e1b0ace667d7 : SUCCESS in 17m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6a61eea5b3e14bdd9a17cfaea9f1b51b : SUCCESS in 52m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/45053ef362884448b3c94371363def19 : SUCCESS in 1h 03m 20s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/2f655a6b9af24f5bb160ebdf8a63e8bf : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bfe7451a10af486cbdde3f493278f995 : FAILURE in 54m 17s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/df8ebe66f5f94cc3881e33c8d1dd6d07 : FAILURE in 58m 16s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1aa6755f11d848bda015d2b830585f41 : FAILURE in 38m 00s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/29e43128216c44a8ab4ccabca41e17e6 : SUCCESS in 1h 05m 35s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/5a308787104443578b5ffb189a573050 : SUCCESS in 20m 17s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/04ac346233bc49b8b75ad500a983cc4b : SUCCESS in 59m 09s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e408f85390904bc2badffb3e7ed9d66e : SUCCESS in 1h 18m 10s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/16fef0dd7c404f199cd45ca3d248c76b : SUCCESS in 1h 23m 47s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d0ba765a96ec4d1bbdc6e9f275645302 : SUCCESS in 31m 39s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9c716b838e32409ca16daf3a80706d57 : SUCCESS in 21m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f0b08254d19248a28c70f8abbc8b0946 : SUCCESS in 51m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9aedbc859ac3451d84fe43017f1b8d2b : SUCCESS in 53m 33s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d17abd0904fe495487060a5404c8e332 : SUCCESS in 1h 17m 20s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f813e45809b042c388064fc3997735c5 : SUCCESS in 1h 28m 18s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/425049e794d74d29b0afbc4fb2a07a57 : SUCCESS in 1h 19m 47s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/7c27eb786dee458aad3b1c283868cd3c : SUCCESS in 57m 40s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/22f9e841b56a4cbaa0a650f0f5d4dda9 : SUCCESS in 1h 20m 34s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/9a0ed6aa1be141f6a8174a685b3b6b9b : SUCCESS in 25m 32s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/33c2acf2712a414b8c9e0440c72be5cb : SUCCESS in 56m 41s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9211afdd174849fd856c3fecf4bf282b : SUCCESS in 46m 19s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b2ee311b1bc643e597a7173549446efb : SUCCESS in 1h 12m 43s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d45b391ebaf4454e91eb3463ac117b87 : SUCCESS in 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"03364bdaec7017bc036304488d649b1a1649e45d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 04:21:32.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aade419a1e8c4862a25ddd26b3ab1d85ed9442bc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 04:22:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b2601cc308b26437ab2409e56246f44365e5e861","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 05:35:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/958aeeaa63164fa2bf71cf023d2ee289\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/db92a9a4e2f84984b8eee3bdcf2f637f : FAILURE in 56m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/069d063b4b96478d8eeb0b1e97476756 : FAILURE in 14m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/565a639f15974444a693fe16b1b84925 : FAILURE in 18m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5255c3964b7341caa88d30cbe48445c5 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b55726c7d9094e6183e2187181d766d9 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95830b1ed61648efb0592d7c4d92d2b8 : SUCCESS in 3m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cf42a61c29fb414cb56845d12aa8ff61 : FAILURE in 15m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1be1999b107741b1925f74d1e1ee9024 : FAILURE in 1h 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3c1072411609dcc49370b6677e1f64c69466d6c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 05:49:07.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dbeada80467144e6bf461f2ec787a4cb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/354cc4e00c194867be3175933bd6dfff : SUCCESS in 22m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8034cd376d33406a9d9549818597dec8 : SUCCESS in 51m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ef7078c334db429b80d17bb2d19b471e : SUCCESS in 18m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dccfe8c786c54fd7a3069921479f79da : SUCCESS in 1h 00m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4ab2915997634ea8a97469cf6770142b : SUCCESS in 1h 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/055bb875b0064a13a2add7267802807e : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41aaea719a204efe9c86b2aacbf3e60a : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8058383b1b014ad59985f1583ecd2ea4 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79e36ed8717b4a5f9bab7a88e1185e86 : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39186882b4924b4f8e246f275095ce61 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d85f38b258ad47a59bca03f7410e2d36 : SUCCESS in 4m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3ad9fde60e344dec8a08ec90d2e2629f : SUCCESS in 21m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0529e8f158a946c6937bb507b1ea59b3 : SUCCESS in 54m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/741888e246e0462697c67acee44d768c : SUCCESS in 1h 04m 27s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/50237516653b4ad0a659f257e33d18ef : SUCCESS in 1h 15m 33s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b476ad17d59a417e9a405847e64ff35f : FAILURE in 1h 02m 31s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/5c4eeb57c06b40068234e681fc873269 : FAILURE in 1h 13m 04s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/d55b6132001849c6b9bbbc413be10a03 : FAILURE in 42m 53s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/d999efeeecde48e6bb79ca46541f6f92 : SUCCESS in 1h 12m 04s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/c0923e627b5a42e1913762beee779408 : SUCCESS in 23m 30s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/206e919b945b40daa9acc4d36cc94c66 : SUCCESS in 1h 04m 31s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/35875ea1fc0844e8a3f0e355d797f52f : SUCCESS in 58m 08s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/92cdc17cbfaf4a89a677f2fb30f35162 : SUCCESS in 1h 21m 10s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2e2cd5305ef94502bb667786b8886ab4 : SUCCESS in 26m 34s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/997a1045ef28407d90403ca2f92aada6 : SUCCESS in 19m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c5adf4debe6a499a9fb0517a4f6925d4 : SUCCESS in 48m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2a02cc81e17546939bbb75a7ceb4c021 : FAILURE in 49m 03s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5aefe21ccf0e411994e2593f3745c06d : SUCCESS in 1h 09m 25s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/64a25ca19c1646a68ed0f82b89252cf1 : SUCCESS in 1h 18m 43s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2597bf5790d141deb2ad47194ef48bfb : SUCCESS in 1h 17m 10s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d05b2517d628417ea4be20589f1cb58f : SUCCESS in 55m 56s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/e69f798d7bf043928cdf3583565e989c : SUCCESS in 1h 17m 40s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c179cf6cf0c5428e8bbf410d0d53c948 : SUCCESS in 21m 54s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/296edc4d1be94aabb4a63db6d64f9591 : SUCCESS in 51m 52s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ae289671ea894d848b6f7275ab31140e : SUCCESS in 1h 00m 54s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0ef5250c211d47f2b041f15a45cf5d25 : SUCCESS in 59m 03s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/289be65f374742bab24961a5ed7b29d9 : SUCCESS in 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9df353bae9c39dc1d3710a845b1249b843054c54","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-15 07:31:08.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9359e7b2f829f6ebf19fafa91e56a4edc3d1b2ff","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-15 08:58:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f738b3c9238236b88e1f6b69dae44081dfbdd239","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 10:48:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2b349e938f545cd896b28e67a2972e2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/645b4a37fbd8484a9eb227266d71d985 : FAILURE in 14m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/274f69635aba49a4b76d8a38dd96ee25 : FAILURE in 1h 46m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f250e057d09747be8e6a5b1930f1e261 : FAILURE in 1h 26m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd39e6c26da144f783c9504abe3e9ec3 : SUCCESS in 2m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53ce4a56a7ec454b8215ad869cfe7d6d : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/575a529f0a7745b185fed8b64a565207 : SUCCESS in 2m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fd15ffca941248218fbc13e699989f8b : FAILURE in 1h 38m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d919426ab2554084950142b67b4048c3 : FAILURE in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"31f671b828f89fd531a602420dc0bf820aab73f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 11:51:03.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/edd854a64609469a806f696885b7749e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/61551d7be59444848af78a5bee74a902 : SUCCESS in 28m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f56cc15dc7e3485181ade7b34bd88227 : SUCCESS in 1h 04m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7f7191ad0376499eaa931614a531dbec : FAILURE in 2h 51m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0cb8201adb4440cab1422fc08b82ecec : SUCCESS in 1h 05m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e12f090b99124c319c3ed88bed0a0ec1 : FAILURE in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1a89688fb6141dd9ac887cadeed84d0 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8648bed3c0e4db2b34472cc4936445b : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd8534e975cc4d8dbe2e40de76d4003c : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcb50fc31eae43418a28193ae42015eb : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0885dc6b19434e55806dbbaac2924ca1 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5cf00c562934390875a006045c81639 : SUCCESS in 4m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2e9a7e3003264db7b0667098252f4cfb : SUCCESS in 20m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7b81a769fa7b423fb8ff7b43ab382954 : SUCCESS in 51m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/66278529620f44708a535db80c740362 : SUCCESS in 1h 40m 59s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a5400cfa84b244d89c0fe98be94a0627 : SUCCESS in 1h 37m 49s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/246b7307dcfe46d1bf212eff05695ad7 : FAILURE in 1h 06m 40s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/f855ffacd316428ea88b9a3c7c3abd29 : FAILURE in 57m 09s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1596a0ba50c744c6a741752bd7c0fb02 : FAILURE in 39m 52s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/e111539374cf4d1791b35add1384a0a9 : SUCCESS in 1h 18m 20s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/5b20d9518d194fafbf7757fb5da8cb51 : SUCCESS in 26m 27s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/78803dc71cd84e708199836f7a86027a : SUCCESS in 1h 04m 18s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/dd2241b408da4236b6f3fd7eb43583b4 : SUCCESS in 1h 10m 56s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2aeae9452a36489281abb2c0f705590d : SUCCESS in 1h 20m 21s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dcfce7cdb1b3492a8ce9e1c21d88de06 : SUCCESS in 30m 31s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4df97b78dac2478c80c7452bebe9010b : SUCCESS in 27m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/80aef042174d409e92b616faad60d523 : SUCCESS in 55m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0f848460d5b946f78244294e6ac309cf : SUCCESS in 58m 36s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d40e2bfde07744559087c88c456bbaeb : SUCCESS in 1h 27m 20s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8e91f68b6cdf4d688586e031568772e2 : SUCCESS in 1h 29m 58s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/39ee9e8f099545b4825e8d967cd09029 : SUCCESS in 1h 32m 01s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0daf6697c42d40068fe014de285241aa : SUCCESS in 1h 01m 39s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/c3f0e0217adc4a7697f6c21b98b135e3 : SUCCESS in 1h 12m 39s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/d8619ca8c725475090306b054d063e3c : SUCCESS in 24m 01s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/0bdc3e95706a4fc08842f6ca0325e7e3 : SUCCESS in 1h 01m 29s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/aac2cb5ed29d4517b586e1ad370a83b3 : SUCCESS in 1h 01m 14s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/efeeebdae525414caa0c1aca74fe7de5 : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1023606fc953440695b215af338427b1 : SUCCESS in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3e617512bb582f72d8847f93f435b8a3301e6281","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-15 11:52:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"473a54c3e789b7b71d3ab8ed05dc626f322e17d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 13:08:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/812b8ff0b0d04e7b9a30d82473970ce8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c13e09f9083245e098333444e2d13ebf : FAILURE in 18m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e78d94824a2743af8f00b3257740bfcc : FAILURE in 18m 15s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/afa5978b08794f3a9d7ffba0f2a1eef6 : FAILURE in 17m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f46a0f440d3f4f8dbe86977373c492f5 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48803e1d791844c1862c4af8ed56e469 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0c1a18c4ba9f45848faa82b5f0ef4ef4 : SUCCESS in 4m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d90cef5404254c8481ac71ef2f70d972 : FAILURE in 16m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1d831793409e4e239d04370c68ec2721 : FAILURE in 1h 13m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d86ab2f0522c7374f7ef391db7017650f3c811dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 13:40:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/138f797fdd2741769afe067e3507a95f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bcd4e0c830f74e9d90daad4f9a8f520c : SUCCESS in 26m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4a87e7ddeb754edfaf81c671d65c7e0d : SUCCESS in 1h 14m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b636221f9d7346d7be60e982b995da26 : SUCCESS in 21m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/076f4cd2486344d988662144e8a80c22 : SUCCESS in 1h 02m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c9c7efc0f08a49b78a364593284615a0 : FAILURE in 44m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1654f69437da4143bc26f1054e07d846 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efa697fffb1f480484db09a26afa8c81 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16f04a2ae7b8488aadf88d2c00a7e380 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2336785f4ce4ba5a13ac7255c9d63dc : SUCCESS in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/729d5bdbff114e1596b03c41dd25971c : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da6eb8c24334ee597290cf59014ed36 : SUCCESS in 5m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6267711bb6d64a9794ee043332a28191 : SUCCESS in 19m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6b569280e4734ae6801ba170cfff86ec : SUCCESS in 1h 13m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5f0c246c3b86476b85aa1109186938c1 : SUCCESS in 1h 23m 26s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b7248255461a47b8b4e4257b7c3e628f : SUCCESS in 1h 41m 21s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a0d66420562c4622acbbe4cc9033db0c : FAILURE in 1h 18m 39s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/5e8a064ebda74c8393cea785f9dadc0a : FAILURE in 1h 00m 00s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a34109141e8d45789048ae5ee895860c : FAILURE in 43m 02s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/1adc9c266ce24e44a8626bf782410855 : SUCCESS in 1h 27m 28s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/a1f7b9cacfe142d8860f0419096451ce : SUCCESS in 29m 35s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a2eb8e72611b4b97a8e585c83a45eb3d : SUCCESS in 47m 45s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/3cc02a54604e4ba4a39b03292fc089e6 : SUCCESS in 57m 24s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8e73cfe5671745bc9dbd6d591a28e7e3 : SUCCESS in 1h 15m 46s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d6938816d3dd4256a0efa06b860a890a : SUCCESS in 25m 15s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c1ad2102d67e403ba9b9b95642babe24 : SUCCESS in 22m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/35afcdb582e5484bb1970f4ffb0da0e2 : SUCCESS in 55m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/302954d12dfb42d3988622b7e6bc046a : SUCCESS in 1h 01m 53s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ce86378d19e64489a2456379ea1e9fc8 : SUCCESS in 1h 23m 56s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/953c1acef97c479c94b41ccdd151c167 : SUCCESS in 1h 31m 26s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/12a5cf23b616473d848ec1722039203d : SUCCESS in 1h 22m 47s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9339d93b4ec144458e38bb107e8bbc35 : SUCCESS in 1h 02m 10s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/68e38bb360d847b6ad688ea9104b6bf1 : SUCCESS in 1h 08m 41s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/4270a76bc66d47f7a8212078bf8341d7 : FAILURE in 31m 59s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/c05c837689da4395bf10f611561d2974 : SUCCESS in 53m 14s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/fdcbebbf491045fcb86f34323f36fd9d : SUCCESS in 58m 02s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/74613003c4174da48ff4036bc546fd17 : SUCCESS in 1h 28m 43s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/37c06c6472a547c890cd035ae68d184c : SUCCESS in 28m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dd14466f6fd13b02eb7de9ef554356d016b2de16","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-15 13:45:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"29ee4cd561b3dbf5b6f16e742ddea20581ebda6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 15:17:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a080c457c1d848d39e99bb02fbff7c3d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/89a3a9fabb434245b4f06d28ce5d6d69 : FAILURE in 21m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8ede4daeee0a4ca2bf1fc75ea34537a0 : FAILURE in 14m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/89583601478141e5b4b99ffb044d2894 : FAILURE in 16m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/784050b703674a4da59c853effca2ace : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e18cc3ae25474cbd9e15207aab31ee85 : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f11fb28645de4540a0e86da2d47c3f5a : SUCCESS in 4m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/60ef300b81cf48989c714dd126ee2ee9 : FAILURE in 16m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6a6bebb02c6e4cd6b5e98d29dcda9c14 : FAILURE in 1h 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e1979c177dccd01784538b22dee27d8196401dbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 15:25:37.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/853ff69417d342c7b94641aed102970f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a28a7be8c3154b79b3663d9a8cd77523 : SUCCESS in 17m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6cc6d16e788a425c8e95b20215def58d : SUCCESS in 58m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0a1ada69087541bab3f517c793e6db28 : SUCCESS in 19m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c68abd079eae4c63bd0ea4fb4ee76aeb : FAILURE in 14m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/675a7d1d158140a48db71ce74b8be47b : FAILURE in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b386cfba9c79459b92e808031f8487a7 : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ef75db1a81645a48339751dbe402c0e : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f2c26940a08460097b8b5ff9240f5c9 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c5f333efead41b1b43933612bf71160 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64c8cad71f8c4f0ab95fafdff418543e : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9083165a4d04386a062e573f8bc9272 : SUCCESS in 4m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/02b48d0374e74910991f6cd2826759f2 : SUCCESS in 16m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3b004d379cb548ec8acac53c9417d6bf : SUCCESS in 1h 01m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/63dba1e584cc45778129a6ec970a926e : SUCCESS in 57m 59s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c91e7469e09c4c52ad263551b39ad420 : SUCCESS in 1h 24m 09s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8f76b008991748d3a1d77b5a1b45e400 : FAILURE in 1h 10m 55s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/0d74e6d11c0d4cfc86a448f7853ba215 : FAILURE in 56m 57s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a65d21a1590b4e8083af592c5e9c87d4 : FAILURE in 45m 31s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/2b2279a849344a52a1ef9a1db41314fc : SUCCESS in 1h 12m 03s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3c66504e32e745cca67bab50e710a79d : SUCCESS in 28m 01s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0e909ea509404c44bce9dd3fb3cc2f47 : SUCCESS in 44m 33s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/661ce07a7a8342f1b237283e37805eed : SUCCESS in 59m 24s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d2909ba4edab4a90849beb2cd6457a3b : SUCCESS in 1h 20m 34s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4571634b8d0445e5a35dbd003db572ae : SUCCESS in 33m 29s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5b020b519a5b4b4ea312b600e7e6905f : SUCCESS in 25m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/30e44946c0b64afd8cdd392a1b30924d : SUCCESS in 57m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a85d90548b90427fa7d45442e2741861 : SUCCESS in 1h 06m 52s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/3e035dfb677540e78f2a2183251f91b9 : SUCCESS in 1h 32m 45s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0710ead986d2484b80ce0d3345007e74 : SUCCESS in 1h 21m 45s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/fc6e94fdecc34102a878fd5e1fd17329 : SUCCESS in 1h 32m 05s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/da36f04dea7d488fb52d3a5ff6d9b5ee : SUCCESS in 1h 13m 31s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/49ce5ef8caac4abe8cfff89bac272817 : SUCCESS in 1h 16m 10s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/27477af625784e9bb53c38f1e46d321d : SUCCESS in 27m 38s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/15a5ece044ad4e7397355cb73db11fd8 : SUCCESS in 52m 54s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/410d8622abe14a40bc8ed056614e4687 : SUCCESS in 1h 00m 40s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/357860fb53ee4eaa89e1a3ccc24b94d4 : SUCCESS in 1h 09m 12s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a96a1dfb73f844379afb80da8115f39c : SUCCESS in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"220d708386a069b108031e77fe1c61ac86ff9cee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-15 16:22:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"efdcec719f6af26c049731a5212e3fc555e63827","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 16:42:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/975dc5d9e7714013bebf36ae07ee030b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8caed5266fde4f73934376ca59c412a3 : FAILURE in 18m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b9ca8c4783d04358957815d167b52aa0 : FAILURE in 16m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b43fe66cd168495190e7d4c3400831fe : FAILURE in 18m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/012a04bab8ca4669a9d635c8f06b1227 : SUCCESS in 2m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc08ad27f64a43379bdef6664ca557f3 : SUCCESS in 2m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c325a2ce25f44d0be3fa1a685fc7aa1 : SUCCESS in 4m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/09ecd46553164e4d9c3d7118a354d7a3 : FAILURE in 16m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/98a133c86e6744f293097d3bbd4a0702 : FAILURE in 13m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"884b7220ef0ad1b2b92ee575e24e660bb4cc6b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 18:25:00.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/768cee40c6e64fba91e579a3be870d41\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d8d7c96cc6144a32af003e17d84982e9 : SUCCESS in 22m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e38bcabfcbae403fa7fded996aa6e2af : SUCCESS in 58m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6d2cd6e59f984d1fb0346d3f6be60414 : SUCCESS in 19m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9dd483a38c904baab161ba629c2c5bf5 : SUCCESS in 57m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b754191444d84ee79ba73d7500a3dc82 : SUCCESS in 58m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a83eb9058613417f84af596028759c71 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/114495368fdf4e46918d547cdd9ce1f7 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cb8e7cfc6c948f5966c07347e493958 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2d8386e12e3408e92345bc51d66b6ea : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47206a57943e4274b247e63b4551ed4f : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f804779318af42cea67d73bd9e65ab44 : SUCCESS in 4m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dce28b11e6894333b39521372f600a41 : SUCCESS in 23m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6b8880a551c84d77bbbb58cca08a1e21 : SUCCESS in 54m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/43ebd22b0418491fa356a9356ea4a57d : FAILURE in 10m 37s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/38e8fb4ee65c4fcdb6b09d49d0cf0044 : SUCCESS in 1h 32m 45s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ec93835569204383b5a079b7900ea60a : FAILURE in 58m 17s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/f4f37c03a8c746a1a2f130c01df6c88f : FAILURE in 13m 16s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/aef0ce03877a4043a5d4247249036220 : FAILURE in 44m 39s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c1d4a18f6f3f4cf1b75a8ac5f6ef6479 : SUCCESS in 1h 24m 43s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/ff243c2548dd499681c94c6c0c6485f4 : SUCCESS in 25m 14s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/6805fd65acf24467b8df56a93d7f2e46 : SUCCESS in 58m 43s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/4afa1ee5db044ab7aa95a901b6eecd38 : SUCCESS in 57m 27s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/476f2b9f77564198a443091a57ff98cb : SUCCESS in 1h 18m 06s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9b3a9e10d3c24c07b69deac251a8aba7 : SUCCESS in 31m 56s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d3940564bbd149728e283ada23ee9925 : SUCCESS in 28m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a785d2450e884c069bb725dc1abfe046 : SUCCESS in 55m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/91b457e4663e4b19bab85430bfe3fb51 : SUCCESS in 1h 04m 26s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/254ded3b633e4c53a1127014a8da59e3 : SUCCESS in 1h 29m 14s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0c56ed8b2bb44a10bec13e0c819cc09e : SUCCESS in 1h 39m 39s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/63593f92fcb04f96acbf5ee07fa51b7e : SUCCESS in 1h 26m 38s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1c1eb1ffc68642d1bf1347018035cf46 : SUCCESS in 1h 04m 07s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/c552e5a20eba44a5b4ba7c91b9819a85 : SUCCESS in 1h 11m 08s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ae7d32ef77544d59aa6545e8042537c5 : SUCCESS in 26m 21s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/c94fb63d91dd416d8620dc793f6f396b : SUCCESS in 56m 47s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/974999150ee74e1fbf4f5373a78aa4b1 : SUCCESS in 51m 34s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/cc6b090b9291446c99f3da69a55e7060 : SUCCESS in 1h 04m 03s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/741a59a81d2f4396b0d92dc58c3234e5 : SUCCESS in 28m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"661145b4ae39f09f0d22c41a4823d29d68aa9d51","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-15 19:22:36.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_24072\u003e.","accounts_in_message":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"_revision_number":4},{"id":"9e1cb5d8795964a6ad02f067ba1ce9d4d80bc85e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-15 20:03:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1868a3e98b6f42ee1aac55610346f0e42d995dd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 21:30:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ece5c5dc15df42829fbf91c078734eaf\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c55af43a3f774d83a93000d4cfed070a : FAILURE in 1h 23m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a6673379695442289c2ca5c7891a7e3f : FAILURE in 1h 20m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c2d163b5e7864207a476ce7c5e545fe5 : FAILURE in 21m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e547f47953d34296ac490ca2964c1fee : SUCCESS in 2m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af1ab0565b864b0180c922eccd2cf37a : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d47594f73556451c86eb230f242f27e9 : SUCCESS in 2m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7acc052fb90545fc89e91fc98ebb93f0 : FAILURE in 19m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f4529b70d39e4524a9e867356b81ab77 : FAILURE in 15m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1dda8116decd9360d3c7f1d019ec64d6d72d5e15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 23:23:20.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e02c3ff881246d18f363996a8b7b099\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ab1ba2cce34849d1a4495c1b1df12d5e : SUCCESS in 32m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3fbc827745424aa194296676d17b372a : SUCCESS in 1h 03m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7b8d30badfc94b1b8082a29eccf5e1dd : SUCCESS in 15m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8d5b6fb34aae4344b417b04e54fca690 : SUCCESS in 51m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/71391d2cca5b455eb48b5024706c6b6a : FAILURE in 52m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a514a532fb7a4ac0933b3685ebbaf134 : SUCCESS in 2m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c756f6f7c0384d1d93199bb46accfd8a : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99177309d9eb43f5969a9dbae05033f7 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f79507d0a5a43c0a11b83c3e7fa4e7a : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0b85c25b56f4e5d8c4199168d15f2de : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15df7e9fe4f14b6f967bb1fab9cfe0be : SUCCESS in 4m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a1b9718b6d4245ad80eecfd98d3944c8 : SUCCESS in 21m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e0921c4183f846b985202882d69bfca6 : SUCCESS in 1h 09m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f6f1bdf9fc9b465da2d660374ea02270 : SUCCESS in 1h 10m 13s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/aac3128eed684829ae8bc4832faf9fbd : SUCCESS in 1h 28m 44s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ecd71a2b491f4050b44bc82ef03d9d1a : TIMED_OUT in 2h 31m 55s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/51d3a1b091654c478eac88e10dc3dd7d : FAILURE in 1h 06m 43s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a8dbc6e368474af1a368fd59bc9880ca : FAILURE in 1h 01m 26s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/7689fe3c21af41b1acb4a3bb20521b9a : FAILURE in 31m 35s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/f05153e0766b4c0b97517a1b0a573336 : TIMED_OUT in 2h 01m 31s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/55b0e43d9c9449719cf140993a084638 : SUCCESS in 57m 33s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/8e40a51204164b929b43cc42f93072f5 : SUCCESS in 1h 08m 46s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a329719f8c7e475f965a7d332a323dd9 : SUCCESS in 1h 22m 42s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/123666cd3e3c4d4287ac4dc517e45571 : SUCCESS in 26m 24s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d4f0fed4494d4e72afdd62be65028aaa : SUCCESS in 30m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d58ebe3bea4942b7bb8f52b151830d8b : SUCCESS in 1h 07m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c91b63b3322445c8b291fae35d0531d3 : SUCCESS in 1h 07m 12s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e9cafed206bd4de7b60089afc9db4ef9 : SUCCESS in 1h 22m 13s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f9fc7464de0e488d8c67b98ab4957317 : FAILURE in 1h 43m 10s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2e85ce27445a404eb65eb268eb0dc560 : SUCCESS in 1h 07m 38s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/6293c62ef01b470d8f12bbd7b5fa766f : SUCCESS in 56m 49s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/22bf938e17c848499a431c69485fe0e3 : SUCCESS in 1h 22m 10s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a1f7898a295e4297aece03ebd2b5f13b : SUCCESS in 27m 49s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/fef053202f1a4334969c3087a50f2428 : SUCCESS in 54m 57s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/17814dd35c964607a77ae2a0daa1981f : SUCCESS in 1h 01m 56s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3a790583fd4c4f30a4370b2933672ac2 : SUCCESS in 1h 15m 51s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4f156a3552c349deb0f55c892b04ab24 : SUCCESS in 25m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"54d9086df6e9cc3679580c5b60ce27398cb12ca1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-16 07:49:50.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"056d27e0e78af7ed8fb7cf1ec28ed504c7439543","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-16 07:52:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"125e72c822ad653440a756eb044c07479a7a14ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 09:00:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b1c89ac4a2246bc80f2cb13cf5b85e9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a577d20048244d2dab8f510698a5598f : FAILURE in 1h 08m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ff04eacf7ee5466a99e712a25a2e4f16 : FAILURE in 13m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/42cc2806f0324e4ea7e07228259a52b8 : FAILURE in 20m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d72fa165f7647cc81cd9652dd516c1f : SUCCESS in 2m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d29395ecda3e4785900de9f09a2b459e : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3b4937011b084fc2ae6e4eee9deed269 : SUCCESS in 2m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ec0208f6c74947eb97f839f28a9a8ba9 : FAILURE in 14m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/07c9afeafb2b43f39c0a15e7b0914f2d : FAILURE in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"854454f20e1b2926478f865110b171898425bb4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 10:26:33.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fa5fd2c53534836b7eaab8b1911dd5f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/89ca55334e6d484aaf3cb17a10d8494d : SUCCESS in 21m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f1cc537a784c49278d0890f4d66c2a4b : SUCCESS in 50m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/639fe773c79e4a6c9501a917681d7423 : SUCCESS in 16m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ee982b9cc63e43c1ad92af6dd21476a7 : SUCCESS in 55m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/557692be7525467d80efe81ed298139f : SUCCESS in 55m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8c69693154b40ab908d1dce11436fcb : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3637466176db41d49de56c4a43f815e1 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75b0a94b797a4030a62d6eeac6ab4cd3 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84fd5d72e93c4f2ca9ee24e308a84588 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5edbc910602f4e3d84806c1afc4f48e3 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c8ff5363154c0bb1a920cd474d8d28 : SUCCESS in 4m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f224123bb5b042a5bba745c462ccc6ec : SUCCESS in 13m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b01b0c75856d40d588a413fe65fb92a4 : FAILURE in 1h 33m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a81a7ea77b414ca78f0259b5007aba2b : SUCCESS in 1h 19m 44s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a7fdbb2462d747698af88f6ff8e5ae9f : SUCCESS in 1h 30m 12s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ee1a5d3ce0b146c3bfb98183f979a617 : FAILURE in 1h 13m 59s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/c7626235ce8b4414abddd83ce53dbde6 : FAILURE in 1h 08m 50s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/72062f725ec04025afd0f3c2b52b45df : FAILURE in 36m 07s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/cecfbfb74234409f93b9c958ba2ac20a : SUCCESS in 1h 29m 52s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/335e407d05ac4455a56f6f21dfd636cf : SUCCESS in 21m 28s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/c6b6c60365d0430daf210376a9a1e408 : SUCCESS in 1h 00m 45s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/661b1c0ace604e358892a87a766ec9f1 : SUCCESS in 1h 10m 03s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c8bd351c75154f5eba3747d566ff0ba6 : SUCCESS in 1h 23m 29s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8f0673ca5a644b8fb5e3a75826ea50d4 : SUCCESS in 36m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/63854f023a6c403b9522d17b68f48e7d : SUCCESS in 15m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ff374eb16a834ad0bfe3ba428d3a5eb9 : FAILURE in 1h 31m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3108f65439214aeab231cce175f1b6bb : SUCCESS in 1h 00m 25s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/440b99f9c4cc44c48d03381eeb32c642 : SUCCESS in 1h 19m 02s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/99653f69b6a9437eb0d4c2080cb5598c : TIMED_OUT in 2h 34m 08s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/4d966581ed26451994f51cff8b5f5ecd : SUCCESS in 1h 24m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/99f1ae6313614fd8b7f0d8e8c5c7930b : SUCCESS in 1h 09m 44s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/6dd2fb635baf4975ada8da008695aa87 : SUCCESS in 1h 14m 12s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/88ef716e141d4ae394d88d43cf83cb35 : SUCCESS in 26m 22s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/bf40757473ee4b11836676870b0babc3 : SUCCESS in 56m 14s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c5ab99b352a345a78a822e3fa3e5cebb : SUCCESS in 1h 06m 16s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/91b1d6a3151a4e0884e8810965735f64 : SUCCESS in 1h 11m 07s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/61a72e0074d54ea69643ceefeaea0f68 : SUCCESS in 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5dec72f27c54f00db135122649dbcf351b135aab","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-16 12:30:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"77926a15b0be7b70dafed89c53826ae0e58f6481","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 13:38:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26e352394a2f4df0977ce00a92e8ec67\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0494f136450f4cfbaabac53d70e4ed99 : FAILURE in 1h 05m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/631259a00ab843e994f2e91efff60f7a : FAILURE in 16m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ecfe22cbf1aa4fb5a4a9c44c747c3f5e : FAILURE in 46m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79d99285f06049cab3c5bdf41257aee7 : SUCCESS in 2m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76507d0526704a45955a5c8cd9933505 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0a85d5e5e026492a9e4a7a57047d0d75 : SUCCESS in 3m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/efa9fdca0d9946ef98f9a398e01752b4 : FAILURE in 1h 05m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9dbc141a97bc4f838731d7aa17d81fa9 : FAILURE in 5m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2a662ce0a9cce598db9aae34c7c78a89d6a18c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 15:14:32.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42c0b4a672074f909d1ea1a100feb257\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0e30c32969594d6584aeb792934776b7 : SUCCESS in 28m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2772ca4767bd4b84b9d4718f2673ff78 : SUCCESS in 1h 04m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/48a6be8c392d455cba10f73a9ad5f3ac : SUCCESS in 21m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/68cd5f28897b4e3ebc55f91a168e22e2 : SUCCESS in 56m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/81259aa858894b0388826814fc8738fb : TIMED_OUT in 2h 31m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/867d5aec4ec9477f966e08d559d0ef0c : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffd918adb63b40ca8642e13acfa02c4f : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad9495008c9e47c2bf10a4c88478cf0e : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89ee201be83841e8855c01ea92ccf073 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32c7ffeb29c3474eaa3a9651c026408a : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9955c27bf8f2491a87886cdd0d515433 : SUCCESS in 5m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5bd58822016343c69d4b82013df6223e : SUCCESS in 17m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/72cb54977ad843a7bed3b6e6f038ed9f : SUCCESS in 1h 03m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cbdb5bc5149e46d89379325e75e22121 : SUCCESS in 1h 10m 03s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c642bd702b7844a28228eeb8972084ea : SUCCESS in 1h 22m 58s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5e64a02661304d8f89f58602cb22f253 : FAILURE in 1h 06m 53s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/ec1383e666924875a9da1c5576d48742 : TIMED_OUT in 2h 03m 46s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/192f40ddd98d4b118d9522ba7dbf2ad5 : FAILURE in 46m 51s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/d901bd07739c4541be5b6c3c6d98008a : SUCCESS in 1h 17m 24s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/94f05819ced141fbb9cf544690302af6 : SUCCESS in 27m 19s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/ee768cfe9a204743b2137fe2c14f4af0 : SUCCESS in 1h 05m 36s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/1a3bc2f2764a4333ad2169dd64f63c3c : SUCCESS in 1h 09m 04s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/17d0d40d91274113a2f8e82d475eefcf : SUCCESS in 1h 23m 53s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/669c43af25454d6f8620a1fdaf968591 : SUCCESS in 32m 29s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/653d1e71b79e44de8c671c8160f3a18a : SUCCESS in 28m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/337609895aa54fcfa7136080573bccdc : SUCCESS in 56m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ed78f2c3c31b4636aeb89c15e1f0edad : SUCCESS in 1h 07m 00s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/fc9b7128cb6d4054933607215cafdfb0 : SUCCESS in 1h 18m 57s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/00f5f3715b19424a99fa3bf945dd063f : SUCCESS in 1h 30m 39s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/c0a81aa7cef74cccb2ec9a04ab94902b : SUCCESS in 1h 20m 58s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8b5f0c88dd6548598f4ffd9522ba7b5e : SUCCESS in 1h 04m 22s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2f4f2049eff04f92bc2c2c98ae279ffd : SUCCESS in 1h 10m 06s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/4888f06c25d94b85872dab932fc6cc07 : SUCCESS in 24m 23s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5ebecb17771e4b6cb3ecd516f8589a61 : FAILURE in 40m 19s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b41f1b1610cb4371b51a71aa083de088 : TIMED_OUT in 2h 01m 10s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9704019ab7fc42d699c7bd7b244d2e3b : SUCCESS in 1h 05m 58s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/08ef07b7b2b34fef9ab25884a9185b54 : SUCCESS in 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f7fbb5ae158bdbac80b6442ab1416a2dcad38602","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-16 15:33:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7baa0bf39a0b12985f8162a8322cf479c32242b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 15:53:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6af8f4294b354b0b9e2481c7b21c11c2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/755c6589af0b4076aecfabbf44718966 : FAILURE in 18m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/07df9595ce62490cbf6847901c1d2907 : FAILURE in 15m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c9916316839c41d09507ce940bb19b86 : FAILURE in 17m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6beb42590eb444beb8da4176b2763361 : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a711266a751a4cc2a029db79fc781356 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8d77ae0a012c4bad9eded5a13e7f97c7 : SUCCESS in 2m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f21f270521ba49b8bf8a98e894748cfd : FAILURE in 15m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1b74408788734b9fbb34a05445c1dd2c : FAILURE in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"18c9c2d0f37ee8c349c045f9e5e3e7b4a013a54f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 17:45:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f46cd851b8204fed94b185a9548cfd0b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8a356604ae29401da7f4c1aa590a3187 : SUCCESS in 23m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/14831c0e0ec247e9830b8664fbc9a6b2 : SUCCESS in 1h 05m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ed8718ba552c44a0a35c078fc9505015 : SUCCESS in 12m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ff9d73a7c9464e25804cba6f00d07780 : SUCCESS in 1h 01m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d16fad95392b4348b40881c30048cf3c : SUCCESS in 1h 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13391bfb35d9471ba15bbbc238d18a9d : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/890f700333de4eebb02d46ec801791e5 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d05e60176944c448b9468fbcc281efd : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/635d2d285a65433e94e828d57e66db2d : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37896f4bc2ca475bb7ca0d50ed6b0f7d : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d18baaaf3e634ea99a71b0855cc1f926 : SUCCESS in 5m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8245fe55399d417c831fb8bdb83d57e7 : SUCCESS in 19m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6d0df054d57d4050bcffa9aa7e82a848 : SUCCESS in 1h 02m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6836e5564139482b8e2fa2c4f444b5f6 : SUCCESS in 53m 09s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/86079a41483e4135a00d5d501d29e8cb : SUCCESS in 1h 37m 13s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/47617bbbefa54293a93829c758eeef02 : FAILURE in 1h 11m 41s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/84480f9ad4074b9fbe01e296dddd3e3d : FAILURE in 1h 11m 42s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ce7de1b673b847f797fcee37908ac97f : FAILURE in 40m 33s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/e7fb99b065e248cc80699f470b69e226 : SUCCESS in 1h 28m 44s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/065e6be846ec4c39aa261be7788f1791 : SUCCESS in 26m 52s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/ec04d705520b4a7eb74ffd7c671de729 : SUCCESS in 49m 15s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a05d094b65014182a28840657d58480f : SUCCESS in 1h 14m 06s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/abe9a60917014d25a61d83eaeb076e39 : SUCCESS in 1h 23m 41s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d11317c22e0048cdbf92016ef66f814a : SUCCESS in 34m 39s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bc4b77eb7557455c899c214d26d9f01e : SUCCESS in 22m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/577bfea4d1d340b2bd89fffc7942cdb2 : SUCCESS in 1h 03m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1f8e35c36d8648ac86b967a573ea1b3a : SUCCESS in 1h 10m 10s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a1836ee630cc4e8e814939b4a9f47e95 : SUCCESS in 1h 36m 59s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8ab3cff84a1240d9b00a1f7b35674566 : SUCCESS in 1h 49m 12s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/8c161008a70542f8acee8ce1c2ef16ca : SUCCESS in 1h 33m 37s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ff8eae834e2342e1a86396bef9194dac : SUCCESS in 1h 19m 18s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/e7ce1b60aea24405b0db7db76a024bd6 : SUCCESS in 1h 19m 33s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7e01e8f3e5484d59b219570cb934a75b : SUCCESS in 28m 30s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/d9291fdae024461daae5285aa276ec88 : SUCCESS in 1h 04m 24s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/563c3119b02845269ed5555b3153578b : SUCCESS in 1h 07m 02s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/62638daf9fee4409877d4194bab8df57 : SUCCESS in 1h 22m 20s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4bc413b9683e4c6ab4c0dd94511d3775 : SUCCESS in 1h 09m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a2d93b7c76540b5f80f44ae8d554b08f89ec54f2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-16 18:04:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a664054069ce108801ea0180320a72f322e2cabb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 18:05:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ccf4fd57376bc2db524111f0cdde6c336df39cdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 19:58:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/972e47e557bb475e98165f9f4fe89a07\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/69aaf0f0f17146f7a2e494252bd79225 : SUCCESS in 21m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f69443d430004c4fbd5aafd2d382a325 : SUCCESS in 1h 06m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d3837ffe6aa24aeba72f268a86d1ea54 : SUCCESS in 1h 46m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/789091b8511b4011a7a1552bfa1b9c31 : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39138181f19b440799d2cf25cf4a9422 : SUCCESS in 2m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e711b43c56bd4d5d9f1d2c764fd44f30 : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17dea8b7256c4ed7ac0196228e6f2c2c : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f2260bb99334c6d9df22acd7d0fd533 : SUCCESS in 3m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d0f12f30375e4bf98285da33b4b9c6b5 : SUCCESS in 21m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2a60e1ec31dd4d92834bc760f54facfd : SUCCESS in 45m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/da309687ae7647969b37e545da4c28aa : SUCCESS in 54m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/36acb53990bf4bd9a16dee96fa2bcafb : SUCCESS in 17m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0475a4f1d4df4c73aca11b92d36c6199 : SUCCESS in 47m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ed3a86f5957f4f76ad0531732b75054f : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":4},{"id":"12fcf68768b93e1a44031a4bd0c316af871ea8a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 19:58:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"406240d0824a43401c4d335a02cd6cb5e0ef83e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 19:59:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e7a16e2aa2e4652b7779396d2eb877c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2158f9086ba486c8fb5c0f7fa7d1be9 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4950c03e59b8d1af46018b81a49fa028e1b8296d","revisions":{"a2867f5053f35d12ae8b92bd695b150cbbdcb2b3":{"kind":"REWORK","_number":1,"created":"2023-06-12 10:15:01.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/57/885857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/885857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/885857/1"}}},"commit":{"parents":[{"commit":"9035251443ed6cbb110e9dc5efdbc9d4be797ef5","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9035251443ed6cbb110e9dc5efdbc9d4be797ef5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-12 10:12:43.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-12 10:14:59.000000000","tz":120},"subject":"Fix rabbitmq image build, pin rabbitmq 3.12.0 and erlang to 1.26.*","message":"Fix rabbitmq image build, pin rabbitmq 3.12.0 and erlang to 1.26.*\n\nBecause of launchpad ppa release new version of erlang [1], it\u0027s\nneeded to bump rabbitmq version to 3.12.0 and also erlang-base.\n\n[1] erlang-base_26.0.1-1rmq1ppa1~ubuntu22.04.1_amd64.deb\t2023-06-09 15:53\t10M\n\nChange-Id: Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2867f5053f35d12ae8b92bd695b150cbbdcb2b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2867f5053f35d12ae8b92bd695b150cbbdcb2b3"}]},"branch":"refs/heads/master"},"d70dd6d75b483d098507f6fbf08ddb1a4c8e939b":{"kind":"REWORK","_number":2,"created":"2023-06-12 11:05:29.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/57/885857/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/885857/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/885857/2"}}},"commit":{"parents":[{"commit":"9035251443ed6cbb110e9dc5efdbc9d4be797ef5","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9035251443ed6cbb110e9dc5efdbc9d4be797ef5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-12 10:12:43.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-12 11:05:20.000000000","tz":120},"subject":"Fix rabbitmq image build, pin rabbitmq 3.12.0 and erlang to 1.26.*","message":"Fix rabbitmq image build, pin rabbitmq 3.12.0 and erlang to 1.26.*\n\nBecause of launchpad ppa release new version of erlang [1], it\u0027s\nneeded to bump rabbitmq version to 3.12.0 and also erlang-base.\n\n[1] erlang-base_26.0.1-1rmq1ppa1~ubuntu22.04.1_amd64.deb\t2023-06-09 15:53\t10M\n\nChange-Id: Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d70dd6d75b483d098507f6fbf08ddb1a4c8e939b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d70dd6d75b483d098507f6fbf08ddb1a4c8e939b"}]},"branch":"refs/heads/master"},"aa913ecc34dda3883948e87a6e0e8167ee485c5e":{"kind":"REWORK","_number":3,"created":"2023-06-14 10:17:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/57/885857/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/885857/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/885857/3"}}},"commit":{"parents":[{"commit":"9035251443ed6cbb110e9dc5efdbc9d4be797ef5","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9035251443ed6cbb110e9dc5efdbc9d4be797ef5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 10:14:02.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 10:17:16.000000000","tz":120},"subject":"Fix rabbitmq image build","message":"Fix rabbitmq image build\n\nThis patch is replacing ppa.launchpadcontent.net\nby ppa1.novemberain.com.\nThis repo is based on the official Erlang/OTP package\nfor Debian. The RabbitMQ team adapted the upstream package\nto produce packages for more/different combinations of\nErlang/OTP, Debian and Ubuntu releases.\n\nChange-Id: Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa913ecc34dda3883948e87a6e0e8167ee485c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa913ecc34dda3883948e87a6e0e8167ee485c5e"}]},"branch":"refs/heads/master"},"4950c03e59b8d1af46018b81a49fa028e1b8296d":{"kind":"REWORK","_number":4,"created":"2023-06-14 21:28:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/885857/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/885857/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/885857/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/885857/4"}}},"commit":{"parents":[{"commit":"9035251443ed6cbb110e9dc5efdbc9d4be797ef5","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9035251443ed6cbb110e9dc5efdbc9d4be797ef5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 10:14:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-14 21:28:05.000000000","tz":0},"subject":"Fix rabbitmq image build","message":"Fix rabbitmq image build\n\nThis patch is replacing ppa.launchpadcontent.net\nby ppa1.novemberain.com.\n\nReason for replacement is that only latest erlang version\nis available on the current ppa source.\n\nSetting Debian ARM jobs non voting for now, since the new ppa\ndoes not contain arm64 packages.\n\nThis repo is based on the official Erlang/OTP package\nfor Debian. The RabbitMQ team adapted the upstream package\nto produce packages for more/different combinations of\nErlang/OTP, Debian and Ubuntu releases.\n\nChange-Id: Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4950c03e59b8d1af46018b81a49fa028e1b8296d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4950c03e59b8d1af46018b81a49fa028e1b8296d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
