)]}'
{"id":"openstack%2Fkolla~886088","triplet_id":"openstack%2Fkolla~master~I45b5a5980fb2318f1aab23a3a6aaf0922563b171","project":"openstack/kolla","branch":"master","topic":"for-debian-bookworm-upgrade","attention_set":{},"removed_from_attention_set":{"34579":{"account":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"last_update":"2023-08-21 14:26:49.000000000","reason":"\u003cGERRIT_ACCOUNT_34579\u003e replied on the change","reason_account":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-08-21 15:11:35.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-21 15:11:35.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-08-21 15:11:35.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-18 07:40:42.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-08-21 09:22:14.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-08-21 15:11:35.000000000","reason":"Change was submitted"},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2023-08-21 15:11:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I45b5a5980fb2318f1aab23a3a6aaf0922563b171","subject":"Move to Debian 12 \u0027bookworm\u0027","status":"MERGED","created":"2023-06-14 12:07:49.000000000","updated":"2023-08-21 15:13:53.000000000","submitted":"2023-08-21 15:11:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":37,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"886088-for-debian-bookworm-upgrade","meta_rev_id":"272d31df3164e000dd1042aef7ce450c39b20acd","_number":886088,"virtual_id_number":886088,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-21 15:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"all":[{"value":1,"date":"2023-08-21 14:26:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"value":2,"date":"2023-08-18 13:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-18 07:40:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-08-21 14:06:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"value":1,"date":"2023-08-18 13:49:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"value":1,"date":"2023-08-18 13:49:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-14 12:35:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-14 12:39:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-14 14:39:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-22 10:47:55.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-06-28 15:50:04.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-07-10 12:37:42.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-07-13 01:26:28.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-08-16 13:00:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-08-16 14:31:36.000000000","updated_by":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"reviewer":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"state":"REVIEWER"},{"updated":"2023-08-17 07:04:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-17 07:11:06.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"c7499ac31ab7ef648c8082fd0d1422d91c09731a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-14 12:07:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8da763db2e26df2fd6678d08007560d493009a61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 12:35:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/618c1ae1ef0f4a71a1fb01ef6b1b55ce\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5ee16abc702c4808906a7b25fdb66300 : FAILURE in 24m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ba6308469f434a4c9be5c51ea8dfcacf : NODE_FAILURE Node request 300-0021458795 failed in 0s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a89174e44ab3448bbff2746cc8ceb3c3 : NODE_FAILURE Node request 300-0021458796 failed in 0s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77e535f841f146d8bc76d94c747b2a30 : NODE_FAILURE Node request 300-0021458797 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18d8b0e3676e432a81469d4350ee37a3 : FAILURE in 3m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b09ac9d3bf05499cb4fef689718e5f93 : NODE_FAILURE Node request 300-0021458799 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/72682335006a4156ac4424fb72edc824 : NODE_FAILURE Node request 300-0021458800 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b7112cdfe83e45669f0b8e1e3b482505 : NODE_FAILURE Node request 300-0021458802 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d45ea88264b2bb5d85d58b1d3a78721b7cda058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 14:39:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/16fb327f5c9c4cd5b2d43619ebf7992b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/741858fcd1f04442811179916962adb7 : FAILURE in 24m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f5c5bc52a2894bd78bde99ac81a4fee4 : FAILURE in 46m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ad16477b4cce4912ad56446f0f8a80b1 : FAILURE in 4m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/609061d50f1a48e88a7fb92962cbe048 : FAILURE in 12m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cfb005661e8b406891fbc50e5e9add1b : FAILURE in 17m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3884d95b7af741b6a2ae1de3998ac867 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22a9e2d7a9644e99879b8ad673201cb9 : FAILURE in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b59e9d08beb43348d86d5779a4b227e : FAILURE in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06a5b3a997a0478e897d29973d396b3f : FAILURE in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8180a23a78214ba0b87e512ced542ece : FAILURE in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b0c5bcb87445c78283f181e29d5fbb : SUCCESS in 8m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6d396e493d98490cb9b9a4bfd43f0685 : FAILURE in 24m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dcb4b2ea55d24e8aacb10aab4be4ad7e : SUCCESS in 49m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dcb2da44b1784416bdfa0b84178f33ea : FAILURE in 49m 26s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0713d159449a42eeabbefa13b2e097e5 : SUCCESS in 1h 32m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b979d685e71648cb89c94fbda45e4f0c : FAILURE in 1h 05m 10s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/eb756ff609e04341be8082b890f42e64 : FAILURE in 59m 59s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/676b85f3039b49ecad2603f723157140 : TIMED_OUT in 2h 02m 33s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/41fea5d7c5d64adb96c5670271de6e64 : SUCCESS in 1h 24m 22s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/874230a008754672a93ba78397f065dc : SUCCESS in 48m 35s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9a8eccdca7eb4ff9936888c114b87601 : SUCCESS in 58m 56s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e566be7b464b4749a00d3fe8334b6713 : SUCCESS in 1h 02m 02s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a01f5101b869401789eee99243dd213e : TIMED_OUT in 2h 02m 44s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4a9f2e39a94948999c8d6648f62f8194 : SUCCESS in 34m 13s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0f8737c9c94a4bc7af400baf77a723c5 : FAILURE in 24m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c09f99dda47a4dc69cfdc516673861b4 : FAILURE in 17m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1ec9c20dbc084c2c878d5aea18ffe5fc : FAILURE in 18m 26s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/84ad54ad395c4336981187c266bbc75c : FAILURE in 17m 56s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/885a141f9c4047a09e202fbadbe24496 : FAILURE in 17m 47s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/38f5c6b8077c403c926f531ae56cf0fc : FAILURE in 19m 33s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/7fadb250f8b44208ab75d46f8d54841c : FAILURE in 18m 32s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/6edecb22dc924cd3bb7c7276dd4baff6 : FAILURE in 18m 44s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/fe4f236ee38149bfa81a150ac68ef914 : FAILURE in 11m 47s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5b4b3ddf4367445a80a2fd0445b502c3 : FAILURE in 18m 15s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/0ae55a3577434eb4a3cdd14d16fc1330 : FAILURE in 14m 41s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c1c58582c0984d62abf3564b74b1865b : FAILURE in 18m 50s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c533b9163653474cbcd27289d3697b37 : FAILURE in 13m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8dc346224b0f49fae64c0c720c27ca9086b175b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-22 09:39:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"45898fe5b1f84e1b9604dcb5ec00462e5e9e5e5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 09:48:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2084e6969d324c8f9a05c8271b09f01f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7c725351f3da4573bcd44e637e4d82e2 : NODE_FAILURE Node request 300-0021515867 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e7e66fad23754ce09cdca1612238ad2a : NODE_FAILURE Node request 300-0021515868 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a344f164d8414af0a4b534812e632bf2 : NODE_FAILURE Node request 300-0021515869 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a39969a4e01845d0b74f210a7eb1352e : NODE_FAILURE Node request 300-0021515870 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66a3800067574d489f29f578b0880133 : NODE_FAILURE Node request 300-0021515871 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f7836682bf9e4bb9927fb8fbc111fc75 : NODE_FAILURE Node request 300-0021515872 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f700cb3cb0d149559eea6dbec2d1347c : NODE_FAILURE Node request 300-0021515873 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d8482b7b63bc4c24b610763210a5b5e9 : NODE_FAILURE Node request 300-0021515874 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a2aadebf3f9fe4a7727a6616c188f6d381c1486","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-22 09:50:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fe47099e334c3dd24eec6cc314798854b4b5c43a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-22 09:57:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"53ed7cc7175425dd46d438ceabc0e3df6e9650fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 10:02:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91529add9b154889892cff1b9ade02e6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/26e5e7cb48fc4f08a1cfb4f9ac6aeb37 : NODE_FAILURE Node request 300-0021516363 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/425eccb83935465cb75dabfae2178c01 : NODE_FAILURE Node request 300-0021516364 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ba8b544367a64addb685a1c6878a5d9a : NODE_FAILURE Node request 300-0021516365 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c89d19018ac14a5a821d4f1298cf0222 : NODE_FAILURE Node request 300-0021516366 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c064b12899b4eefb0fcd24f430a935e : FAILURE in 3m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec4b1a616acf443bbdaa9ec6c0804ef3 : NODE_FAILURE Node request 300-0021516368 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a213d06451b348d4b0175864ba31b6c6 : NODE_FAILURE Node request 300-0021516369 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b12a0c809c39410ab136a79a109274e2 : NODE_FAILURE Node request 300-0021516370 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4c68573bc5a2ce8aa8947a50b504bee974241398","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-22 10:46:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8fbf31e24bbca75f9d9e41d09610d7e6dfedffee","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-22 10:47:55.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"788ec22bc88617b625944d7eb5362b73e3abe9f6","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-22 10:49:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4c1b7733da48c7ea038a2b8e53d1810672a2e92e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-22 10:58:54.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3a8273cf5831a29b3182245e4c672e6de86a5ca6","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-22 11:01:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5d3fc647537e3c08741154456ac6fe88649cad6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 11:09:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39c03fbc29e04465923a1d9c27ebead8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a93821806eea448e80c9cd71f4519d83 : FAILURE in 19m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/983092810f7e4583b029fc761d3b1edb : FAILURE in 21m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/678265b8f8214b4f9f00223bd301e67f : FAILURE in 21m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3b86edefbdc4f4c8e92f0f7995d9e01 : FAILURE in 2m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9d41d7f4a2f40c7a892c0d164fda2f3 : FAILURE in 4m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a320f1402a1f405ba686dae9b0cd0fa5 : FAILURE in 2m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/50a5fd7251a149b69fb9a9364d72003b : FAILURE in 20m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c5eb97824fc544d3a735f8d0414941c6 : FAILURE in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ecc2d8f1674d27f96e047a4aaf42a0b7627390a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 13:53:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6df52199e6bb4fb5a14ddbe2184458bd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ac706d0290904e5aa66dd7325151a835 : SUCCESS in 22m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/315aebe39af3473b970214f0322f69a9 : SUCCESS in 1h 02m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c534f2bdc4724369bf75f61d95023423 : FAILURE in 24m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/92c064d51a7c4c5aa9e8df0d8a6e45a9 : FAILURE in 25m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/85e18f9f35fe4b058b797d3f7280489b : FAILURE in 26m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9335c9aa90584da8b76cef6f3d200ff1 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79e7241fe76a42b1abbcb59353ea5112 : FAILURE in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7891a8bcb8d41c9a564ee07b24b5011 : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52aa6e4b51774759a94ff11773721983 : FAILURE in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a1f68175f8e40f4952534ccc45abf7f : FAILURE in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057b333a4fb34e8983c4d01b505a483d : SUCCESS in 4m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f70b34feb29644de9b5bcc534c722be5 : SUCCESS in 15m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bba4b921deab4bc8be8da52a344680a3 : SUCCESS in 1h 15m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7d75c5a42a034f7cab2969d9926acec9 : SUCCESS in 1h 17m 40s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b05327b81b1345e4ae0266289e78a55a : SUCCESS in 1h 34m 10s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/46ba52f93ae647e1b24c61351a179e43 : FAILURE in 1h 11m 31s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/257a40e4d3d74350a573bde1b6170b8c : FAILURE in 1h 12m 55s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f4054d3793ed4c4db5d41757d410924f : FAILURE in 55m 50s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/9fe45be7f2304f9eb6397e7433043a95 : SUCCESS in 1h 32m 40s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/0857ed9feb9c48eba76102c0429018a1 : SUCCESS in 24m 54s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/fd1076ae7cda487d9438152f9ec3e6cb : SUCCESS in 1h 11m 42s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/4f6bc960715e494ca609826f14867cdc : SUCCESS in 1h 07m 05s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/98701fad143c4537b4e6a7ca6d71b151 : SUCCESS in 1h 38m 22s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3bba69cf72634de593fed106f14839bf : SUCCESS in 32m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/13fc88b04dd6419d979f1b78d5c9652e : SUCCESS in 25m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fe1c2e72d7b644329e42d251552597cc : SUCCESS in 59m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/961ce5d44d3949158a9b59d3c1f73a02 : SUCCESS in 56m 00s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/fec3c0726efc4af19e824d2c3bb872d2 : SUCCESS in 1h 19m 34s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/76b1994cf1e2477484311a0859579c99 : SUCCESS in 2h 00m 50s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/507c0b3aa7794a96bc2d883097f8883e : SUCCESS in 1h 47m 06s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/57a9e9046e414b2289946576677d218e : SUCCESS in 1h 24m 40s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/e51a510ac57b4851b7c6813876064569 : SUCCESS in 1h 31m 02s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a30227077a1f4a968390bb5a14bbde0d : SUCCESS in 23m 09s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/0933974b3c0e4401bc76b35c00229d65 : SUCCESS in 1h 05m 41s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6908c6ebda24453d877d943d5bafd28f : SUCCESS in 1h 19m 52s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/28db3c92a36f480e83610bbaa4e7299c : SUCCESS in 1h 17m 37s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/106915a524b24cac89b0deb18ed8351e : SUCCESS in 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"43377705f4f4c078761c630b888f501a2a789822","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-28 13:17:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8bedc6a19092fbf5aec890f99f161028ba3cc7fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-28 13:25:12.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1822125ca682f932f6f5b6b105fba84d88d2c8b1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-28 13:25:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8cec5ad0e55ed4220017758058ed8c187fb42a27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 13:46:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbe5e8638dd042b1a6c75dcd18dc5e0a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/caa16ef2ba9342fcac5ea2fe85c52005 : NODE_FAILURE Node request 300-0021565244 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/01bab145419447c6b4c8d0f6e91d8f1a : NODE_FAILURE Node request 300-0021565245 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3759a220a32c46f382e5d88e293e147f : NODE_FAILURE Node request 300-0021565246 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a874a2e37fb4e26829b1e17387e6f90 : NODE_FAILURE Node request 300-0021565247 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3dcdf2b480e4e0e9e5ba2d4b0551f83 : NODE_FAILURE Node request 300-0021565248 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7811b360847468e9a63318bd23274f8 : FAILURE in 3m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3b8dbcf3a22840c7870fa5501da9e4d3 : NODE_FAILURE Node request 300-0021565250 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d530d3a37cc143ea91cf90874b9ef433 : FAILURE in 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9449bc0c60912e7e5139749d3d5f9f05b01fbb7e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-28 15:50:04.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"87260b2d13f8a52e94f5e63f4db3ba628a6106a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 16:13:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/219a1ddc30f04120bb361df5b4a9e71d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/37d3046f29354267af2855d5ff63a257 : SUCCESS in 18m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/95625f5620ff461f94a55c5153cdd04f : SUCCESS in 1h 01m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bc69cd85b0184edc971fa4c4ee7ae6c3 : SUCCESS in 21m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d3bc05792269422fa817463421de6a00 : FAILURE in 40m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d8663a3c92e641578417bd8d7a694166 : FAILURE in 1h 01m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b97a881e0fe4acdbf2fa78b888e6d01 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51be129cd41749ae92e6eeba7ee0c826 : FAILURE in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b29d77b75004fffa40b84f47e20f9ea : FAILURE in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f7fa6a1065148e093a2f70e721a6c9e : FAILURE in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94922424e4fc416eab93426583ee5ac8 : FAILURE in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6fc53338c1248d79e2223108c04f443 : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/04bcd03ae9ec4d13a8ea2ef837565b55 : SUCCESS in 18m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/597eb624e59b4d0493d55322aa1a2b29 : SUCCESS in 56m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/69d5597d11834153ba2da1b141caf855 : SUCCESS in 1h 19m 01s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/644d2f9f2dc349588e3bd305d677b787 : FAILURE in 20m 59s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/316892d2f9eb4dce8fa4fd609d9e6459 : FAILURE in 16m 42s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/81c7ab32525244cda5ce61c34c303567 : FAILURE in 18m 45s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c6c5425498614cd38cd9fdf2260bdc64 : FAILURE in 18m 18s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/5fd22e6c2757418882f765933cbb2e21 : FAILURE in 18m 09s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/97d8e95980624280ac9e1667e48a0ef3 : SUCCESS in 24m 46s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/1ab20ab20a3b4dbaa3ddb5377eb63bff : SUCCESS in 52m 26s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b3bacc9832ae4c45a3780f1d27fd7175 : SUCCESS in 1h 15m 38s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/f5076b99255847f3ad574537e091fe7b : FAILURE in 53m 46s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0c26d8c7e85248e6aad5cba4d5ed5d77 : SUCCESS in 31m 57s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ad87b82513104896bfa0ebdb0af24246 : SUCCESS in 18m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/07428837245145eb9f96d49de01a230d : SUCCESS in 1h 02m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/87ee25a12f1f47dd9decdd9d3f57b9e9 : SUCCESS in 1h 11m 45s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/167b1685f76e48f3b8a96553b29a687e : SUCCESS in 1h 21m 23s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/180bd21691134882aadd530d26141f93 : TIMED_OUT in 2h 34m 05s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/02086c9990484e55b43632e6b87d17ff : SUCCESS in 2h 02m 58s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1c9bfe7c697046a6a84ae5ce277ae591 : SUCCESS in 1h 09m 27s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/ed81eb130775481c9e492bfdbed2ee99 : SUCCESS in 1h 37m 02s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f4d405af02004bbc900abc5528dbb170 : SUCCESS in 38m 38s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/3ac2b3c39fbb42898dd568786802443a : SUCCESS in 1h 21m 13s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ce7d6bbdb6a544bc838b2c56fa5928d1 : SUCCESS in 59m 34s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/437a1070cd734654bd87f62d4b69b327 : FAILURE in 1h 17m 42s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fd23d1623fcd4ac8be10a4870066e83e : SUCCESS in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fe2f005819921ec93fd441f7638630a37e52e6d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-28 16:17:51.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"9abb3ce9790c8fd948c5951ec1aac897c19de347","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 16:18:13.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 886088,7","accounts_in_message":[],"_revision_number":7},{"id":"332fdcde1312c8d66b4c5cbc142d37110d233f04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 16:18:13.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 886088,7","accounts_in_message":[],"_revision_number":7},{"id":"21453087fcf21101dec76cd3a96aa869a0b25b1b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-28 16:18:17.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7ccaec731acc7f6c886d436b7b01a9494a2eccd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-28 16:21:04.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"a722e036def5fd93bcc9e068ea76b59918b9d05e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-28 16:27:22.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"f9beb80b2bbcc0cb548613bede409e750d21c59a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-28 16:34:19.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f6a330ec61610499af9220cf6122088c928d98a4","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-28 16:42:47.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d6ab5b2efd4fb4150f524e6a756d7597167bd16a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-28 16:54:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b96414fb52ed61eb67bae7ba833780c01aa802de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 17:46:52.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7955695431d4575976df7f03beafd5c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b4a0150e2df0453f8bfa7618bb30b40c : NODE_FAILURE Node request 300-0021566762 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5705036c9fcd4587afb5c5a0f55ece83 : SUCCESS in 28m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/46b411552ac84791aa6939ab07370132 : FAILURE in 44m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/310ac4d67f1240f9a05d7391e144311a : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3f2a93c0df1433c9cd62ce8ac322ad0 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/40eae05168f249be968b3ba36bdf6bdd : SUCCESS in 3m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e8c9da51389641598054455543961dcd : SUCCESS in 1h 10m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9f2abb1b528946c9be4c8e6493bff315 : NODE_FAILURE Node request 300-0021566769 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"30dd6257915e275406fad346433a18e5565aec05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 19:12:55.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3b3ae43c7e54741bfce7a52e14fa9db\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c42cdfd307c54c2981acb7a4b65eebb8 : SUCCESS in 25m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/27f4ee3d768e447e9d8bd3d7668d52f8 : SUCCESS in 1h 18m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/518af7c997c241ea98de8eeb71e7a9b8 : SUCCESS in 27m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1178321b333a45359838ac3fd23394a5 : FAILURE in 33m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/785a2912010e43f2b98c42ee939feac3 : FAILURE in 1h 13m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e857f40b0f7b48b48701be6ce39015d2 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba7faf44e0004de4983e4eedf7221f68 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f77f9306cdb4c368b2eaf9a305c87d5 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd0a7f613ea54a9ca41662b47b76111b : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ae56a1f1fac48ee851e511bb6f8b65d : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d6dd5f112047c2adcb82bf3d493ca9 : SUCCESS in 5m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/957836c9272f486db57b76ac8a4be082 : SUCCESS in 19m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1a60d143e07e4e989217c7c352ff45b8 : SUCCESS in 1h 12m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/96ec3edf8956469d8d8acb900c74c850 : FAILURE in 29m 23s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/52083255cce8459187c1eb24aed6480a : SUCCESS in 1h 33m 45s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/97e9c6d82dfe4b1ca47131555d6c6ff4 : FAILURE in 18m 36s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/87265af4f34a4b6fb27995d28c9b9c8f : FAILURE in 18m 28s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/9947e65efd0646ce910d71d32a1f8f56 : FAILURE in 22m 20s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/d3be4356fdb34fb5b5e6b0f0b3626756 : FAILURE in 32m 32s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/02b44c923d154c87b4382f55fd629010 : FAILURE in 21m 37s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/e55bb4d6baba44fcbee75277d179fdda : FAILURE in 32m 30s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d53dc69c92f34c169aab9dd97c7b128f : FAILURE in 17m 27s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/30d4d057f6484da49a5c341f68a08d36 : FAILURE in 35m 52s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1c21105b9a3645d4a7a2270f759ff882 : SUCCESS in 49m 01s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/59f637b67aea4f2fa627f7ef2bdae68e : SUCCESS in 52m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/996b68dbb0ac4419800eaaa2ee93d784 : SUCCESS in 1h 37m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a1f65656f595483f8b04dd7a8d34d1ea : SUCCESS in 1h 45m 50s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b9f89348f7f0464eb2680dadb83fc110 : SUCCESS in 2h 05m 18s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/530f59d5a829438fbbb969e4fae1c514 : SUCCESS in 2h 24m 37s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/1434a38e34b942f7830e7462ca3d3168 : TIMED_OUT in 2h 03m 24s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0ec997522e9a41a09428ecf0a57e7107 : SUCCESS in 1h 11m 23s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/abacd2539bf44c99986c2eddd9277359 : SUCCESS in 1h 54m 12s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/94362d202a814c1cb1716a5052f5e2f3 : SUCCESS in 39m 47s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f50f84e2d1084dc2a5a506b6fb347245 : SUCCESS in 1h 25m 03s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/09bbafe134b047838c3401175c7f073e : SUCCESS in 1h 26m 18s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8d4d9405e3ec4726aeff5ee4c426b42d : FAILURE in 40m 34s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/43b21c26b8de4a3d94fe3b6ba6f5a9da : SUCCESS in 45m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1d00681a54d219f4ec07d18db4b3590cfb44c414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-29 14:13:26.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"9e2f84cd7b5178f44f5ed0616b7a5eab1d002949","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 14:42:36.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f1bbac0e56f4b76b26e0e95912d6754\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e6d49eef052b4c6e923b636be944eb2b : SUCCESS in 26m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d7a1ea1f7cf8436ab96aa6bd1d65245b : NODE_FAILURE Node request 300-0021577310 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f384fad02ef34dddbaa1f699279c326e : NODE_FAILURE Node request 300-0021577311 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f13adeee08e5405cb024b59cf6bb0aa3 : NODE_FAILURE Node request 300-0021577312 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/198e68456c6d499b8b020afc2ea58a5a : NODE_FAILURE Node request 300-0021577313 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa623f667e914fe7bd2d3e85c1386ebb : SUCCESS in 3m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bd2ee376c35c420f8873edb16ac15bf1 : NODE_FAILURE Node request 300-0021577315 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1e339052b4574715a506ca7949ff82c7 : SUCCESS in 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d7984321f31c98137330476c5c3e1b19b113f0bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 17:22:57.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d9a4a65b6dc47a5a59e778a6e1ea1d6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5be8f2788ae549cba58378518fefb0c7 : SUCCESS in 17m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/18ef36a3092f46719943c64ea5f23a7d : TIMED_OUT in 2h 01m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3a9356b81e4345d4b7b4d74bb7d0570c : FAILURE in 14m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3807e911f08248a9bfb58e8dff0c0597 : FAILURE in 33m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1ce487828141480f92e8344e22535a86 : FAILURE in 1h 04m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81ce824e6f1b413aa69350e2fc8db87e : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e54730a0cc7642998f00c2a09631c858 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9157adcd1a214ac1ae9c431921142826 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4e4a2d6c54c4dd89aee1a011774c8ed : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf7fc8ead454485e892ed264e994423b : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1078f4d8d2074b3098537542bb606a17 : SUCCESS in 14m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6449d768168a4869b2f5bd123897ec88 : FAILURE in 12m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/40caeecf788b435e9c4a00fa5028fda5 : FAILURE in 33m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/992e3f4d4f9644babfbe9cbf9a18877f : FAILURE in 32m 27s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7a204874fed6486e8fb16ede4cb26b57 : FAILURE in 40m 10s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f2dd2640e8704e9e89737704d5685b13 : FAILURE in 2h 04m 58s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/c734ab8c80c243b6ada0ce6f09802edc : FAILURE in 1h 53m 18s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/d23b4c1247f5485dbec3625a9be49ff2 : FAILURE in 47m 16s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/cfc264dd521647c5bec38ae165699f3d : TIMED_OUT in 2h 03m 26s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/9ce867d91ab2471bb86c225337d2e561 : SUCCESS in 1h 12m 10s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/1966b4bb2d9f4ec6847f296e20184dac : SUCCESS in 1h 43m 16s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/03bf5e3aa8c843af9552db53e615f400 : SUCCESS in 1h 06m 56s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/1e99cbcdc2f7484b88718d337cd0776e : FAILURE in 1h 45m 35s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/cf37772528a64e3b9c1a64557e54072d : SUCCESS in 59m 01s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/87703f8bcef7417cb99a6a62bf23a225 : SUCCESS in 30m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/35a974c151b3407980a8a91046e8c2b3 : SUCCESS in 46m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/001733f89942411885acff2de88e269a : SUCCESS in 56m 49s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/727f0ae148b0458dbfe8468a8e433785 : SUCCESS in 2h 30m 21s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9f6ba7a6a49b48219b95b114bbbfaba5 : TIMED_OUT in 2h 34m 05s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/4f18d9be51fa40fe98ce4fb30df913fb : TIMED_OUT in 2h 03m 57s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/266eaff58ccc423690053579b59f6571 : SUCCESS in 1h 11m 17s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/d12213fc022047dea0876856cc60ffe2 : SUCCESS in 1h 09m 03s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/41b7909c245b42eebd5e41106decb95f : FAILURE in 39m 47s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/d9aaa03d641f4137abfe8dd6d7264327 : SUCCESS in 1h 06m 35s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ca8737916fff4d05aee1e9f343acf1ae : SUCCESS in 1h 06m 21s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/46b54972e6f940638395aebe9ca5f908 : FAILURE in 51m 18s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2d4f70899cb84eb8aec2ff4289e8c3dd : SUCCESS in 28m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a0ea9e1641f9c1f1dcbae04ec3e56dfb3b2b9fe1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-07-01 07:28:22.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4062fd224ecd14ffaf8ce6364cdae2ba55536173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-01 09:23:29.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50e2c4433e1c48629086c7a17723d35d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d540982ac5c84811b3472630638b6ea9 : SUCCESS in 20m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/859c1b5d860440de8705de6d9b507d1a : SUCCESS in 1h 18m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c8285eca183743babc225142fa7ce235 : SUCCESS in 24m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e37e79209f51484a934ed2c041d1aa5f : FAILURE in 39m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1b7c0ebfead347cd998c3c02624502a4 : FAILURE in 54m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cca316a5c04443880c906bc4aa7ee4d : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a767aded533041dfa26f986e4fa98510 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd39c2215afc4f8a8bc6ff521fb0f971 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adc2268b574946539f66f8b662469793 : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce35db55f9a64574942411908da28bce : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d659be7ff24f50b4ff89f3c1162d82 : SUCCESS in 4m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/958ab5d56b3644d29ad124101bf4c8d5 : SUCCESS in 18m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1f87b6085f514dae9fcb71dcc1fefb72 : SUCCESS in 1h 01m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bc7e6c710ccf411d90199533e96a2b28 : SUCCESS in 1h 12m 30s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c0c2c3213574452a9bf765495eed4623 : SUCCESS in 1h 30m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f07acade86dd4ccc86239fc69e9b52ea : SUCCESS in 1h 29m 54s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/08888eff982d45b8b3a37bbe03f3f2db : FAILURE in 1h 12m 45s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/27c01185a99544a0928136958c8b499e : FAILURE in 1h 07m 37s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/85c32d4582284aefae1c59a130ddc061 : SUCCESS in 1h 26m 12s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/36ab8e8f8624455ca9b23d29fa321565 : SUCCESS in 27m 08s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/d4fa350c5af842f58d0f8c86b857042f : SUCCESS in 49m 14s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/aa54916f6ab849ab9112949d6e74f256 : SUCCESS in 1h 10m 42s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/20b6c7eaf8f444b9b94be68397e3543d : SUCCESS in 1h 25m 01s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7a1721a3632d435aa3871273acfb86de : SUCCESS in 31m 33s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/93239e7faf7742f89499515c99ed0f8b : SUCCESS in 22m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8f48e955688a47368d54326208c69e3c : SUCCESS in 54m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e110325dff94437d93e7e68e8d4cf0a4 : SUCCESS in 55m 11s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5300bd4eb3a443f8be67e706ba92da9b : SUCCESS in 1h 18m 29s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8510206b9b17427daf3bbd8212b9df45 : SUCCESS in 1h 28m 30s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/68fab3b15b7b4bb491805e7b52e96ffa : SUCCESS in 1h 21m 29s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/19f88c9d392e4f29b69e042e0427ca06 : SUCCESS in 54m 09s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/d4832aae925b40f19187e5bb9f0a976c : SUCCESS in 1h 09m 09s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/0d6a572b36f9457eb8d399a2758d6770 : SUCCESS in 40m 15s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/bf91108dfae840438d96212db662a2c3 : SUCCESS in 53m 23s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/646a5812534c4934871c1bd4ee904a63 : SUCCESS in 1h 28m 09s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e9889ca72c7b4e1d87b5d8843019a558 : FAILURE in 39m 06s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/97dc297a50274bb3a5c2ade4feea5949 : SUCCESS in 32m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0f6a6aca2248ceb773d6aa906764ec2a601f0425","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-01 10:03:18.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f503034176740118a24c853159b9b29\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2fc3404008cb480380d28408bc1ad832 : SUCCESS in 18m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ee90ee2d858547249805f95556796a96 : SUCCESS in 17m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e644274a39b04bf4bb37fa1c30e9b1a0 : FAILURE in 2h 31m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e68780694504f0aba647e1eb03a9119 : SUCCESS in 2m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdbe426fd63c45e68ccff16335b5c912 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/045288c8ee89467d86d1f24ac29bff58 : SUCCESS in 6m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/37468380f3a9453e9f65b2d09c197be6 : SUCCESS in 16m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/57c6f9ec7c56451d908b0375e9c507cb : SUCCESS in 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e5fb85fe5e4751d9bac21fef88de44510c4cac18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-07-10 12:00:25.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"ed140602862df94d5bf9b0b061edb18ac6d462e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 12:22:22.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f66c0abf5e6b4a3eb5726620226c35e2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4cfe352b5daf4522ac740523246a61b2 : NODE_FAILURE Node request 300-0021652465 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8d8a5c36edbc4455ad8bb31ebbc1d322 : NODE_FAILURE Node request 300-0021652466 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/24ebeb31d120479a9a8f3125c73fc35a : NODE_FAILURE Node request 300-0021652467 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4927ae6685dc41ccb9957e473a95e815 : NODE_FAILURE Node request 300-0021652468 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a24566361c33492296beb979d0aa0703 : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e90f6ec07b9b497f9ad41bc4c743107f : SUCCESS in 2m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/89e48088c25a45b186a74f2aaff9320e : SUCCESS in 16m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/18062c55dd2046989aceb251ccb94ef5 : SUCCESS in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"42273a043a32a394b7334a4399ede0e107fce664","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-07-10 13:12:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"82b4685591c9627466fc77e8d93f775a349bd869","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 14:44:26.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6009d51519841a097c3eac453e40802\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7d743a9822a7484580c8affb0aac92ae : SUCCESS in 20m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f9d5ed517cd141d28735a787e2a1be0f : SUCCESS in 24m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/43423cbe595942f195eeba7d090d6365 : SUCCESS in 1h 31m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9045bfcf0e6477ea6447ac7cabbb975 : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e974b80053a4c8cbf572136c2911672 : SUCCESS in 2m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b972de0faaea476ba92be6054f2ffe7a : SUCCESS in 2m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8dee1749a6a94a3fb3fcc1cb4bac8817 : SUCCESS in 22m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/39a74a6d03954aa28f718c28c5bf9d9c : SUCCESS in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e591a1624292f2de2e1eab63a09810294de86c59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 15:34:33.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/781f6097811d4ad683a52973c1424eee\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/738d7f349fa9453b9224fa0359ab6da8 : SUCCESS in 17m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/747d627067d249fc84e7fa50caa1dc3f : SUCCESS in 1h 17m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f088337f401e4423b96707be140f1712 : SUCCESS in 33m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e72b2466a98d4c00ba31fa27c9616eec : SUCCESS in 1h 29m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3a0711911a17443ead0b70251216f790 : SUCCESS in 1h 42m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41bc80f43d474a5e9915296f83fd7b79 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6141ec60c51646a599b28c2d1494cbc2 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c939cf46b9c7479fb76ba04015679012 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb3ef9dc389440caa4f6c42e80e270e0 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e710e53bfae2432881a558015fc52b64 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cbb582eb3664bc6b433f78759f1a489 : SUCCESS in 4m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7d9212bfd391401bb90a3c61bad54d2d : SUCCESS in 25m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8a40b0be378945769d643dd1eaf986b5 : SUCCESS in 55m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c7d89f3f4e3844f9999a4fcf5608ee44 : SUCCESS in 1h 30m 32s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a1e2212bcc19455ea99bcf28048c216f : SUCCESS in 2h 03m 42s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4a38a05b7c9a4279a7bfc56d4ed7fe0f : FAILURE in 57m 25s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/0f544168eff7415fb039fdc54851a178 : FAILURE in 1h 32m 59s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/50096634f8f3498ebaa162393e3697cf : FAILURE in 1h 01m 49s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/0c9ff37a741340869da33e85a83b8ac0 : SUCCESS in 1h 09m 32s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/9e4f6f72e5cf465b8914d67effa8b66a : SUCCESS in 36m 35s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/ba61eee542484164949cc2c0071f6a34 : SUCCESS in 1h 21m 23s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a5bc6e05eb18444887c40f60eac85568 : SUCCESS in 1h 31m 28s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/7fb6d677300a45e5a077c644c210a6aa : SUCCESS in 1h 46m 14s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/118ce011188d4bfa8ce3e79d46921631 : SUCCESS in 25m 00s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a5e1d53c1e4d41bfad4a2ed4ab996d0a : SUCCESS in 14m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/71c628cbd10c43ceb3652c76696485ef : SUCCESS in 59m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0cdf686810ee4d25bdb66f403e55d458 : SUCCESS in 1h 32m 35s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d20e0444891744f18f44496e2c9dddbc : SUCCESS in 1h 25m 21s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3613b245d98849afb1079f28119e39ba : SUCCESS in 2h 18m 48s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/0e36fa8d456347a3bc7e12e8da4cef92 : SUCCESS in 1h 58m 39s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/4022a8da307a435a8b4fa043a8a47603 : SUCCESS in 1h 16m 44s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/fe5a0f03493a436d80cdd5b0a260d296 : SUCCESS in 1h 43m 42s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b6aa9795e4264939a417564783b1e340 : SUCCESS in 35m 17s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ca605831d3f74c4fb45c0790000b8714 : SUCCESS in 54m 46s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/88851fee2fbf4ca2908e648cc75f669a : SUCCESS in 1h 27m 20s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8184583783104b8cb294b5d2476a660e : SUCCESS in 1h 49m 21s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3df06417f3f445b799d7b7402b999c62 : SUCCESS in 38m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"89bff9fcc3cc327ff66b3546a7a6b981be1fa505","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-07-13 01:26:28.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"e8f0895c21984f70195812a4712009eddcafc8c8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-07-13 07:26:20.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"23f97b83e7cdb891cc99b226be555a66fb9b7754","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-07-13 08:45:39.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fcd86f4fb5ec72f069e6149f0f1e4ee2d8b4f940","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-07-13 08:54:25.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0673560d22719bf7b2842546472b3d47de643c22","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-07 15:18:12.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"2d387efff206fe53b1b0b7b6ad2cd889380b3291","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-09 13:02:07.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"0b5c3c2f331d7ddc24e7135ef1e0cf17f7585047","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 13:02:24.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 886088,13","accounts_in_message":[],"_revision_number":13},{"id":"fbec96bf2bcab6c88a71a6e6b0c00127c7e3433f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 13:02:27.000000000","message":"Patch Set 13: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 886088,13","accounts_in_message":[],"_revision_number":13},{"id":"2defbcde376679d50c905415c079633fb5677b95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-08-16 09:10:07.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"39ae770a15fae01f210e5f4ca14431f929de7156","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-08-16 10:13:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"66aac4bca86c337062ae30ff85177d231c71c50b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-16 10:37:44.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"cfd414ef916f6d732e8258b228d659e743bcdf47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 11:28:36.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8295ccf2414d43d5ac2c21b1a20902e2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fc6e513e6b294a3f8149b0cda07c218b : SUCCESS in 24m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f1245d07ba7a49bebd941727eca61061 : SUCCESS in 1h 03m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/301840efd84d475199d1af8120710aa7 : SUCCESS in 33m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/56bff5c4f7b5447db3937706fc411d8e : SUCCESS in 1h 23m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5b88394ad1ee445280329855b7cc6c05 : SUCCESS in 1h 34m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09fc6fd1dfdb449c858f4e9ba6b8b6ef : FAILURE in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbbb3fd0167a425698619f9af48e4b5a : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39f650663c88427fb816f0bf9da47335 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e18199611b847c486e4433aa4c62516 : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d05689337fe4beabe0b4651893586b3 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a05d4603574724956cc7237943e223 : SUCCESS in 4m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/efaff7afca0d49ae998cdc03d9aec640 : SUCCESS in 16m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/906acc3af18d42e8a94fe4abe4f34090 : SUCCESS in 59m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e7b02ee090674e4e87b31d4d8d28106b : SUCCESS in 1h 40m 49s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/4e3923c9a1094fa69993b18b81843080 : SUCCESS in 1h 34m 31s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b262d99eb3c749c09d22b25c888c9b76 : SUCCESS in 1h 44m 04s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/08e5c658264944f4bc8ce767efb07306 : SUCCESS in 1h 59m 01s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/418bb992683b43bf892e6173500ec69c : FAILURE in 1h 07m 44s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/aed887ac0bdf4cd4aad71898f20965e9 : SUCCESS in 1h 22m 50s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/7f9a1c9d56964f989b64035363f5acd2 : SUCCESS in 27m 02s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/d3552e0f9ff8492ab367d78856d50c0a : SUCCESS in 1h 09m 07s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0c57cdb3615946f9a0b8d4130680fe36 : SUCCESS in 1h 03m 38s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2c9333ea3c4448deb668772559a2f7e5 : SUCCESS in 1h 21m 02s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/589e58b97c7d419f893acb1396fad64a : SUCCESS in 33m 12s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cb8a5214bfdc461e958bbda609b9db5b : SUCCESS in 26m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e46c2bab45fd4eb29ab5915b7a0d5dad : SUCCESS in 1h 11m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9614ace3c4bc4f12a6c820343ee88896 : SUCCESS in 1h 26m 23s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/fda57205939c43798c9f70559492ef2c : SUCCESS in 2h 07m 13s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/20d832b478fc40949be1cd02a756a39e : FAILURE in 56m 56s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/4d28399b0ed34958a04b6643043391b0 : SUCCESS in 1h 51m 19s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/711a459af5b34a89bacfe66f56745f37 : SUCCESS in 1h 29m 17s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/5fe30c16a5844100b619e4fc467a5a53 : SUCCESS in 1h 21m 25s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/5a9c8a73cd2440eaa96ad72584f58c3e : SUCCESS in 21m 54s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b7dac24432dd4ceabacb7ea8610b089f : FAILURE in 39m 53s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d7bc1656525b4a38b2f59b103e9459c7 : SUCCESS in 1h 22m 03s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/452d259221d040ad8e492e09731bfb50 : SUCCESS in 1h 17m 26s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c2ae0bbb4d184e08bb8e2de567dc585d : SUCCESS in 31m 58s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f34db1766da7c631baab35741efe57fae65053e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 11:33:38.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d277a381e2cd4b90b0d17dae2acd343c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/69bf1a5ca6f7456daf59e7b4a97cd71e : SUCCESS in 1h 22m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/455abf21ce914e2a87eac525b6dd059c : SUCCESS in 15m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/97985d842b7144f588af0d13c4e4010f : SUCCESS in 2h 21m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f7064e52ade4869be4e8fffb014a5e3 : SUCCESS in 2m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a67b78e7d7ae40949d8bbc0e187ac69a : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b9b58cfdabee4d42809737885d096867 : SUCCESS in 2m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/60e03ee0ab2749af9ac3d129089e8958 : SUCCESS in 1h 15m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bea4103243df4624a9f5cbb1114244af : SUCCESS in 14m 38s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0e078ce2c7ccf445ee602f85d2382f415b064a6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-16 12:38:35.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"7b68883549fc607f9c435b051f83c6b2df89e61c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-16 12:48:31.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d56386eb8c319e8130d95cbd9bc45d76c146ef3e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-16 12:52:02.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"de975a0699faa4bec738441afd611a87f84a4bc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-16 12:59:18.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"defb65a04a00243fc3dcdd390a1937f9bca470ff","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-16 13:00:31.000000000","message":"Patch Set 16: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"f7a8d513ce4683725bc592b44d2b6ae04c7eec89","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-16 13:00:56.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"7e9245cdaa19d9676cfd576b40c7a8f44b048ae9","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2023-08-16 14:31:36.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"96deba8e362661a7bb90755d49f9b7743cd4fcaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 15:00:13.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76ac843986a4e828a5822cfbd48656f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fd0badd2ac37492b887a3672f5a3e52a : SUCCESS in 25m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b0b756e8eb454bea99c467823df108e1 : SUCCESS in 1h 12m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1f317e2f550a453e80de1f3b7cc2f1ac : SUCCESS in 22m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ebaf98a8192d46d4a45f5ccd64cc8f06 : SUCCESS in 1h 30m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fd5f2eb2b8e3456e9fc87ba8dd6f73d0 : SUCCESS in 1h 41m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6f2e2fcecd4ef19f5bff186761c601 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d476b1548e24844ab6bd4abdbdd8fa7 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/447a60c0c8ab47e7918e778d79735e6b : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/120132fbf5074b0d82c76cd5f2378d5b : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e29aa1b9dda949448305a0a0b49b9522 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a49d0b7c08458cad1db3732cb779aa : SUCCESS in 4m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6faf137fb4c5457ba5d839edf9a167c2 : SUCCESS in 23m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a5409f5a24044b449709d0a75172fb4f : SUCCESS in 1h 15m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8a8469e33e284b9ba1aa53dcfff70694 : SUCCESS in 1h 25m 52s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0f86c6280324471e9cfccc6522ce4692 : SUCCESS in 1h 36m 53s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d8ca463a3ba5442c95511f1b52822221 : FAILURE in 53m 46s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/7d7839c150fa46a5a01afec5e24b9615 : SUCCESS in 1h 49m 50s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3bba453553754228a17a6a5ae931c8cf : FAILURE in 48m 12s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/65b29c4a63d249dfbaa313aef290e321 : SUCCESS in 1h 27m 23s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/491c5894339842c4b5a747328a21f3a0 : SUCCESS in 33m 31s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/7072305d920e461c8231ce43d1864f7d : SUCCESS in 1h 15m 48s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/9b9c5749faf84bf387c72e280c130434 : SUCCESS in 1h 05m 25s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/920d2ef297634f8c87c62ec30dc00784 : FAILURE in 1h 12m 46s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/aa8f5d31135540efad374a9a099eda16 : SUCCESS in 36m 58s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3cb22308eac441cc93c1892549329a8b : SUCCESS in 24m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6c3b6d342e4649e3996deb8443285e78 : SUCCESS in 55m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fd027743d0db4f4aa82c6b84ce02e1f0 : SUCCESS in 1h 03m 23s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/79d429a3fb7042cc8e873b092bdedd7c : SUCCESS in 1h 19m 29s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/452bf9f055d34ef88673351dc226bb29 : SUCCESS in 1h 39m 02s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/98a6523481a642f990fce8eecae42799 : FAILURE in 1h 28m 04s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/482aea5512dc408dba00a55002090917 : SUCCESS in 1h 25m 21s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/f00b71458ea74354beae4a1a00f150bb : SUCCESS in 1h 29m 54s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/43ae1b469d06440584420f7235f5dbe8 : SUCCESS in 31m 33s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9e43cacd406843358db19fd60de497df : SUCCESS in 53m 49s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/044cae346a544a1eabaded5dabf352b9 : SUCCESS in 1h 03m 46s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/da538d35437c4e1a9f823c29fe5649c1 : FAILURE in 32m 47s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/70b3c6c23f6c48a2bf4297aa8430dab9 : SUCCESS in 32m 29s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0ecc300830e7357bd5d36dd0f4041b56528fd65a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 15:58:44.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53fb630a877744e6979f337512eab7ca\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b38692f4024e4520b90b9c03977185a4 : SUCCESS in 1h 46m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fbbc69b430034391bfaa48d216555d81 : SUCCESS in 2h 00m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3df81f8a2ded43938b7837c15d79022f : SUCCESS in 2h 50m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d42971242ce440838ca4b4a27b5c7b39 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53a02c533e694999b1422e29911e904c : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ee8e1b2caf849bfa5e81ebd8eaa3296 : SUCCESS in 3m 04s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bc426a63150a4b478e46233f7ebb9d09 : SUCCESS in 26m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/860df9c6eae847e0a71e7370f71e3947 : SUCCESS in 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"cd17a9aab85c961a9a954abb5c731ad1f9ffa6c2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-17 07:04:46.000000000","message":"Patch Set 16: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":16},{"id":"caa006c817e9f6a73ae7f2b620d76c080277a014","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-17 07:11:06.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"7454822da9d1d0a5aac9f1ffe8fd031a307d3ba5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-08-17 15:48:13.000000000","message":"Uploaded patch set 17.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":17},{"id":"9500d5ee53585d0f4118b82a2fa5cf42b49ed3af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 15:48:33.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 888034,2","accounts_in_message":[],"_revision_number":17},{"id":"a6c2d5f556c986592d0ed32ab70c6d4f35776b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 15:48:36.000000000","message":"Patch Set 17: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 888034,2","accounts_in_message":[],"_revision_number":17},{"id":"ab0ef4dc98ee25e7acc699bd9122ff98ea5e7630","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-17 15:53:12.000000000","message":"Patch Set 17: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"6f442905eb0de958d0d52710655ba4eed0f654ae","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-17 16:53:35.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"cabda442eac82472df12bc0c7141f74f69ab4caf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 18:27:06.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a28b1408bd04460b2b69fc1cb5107e6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d6d58edf11f246df9c009672c6ef57d3 : SUCCESS in 1h 38m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eed1b173f74a4f5fa4cb5d3b0440f379 : SUCCESS in 1h 34m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1563c912fd7540358e397d072f3046d0 : FAILURE in 1h 53m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c6f0a644ed64a48bec03a712d81cfb2 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7f9af61c9e74e67a8ad6551fdfb3073 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e1ba567049b54c67920d5f5560f0dc83 : SUCCESS in 5m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3c4d38731374486096b748f46441dcf2 : SUCCESS in 1h 23m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/96c8027f242d47e083ba6f2fc2c8b36a : SUCCESS in 1h 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"7994672b15ccb91f7d42d4fcf6d68d4bafb4131b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-17 19:08:12.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"45d84ea9323a1e3d0d8d8f9b6df9c045263abe24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 20:04:19.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a0837eeec304b73bb927f4b88d0c842\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3f1201bdac5f411088d1b147a8b29c56 : SUCCESS in 18m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/03e9e0c10a584b64abc9db2d0ed648d9 : SUCCESS in 18m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2498b0e9cffe41339c32591618ff2ab9 : FAILURE in 35m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/69fa2c54b9a84327bba5e7bdbd56c6ec : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fc922bbf0a44d9aa1026489d8c14cb5 : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91b987e49f3041ac888d7178588722ea : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7ddd5b4b646b4d4d95ad3d108286a0b8 : SUCCESS in 2m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f1a3d4a463b948c89cb7273f65e83c66 : SUCCESS in 53m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/53ccea31e8ef47869ff4b5a4ff8d5fe2 : SUCCESS in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"b454575bf1f3a6aa33a8dccf11aa543464e759bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 21:17:00.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/28380f2bdbda47ae8ecb5a133b6ad9d3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e4204433174b43938acbfb80e6fe4d36 : SUCCESS in 24m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9b279ad5fe5e45a5b272093b48573a03 : SUCCESS in 1h 15m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2940b8f25b674e009125a5f359ca65ed : SUCCESS in 26m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/40d1b602faa74319942b0528541ab3ea : FAILURE in 42m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/52541d7aace54d9b9f19747b53b70dc3 : FAILURE in 52m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad4ec363357242b5a25f7446fb139892 : SUCCESS in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc1d855f147e494c987cd1da9d03bba2 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9668e08e74846f5ab98efcc2923642e : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/437f418f919e42deb3b079c233ba2276 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5346f32f7d74557accccc902ebf06ac : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cc34679890744a0a9421179f8031a03 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b94eec085ecc43c89f99655d500756f0 : SUCCESS in 4m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/01bdd04eced44e36ae2d99aa6aefe4e1 : SUCCESS in 15m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/01cfb523819b4ee08b9aa7177b95c4f5 : SUCCESS in 1h 06m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b596688e36f348e4a432d8a6fc077cfa : SUCCESS in 1h 19m 58s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b26a2ccb4d424cc495202e6ae837f2ea : SUCCESS in 1h 56m 44s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e2aef75271bf4feb9df931e755f76552 : FAILURE in 1h 22m 46s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/c4249a5cb12341e095fb4db592045f12 : FAILURE in 1h 05m 47s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/64f6cf1ba70648bbba9c3fce81a1ee97 : FAILURE in 50m 02s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/458839b071cb4e1087d72409d3ade4df : SUCCESS in 1h 30m 26s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/b21f40926afd43189d8a4c8d3f5e969b : SUCCESS in 24m 29s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0c1dfd07bc1a40029c9de4290a277059 : SUCCESS in 59m 04s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/fe934cafa8964c7493a37227dfb3155a : SUCCESS in 1h 13m 30s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9cdba38d91a940378f72235a613b8cc7 : SUCCESS in 1h 39m 32s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/68a8df83f10e45b68547abe9094d20e5 : SUCCESS in 35m 42s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bfadc4988c6c468ca1c42e94b66facc6 : SUCCESS in 24m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/eff5dda61bdf47ecba1ca986ee6c1759 : SUCCESS in 54m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e1e5679ec58047c7bf42ec1efe80e620 : SUCCESS in 1h 06m 25s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e15b1bed3b94450dbf9b5f05613278b0 : SUCCESS in 1h 27m 19s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6a949fc717384e2a85dfa09511804830 : SUCCESS in 1h 52m 39s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/57d12b975f6a4e2fa78ecbfc5f043e44 : SUCCESS in 1h 45m 16s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/6ea8c317a853485fb552209b44ccab7b : SUCCESS in 1h 08m 45s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/e8cc2140a4694e2d94c10a67b28d1dc0 : SUCCESS in 1h 17m 34s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/962e194276c44deb99123ee066dd5c0a : SUCCESS in 24m 06s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a1d45b59d3cb4e0394e216343942202f : SUCCESS in 57m 26s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6f9a57e126134a39847a15172faf862c : SUCCESS in 1h 15m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b64cde32eace41d0b681b71beb554343 : SUCCESS in 1h 18m 51s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e12292bb95894a0c99826f271d3652fe : SUCCESS in 26m 51s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"12cd005a0b1d11dfd10febf4bcd4c14a541d8e00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-18 05:26:03.000000000","message":"Patch Set 19: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"ea57584418645113d378f2347bb087b2ddb50ab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-18 06:48:11.000000000","message":"Patch Set 20: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":20},{"id":"7106b17e1417f9da48ab5414daf242490e7af032","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-18 07:40:42.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"0f280794d8abea329b912c2c6a51ed0c99f9a21d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 10:15:15.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac31005f530949c6844082397b8fb6e3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/657956ce12684f4aac6e576152651222 : SUCCESS in 22m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5f0dae66af9349b391582da9f346705e : SUCCESS in 1h 11m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/97576b4babe34fbe8a0de441e26b9b43 : SUCCESS in 1h 39m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6c90d4c9a774303a2e5dd641cbae888 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3fc3f7010c4497db9f5b394094f2a7d : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a1a5e959abcc44b4aa118907858b9dd0 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9fe5f23ba3864c8595a82b0a40e357c2 : SUCCESS in 3m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9f8f08474cc94cda9d8e754147e48101 : SUCCESS in 1h 08m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/029f250bfab3488d9c376e99c3b72d19 : SUCCESS in 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"7c6831105731ed1ce43b90147e3415b86380d3fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 10:40:19.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/981ad416e53c4b84aea5dd9fed0b469d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b7b42794a65648be97c7fad25ca2e68a : SUCCESS in 28m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4fabe18f11f849829ef4cd2e4ff48f0e : SUCCESS in 1h 22m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7c0076f5694743229e8935c9da567916 : SUCCESS in 24m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/db3a9435c646459bba9d83fa593fe063 : SUCCESS in 1h 19m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dbe36b094e04446aa8a124653e798aa0 : SUCCESS in 1h 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d44a7931adfa456f9aa2e5b55262d40d : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ae0fcadd60e4f43823dd94bb429f47d : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/239accd7dd5b4482808151bc5fabce9a : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71843e7d42114c5682cb13b1d3a5ebb5 : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/00c55416593b463d992eb2b596ef524b : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cdb84fff5344c41b5380d728b7951ef : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82e880a8da18433db06660f16b12f793 : SUCCESS in 5m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a9e39e6f3ddf4642b327ca3b520cb861 : SUCCESS in 14m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ce8715802bc84bfcab1c0dc241b83c8b : SUCCESS in 1h 05m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/77203daede8d49959db198cf87cddd0c : SUCCESS in 1h 17m 03s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/78163a8209864c8c8f72c690d3bcedf5 : FAILURE in 52m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d976834e8e1241a89ff590e8beabbb36 : FAILURE in 2h 00m 38s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/0d0006cfcaae48ba84e6ba3ba1505418 : SUCCESS in 1h 38m 33s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/feb4d2bff5f847949102932b7c687821 : FAILURE in 40m 58s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c5275280a08d46a7a45f71d517f479db : SUCCESS in 1h 11m 46s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/69aab74d97524900a48e7d386d29056a : SUCCESS in 31m 02s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/1443762f7a034847bfffb3a2b65ac27c : SUCCESS in 58m 54s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/40c179666d1e43f2a51a24ac81a8221c : SUCCESS in 1h 15m 38s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/646dc4333bd446908f5f9fc6e7dd1828 : SUCCESS in 1h 23m 05s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/affb324d197a43a9a3cc13396b9ef807 : SUCCESS in 34m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8f4f902c5e234d6981e6af31b3d06a05 : SUCCESS in 20m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e049edd1230a4a49a560ef02f7c30a0a : SUCCESS in 1h 14m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ffeb91d32f5e44b39a1eb25abb533fad : SUCCESS in 1h 09m 12s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ffc2434cbad9405aba0f9e55f2b2e318 : SUCCESS in 1h 42m 29s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ca4d8abdb5dc49d69a2852fffd567ad7 : SUCCESS in 1h 45m 48s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/d8087c12eeea41969a84579ef58f8e1e : SUCCESS in 1h 47m 33s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1a01da4283a843308cd40fca3bc78893 : SUCCESS in 1h 09m 06s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/b09158b2872e4cb2aebdbf18f3e8bb22 : SUCCESS in 1h 13m 08s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/d028d4d775cb4bb49139570f9a46a686 : SUCCESS in 23m 20s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e226b17fde50429ab1f80e9888a5fbdf : SUCCESS in 1h 00m 15s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e981a98d12df45caa8a2423fbeba594a : SUCCESS in 1h 11m 03s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/91976e24bb994f14bc71afa71aebf9c5 : SUCCESS in 1h 34m 03s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fbbc2489a4a84e83bc363bc09db71c54 : SUCCESS in 32m 16s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"dc4b116249f80ec6935ece7683202a1fb76cee66","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-18 13:49:50.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"6ad604d5e367ba197c0094e2397742bde8500c31","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-21 09:22:14.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"6456db15c7278a01833389c69bb2a77618399fe9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 13:51:32.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"a2df401435a7936bb64077d011f70a00d2437d2b","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-21 14:06:40.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"1b21905499041617d4ba4ad63b49cc35316daccd","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2023-08-21 14:26:49.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"10807eb4b9c0e3b05892cdbcf15fa1cf667feb6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 15:11:35.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc4997b90d4a49dc902a88d125d181af\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1095536d88cb439b9edd428eb35f42d4 : SUCCESS in 29m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7c5d23ebb17e4f24975d1d89b252d2ef : SUCCESS in 55m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8e7f7786304a435780521cfe9e778db8 : SUCCESS in 1h 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a9b270df7534b5e9ff9a38d885c235d : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c100122ba7440c9bedaf80a7159339a : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec290acec15d4223b28f9238a743e57b : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fc91820169646228f87f23bfcb74986 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9b873c12934ca8bdd1a9f028ed2230 : SUCCESS in 4m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2e7332c7830b4525a368b38c48826ac6 : SUCCESS in 13m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/de68a741bdcc4c628c6e187c05b79258 : SUCCESS in 1h 02m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2e297e77c1804b339571bd018b6b744f : SUCCESS in 1h 08m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/833a69f127bb4975a40ec487a0a59281 : SUCCESS in 18m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/09769b2cdfd14a14b25a5d071c88fbfe : SUCCESS in 58m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9486273636a14be98d6346583408b582 : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":20},{"id":"4e37666ea6b49d8c53c6f1e4d96d2bed6e5acf7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 15:11:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"272d31df3164e000dd1042aef7ce450c39b20acd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 15:13:53.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ba084217c1b49268e7c6a7f580ef44e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef25f4ac4ef44bf8d2fb23bf30be080 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"7fa2ac132cce94aae8cd3d196abba342394f6c70","revisions":{"cbcf78071718ad50959a162afad489a239059806":{"kind":"REWORK","_number":1,"created":"2023-06-14 12:07:49.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/1"}}},"commit":{"parents":[{"commit":"8bfc5bedc42b25ef504f8387f4f3366e0a05b396","subject":"base: Debian 12 uses new APT sources file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bfc5bedc42b25ef504f8387f4f3366e0a05b396"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 12:07:00.000000000","tz":120},"subject":"WIP: Move to Debian 12 \u0027bookworm\u0027","message":"WIP: Move to Debian 12 \u0027bookworm\u0027\n\nIt is work in progress as several 3rdparty repos are not yet available\nand some of them are replaced by either Ubuntu/jammy or Debian/bullseye\nones.\n\nDoes not build yet - no MariaDB repo, some images need work.\n\nhttps://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbcf78071718ad50959a162afad489a239059806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbcf78071718ad50959a162afad489a239059806"}]},"branch":"refs/heads/master"},"a1db76f60577c0a67282db27d576a5fae4954120":{"kind":"REWORK","_number":2,"created":"2023-06-22 09:39:20.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/2"}}},"commit":{"parents":[{"commit":"abdd982a5397b4bbe5e6a1cda3881bc3431c07aa","subject":"base: Debian 12 uses new APT sources file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abdd982a5397b4bbe5e6a1cda3881bc3431c07aa"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-22 09:39:08.000000000","tz":120},"subject":"WIP: Move to Debian 12 \u0027bookworm\u0027","message":"WIP: Move to Debian 12 \u0027bookworm\u0027\n\nIt is work in progress as several 3rdparty repos are not yet available\nand some of them are replaced by either Ubuntu/jammy or Debian/bullseye\nones.\n\nDoes not build yet - no MariaDB repo, some images need work.\n\nhttps://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1db76f60577c0a67282db27d576a5fae4954120"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1db76f60577c0a67282db27d576a5fae4954120"}]},"branch":"refs/heads/master"},"358f7069cf6c76286592596c644509836a59e324":{"kind":"REWORK","_number":3,"created":"2023-06-22 09:50:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/3"}}},"commit":{"parents":[{"commit":"382e1db8c60b2944a0361a25fa1e47fb4a8fb8b2","subject":"base: Debian 12 uses new APT sources file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/382e1db8c60b2944a0361a25fa1e47fb4a8fb8b2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-22 09:49:50.000000000","tz":120},"subject":"WIP: Move to Debian 12 \u0027bookworm\u0027","message":"WIP: Move to Debian 12 \u0027bookworm\u0027\n\nIt is work in progress as several 3rdparty repos are not yet available\nand some of them are replaced by either Ubuntu/jammy or Debian/bullseye\nones.\n\nDoes not build yet - no MariaDB repo, some images need work.\n\nhttps://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/358f7069cf6c76286592596c644509836a59e324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/358f7069cf6c76286592596c644509836a59e324"}]},"branch":"refs/heads/master"},"0886773f372f8e17a1a89ce389edcd4dc1ca6215":{"kind":"REWORK","_number":4,"created":"2023-06-22 09:57:23.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/4"}}},"commit":{"parents":[{"commit":"382e1db8c60b2944a0361a25fa1e47fb4a8fb8b2","subject":"base: Debian 12 uses new APT sources file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/382e1db8c60b2944a0361a25fa1e47fb4a8fb8b2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-22 09:57:16.000000000","tz":120},"subject":"WIP: Move to Debian 12 \u0027bookworm\u0027","message":"WIP: Move to Debian 12 \u0027bookworm\u0027\n\nIt is work in progress as several 3rdparty repos are not yet available\nand some of them are replaced by either Ubuntu/jammy or Debian/bullseye\nones.\n\nDoes not build yet - no MariaDB repo, some images need work.\n\nhttps://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0886773f372f8e17a1a89ce389edcd4dc1ca6215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0886773f372f8e17a1a89ce389edcd4dc1ca6215"}]},"branch":"refs/heads/master"},"133892cdf5533122c2f552071fb74e980b30d6f8":{"kind":"REWORK","_number":5,"created":"2023-06-22 10:46:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/5"}}},"commit":{"parents":[{"commit":"382e1db8c60b2944a0361a25fa1e47fb4a8fb8b2","subject":"base: Debian 12 uses new APT sources file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/382e1db8c60b2944a0361a25fa1e47fb4a8fb8b2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-22 10:42:45.000000000","tz":120},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n- RabbitMQ 3.10 (on AArch64)\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/133892cdf5533122c2f552071fb74e980b30d6f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/133892cdf5533122c2f552071fb74e980b30d6f8"}]},"branch":"refs/heads/master"},"c3ef7b9425d4374145683b676187f3b087fb4432":{"kind":"REWORK","_number":6,"created":"2023-06-28 13:25:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/6"}}},"commit":{"parents":[{"commit":"382e1db8c60b2944a0361a25fa1e47fb4a8fb8b2","subject":"base: Debian 12 uses new APT sources file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/382e1db8c60b2944a0361a25fa1e47fb4a8fb8b2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2023-06-28 13:24:44.000000000","tz":120},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n- RabbitMQ 3.10\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3ef7b9425d4374145683b676187f3b087fb4432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3ef7b9425d4374145683b676187f3b087fb4432"}]},"branch":"refs/heads/master"},"433234e55f1e41853677be6c4023bf10c5d57b1e":{"kind":"REWORK","_number":7,"created":"2023-06-28 16:17:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/7"}}},"commit":{"parents":[{"commit":"382e1db8c60b2944a0361a25fa1e47fb4a8fb8b2","subject":"base: Debian 12 uses new APT sources file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/382e1db8c60b2944a0361a25fa1e47fb4a8fb8b2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-28 16:17:51.000000000","tz":0},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n- RabbitMQ 3.10\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/433234e55f1e41853677be6c4023bf10c5d57b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/433234e55f1e41853677be6c4023bf10c5d57b1e"}]},"branch":"refs/heads/master"},"b62cb68489db7d48863e81e650032f9f65b19849":{"kind":"REWORK","_number":8,"created":"2023-06-28 16:21:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/8"}}},"commit":{"parents":[{"commit":"3e4d291b12719e8ede9b2801847da987db2885dc","subject":"base: Debian 12 uses new APT sources file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e4d291b12719e8ede9b2801847da987db2885dc"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-28 16:20:10.000000000","tz":120},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n- RabbitMQ 3.10\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b62cb68489db7d48863e81e650032f9f65b19849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b62cb68489db7d48863e81e650032f9f65b19849"}]},"branch":"refs/heads/master"},"ca6f8b078f1c517a63ad9eefaf14cdd9b19e7b0a":{"kind":"REWORK","_number":9,"created":"2023-06-28 16:27:22.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/9"}}},"commit":{"parents":[{"commit":"3e4d291b12719e8ede9b2801847da987db2885dc","subject":"base: Debian 12 uses new APT sources file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e4d291b12719e8ede9b2801847da987db2885dc"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-28 16:27:21.000000000","tz":0},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n- RabbitMQ 3.10\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca6f8b078f1c517a63ad9eefaf14cdd9b19e7b0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca6f8b078f1c517a63ad9eefaf14cdd9b19e7b0a"}]},"branch":"refs/heads/master"},"d715f8f303c2de2a674e32f6f93c3a98a3e24338":{"kind":"REWORK","_number":10,"created":"2023-06-28 16:34:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/10"}}},"commit":{"parents":[{"commit":"3e4d291b12719e8ede9b2801847da987db2885dc","subject":"base: Debian 12 uses new APT sources file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e4d291b12719e8ede9b2801847da987db2885dc"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-28 16:34:11.000000000","tz":120},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d715f8f303c2de2a674e32f6f93c3a98a3e24338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d715f8f303c2de2a674e32f6f93c3a98a3e24338"}]},"branch":"refs/heads/master"},"428d9c545a5151a7cdd8237848004d8f93f6c509":{"kind":"REWORK","_number":11,"created":"2023-06-29 14:13:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/11"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-29 14:12:58.000000000","tz":120},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428d9c545a5151a7cdd8237848004d8f93f6c509"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428d9c545a5151a7cdd8237848004d8f93f6c509"}]},"branch":"refs/heads/master"},"62c1c6d784483a4c613eb2a21a2f6e9c895f16d1":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2023-07-10 12:00:25.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/88/886088/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/12"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-07-10 12:00:24.000000000","tz":0},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/888034\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62c1c6d784483a4c613eb2a21a2f6e9c895f16d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62c1c6d784483a4c613eb2a21a2f6e9c895f16d1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"87c21e66a365b05866c16babcee52553dd0cc7f3":{"kind":"REWORK","_number":13,"created":"2023-08-09 13:02:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/886088/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/13"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-09 13:02:07.000000000","tz":0},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/888034\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87c21e66a365b05866c16babcee52553dd0cc7f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87c21e66a365b05866c16babcee52553dd0cc7f3"}]},"branch":"refs/heads/master"},"5dec67f544f56a823d90e98d71f3366060b5f866":{"kind":"REWORK","_number":14,"created":"2023-08-16 09:10:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/14"}}},"commit":{"parents":[{"commit":"7f12d216dc4de2c8d32291c3d6223185ecf2b510","subject":"Merge \"Revert \"Disable build of Gnocchi containers\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f12d216dc4de2c8d32291c3d6223185ecf2b510"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2023-08-16 09:09:10.000000000","tz":120},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/888034\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dec67f544f56a823d90e98d71f3366060b5f866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dec67f544f56a823d90e98d71f3366060b5f866"}]},"branch":"refs/heads/master"},"1c76e8831aa601e7a1efc18280d2788bad153d46":{"kind":"REWORK","_number":15,"created":"2023-08-16 12:38:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/886088/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/15"}}},"commit":{"parents":[{"commit":"7f12d216dc4de2c8d32291c3d6223185ecf2b510","subject":"Merge \"Revert \"Disable build of Gnocchi containers\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f12d216dc4de2c8d32291c3d6223185ecf2b510"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-16 12:38:34.000000000","tz":0},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/888034\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c76e8831aa601e7a1efc18280d2788bad153d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c76e8831aa601e7a1efc18280d2788bad153d46"}]},"branch":"refs/heads/master"},"db735ce3b7ce86756c1258869e73a4e6601847a5":{"kind":"REWORK","_number":16,"created":"2023-08-16 12:59:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/886088/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/16"}}},"commit":{"parents":[{"commit":"7f12d216dc4de2c8d32291c3d6223185ecf2b510","subject":"Merge \"Revert \"Disable build of Gnocchi containers\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f12d216dc4de2c8d32291c3d6223185ecf2b510"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-16 12:59:17.000000000","tz":0},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/888034\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db735ce3b7ce86756c1258869e73a4e6601847a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db735ce3b7ce86756c1258869e73a4e6601847a5"}]},"branch":"refs/heads/master"},"7679b777f543c12d224e8e35bda0683056bdb0f8":{"kind":"REWORK","_number":17,"created":"2023-08-17 15:48:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/886088/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/17"}}},"commit":{"parents":[{"commit":"a004abbdc36b425576c7a9a8c480df938f5c0a1f","subject":"Merge \"rabbitmq: bump version to 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a004abbdc36b425576c7a9a8c480df938f5c0a1f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2023-08-17 15:47:52.000000000","tz":120},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/888034\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7679b777f543c12d224e8e35bda0683056bdb0f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7679b777f543c12d224e8e35bda0683056bdb0f8"}]},"branch":"refs/heads/master"},"bb7de26d324d913b5cdb3cb7e2c801006da284f7":{"kind":"REWORK","_number":18,"created":"2023-08-17 19:08:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/886088/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/18"}}},"commit":{"parents":[{"commit":"a004abbdc36b425576c7a9a8c480df938f5c0a1f","subject":"Merge \"rabbitmq: bump version to 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a004abbdc36b425576c7a9a8c480df938f5c0a1f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-17 19:07:57.000000000","tz":0},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb7de26d324d913b5cdb3cb7e2c801006da284f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb7de26d324d913b5cdb3cb7e2c801006da284f7"}]},"branch":"refs/heads/master"},"50de88b4a76f4f698cd6525e4884e6dd4efd4a83":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2023-08-18 05:26:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/886088/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/19"}}},"commit":{"parents":[{"commit":"a004abbdc36b425576c7a9a8c480df938f5c0a1f","subject":"Merge \"rabbitmq: bump version to 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a004abbdc36b425576c7a9a8c480df938f5c0a1f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-18 05:26:03.000000000","tz":0},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nDepends-On: https://review.opendev.org/883982\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50de88b4a76f4f698cd6525e4884e6dd4efd4a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50de88b4a76f4f698cd6525e4884e6dd4efd4a83"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7fa2ac132cce94aae8cd3d196abba342394f6c70":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2023-08-18 06:48:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/886088/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886088/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886088/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886088/20"}}},"commit":{"parents":[{"commit":"a004abbdc36b425576c7a9a8c480df938f5c0a1f","subject":"Merge \"rabbitmq: bump version to 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a004abbdc36b425576c7a9a8c480df938f5c0a1f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-14 11:54:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-18 06:48:11.000000000","tz":0},"subject":"Move to Debian 12 \u0027bookworm\u0027","message":"Move to Debian 12 \u0027bookworm\u0027\n\nWe move to use some components from distribution instead of 3rdparty\nrepos. At least for now - this may/will change later.\n\nPackages from Debian:\n- Erlang 25.2\n- MariaDB 10.11\n\nSome Ubuntu/jammy repositories are used in places where there is not\n(yet) Debian/bookworm one.\n\nopenvswitch-netcontrold image fails as it needs to be converted to use\nvirtualenv - added into unbuildable images for now.\n\nSee https://etherpad.opendev.org/p/kolla-bobcat-ptg#L199 for more information.\n\nDepends-On: https://review.opendev.org/891946\n\nChange-Id: I45b5a5980fb2318f1aab23a3a6aaf0922563b171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fa2ac132cce94aae8cd3d196abba342394f6c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fa2ac132cce94aae8cd3d196abba342394f6c70"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
