)]}'
{"id":"openstack%2Fkolla~886093","triplet_id":"openstack%2Fkolla~master~I6a67910965e38b53958fc6f8546fa6f665a47a16","project":"openstack/kolla","branch":"master","topic":"bump-proxysql","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-21 09:32:01.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-21 09:51:35.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-21 12:36:00.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-06-21 12:36:00.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-06-21 12:36:00.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-21 12:36:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a67910965e38b53958fc6f8546fa6f665a47a16","subject":"Bump proxysql version","status":"MERGED","created":"2023-06-14 13:15:48.000000000","updated":"2023-06-21 12:38:22.000000000","submitted":"2023-06-21 12:36:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886093-bump-proxysql","meta_rev_id":"97a767c1522d29a219426f1bccbf62d8635f657c","_number":886093,"virtual_id_number":886093,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-21 12:36:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-21 09:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-06-21 09:51:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-20 13:18:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-20 14:09:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-06-21 09:51:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-06-21 09:32:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-14 13:16:08.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-06-14 13:31:02.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-14 13:35:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-14 15:38:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 12:29:39.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-19 12:29:39.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-20 05:50:31.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-20 14:09:32.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"ef55dad9a785384231d77025c585faac9c915089","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-14 13:15:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ee9243ff12095cb67682ae63fcf149a5dd8799a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-14 13:31:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"710b28ed46bdab5a8049ef673603539b11cee2d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 13:35:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce730546bca74551b5136690f271e821\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7615c59c6b4c4e1a8c0968a8dbd11bc6 : FAILURE in 16m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c3342820a16343098f09d05d82923463 : FAILURE in 15m 59s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/15eea20fbdb2443e99eb15f3d4c30bb8 : FAILURE in 17m 18s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1569df471d244cc4b2ebfce91e56c024 : SUCCESS in 2m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fe6fe0b0bac4c2fbea4cd2270cb6d2c : SUCCESS in 2m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/53501e6c731c4feb9adf0627d7683c5f : SUCCESS in 2m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5775aceb8f714cfba3dd1620ca663be6 : FAILURE in 16m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d6419a20225045dfa588016c921540e3 : FAILURE in 13m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6dd8d70eace44788fba7b9c2f852967aefbabad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-14 14:01:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a28d9f2ac1c3d19c1a8cb7aad7538db9e7056128","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 15:31:16.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6a6a04405184d1c969083003732f6eb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ece77f054af54ed990df08c5f2cb1d9e : FAILURE in 1h 27m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/37f447c02797497f86015ee87ab01ef9 : FAILURE in 21m 03s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c8c73062d1a242958f28925ab3d1e14c : FAILURE in 23m 25s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b9f60143d9e440a888a37aa16e6424e : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19f1c548bb04423abb1b3b3960218066 : SUCCESS in 2m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f621bdffac2449159b72921bc5092fee : SUCCESS in 3m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f6b3f37f82f04eb886c9d9bac05e54fa : FAILURE in 19m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ab9ffaf2d63d44de941a100268901d57 : FAILURE in 17m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6c0e25496b352775f6d2fc0201393d4e54c390c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 15:38:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7457c2940f72443fb2819de6c806d9c9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/57923637ca014518927d661edea8ca9a : SUCCESS in 27m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/850f7ebc5d3e48a48461e1217e1038d6 : SUCCESS in 59m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f4789753eb2a42eab186a3378f1b698b : SUCCESS in 14m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a602b75d4c644e1d8aa581ca64ff4009 : SUCCESS in 52m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8e08341676624088a3e5520accf703b4 : SUCCESS in 1h 17m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55e285f3e4d5453e9313b7f29bf4558a : SUCCESS in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9c45cd6a8904339a5b3a1f7f0b7c8ea : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a88bff5f26b54c6182c42fb00f747aa6 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2038d7b685e443eae81c47732d35603 : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f5360a1a4754046b67e711eac835cc6 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465a0ad54a8141c1ab29590d97210575 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d3cd54d308348f896d044e5999647f8 : SUCCESS in 3m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/554f931337044c9f832baf90b278de97 : SUCCESS in 20m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/665a66d6760b4754b16f1a04cfb4a8ff : SUCCESS in 1h 01m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/559e004dcdfd4520a9de5d28b5ba7f02 : SUCCESS in 1h 08m 06s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c8b81ad8b9304a34a3dc4beed1f9c431 : SUCCESS in 1h 29m 25s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/19246af2034249189ed46ef4c8338d28 : FAILURE in 1h 05m 58s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/422d13f27f714bc0bb37f835843a713a : FAILURE in 53m 09s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/2d78f902c5c041ff8782bfe5430a1642 : FAILURE in 1h 02m 30s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/6dfa44f9224a4b0ea9faf58125deb53a : SUCCESS in 1h 10m 25s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/e6f7436f4f6748a78e938d8e476dadf8 : SUCCESS in 19m 51s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/72ce67a4f8c34d718d6ae5a7c745b01e : SUCCESS in 1h 06m 50s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/bceaa6700c774d58b1a69a6507668eb7 : SUCCESS in 1h 04m 47s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/26e064f7447e4edd9ddf1b2ae10bc373 : SUCCESS in 1h 21m 13s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7c496dadaed84d31b6824089577cd752 : SUCCESS in 31m 26s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d5a47ad7e92d45e6a17ae8f58f0b2805 : SUCCESS in 26m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c3f20b6d84d44852b59f136fbbce62b6 : SUCCESS in 53m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a10fa550525c407d9add1a4c2165fbf9 : SUCCESS in 1h 02m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7705e8614cfd4310a88a7c3ec577e898 : SUCCESS in 1h 21m 20s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/40d9c80899a14a6e96205cf287aec181 : FAILURE in 1h 34m 21s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/b2f8b6c93a384c9fb0820b81ea026798 : SUCCESS in 1h 20m 27s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1cded2b9aa504a35a1bb5d6a1f01c7c1 : SUCCESS in 1h 07m 07s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/e79505060edb43378040ce2ef07c943c : SUCCESS in 1h 05m 53s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/8a8ea497952748478adbba2a91a12bdb : SUCCESS in 26m 25s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f39e6551ae9d4cdda14f8a273e62dca6 : SUCCESS in 56m 16s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7a4214a4d1d740a2983b11d3d297bc1f : SUCCESS in 1h 05m 32s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/7d3bc72db3c54dc183bf93ff941c0f67 : SUCCESS in 1h 32m 03s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ea9890e1c31f4310b2ba0a566c688033 : SUCCESS in 26m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"720bea6b2192a2382c8dcc89c3675b08d5c256ed","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-19 12:29:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aaac8aa97d1d10ecf5d3341ca9458195317f4eb9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-19 12:33:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"02f999859f2210c5afb8cf287f072692748d34f9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-19 14:30:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"38455c9c71d091cbed24faf73adf4848616de55e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 17:20:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"740d9cf7d0e514b1ace92b81d177caddf2722512","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 05:50:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7728a08ef31ea188dd9ea7aa2af5a4a439921449","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-20 10:40:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"759008e3d50682d2ec629ecd68d77cdb571cba10","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 10:42:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d6f433d2c7105f869b460c3bacfb2157d56e3f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-20 10:43:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3809766874d17924f3eb0056d043aa9ce0042b4f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 10:45:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d67e2d9d5cee274bb66e27e4731f18c21b9995cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-20 11:00:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"1c57c603ed2e5ccddfb402ff3d9f01cdddd305a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-20 11:05:54.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ea50177c5e2380bc1c9a17981e91bd9376888972","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-20 11:06:25.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4c62ae730a0cc393b840474e6f81d402bee08c95","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 11:09:51.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c28f133b20cf85e13b0afc6779528cb84c17899d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 11:40:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfc73919678c4b8c86e269cf0eb459e5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/601278065a384715b67d33f1ea38f5cd : FAILURE in 22m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f509b65cf73c4a4881cc691f5457f567 : FAILURE in 24m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6effda8762664145bd15097da1c3934c : FAILURE in 29m 41s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ed13830f659483ba36e441aebe8cf83 : SUCCESS in 2m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b15a77106544181a8908d96184a0ffd : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d6be4576ef8c4c21ae5308f4d185c0cc : SUCCESS in 2m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/308bc8b23e124da2b80fc26fd3a0256e : FAILURE in 26m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f48f5d9f832c4223ad88caa80f0e5296 : FAILURE in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a79242ec246013805aa9abec020baf3097bac842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 12:58:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26dfa1505e8849eeaca73d19153eb456\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3a336ceaa00d44b0807e6ca4f667012b : FAILURE in 30m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6236e655aeb74f2fb4653ccbbfd95a2c : SUCCESS in 1h 02m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0671c768762b4faa823b882e9e59b41b : FAILURE in 16m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/befb46e857a54c8785c425c47c29df09 : SUCCESS in 58m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e4d6fefc98484409aa21cb587fab5042 : SUCCESS in 1h 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fef31db018d94fa8a09947b38e834763 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c63cdb5029324a09b5896f8831dfacbb : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0b16785ea5340b4830efa69e0e1bf0b : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd81ada3152448eabc77080308298f87 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5105c0f8981848c7b48e49a097efcc3d : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e293d5dcc444798309dd54e34f18c7 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/210eb565a5ab4076a7cabfe9d7b55462 : SUCCESS in 3m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b52c43dea8fb46feb4f2daed636b68df : FAILURE in 22m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/090df6a046b9478b907dddde2fbd50c2 : SUCCESS in 1h 02m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/567f22693664415fbd4901f52110e108 : SUCCESS in 1h 10m 12s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/dfc34bcf2ee64c7f81aa5ed0582b05f3 : FAILURE in 1h 35m 46s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f81e3b6bcb524067b63578549df6c79d : FAILURE in 1h 32m 31s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/8c4d3585cbdb47ddaf4e3bfe8d83925b : FAILURE in 1h 09m 30s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/dc17822ba36947ef9a501085170d2596 : FAILURE in 1h 05m 44s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/e895ad7746964e439214da145a1533f0 : SUCCESS in 1h 37m 31s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/fa734f1f6b894033a0c1bb84f33fa951 : SUCCESS in 24m 35s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/346ee04add3d4235b5cea42ac2921b23 : SUCCESS in 1h 09m 48s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/1ac4fa96fe534d618dec97569c378c07 : SUCCESS in 1h 09m 27s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d6d0ebc075d34a799538ec00ebeb8c69 : SUCCESS in 1h 16m 56s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b59afc54fe2e468bb77f83a627fb63c6 : SUCCESS in 30m 28s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/0437aeac6ae44bc88d8af83283e7647c : FAILURE in 1h 19m 21s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/17ec802a5e554d5e8cd20a90e1e61231 : FAILURE in 30m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f4a1fe9b4b144886b594b8d2bbf6ae9d : SUCCESS in 56m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c86475d82bb64df59650b9a8b930bbc6 : SUCCESS in 1h 00m 57s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/bd96abc458e746b298fc67be945ee74c : SUCCESS in 1h 24m 03s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b2a66a5dbd2641f3b783276120e41f3c : FAILURE in 1h 35m 12s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/c6fb6a839c5c496291702646a1c2c67c : SUCCESS in 1h 24m 03s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/76a27901176f42d0961a80fcffd4388d : SUCCESS in 59m 51s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/c8070d7531414ae89e7f813928ce00e0 : SUCCESS in 1h 09m 49s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/4c29efd6fc2048dbbf979b24e72d170d : SUCCESS in 23m 45s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5dfea1edb06246849991856875ba1d49 : SUCCESS in 51m 06s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1bd2c416b89a45b0ba3f4241c68a5131 : SUCCESS in 57m 41s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/bb274ac8bba54273a94f74aafe32c47d : SUCCESS in 1h 07m 18s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3396bb6df14b4514821990896642d717 : SUCCESS in 30m 39s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/eaef5f687892467fa9d446929c0ea17c : SUCCESS in 1h 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f7260a97fcb66b873ca5805322c3f04f1a3a9c51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 13:18:18.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"60daf3b8db8773f7bb00c67a0ed5de28221e88c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 13:44:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48525d75117147f49e7ff2540acc03cd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a6704d20ffad40c6901e8a2d37e7e5a2 : NODE_FAILURE Node request 300-0021495527 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3dc2b008b06443309b41d8d5ae3b1386 : FAILURE in 25m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6fa60b50f395449d8d01e34aa0c0ad44 : FAILURE in 23m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/19fbb33bfd8f45fa8cc1cbd93be8eea8 : NODE_FAILURE Node request 300-0021495530 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f79d25722c5401ba374c90ec5f22a0f : NODE_FAILURE Node request 300-0021495531 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ee5dbaf80e24e32ad36308b971d27ac : NODE_FAILURE Node request 300-0021495532 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a1f65fb802f445ef9fb6db23ce7a5e24 : NODE_FAILURE Node request 300-0021495533 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3d07ea30619b4384bbf15f99f11c7781 : NODE_FAILURE Node request 300-0021495534 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7b318940361d31fd10e0ac6556cb9facf293f51e","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-20 14:09:32.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"35e46ae4381a9a87c5c04df8a1eec2654018fad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 16:35:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0503484b4eac4a0180ab817307ac5a53\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/09d41b9de988460185fd3f73a0225cec : SUCCESS in 29m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9a234d213f434f59827f90de7dc11f75 : SUCCESS in 54m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/17b71ba90c5e4939b5b4288f4c7803c0 : SUCCESS in 19m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1b7e4426e0af4b1386130b10a315a185 : SUCCESS in 1h 07m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0d2f46b4de6e4b74825933a1e1aaa071 : SUCCESS in 1h 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/285aed2d46344c509728e019bc1a9515 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/012cf21f542243e2a69c603b37637e1a : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0891f8082e3947b995aaae4c4f7467c4 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d3b15bb35d04aeb911a732b3b051067 : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b609bcbbbbe845adb2259246f419bba3 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68acdfbeb96742568949d09a020feb85 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/873a1c9d73794ad3b84b1acdcf3558a7 : SUCCESS in 3m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f7dfd02d7a024812894ca609867aa9db : SUCCESS in 27m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/435e0118ec724b78ba927d73c7ec1289 : SUCCESS in 1h 05m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/50fc0b57267b41569390699ce48e99c8 : SUCCESS in 1h 07m 57s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/aff1a33f39eb445bb7f1d65048c2c3dd : SUCCESS in 1h 53m 38s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4f5103b1edb74d70aed8b286369db001 : FAILURE in 1h 27m 02s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/ca390abf2eb742bab2038bcdf2e69da8 : FAILURE in 1h 31m 41s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c83979b401a3471ba4d78dc6d883b4a9 : FAILURE in 1h 18m 32s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/078114fe6f124da7ab61f9701748be96 : SUCCESS in 1h 45m 38s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2186cc25ca254255962a73080b5a6f53 : SUCCESS in 29m 07s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/2a7c4ccec47d4cf88b6e42fe9f787adc : SUCCESS in 1h 21m 00s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/35e4a34254954bbe8e51d714d08d09be : SUCCESS in 1h 10m 43s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/834b73e402c24962b2ce7d6ec3401ce4 : SUCCESS in 1h 26m 57s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1fa07abe4e7a4f28beb58aba36d79816 : SUCCESS in 34m 38s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/ca17f87e6dcb49b9ad5d8bb64d952837 : SUCCESS in 1h 35m 15s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/19f73ecd79d34967aecd284fedc6ab5f : SUCCESS in 23m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ad5d48c2cb2c4ac3928d7c33fa3ff314 : SUCCESS in 57m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/789558f47b384de593a8e738f36eed94 : SUCCESS in 1h 08m 01s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/275dddf6d2234569bfe2d5125cd4cff5 : SUCCESS in 1h 22m 48s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/49f3b58b33d741c5b796cd70a8c6406a : SUCCESS in 1h 41m 18s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/80b83c97060f47d5801e41cd9d810df0 : SUCCESS in 1h 28m 08s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/432be1312b11477c8347984144914034 : SUCCESS in 1h 36m 20s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/f799fa1b5a504619a98acf549b94491e : SUCCESS in 1h 12m 19s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/6ee32d5afcac4bcaa31a982166719286 : SUCCESS in 32m 40s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/8d095c92a48a42319eab2b0b784e8f37 : SUCCESS in 1h 14m 48s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/17c1c8db73484a93909c8c0818c474a6 : SUCCESS in 1h 07m 54s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d5a32296a1e0414c8f6a378c03ae9b1f : SUCCESS in 1h 41m 34s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b22a72676d01436daf29b91bcc1252ed : SUCCESS in 31m 10s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/d31032e919834dd1aa7dc589d30fb8a5 : SUCCESS in 1h 48m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1b598a1cb2285404adc387c1d5b9148a715e65af","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-21 09:18:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"94bf7323eee001aed293b687469529f7cbc6908b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-21 09:32:01.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"c1bc491b2acf77ca3b39cf9ae0e82f2a69b5537b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 09:51:35.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"cd9b4e4b1a4a3a940e913c2679da8710164b5e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 09:52:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"10d079c1e4524377df6e5c6088bd4573a433ce11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 12:36:00.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/531105df1c3f418ba2804a0174b71e06\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b3d1e4b2cb134e128e9e87e9e5dc04a4 : SUCCESS in 17m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/60a525f67803463481de1a58a95e98ac : SUCCESS in 1h 12m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/13ea7ff04aa744afb0fb0b3249e1fc10 : SUCCESS in 1h 21m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ae2dc1269d4654b3a9250c93b4f6e5 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/599c67cff7464ecea7291919682de2de : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/182cd577ce724d309d5814b7c4eb53da : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c82c65755ef430a9b38c1d62ccd45a6 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2474bc5ef147e182c510aebe231675 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad792248ac344713b227e98cb5671eaa : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/19d1df74623f4d69a60a2f27701a6c3d : SUCCESS in 24m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4e14445cb99d47e387aba40609784044 : SUCCESS in 1h 26m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/35064ed135094cf3ab900b38b06996b5 : SUCCESS in 1h 31m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0b6d97af327d4790ae1baabd082763f2 : SUCCESS in 23m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/aee4f0dda7294297b71112a270d4f0e1 : SUCCESS in 1h 13m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1c7a2839a5b7498b8906897443a13cd3 : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":6},{"id":"1f5745d798963922d0c0f63478a36dacbd303691","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 12:36:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"97a767c1522d29a219426f1bccbf62d8635f657c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 12:38:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9cbdd1189f446de9b08ced3087c273a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3712752475f4fbcac5009efb5dd06fb : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c23942a5252467eb4d82c9d121a619f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"60603d240cf26fa9bed8f7883ccfd5ae51c36f2e","revisions":{"ce88a609d592a6c391d3225394aa42ce7762ee1e":{"kind":"REWORK","_number":1,"created":"2023-06-14 13:15:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/93/886093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/886093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/886093/1"}}},"commit":{"parents":[{"commit":"9035251443ed6cbb110e9dc5efdbc9d4be797ef5","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9035251443ed6cbb110e9dc5efdbc9d4be797ef5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 13:01:50.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 13:15:46.000000000","tz":120},"subject":"Bump proxysql version","message":"Bump proxysql version\n\nThis patch bumps proxysql 2.4.x to 2.5.x\n\nChange-Id: I6a67910965e38b53958fc6f8546fa6f665a47a16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce88a609d592a6c391d3225394aa42ce7762ee1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce88a609d592a6c391d3225394aa42ce7762ee1e"}]},"branch":"refs/heads/master"},"cacedc3a6524a07f2394431237fe377d367e75f8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-14 14:01:20.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/93/886093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/886093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/886093/2"}}},"commit":{"parents":[{"commit":"aa913ecc34dda3883948e87a6e0e8167ee485c5e","subject":"Fix rabbitmq image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa913ecc34dda3883948e87a6e0e8167ee485c5e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 13:01:50.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 13:53:53.000000000","tz":120},"subject":"Bump proxysql version","message":"Bump proxysql version\n\nThis patch bumps proxysql 2.4.x to 2.5.x\n\nChange-Id: I6a67910965e38b53958fc6f8546fa6f665a47a16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cacedc3a6524a07f2394431237fe377d367e75f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cacedc3a6524a07f2394431237fe377d367e75f8"}]},"branch":"refs/heads/master"},"77d83547f34c1d8943a1e4a5358417ecea111d7a":{"kind":"REWORK","_number":3,"created":"2023-06-20 10:43:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/93/886093/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/886093/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/886093/3"}}},"commit":{"parents":[{"commit":"232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba","subject":"Merge \"Fix cinder-volume: Add forgotten lsscsi rpm package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 13:01:50.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-20 10:43:35.000000000","tz":120},"subject":"Bump proxysql version","message":"Bump proxysql version\n\nThis patch bumps proxysql 2.4.x to 2.5.x\n\nChange-Id: I6a67910965e38b53958fc6f8546fa6f665a47a16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77d83547f34c1d8943a1e4a5358417ecea111d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77d83547f34c1d8943a1e4a5358417ecea111d7a"}]},"branch":"refs/heads/master"},"b6493797cf0fb36d00e42e3dc207e36dd8fbee60":{"kind":"REWORK","_number":4,"created":"2023-06-20 11:00:45.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/93/886093/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/886093/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/886093/4"}}},"commit":{"parents":[{"commit":"232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba","subject":"Merge \"Fix cinder-volume: Add forgotten lsscsi rpm package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 13:01:50.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-20 11:00:43.000000000","tz":120},"subject":"Bump proxysql version","message":"Bump proxysql version\n\nThis patch bumps proxysql 2.4.x to 2.5.x\n\nChange-Id: I6a67910965e38b53958fc6f8546fa6f665a47a16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6493797cf0fb36d00e42e3dc207e36dd8fbee60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6493797cf0fb36d00e42e3dc207e36dd8fbee60"}]},"branch":"refs/heads/master"},"c18dcfadfd5deea6c1ea3ae6cf8b9d075bde273a":{"kind":"NO_CHANGE","_number":5,"created":"2023-06-20 11:05:54.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/93/886093/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/886093/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/886093/5"}}},"commit":{"parents":[{"commit":"232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba","subject":"Merge \"Fix cinder-volume: Add forgotten lsscsi rpm package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 13:01:50.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-20 11:05:52.000000000","tz":120},"subject":"Bump proxysql version","message":"Bump proxysql version\n\nThis patch bumps proxysql 2.4.x to 2.5.x\n\nChange-Id: I6a67910965e38b53958fc6f8546fa6f665a47a16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c18dcfadfd5deea6c1ea3ae6cf8b9d075bde273a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c18dcfadfd5deea6c1ea3ae6cf8b9d075bde273a"}]},"branch":"refs/heads/master"},"60603d240cf26fa9bed8f7883ccfd5ae51c36f2e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-06-20 13:18:18.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/93/886093/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/886093/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/886093/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/886093/6"}}},"commit":{"parents":[{"commit":"11f90f3746f6affd34885019a82eb94d3c098ee9","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11f90f3746f6affd34885019a82eb94d3c098ee9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 13:01:50.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 13:18:18.000000000","tz":0},"subject":"Bump proxysql version","message":"Bump proxysql version\n\nThis patch bumps proxysql 2.4.x to 2.5.x\n\nChange-Id: I6a67910965e38b53958fc6f8546fa6f665a47a16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60603d240cf26fa9bed8f7883ccfd5ae51c36f2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60603d240cf26fa9bed8f7883ccfd5ae51c36f2e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
