)]}'
{"id":"openstack%2Fkolla~886346","triplet_id":"openstack%2Fkolla~master~Id34d9b8d6ddf7ca363ded63a18fdfd11173ddc39","project":"openstack/kolla","branch":"master","topic":"fix-bifrost-build","hashtags":[],"change_id":"Id34d9b8d6ddf7ca363ded63a18fdfd11173ddc39","subject":"DNM test bifrost build","status":"ABANDONED","created":"2023-06-18 20:39:47.000000000","updated":"2023-06-18 22:29:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"24e8dd5329926e4f7d8503dc79c8a7479ebdf393","_number":886346,"virtual_id_number":886346,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-06-18 22:04:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-18 21:09:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-18 22:04:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3aa9797a4fffff92d9cf043226a71974ee6fd58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-18 20:39:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65a683cf954494409c67c116d7f6556dfc56c118","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-18 21:09:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4dc135172194d2ebcf7c4e850156a9e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/83ceab380dba42bf96c942b46cc623f0 : FAILURE in 25m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4f74bdf5e67c4e7c8d43fbba94f5a8ea : FAILURE in 23m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/183a4157127d427ba18788649e5a0151 : FAILURE in 25m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a68c09c9519244f59f9bc4c7b177f64a : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff3b20a32b1642999d880d25daf723bc : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d1568ea123e24699ad8243e74d2def0f : SUCCESS in 3m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d9e2e9ece6cd46b9b7fc69e25975ca85 : FAILURE in 28m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/13bc713b924743fe9be45e5650d0604d : FAILURE in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cde69ada475c61746cc71e8a60fe2adb0b275c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-18 22:04:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d7a513f5b3c4d4f96d50c80eac13be9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b3bd6f0a6af0405ebd5ff3904cf58283 : SUCCESS in 18m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/70c3b6caf809481abe08a08349067d76 : SUCCESS in 51m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/278fb7765df84d2796b17a82b60eae5f : SUCCESS in 17m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1a119935587c4fa78cdffc9748e1fba5 : SUCCESS in 52m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/92383d1edd1345bf8c6b6f575d4ef70d : SUCCESS in 56m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f8779a82dd04b4292bac0f9591fc47a : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9c26af26fba474daefd44eac59a5ad9 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1df08fef0e7b4b0fbe06c82a8bc8a4fe : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47b8d151bc5e485f872a663ea8dc16b2 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db3de8d087e54e06935fa70fff8c35cb : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/960cc43a7990470d81faa0f8655e2426 : SUCCESS in 5m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ca5a2164d77542cfbb17a7c3e37ce500 : SUCCESS in 16m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ceb3c9afdaa44fe196485095ce210a5d : SUCCESS in 1h 09m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/edb6f28a46a64ca9b12a0c3401274ce6 : SUCCESS in 59m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/589ee554297f4f2f9695998ca6efaefa : SUCCESS in 17m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f2f277bcadb943439de27293d5cb8c5d : SUCCESS in 1h 00m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4ccdbde785f84feb82e9055254834595 : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":1},{"id":"24e8dd5329926e4f7d8503dc79c8a7479ebdf393","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-18 22:29:26.000000000","message":"Abandoned\n\nwrong test","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b66ac5dd6dfb653b5138cee7de5fd29c3593f6a","revisions":{"6b66ac5dd6dfb653b5138cee7de5fd29c3593f6a":{"kind":"REWORK","_number":1,"created":"2023-06-18 20:39:47.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/46/886346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/886346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/886346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/886346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/886346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/886346/1"}}},"commit":{"parents":[{"commit":"4950c03e59b8d1af46018b81a49fa028e1b8296d","subject":"Fix rabbitmq image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4950c03e59b8d1af46018b81a49fa028e1b8296d"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-18 20:37:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-18 20:38:52.000000000","tz":180},"subject":"DNM test bifrost build","message":"DNM test bifrost build\n\nRevert \"rocky/centos: mark bifrost buildable\"\n\nThis reverts commit 878e9c0524e97dda7d61c6161d33e26052d6285f.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/886344\nChange-Id: Id34d9b8d6ddf7ca363ded63a18fdfd11173ddc39\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b66ac5dd6dfb653b5138cee7de5fd29c3593f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b66ac5dd6dfb653b5138cee7de5fd29c3593f6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
