)]}'
{"id":"openstack%2Fkolla~886357","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I76ac61604f5b0739ec27973a233a65e4c55c8c32","project":"openstack/kolla","branch":"stable/2023.1","topic":"fix-venus","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-19 15:13:24.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-19 15:13:24.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-19 15:13:24.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-19 13:34:59.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I76ac61604f5b0739ec27973a233a65e4c55c8c32","subject":"Fix Venus containers built from correct branch","status":"MERGED","created":"2023-06-19 06:45:55.000000000","updated":"2023-06-19 15:14:36.000000000","submitted":"2023-06-19 15:13:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886357-fix-venus","meta_rev_id":"85f6adf88383841ee7c5ee148e27b21867fe791c","_number":886357,"virtual_id_number":886357,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-19 15:13:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":2,"date":"2023-06-19 13:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-19 13:34:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":1,"date":"2023-06-19 13:55:26.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 06:49:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-19 06:49:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-19 06:49:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-19 07:12:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-19 08:30:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 13:55:26.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"8022020dbc8522bdb8cca769a352625037de719b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 06:45:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c29119ea1273950a570b873bb9cae17267227318","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 07:05:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"05ca31c4e88e393997a2ea1bcb0d06285177c6c7","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 07:11:27.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"1d319539eb20d56b0418b5efb9c5dfa5d3567546","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 07:12:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0316a7e4e95e4bbca2161223b336f7c3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0596932c7bf34ea7bdc3d3969363f600 : FAILURE in 20m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0a792dcf589043f0a23ac146de057d17 : FAILURE in 20m 11s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7e6c22e283f74e52aa4be0ccd36530c4 : FAILURE in 23m 26s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff9acc5fc0634a77a3e4c6de8f5c9af4 : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/93ff5331077f4c95a1cb6f848e1ea264 : SUCCESS in 2m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/637bbc810bf642de89686eaf3275c13c : FAILURE in 22m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/188e41878bcc4c31885bdce67c93f482 : FAILURE in 6m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c362b24f27adb7703dd51ddd37443064d4f88eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 08:30:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4ac9650e6b949a28958149318077ca2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ae231377972d481db3a06e8bd209a6a4 : SUCCESS in 22m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/02bffa09631b4ba1a967d8ae230317d8 : SUCCESS in 50m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/59f574946cfb46019ba2491b265f8dc5 : FAILURE in 22m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8423ac21fa114abfb51f1507ff2b5c92 : FAILURE in 16m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/82c1d02e65754f92af9fb7f9caf888df : FAILURE in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e788fb19ae4d6ebdbb412d462d4e07 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/329f52606027414cb17d2545bf6733fb : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2a008324bf04584ab3052939849963b : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddb1b7ea086c4b65b7c6b1c553de3ab1 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e82c62d6ef24dba9a669feb8e6be4bd : SUCCESS in 3m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d7e1b7adf8534f26a4ba8b497297401e : SUCCESS in 21m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b6ed5a34f6184b20af956b5f1f6acfb9 : SUCCESS in 1h 07m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4798f0cf3e4a43c3898cc735948b5c42 : SUCCESS in 1h 10m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/08e3097e61624bb291a73603c8cd3204 : FAILURE in 14m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/288132788be54c6aa7cef15c97293ec6 : FAILURE in 13m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/72d7f724b7a0416085af74b341fbbf86 : FAILURE in 13m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ee7d3bd9b6a347d90cef04176165cd1483b43f31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 10:40:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4e4bb7c34bc9f638daf742acdff7ecc28ec7ab2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 11:04:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e06627880faa4e2397ec10b27e8579da\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/665c972537514a92b4c0b715b54b3917 : NODE_FAILURE Node request 300-0021486078 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3ec37fece2704711b5d49578f66da604 : NODE_FAILURE Node request 300-0021486079 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9e635464e39645ea8326de3d3b446f23 : NODE_FAILURE Node request 300-0021486080 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36c929eb4bda4233979aeb5cc3219559 : NODE_FAILURE Node request 300-0021486081 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7044d057d7bb445c953c7e708c560716 : NODE_FAILURE Node request 300-0021486082 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/121c83b29c3b4401bfc36078475d8f85 : FAILURE in 20m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c51cc0fffe0f433cb063fa83fc2d77b4 : FAILURE in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"85afc83a321553c10f710ed352cb93c64f306638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 11:52:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1102e59683244f1088aed3e27998667e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/53c8b792c3694868b7886a38a4e880c0 : SUCCESS in 21m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d11878c77c024ddc8a6c28781e3e84eb : SUCCESS in 58m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5baaff49822f450d8595bd58424ffe28 : SUCCESS in 16m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/352fbc9d49d44c03a8f9b186e3e68bf6 : SUCCESS in 54m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dd5b2a749a5043e59a39e9a188e8a451 : SUCCESS in 1h 01m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a744b0db9394b9a9b20828743704957 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5248dd38618e49d1a75bf91f0fbd2f23 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9052d7e7b75f4d09b3a2acc9966fd5fd : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edbf69a8d5c8426f859b99a3890dadf4 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec34411282174c4392aa244eabdbb056 : SUCCESS in 6m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6b19fd6b84b04b36b04dbe360dbec62a : SUCCESS in 21m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/43d22bdd2e1343449a6738282418aef4 : SUCCESS in 49m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/464ff38283494d1bb6b17b7b584059c9 : SUCCESS in 58m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e51d79a8313d4e44be7fa864c43d8703 : SUCCESS in 19m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f4802f5e69ec4e62aa25f8ea493de32f : SUCCESS in 1h 06m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/683266957ebb47b4abb0e15588ee6dab : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":2},{"id":"3adf6606f7821615fd9d4d3ca8c1b12af98dbeb0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-19 13:34:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"47409b369410b9e62cf3be4b8dd94622459ad88a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-19 13:55:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5e76a84c5193b891956a262bea03ef3986ffb199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 13:55:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4c7b3bce9183e44cc14229b19766e1a220c5728","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 15:13:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b07d0cae13ea4c739f7d48db24b3c430\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/04e9ab33b42945f5aca4bc80310d1550 : SUCCESS in 19m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/369bb11116784b5880a9a29b4027c603 : SUCCESS in 55m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e91bf59d34f94be5a2fa65fc56a1d621 : SUCCESS in 1h 07m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63eb38d148ff4ef49e548090be9bba47 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae07203a507a4695a3754fac25152c38 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d32b85dbe85c41159dab0b07f37db37b : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a3ee7cebdf4d92bd689d7376549fc7 : SUCCESS in 4m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0de3a9873c4441afa9045efad22bd8a5 : SUCCESS in 25m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/51f5093da1464858a1d045ef0d4b811c : SUCCESS in 1h 01m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5e240ee77d7f4142ab10da82103f5e8d : SUCCESS in 1h 11m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6c04a93453434a7189fc64b338bec347 : SUCCESS in 18m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5405f90a466242b294bd63dabee6fc02 : SUCCESS in 54m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/805aeff8c66647c0af23c4a50f143385 : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a43e6975c119b94d7863dd050415b76ef8d8d1bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 15:13:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"85f6adf88383841ee7c5ee148e27b21867fe791c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 15:14:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a363fed4eb7847f39b9aeffcfef266e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/653d5bff81414ee3a157baf083c73cd4 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"30802ae09487974e9f599b677790894f9fcb9adb","revisions":{"b69716928b6817e1b10951959a772691f126bf01":{"kind":"REWORK","_number":1,"created":"2023-06-19 06:45:55.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/57/886357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/886357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/886357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/886357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/886357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/886357/1"}}},"commit":{"parents":[{"commit":"2ca7ec16135bf62637ca1115e122f66cb1eefb27","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ca7ec16135bf62637ca1115e122f66cb1eefb27"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-13 10:39:30.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 06:45:36.000000000","tz":180},"subject":"Fix Venus containers built from correct branch","message":"Fix Venus containers built from correct branch\n\nCloses-Bug: #2023628\nChange-Id: I76ac61604f5b0739ec27973a233a65e4c55c8c32\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit b0804cd04cd2e47cf18666fc420ebd42e69aae75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b69716928b6817e1b10951959a772691f126bf01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b69716928b6817e1b10951959a772691f126bf01"}]},"branch":"refs/heads/stable/2023.1"},"30802ae09487974e9f599b677790894f9fcb9adb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-19 10:40:15.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/57/886357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/886357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/886357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/886357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/886357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/886357/2"}}},"commit":{"parents":[{"commit":"8dc32be6f1b99eccca5f4f6d15a4e654d4519950","subject":"Fix rabbitmq image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dc32be6f1b99eccca5f4f6d15a4e654d4519950"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-13 10:39:30.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 10:40:15.000000000","tz":0},"subject":"Fix Venus containers built from correct branch","message":"Fix Venus containers built from correct branch\n\nCloses-Bug: #2023628\nChange-Id: I76ac61604f5b0739ec27973a233a65e4c55c8c32\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit b0804cd04cd2e47cf18666fc420ebd42e69aae75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30802ae09487974e9f599b677790894f9fcb9adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30802ae09487974e9f599b677790894f9fcb9adb"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
