)]}'
{"id":"openstack%2Fkolla~886358","triplet_id":"openstack%2Fkolla~stable%2Fzed~I76ac61604f5b0739ec27973a233a65e4c55c8c32","project":"openstack/kolla","branch":"stable/zed","topic":"fix-venus","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-20 09:46:17.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-20 09:46:17.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-20 09:46:17.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-19 13:35:15.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I76ac61604f5b0739ec27973a233a65e4c55c8c32","subject":"Fix Venus containers built from correct branch","status":"MERGED","created":"2023-06-19 06:47:07.000000000","updated":"2023-06-20 09:47:18.000000000","submitted":"2023-06-20 09:46:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886358-fix-venus","meta_rev_id":"44100c5d62485ffdc6ebc05b97a9402213e397e8","_number":886358,"virtual_id_number":886358,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-20 09:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":2,"date":"2023-06-19 13:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-19 13:35:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2023-06-20 08:57:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":1,"date":"2023-06-19 13:55:39.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 06:49:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-19 06:49:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-19 06:49:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-19 07:11:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-19 07:56:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 13:55:39.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-06-20 08:57:41.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"71b071d5fb0190b369f1fb714bbab4068c320be9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 06:47:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a1763911b1d2ecf65b0d567d3f994b9142875f2","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 07:05:37.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"f7a9be43032c7d53d148aa8f575b2fd8670d052a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 07:11:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d906a71ff7c841c6b9f6e15ee661d7b2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/58cf7e66335048868a50506327650264 : FAILURE in 21m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/07b87822bfd74128aa1dfed3ec2e43cb : FAILURE in 20m 25s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc674df01177446abb7bdbffcd027ad2 : FAILURE in 22m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/88b39584b7564b048e505f7362250366 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ca9e2dc6e244a69b8d3b0f1ece8ad72 : SUCCESS in 2m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2b2e0d4455fc4aca872b3646b574a410 : FAILURE in 18m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/211e538125244475bab6e6dbd98efd22 : FAILURE in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc4437e115e534fa7bfac04094d5c29762f567ac","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 07:11:21.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"8695d249f267540cab1e9c6a9f93075aed18f48e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 07:56:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cc266cdaa49471a904ad06958f255f6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/49611e0006aa446e9d30e2c182c9a911 : SUCCESS in 28m 58s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/081b52188e794247953e8e27ffacda4c : SUCCESS in 1h 08m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1784ec8aa46544edb1796fc48ad80120 : FAILURE in 25m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e5e1a2c1e8a64cde898f91e173f9d018 : FAILURE in 22m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8b573577ecee486894d7a0ec2bb8aa49 : FAILURE in 22m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ba4e4e820247f89fad96b9d6926838 : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5a2ee00c7a14c29a60aa00b2352d485 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a435e2fdee0847b4b0f41ed56dd43bca : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/466d9b08dbe44f2fb92ccc7197ace0c2 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f97fe5508bc542a1ae41ea15b0ad8af0 : SUCCESS in 8m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fe5f3c38177a4e56a09cb1c7a8d69e63 : SUCCESS in 17m 08s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8859b7d41eee44a69851a59ed1f032ce : SUCCESS in 1h 03m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/890615f06aae47688f5ab70b5bf87492 : FAILURE in 14m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/33117670bf864e8cb6f9966687150887 : FAILURE in 14m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d6f54a7fa96f4464b4cab04851e319ed : FAILURE in 20m 03s","accounts_in_message":[],"_revision_number":1},{"id":"214d1136a53dc59b6b2de0d4114ede4d747bad7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 10:41:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f574b1c89215bd260c89f2ce621d026f4ff4c405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:41:53.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"bc9bd41100427e50ac2f48ec02b2702b263e17ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:42:03.000000000","message":"Patch Set 2:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"bf611926b5c2e42df1abdc7d61f75dbe4d246839","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 10:59:11.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2ffd0c0bcb2a88d377ec9d284f3fdb81fefa2bca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 11:40:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d25ce2662c2428fb63cd882a820c6a2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f2eddb9158a143009a49d74c2a9f845b : FAILURE in 39m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8e9328c74e1c434281570aa02b7af61b : FAILURE in 34m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b27839e03184a8db65b7ae252b7414a : NODE_FAILURE Node request 300-0021486164 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/abbcd49a081444a299d2bf7c10b5a676 : SUCCESS in 2m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0666159157d447fbb98f42a990cf5a4 : NODE_FAILURE Node request 300-0021486166 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1d0b4257f09e4a57b975d984e2a8c205 : FAILURE in 33m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a6760c31df924a638e249ddf17a113d1 : FAILURE in 7m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"34eabe9b8c69faff1a8169201622ffd7abd884c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 12:23:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6373bbf7c7e44acdbfc23d2c4d17260e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/718cc53179db449b8eb6bd1f02a35b59 : SUCCESS in 31m 17s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f3b439a6c7574c0caedf7b4e9044f538 : SUCCESS in 1h 00m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c7940d463acf4b869610e1c02d722d96 : SUCCESS in 22m 21s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/49d5d1d9562042a897f21181172c92a4 : SUCCESS in 58m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e2854089c2c144f2b1bc7b98b44d0222 : SUCCESS in 1h 17m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6832eeb9d9d54e8ca56a25e28ab285f9 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dae674ebbd744749603409378f830d8 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db3fcf98ed734f7b95bf4a24cd9606cf : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fdf5127bf9b448d39f8c3a2f2ba20bc2 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026620f112f54eb59c4d1e8ccccce675 : SUCCESS in 5m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7cc404897604432194581607846d3381 : SUCCESS in 21m 04s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/b5cabc8e2f4245278052598687bc6be9 : SUCCESS in 55m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/aa23e004ae62476084c915491c85cdbb : SUCCESS in 12m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/920169e59b66466aba5c511294476bbc : SUCCESS in 48m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/633107acca4f4b04952fc29904de0246 : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":3},{"id":"657871751a5ede01f04e2a6754c9ea696c68b7e6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-19 13:35:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"021b82f5542d76713269e3df4355ca9c63ca9ecc","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-19 13:55:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"37db8f0089eea78292b7e4c0ca4959a956934e3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 08:19:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"47c0c3c8f32ebff301acfbe9e5e10eacfe4dfc0e","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-06-20 08:57:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6f25d8b392a2f14c331826d5942f71b9628e466f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:46:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b92694026f94700a37942c91df08072\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2a7a598263ce4037bfedfd8f3d7b5903 : SUCCESS in 17m 23s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8ce290e26e1044c1a111d821386e7f52 : SUCCESS in 1h 01m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3d35f4c3417f474b9f456c3a382d32b1 : SUCCESS in 1h 08m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b66c6e6f274a46af9d93d20215e85f : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53cfa69a91f9473bbbd4b1499aa1dbc4 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/621af560847945f890c6bdb79c0faa8e : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ff86e1e20144bc999572b097a074fd : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/28869f724506472290e1e7f5c94912df : SUCCESS in 26m 50s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/87e1a7b519aa4eb9bf5824e9f567714e : SUCCESS in 1h 19m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ab98afe4516f4abfabd12f9e18aad2c8 : SUCCESS in 25m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6faa25346e364a85ba47a6e2758ad7da : SUCCESS in 56m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/21b090ef07484ba88f7d8ddcd187ea03 : SUCCESS in 1h 11m 58s","accounts_in_message":[],"_revision_number":3},{"id":"4c87800cfcbb75c15d60e1fb401db68607239e4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:46:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"44100c5d62485ffdc6ebc05b97a9402213e397e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:47:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d66b00b28be40a8967c83d04de71083\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd8b5706993418a981b2586c4000d28 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3aecfa93a8f8a2e73f49139346d8a66fbb7154be","revisions":{"157525b03603d6acdb1d579065c278bc8b8146e9":{"kind":"REWORK","_number":1,"created":"2023-06-19 06:47:07.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/58/886358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/886358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/886358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/886358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/886358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/886358/1"}}},"commit":{"parents":[{"commit":"fe9d33c64fff9b4d4607cabd090ead62b052a3ab","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe9d33c64fff9b4d4607cabd090ead62b052a3ab"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-13 10:39:30.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 06:46:57.000000000","tz":180},"subject":"Fix Venus containers built from correct branch","message":"Fix Venus containers built from correct branch\n\nCloses-Bug: #2023628\nChange-Id: I76ac61604f5b0739ec27973a233a65e4c55c8c32\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit b0804cd04cd2e47cf18666fc420ebd42e69aae75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/157525b03603d6acdb1d579065c278bc8b8146e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/157525b03603d6acdb1d579065c278bc8b8146e9"}]},"branch":"refs/heads/stable/zed"},"d27bef4e6e1c9c8db4e2b65785ca50647c74e0bc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-19 10:41:29.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/58/886358/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/886358/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/886358/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/886358/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/886358/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/886358/2"}}},"commit":{"parents":[{"commit":"70ce6050e297888e94ec3c6c7c60e795772a6986","subject":"Fix rabbitmq image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70ce6050e297888e94ec3c6c7c60e795772a6986"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-13 10:39:30.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 10:41:29.000000000","tz":0},"subject":"Fix Venus containers built from correct branch","message":"Fix Venus containers built from correct branch\n\nCloses-Bug: #2023628\nChange-Id: I76ac61604f5b0739ec27973a233a65e4c55c8c32\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit b0804cd04cd2e47cf18666fc420ebd42e69aae75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d27bef4e6e1c9c8db4e2b65785ca50647c74e0bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d27bef4e6e1c9c8db4e2b65785ca50647c74e0bc"}]},"branch":"refs/heads/stable/zed","description":"Rebase"},"3aecfa93a8f8a2e73f49139346d8a66fbb7154be":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-06-19 10:59:11.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/58/886358/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/886358/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/886358/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/886358/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/886358/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/886358/3"}}},"commit":{"parents":[{"commit":"1314b297fd817c1ab68110e2c820bfa85afe4e5a","subject":"Fix rabbitmq image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1314b297fd817c1ab68110e2c820bfa85afe4e5a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-13 10:39:30.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 10:59:11.000000000","tz":0},"subject":"Fix Venus containers built from correct branch","message":"Fix Venus containers built from correct branch\n\nCloses-Bug: #2023628\nChange-Id: I76ac61604f5b0739ec27973a233a65e4c55c8c32\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit b0804cd04cd2e47cf18666fc420ebd42e69aae75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3aecfa93a8f8a2e73f49139346d8a66fbb7154be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3aecfa93a8f8a2e73f49139346d8a66fbb7154be"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
