)]}'
{"id":"openstack%2Fkolla~886380","triplet_id":"openstack%2Fkolla~master~If4e346c2db841aad9b2ddac049dbbbc1ba5782ec","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-11-26 20:05:53.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-26 18:23:35.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-26 20:05:53.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-19 10:52:16.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-11-26 20:05:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4e346c2db841aad9b2ddac049dbbbc1ba5782ec","subject":"Swift: Remove unused rootwrap","status":"MERGED","created":"2023-06-19 09:00:52.000000000","updated":"2023-11-26 20:06:50.000000000","submitted":"2023-11-26 20:05:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886380","meta_rev_id":"94e6bc37946adfd5bba6ec41628347e03312adcb","_number":886380,"virtual_id_number":886380,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-26 20:05:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-26 15:11:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-26 15:11:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-11-26 15:11:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-26 15:11:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-11-26 18:23:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-26 15:11:44.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-11-26 18:23:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 09:22:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-19 10:01:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-19 10:38:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-22 06:47:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-22 08:12:26.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-11-24 10:13:56.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-11-25 11:32:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-11-26 08:35:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a2b88af83cfd6161411c1aadd200e95b6b2b48a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-19 09:00:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b97b8d70bec9366414df9a1b14215ca21603798f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 09:22:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a19b7a0a0d8249e78de995b73ef4e30f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/41db7ec6b6de4db58015c4cee2e7eb68 : FAILURE in 16m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4ced71c322ab4bbfa082c2b179071764 : FAILURE in 20m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3524c151a1b24da4ba31610ad94624e6 : FAILURE in 19m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1cba11c3de7469a8e8f107cc4a76440 : SUCCESS in 2m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2adcb8207184c3a9cac71569962c694 : SUCCESS in 2m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1657cd68e21f489f8b5ca6a8de5126d0 : SUCCESS in 3m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ddcdd16fd0554f52a05164b2e672d287 : FAILURE in 20m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/77726c3854ea4e5187107a8f36e09466 : FAILURE in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2fdff86f7c82539ed1d451568b56211565dcde7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:38:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/026f234c53c44d7a8da1fb3d35cc6cea\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f8c1b0d0a0cf42c7b0b389ff71c4e1c6 : SUCCESS in 24m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/aab9ab86687243c7ac7c645cc003c97a : SUCCESS in 58m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/48bf83cecc6b4b6ea1df1ae5efa389f5 : SUCCESS in 17m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6e0735b4334e4272932f29fb6077fda6 : SUCCESS in 50m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/317284c950a9462b9afa42b7c6312bb7 : SUCCESS in 59m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f31a6f7d4231475a998b98c29bda1c96 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81c1994bd6614487872d4213ff4bdb54 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89e5d1da27f64840bd6aa26987b9d61c : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/badc0308d3f74401a66dbfc22efb59b7 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4ba4f697fd44adcbacf7a2f34e09808 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/443dfb50c7d64465879744ed21cb6b5c : SUCCESS in 5m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/950e5efcf3a240ef906384753a2ea6d5 : SUCCESS in 20m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/927fde12f1b34f64bec6f820c33b2de6 : SUCCESS in 1h 04m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5a8ef09e0f6240c6a22964e634ef5c78 : SUCCESS in 1h 10m 00s\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/2d6c492b6dbb452187861b714a96174c : SUCCESS in 1h 31m 18s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/49d8edfc7beb4567b944d1ea5ec3bb47 : SUCCESS in 18m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/776fc483e618420fa1ab51e821448c2d : SUCCESS in 53m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a90860c7bf1849aba5c5776337bf7037 : SUCCESS in 58m 50s\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/a864e7b0b5964980b21435e8814395df : SUCCESS in 1h 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89c000a8ecd5668b675da4b96a12579ca3f21bd7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 10:52:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b5c1e654beb2a3cdabbf3261c9da0bf1201f646d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 10:52:23.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"378f471988fc65758cf2934ae4982b5bc9c0bea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 06:47:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"d631685faeefe0d906ed75bc6207790cf30824e7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 06:47:45.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"31c81aa4136c3e47511c733cc33768a4d860a31f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 06:47:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3ed2b0a9763d4cfd8baa6d11443bcc53748dfbe1","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-11-22 08:12:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"76b3500e76676a3fc628ef6012fa5f7c32d77cdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 08:42:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bc7fcc12d7a4edcb45a35cb69f87556\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6e61b97033864d71979822ef7101fc53 : SUCCESS in 21m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f565950917494ffaa0a2251896e162fe : SUCCESS in 1h 04m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3841eaaa2b914371a959930843ded5b7 : SUCCESS in 23m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cdcd239e599848bb8b6e397f27cb1521 : SUCCESS in 29m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8036a0b3237240bfa40ba786d5e2d167 : SUCCESS in 1h 13m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/36ef26a0f53947a7bb0fd47af261a480 : SUCCESS in 1h 49m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a47596112a34a9c816ea53a348a3eb8 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c9118f1e6424c01b53cfaf3bc11efa2 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7281d493ea094d5485ad30ae75ec3ce7 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8940901ed37949d2a95795d0a28e03f3 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28607598fc13407cb69fbd4d875b9a4b : SUCCESS in 4m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1eb6acbe7415478f8d3cb798db7ca3f0 : SUCCESS in 18m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/29da24b145294952b5923b0e46421962 : SUCCESS in 27m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cb4ab23b904349e182b4ad72a0702ae3 : SUCCESS in 1h 05m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bc21652de1344d6ebb384a6d8bb5bc14 : SUCCESS in 1h 39m 58s\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/21615c62ad1f45e0bf34f70dc343a213 : SUCCESS in 1h 31m 09s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/351f4a04a075452eb90b4e3a6d4d8563 : SUCCESS in 21m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3de038b945794945bdc70c2ebabec3a3 : SUCCESS in 25m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bbe0459323ca46ac805887978690b1cf : SUCCESS in 54m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/00c785c03a3341f1a59a1b8cff7ac86f : SUCCESS in 1h 40m 53s\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/a822dbf069bc47f0b2c7537e33639bdd : FAILURE in 46m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"27f8725d4a4b27e6b4f9999ef927fbf53f91c230","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 09:40:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25cd9f1c88894a6594d1308093a0ddf2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7cde2b6712e8402e896dd27593dcd960 : SUCCESS in 30m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b6502ddf327e425492550b41a83d6688 : SUCCESS in 2h 13m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/337718db54174f94a8abeffca8e5a07c : SUCCESS in 2h 49m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07ffbcd39a884f9aa920c27b2cbce109 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c403d415aa8347c8ad2d08f485000920 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/76aa5fbc4c3b47cfb12a8e4f1a77b1d4 : SUCCESS in 5m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/153843d045ee49b8a236e02eae287124 : SUCCESS in 30m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/08644e44a9e04c49bc6b9077752ee6c9 : SUCCESS in 2h 10m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1f42d3b67b8bdc4acb12b6e55e667c905f8847bc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-24 10:13:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8093107290a6cc48d7a5e0b6e0755aa301252235","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 10:14:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ddb78aabdbfa336af2a787ce6b9d0a3c1e8899f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 12:28:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b04c4bf5c254885a3971d057f3065b4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9631244caac44221976f29b5935b84c3 : SUCCESS in 18m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/475d4e0a96814e8fb4bb1e8902058096 : SUCCESS in 28m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/38a557287a9c4c67a55395fb0d1f4abf : SUCCESS in 58m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/999dc948f64848f39609f8769d567fd9 : SUCCESS in 1h 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b713edae4645b790b3521b31ec9f7b : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20a7fd761e074f8ebd682d7038f502f9 : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f29175cf795433c9aa57bce7ea995d8 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/386674062cd04bcc91a9f9e2d495ba3a : SUCCESS in 3m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/de2d5def8a4948d3a4a64a9b0acac8fb : SUCCESS in 18m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9bb4005f79fb43f4b3f97f0a0bdfb298 : SUCCESS in 28m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a4497b70bca140c3a1095219c4ac1a55 : SUCCESS in 1h 10m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c5d6f3f03b4e4537add840118e7e349b : SUCCESS in 2h 06m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/99380c847ebf460a986814888d37680b : SUCCESS in 27m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9b7dcf3585ef4d5b8a6a9098372b34ae : SUCCESS in 29m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c7822edc776c4f09a7275d55ed3545dc : SUCCESS in 50m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/80456c86a9db4cc586f4a3d43b8a32c9 : SUCCESS in 1h 34m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5f59382905b87552770d7e13489c6a5b1948929f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-25 03:03:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c68aa087182b5380fb79abfa918f7745a9a35f04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-25 04:55:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/943b057f870e4295b334b1afb27697ed\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1169b54b927a46d5853d36f5173eaa59 : SUCCESS in 18m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fb1e60878e0b4daea375a7d8be8abc8d : SUCCESS in 1h 15m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d6cad755bb564a3c859c958f9e789f58 : SUCCESS in 27m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1c2d2605a36148418f5b5e8ebded66ec : SUCCESS in 37m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e377642016864b44a5ba099e3b73d742 : SUCCESS in 1h 05m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8cf52d97b71d42e58bd89b9f531a41e2 : SUCCESS in 1h 18m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8217bd502eb4385b849dc3e8dcff132 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f25cdedbf4f4927b029d6498e3fccdb : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/113c02ce94ad46d7b2bd39187334f807 : SUCCESS in 4m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a2ddc70c75549d9a9c68145f1273740 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a8bb146c254d6497a4d123c4b9b60e : SUCCESS in 5m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7e5ba996a7bb4e63953ce7024442b61a : SUCCESS in 20m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/40c1e1c700b541a1b07b05a5d8519916 : SUCCESS in 31m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b5800b3b22ba4a86bbeea72cca590422 : SUCCESS in 59m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cb6baea389a949eca04f7af143858618 : SUCCESS in 1h 31m 15s\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/2a73cdbf233444b9953975cf544786af : SUCCESS in 1h 38m 56s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6dbc8580f14644b68464d3bf01c61e50 : SUCCESS in 19m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2dfd9e1e7fc647fbb037389d0bf75e56 : SUCCESS in 28m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a418ddbfd96749d4991e7fea71a92083 : SUCCESS in 54m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8e1cd0d1baea48b48d0775505f7bcf49 : SUCCESS in 1h 30m 34s\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/a691c656348b469baa8b38b632a10ae3 : SUCCESS in 1h 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64ad8a1eaef347cf089401d17308647ff2282da7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-25 04:55:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3df4365dad96ec047aa1b8dd4233d1f4044adc30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-25 05:06:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62641acaec7d408bbe4fe6c1e0849ef6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/766b1944846248e0ad3354473a9d623e : SUCCESS in 18m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5f3fce7c1dc94aca8569013bad02c4f1 : SUCCESS in 18m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e4427f61e1cd46b4820f8b4c04ea2b33 : SUCCESS in 1h 58m 01s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42d5d10fc6734d738bd30cb3ba327299 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bec53dac690e4b159eb654a7c8780d6d : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a750355ab61492db8951178e2d83319 : SUCCESS in 4m 03s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e51d647494264fb58fe657a4c27e55bb : SUCCESS in 1h 17m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6f228a83666b434ca41f724a9a48b5d8 : SUCCESS in 17m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"297c8713f4de18508f50b8d8b979302c6a0303c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-25 07:15:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9ee5de0e3de4e84bf9b0cda4fb63f27\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5cab6f7c9f72442f8e0d485fd03cc1b7 : SUCCESS in 18m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/48ce58c1f9bd4d8faa24710b529f1ed7 : SUCCESS in 29m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/15bee783a4b04e64b30ccf3f8c69ff6c : SUCCESS in 1h 03m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/555ca10016514ffa905800181b9963ac : SUCCESS in 2h 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b85ab025fafe4ef6957d67dbffd49bc8 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69fe2373e0424e88b006a6360b8d82f6 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1c5c3d86a734ba3ae65224a7897e28b : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b3536791a004c84b1913ec9ae105c8e : SUCCESS in 4m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f0bd398f34504c64b33f9fa89f8b09cf : SUCCESS in 20m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b6033838f85d4099a30ff7d03bf9262a : SUCCESS in 23m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ff10cd13624d4f7ba5c4a0ca2f4b0ee3 : SUCCESS in 1h 09m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5fe3174b2b4c4fe69eacef17ba9d6f94 : SUCCESS in 1h 22m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3cbc5f77ffe549268a0c5c9dd1905487 : SUCCESS in 12m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3b63c8a4f26644a5a2883584c1c6c576 : SUCCESS in 23m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/56eea91fb65444c39f30d3e13fe149c6 : SUCCESS in 41m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e30f898ad17848ff99f78cfe7c3d0326 : SUCCESS in 1h 16m 30s","accounts_in_message":[],"_revision_number":2},{"id":"932327cf97ed74e2c5b53d1cd93d4c7743ac5b82","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-25 11:32:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0897e6d6104fbde2bb482f2db88b2850d8f3429a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-26 08:35:05.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5455dc81e0b19dd5e7b7eb09baccd496f1a4b34e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 08:35:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d3ddacf1f0b146a77d1e1d4b0747ffcf9eac8e3c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-26 08:39:16.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_22629\u003e\n","accounts_in_message":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"_revision_number":2},{"id":"96c6d824f5fdecb97c91c40b2f07953cfa780193","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-26 08:39:39.000000000","message":"Removed Workflow+1 by \u003cGERRIT_ACCOUNT_23084\u003e\n","accounts_in_message":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"_revision_number":2},{"id":"337606aeed821cdad8f571f3635263349c8f5642","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-26 08:39:44.000000000","message":"Removed Workflow+1 by \u003cGERRIT_ACCOUNT_13252\u003e\n","accounts_in_message":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"_revision_number":2},{"id":"ebdb34800ba3057159827a0ed40e6307157de855","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 08:39:52.000000000","message":"Patch Set 2: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":2},{"id":"7ce887c5f6d8d8e665ec7e12332e35163b88cf09","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-26 08:39:57.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"37dfcecaf6d5aa45507abd133531e9de8a6358fb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-26 08:40:23.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d32e4ef901928283c23f2afd27f0d36b8b87caee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 09:57:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/431baf3dd0994c2ba077e4428133af2a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/aa37fe559fc74a539ac4c4e7138a467e : SUCCESS in 17m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d4c5b913d9374065ac9ea76d93b76bce : SUCCESS in 16m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/25509fca858e49fbbaea7a7f76a4fb00 : SUCCESS in 1h 13m 32s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b4f9c4d85ad4576b1cc1317dfa0ab9e : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/acd2d144e4c14345b00b5253b2765a22 : SUCCESS in 2m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5663d3a3e1064d619905fdf3aa38f71e : SUCCESS in 3m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/88cd7b79362c41daa4e48080e4308abc : SUCCESS in 20m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/618724b860874671a1ef27854dc1a348 : SUCCESS in 17m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2ce0553755a5be3c2c1ded771543f71bd3ebd1a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 10:12:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/757fa14e471840949f13cbdb681be5f8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/122e341a18664a038a4b795bc863b97f : SUCCESS in 22m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d30ce30ecb55449ca893ef2468e785fa : SUCCESS in 43m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7e7bdf1882144087aa1fa0b136be6497 : SUCCESS in 22m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/159051194c594d36a6f2ae607a48400e : SUCCESS in 31m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/37d30637b64a48bb8b44c44e4a118a28 : SUCCESS in 43m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d34e19d3870044e69b3f7912cfe96c3d : SUCCESS in 1h 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce237c055af1409f82a6863dcb3abbef : SUCCESS in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19467360605b45d886e8cb9701905243 : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b62b32094ce9440ea6ca02f452731afc : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb35d064b069471e882b237c70c6ac4b : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/861c30ac68ac4f0197eeb03ee8549cdb : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1360c73e2dc546c09848184c6814c7ee : SUCCESS in 15m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/212ded73efa1403a86fe29779f35452a : SUCCESS in 29m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4a836164d83e4cf9ab3277e7e412d4eb : SUCCESS in 1h 06m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a908983805f24ecf8fda00f4a36929e9 : SUCCESS in 1h 13m 57s\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/08988ff137764e0aade8b6504f1f848f : SUCCESS in 1h 26m 44s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2bf0422debb64c078a45d56bc764885e : SUCCESS in 15m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/76f058b44c4f40ee979c1e94dc7e3739 : SUCCESS in 20m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cc48609e0aff4caaa0c5848e0e32b6bc : SUCCESS in 59m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3efd086540244a099b1752d8684cb6fa : SUCCESS in 1h 18m 59s\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/39615e19b85d41b392a31332d4c2f0e0 : SUCCESS in 1h 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1d9130cf6abd71c1cffea9881debe50b84ae295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 10:12:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4f5934b11a11f0b9cdf49dd5107a3a5bf3620f2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 12:00:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7d93686df3845feb10bbf391921e64e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e71f89689f2e4ea3802a7efaaf403eb5 : SUCCESS in 14m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/27737636c2784575a0347887397e7c24 : SUCCESS in 31m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8dd092adc5d84f7b893edfc2227eab3c : SUCCESS in 1h 03m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1d4c285b875e47bf891ea2c109ea28c8 : SUCCESS in 1h 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0284df9771847a7912e54e224852aa0 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acc8526036d04556a886e95892a2ae10 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c93b050f9e554b4ab608e61a1dd034a3 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f02b4bfd01a4b51a281682f33c3cbb0 : SUCCESS in 3m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2e5860556d314ebc8a1e88637becd8f3 : SUCCESS in 19m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1673288d9cb9448dab5275594025a6f2 : SUCCESS in 27m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/24cf5ae932ac46aa8898a27ee7354cbf : SUCCESS in 1h 07m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b370daf44c4442c7925dff64db135a3f : SUCCESS in 1h 41m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/29f54f75b0434bdcab5c6c51687eb303 : SUCCESS in 21m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b3efbd00868b48ab870041f65c4abca9 : SUCCESS in 28m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e3a5d7da358e49c2aec987194cfeac94 : SUCCESS in 40m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/28555ec538e04099a3e1f763aeb6da65 : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":2},{"id":"455ab9c4e9777789219605d87d59f2f6a678a869","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-26 15:11:44.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"25f6467c4a1254df29175fb75caec6433eb76e9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 16:29:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cac74c622d284e1980d24d0311999b70\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ab30d92bf55e4be0a4149ab5f33ac0f7 : SUCCESS in 1h 09m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1455ca6bbc3f408f82ea2f0bfbc0b21e : SUCCESS in 1h 13m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b201d4ded70a4d0eb1e793b2609c20fe : SUCCESS in 1h 09m 34s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37b7b4f6162743e0abe618687b13de8a : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4191ccdf553c409db949b2bbe57231e1 : SUCCESS in 2m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/48bd5184f4fe43d4902f98ce939ca299 : SUCCESS in 6m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/91a95ebad30c49c58444c41d889a07d2 : SUCCESS in 15m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fe85b80487a94739bb7732eed2a2354f : SUCCESS in 1h 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b9c98e6f6825fcf70891e97d8189b44ffc836fce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 16:58:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7770fd696d74c78b6bd496def9471ea\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f1b1b89ef50648108defd7a2ad02a88b : SUCCESS in 19m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/324ab2367b22488fae27d230b4352cf3 : SUCCESS in 1h 04m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d511a3e37b104ac890cc9d651930755f : SUCCESS in 14m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0bbf98fb344341309fb7651e22539a59 : SUCCESS in 26m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e2ff13cb9e9c46b48fefd3a4a5e8a2b3 : SUCCESS in 45m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2e7f931894c24d668fc6892a5fbcc6aa : SUCCESS in 1h 23m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e94434269b1429db2af944da1a31d92 : SUCCESS in 2m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf6d5069e0c04612a98518a65ae2dd7b : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c16a45c50386494b993aef476c9fc7e6 : SUCCESS in 4m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bea5a664f7e8440f9b7093cb79fc3ff6 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0e80374b0841bbba9ce7781cf265b1 : SUCCESS in 7m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8e421f2d32dd419e86478a2a49c96882 : SUCCESS in 20m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0bd96ef4b6c3469daac6fb440eb6b442 : SUCCESS in 28m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9abfcb0ecab54cd1b9f5f4424b91ed00 : SUCCESS in 1h 04m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/00c2a96d1d914974aedf78da2253e349 : SUCCESS in 1h 41m 40s\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/b156e663df804c04a893c4bc9b5974ae : SUCCESS in 1h 30m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5f2311b858964a1fa9629854b9ea577f : SUCCESS in 21m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/219d6f604aa843abb118393f1c0e5e9f : SUCCESS in 32m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/96208c38e83d4c9887a9c8d6b8bb7714 : SUCCESS in 40m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b2a522f23682472ca3b94787a272fd6a : SUCCESS in 1h 19m 07s\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/7afe80abf3fa48b6939b99fc1ab41fdc : SUCCESS in 1h 14m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6f266f237a1be38be99755ea9b10f9ef072efb63","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-26 18:23:35.000000000","message":"Patch Set 3: Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6516400146117649f3be1f2d8871bb9231a0b417","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 18:23:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0f21f0043d474e8f2d70eed6f07881e30bf9385d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 20:05:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9326ff370caa4958a7cae8f430bf545f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2c8ac5644c594c64a4bfa661f225405f : SUCCESS in 13m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e1d1825a0e98480f871d7424af2c4f7e : SUCCESS in 19m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c3ab11b6649f4f2baca9e9f6d8aed824 : SUCCESS in 42m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/289c40efb7044ef2989c2064b39429b6 : SUCCESS in 1h 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fdf1114461343499cd4a0b2e5bcf894 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88913fe813cb442da8f15bba25c99c4b : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2089251a35f45758d3b09c00db0d02e : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b98e3c8110847a1b61f182901723fe2 : SUCCESS in 4m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f1cb3c5c073c49b5b52980ad9338b51b : SUCCESS in 20m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/867b25fd6b6642ffbcbc1bb2ad7c301c : SUCCESS in 28m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/63f2459016874d349ca0003b0f56f686 : SUCCESS in 1h 08m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/66c77d641e244861a776d21516c88e79 : SUCCESS in 1h 37m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/68ae5deef1274ab6af62034f710e97cd : SUCCESS in 19m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ee9026fbb6b4449cbee999e2da4100e0 : SUCCESS in 29m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/53b42b5b54d9405598dbd44c7aa20820 : SUCCESS in 59m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f0ccb8bfb5054805962274d3ab91a2a6 : SUCCESS in 1h 29m 14s","accounts_in_message":[],"_revision_number":3},{"id":"11b77c32b61f94d7dd3dd6efa1bf1f946de97211","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 20:05:53.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"94e6bc37946adfd5bba6ec41628347e03312adcb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 20:06:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a7ae5d89bf64707bc07f28c8aebc9ef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ff6c9ba68c4337a30521c0990cd4d3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d7e497ce52cc109c9f7b4a83d85f0bebfa05298e","revisions":{"e53fae7958d8cd62debecbb05739d3f7820ed85c":{"kind":"REWORK","_number":1,"created":"2023-06-19 09:00:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/886380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/886380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/886380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/886380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/886380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/886380/1"}}},"commit":{"parents":[{"commit":"b0804cd04cd2e47cf18666fc420ebd42e69aae75","subject":"Fix Venus containers built from correct branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0804cd04cd2e47cf18666fc420ebd42e69aae75"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 08:59:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 09:00:49.000000000","tz":540},"subject":"Swift: Remove unused rootwrap","message":"Swift: Remove unused rootwrap\n\nSwift does not use oslo.rootwrap. Thus the command and its config file\nis just useless.\n\nChange-Id: If4e346c2db841aad9b2ddac049dbbbc1ba5782ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e53fae7958d8cd62debecbb05739d3f7820ed85c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e53fae7958d8cd62debecbb05739d3f7820ed85c"}]},"branch":"refs/heads/master"},"e9cc2c28eb0fb0b3730ad982f2a017de633db2af":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-22 06:47:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/886380/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/886380/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/886380/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/886380/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/886380/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/886380/2"}}},"commit":{"parents":[{"commit":"d86cba116ef686a7bad6153a5ce865ba77850cc7","subject":"Merge \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Bobcat)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d86cba116ef686a7bad6153a5ce865ba77850cc7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-19 08:59:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-22 06:47:36.000000000","tz":0},"subject":"Swift: Remove unused rootwrap","message":"Swift: Remove unused rootwrap\n\nSwift does not use oslo.rootwrap. Thus the command and its config file\nis just useless.\n\nChange-Id: If4e346c2db841aad9b2ddac049dbbbc1ba5782ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9cc2c28eb0fb0b3730ad982f2a017de633db2af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9cc2c28eb0fb0b3730ad982f2a017de633db2af"}]},"branch":"refs/heads/master","description":"Rebase"},"d7e497ce52cc109c9f7b4a83d85f0bebfa05298e":{"kind":"NO_CHANGE","_number":3,"created":"2023-11-26 15:11:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/886380/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/886380/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/886380/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/886380/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/886380/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/886380/3"}}},"commit":{"parents":[{"commit":"d86cba116ef686a7bad6153a5ce865ba77850cc7","subject":"Merge \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Bobcat)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d86cba116ef686a7bad6153a5ce865ba77850cc7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 15:11:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 15:11:41.000000000","tz":540},"subject":"Swift: Remove unused rootwrap","message":"Swift: Remove unused rootwrap\n\nSwift does not use oslo.rootwrap. Thus the command and its config file\nis just useless.\n\nChange-Id: If4e346c2db841aad9b2ddac049dbbbc1ba5782ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7e497ce52cc109c9f7b4a83d85f0bebfa05298e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7e497ce52cc109c9f7b4a83d85f0bebfa05298e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
