)]}'
{"id":"openstack%2Fkolla~886384","triplet_id":"openstack%2Fkolla~stable%2Fzed~I7eb97c6bfb65f26b044f26fece09521836ac5a3b","project":"openstack/kolla","branch":"stable/zed","topic":"fix-cinder-volume-lsscsi","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-20 09:41:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7eb97c6bfb65f26b044f26fece09521836ac5a3b","subject":"Fix cinder-volume: Add forgotten lsscsi rpm package","status":"MERGED","created":"2023-06-19 10:18:49.000000000","updated":"2023-06-20 09:44:15.000000000","submitted":"2023-06-20 09:41:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886384-fix-cinder-volume-lsscsi","meta_rev_id":"1b332706791b46a01da97fb7e41c58c4fccb57ab","_number":886384,"virtual_id_number":886384,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-20 09:41:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":2,"date":"2023-06-19 12:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2023-06-19 12:31:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-20 08:58:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2023-06-19 12:31:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 10:51:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-19 12:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 12:20:10.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-06-19 12:31:03.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-20 08:58:01.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"06d7847a886ae487846dc1a82f3cb7f70c9a07c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 10:18:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76ef37d343f16e1cda403624deb36ffd19738cc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:51:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/397213861bf24ea697bd6e12a4282a7f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d29e8f9fc89a46dea95dca29f3f9112f : FAILURE in 24m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/65eb725ff6374bff9e4fe3cc17ca858c : NODE_FAILURE Node request 300-0021485932 failed in 0s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/09e6540861b84c1db221260c619971de : FAILURE in 28m 01s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4bf79baeda8249c6b68435e0fdff064e : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9058201c232e441dbae645dc634fd9c3 : SUCCESS in 3m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d97040fa88a14405bf688350b899842f : FAILURE in 28m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c8df5f8715ef4326b419a6b1e8782115 : FAILURE in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64f7f9ffc58d7c96811d9236c6d8602c19788be6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 11:02:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c6e78ab376667a2dfe21999f85eb2fd6c9e9a0a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 11:45:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10d53de02ec0423083613c048a7bc3ca\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0af6708c5751425db0ef39319c15d44f : FAILURE in 41m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/168974e2f9ba45488ad0746b3a561bd0 : NODE_FAILURE Node request 300-0021486212 failed in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/27bcbd02ef54442893609835e8a851c9 : FAILURE in 34m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5432582be76148d3a141612a13655023 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c684afdef5f444cbcbe30647f847f6b : NODE_FAILURE Node request 300-0021486215 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/13395b584e3a4e4182089b94b33d5088 : FAILURE in 38m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/272a071278554939b6d1b84126368a82 : FAILURE in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f26ddc30cb0d130da8bcb3b58d381baa9eefd96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 12:15:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49d5d33b7a4046368ef0382286cf228e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ddbfbb11df074e149927931f07ce17ef : SUCCESS in 29m 09s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f3fea579944a4d83879d6d15efdddd26 : SUCCESS in 1h 02m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/72842b621fc9440a9680da996acfab23 : SUCCESS in 19m 41s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/94f488f4cb0d4261b4d8481679b954b0 : SUCCESS in 47m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c8328dbed3ba4306ad5ea9411f373504 : SUCCESS in 1h 01m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cbb222b527244acadef17133173c951 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af725cde38e9493482f0ee58597ea6ad : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fc537d62d3244ac92f181aca1cd4a29 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4a1fce84d6a4004a48ba82486357219 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c931c282a41d46e78ca083891ca52bc0 : SUCCESS in 5m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/03e3c45a02ca477cad4dcc5e80b13f34 : SUCCESS in 22m 42s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/d54f14dcd1f14b028e44f4075e184b35 : SUCCESS in 53m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8d8d2e74f61b47129d582f1c30a672fd : SUCCESS in 11m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/da2febdd74b947958198ecc77bcd9073 : SUCCESS in 46m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a46782ae2f6346c28ece14cc3687f2c5 : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ae60f4dbc27aec620bb0cb5d6615a105aa11859e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-19 12:20:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ab2a95d6c3b54484f88092163e7f7496e45d249b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-19 12:31:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"faa4f5365b5e0a285effa8152a145842ab98b8ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 08:19:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4f0706e03c0ddff08b7186ef835912f58d61e2fe","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-06-20 08:58:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"61dcf4e315e437d149b90ae69a7e171ed0a43a21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:41:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c511170b2214c4c80a86d7cf0d422f0\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7df9498c42cf436fb4564816a2174f1a : SUCCESS in 26m 32s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/60f9548b620d42fa90b177d1df3d363a : SUCCESS in 57m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3ac3cdb39bcb4dd69580411aeb433e5c : SUCCESS in 1h 03m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28f9085215a442c8f5b4c7ced1d85d4 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1224188a6f874e159505021357656061 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c4aa670fdfe4631930c4e17c10d03d2 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b37c40af39645f293a9f62337b66c69 : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/74aa3bc7ccfd4daea79e31e9fc29c791 : SUCCESS in 23m 03s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/366392977aff427f8b04d5c87665711d : SUCCESS in 1h 10m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bc01c8b2244847919dd1b1011ec20fb3 : SUCCESS in 25m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d793bd9a80d42d395d03a64196ce648 : SUCCESS in 50m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1c7a47445e0a4e4fb68161cf908f91f5 : SUCCESS in 1h 06m 11s","accounts_in_message":[],"_revision_number":2},{"id":"4e5e9b7b5f3f86c6c9d88d8c27176376dc34f8c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:41:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1b332706791b46a01da97fb7e41c58c4fccb57ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:44:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a462d3a9b584f909ca5cda39b1a4e9e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac92036209b4472b88a4a22901c8fe6 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ae289a5e5e9f07e0a2fe4ce5e3fa0aece27b784","revisions":{"0b9d91c1da081e3ab545f6952011990f0eb57b3f":{"kind":"REWORK","_number":1,"created":"2023-06-19 10:18:49.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/84/886384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/886384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/886384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/886384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/886384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/886384/1"}}},"commit":{"parents":[{"commit":"fe9d33c64fff9b4d4607cabd090ead62b052a3ab","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe9d33c64fff9b4d4607cabd090ead62b052a3ab"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-18 20:30:55.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 10:18:37.000000000","tz":180},"subject":"Fix cinder-volume: Add forgotten lsscsi rpm package","message":"Fix cinder-volume: Add forgotten lsscsi rpm package\n\nCloses-Bug: #2024018\nChange-Id: I7eb97c6bfb65f26b044f26fece09521836ac5a3b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 2e2900e00bb88a38ea51c925d2c6c0cfcd5841bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b9d91c1da081e3ab545f6952011990f0eb57b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b9d91c1da081e3ab545f6952011990f0eb57b3f"}]},"branch":"refs/heads/stable/zed"},"7ae289a5e5e9f07e0a2fe4ce5e3fa0aece27b784":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-19 11:02:42.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/84/886384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/886384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/886384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/886384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/886384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/886384/2"}}},"commit":{"parents":[{"commit":"1314b297fd817c1ab68110e2c820bfa85afe4e5a","subject":"Fix rabbitmq image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1314b297fd817c1ab68110e2c820bfa85afe4e5a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-18 20:30:55.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 11:02:42.000000000","tz":0},"subject":"Fix cinder-volume: Add forgotten lsscsi rpm package","message":"Fix cinder-volume: Add forgotten lsscsi rpm package\n\nCloses-Bug: #2024018\nChange-Id: I7eb97c6bfb65f26b044f26fece09521836ac5a3b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 2e2900e00bb88a38ea51c925d2c6c0cfcd5841bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ae289a5e5e9f07e0a2fe4ce5e3fa0aece27b784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ae289a5e5e9f07e0a2fe4ce5e3fa0aece27b784"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
