)]}'
{"id":"openstack%2Fkolla~886385","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I7eb97c6bfb65f26b044f26fece09521836ac5a3b","project":"openstack/kolla","branch":"stable/yoga","topic":"fix-cinder-volume-lsscsi","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-19 14:07:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7eb97c6bfb65f26b044f26fece09521836ac5a3b","subject":"Fix cinder-volume: Add forgotten lsscsi rpm package","status":"MERGED","created":"2023-06-19 10:20:13.000000000","updated":"2023-06-19 14:08:11.000000000","submitted":"2023-06-19 14:07:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886385-fix-cinder-volume-lsscsi","meta_rev_id":"ebf9d272716e16d4c2d2588a93b63ca74b6f5739","_number":886385,"virtual_id_number":886385,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-19 14:07:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":2,"date":"2023-06-19 12:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2023-06-19 12:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2023-06-19 12:31:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 10:59:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-19 12:02:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 12:20:23.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-06-19 12:31:14.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"7898694cc0ddd1a746a102586e9d8584642e295c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 10:20:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c166440fac506dea52ed102cb991b296d33b571","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:59:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56ae9ff71eee495f9acc4d14b6875ac2\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/45f21d5487f14df9ba7c754b9265a155 : NODE_FAILURE Node request 300-0021485954 failed in 0s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/414cdc52fd80462fb94ebb4f39951c3c : FAILURE in 35m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/55f52fc4cf954918aa120f4d240dde1c : NODE_FAILURE Node request 300-0021485956 failed in 0s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b2624b6c62e149a89e75f39a928a140a : NODE_FAILURE Node request 300-0021485957 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e4b684eef944bbabda3e9e8addf5549 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa496a60b5994a08ae6f6cda076567ec : NODE_FAILURE Node request 300-0021485959 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6952654805174a6a8f80a6d65be5838a : FAILURE in 7m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f784c3d138f7feecf0a2a762a32ddca61de37f54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 12:02:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d11f99100ba488585ffb12b2e3859c1\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b900df1fd4b5436eb4e9378b44b6dff4 : SUCCESS in 34m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f25580bea9044915ab83d4aac50b837c : SUCCESS in 34m 26s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/cccebccd25ba4ce68213decc3f0b239d : SUCCESS in 47m 17s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b3827cffffae48a98dd15138b40d6be6 : SUCCESS in 1h 22m 04s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/220bd44672ff44818fff86af410c0093 : SUCCESS in 1h 22m 21s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ecc3cce0f79d4f22b419822321a4cbbf : SUCCESS in 1h 10m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/68bed376626f4172ba31ed25211e4d26 : SUCCESS in 1h 32m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bc6495b1e3af48a5ba5b2a8316aa3eff : SUCCESS in 17m 41s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7180db2281ab4e00af7a5efda9629c2c : SUCCESS in 19m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f107af07c954927954bc6762377be2c : SUCCESS in 47m 29s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7fabce70190646e0b000614656ceb2c6 : SUCCESS in 1h 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c411f22952949619857252e127c49e5 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79d75b2ac6f14524915391e52381bcca : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/874f7081a6174ff29633143362ef8e6d : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c961aa60ca5c4b4eac8b81699870f91a : SUCCESS in 6m 08s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/c0a35cb3e16e418c87e806b53c8a4805 : SUCCESS in 19m 47s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8a065e5d47654a35ad2754ebcab29766 : SUCCESS in 1h 07m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f64408b0953435a94692988d300d41f : SUCCESS in 17m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2f6d1111ca8744f9915061da8b9de345 : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":1},{"id":"54929dab6e34e43b92948b41def4c77b534b1adc","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-19 12:20:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"219588dcff20474cd44e9a53649fb8b7455dbc8c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-19 12:31:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5cbc549f5b362caebf3c1019d332e1dc6d50f33e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 12:31:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a58e23de6b9733dda899ed2007ea37861c478055","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 14:07:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a56ce7020334892937ae3906e04f65c\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/12b77b730ddb47a69e9618781c711da3 : SUCCESS in 37m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f10586f18ff545b59dc0e020765c531d : SUCCESS in 55m 49s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/66da73904da340548bc9ea024e076e44 : SUCCESS in 1h 31m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b257a681aa2f4baaa1de87095fd3f7b9 : SUCCESS in 31m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a8a8c3f2f46d4f7684dfb929fdb1d86e : SUCCESS in 1h 02m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6591ed01bf834ce88c714ed7d36bfacc : SUCCESS in 1h 27m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c29fddd8d41445e84915ee583438eee : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/770e1c56275c4c95a241800d9aa69454 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ca60d80647e4d4da2aeb29871961837 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3fecdf7fb34707a60ee882545068a0 : SUCCESS in 4m 54s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/3f26a3c12eac43b88609f1e41f92e85a : SUCCESS in 1h 19m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/876a489ec16e494c826c5b463843b853 : SUCCESS in 22m 33s","accounts_in_message":[],"_revision_number":1},{"id":"eb01617ee596040cf2708b69d28a7759265382af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 14:07:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ebf9d272716e16d4c2d2588a93b63ca74b6f5739","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 14:08:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/587a13a9258143f3a5d68b17ea1846ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3b93d9f784425fa0de9e6cf6663d2e : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f35699e33c8a82855ce412f04b01db8b1a8c8a7a","revisions":{"f35699e33c8a82855ce412f04b01db8b1a8c8a7a":{"kind":"REWORK","_number":1,"created":"2023-06-19 10:20:13.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/85/886385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/886385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/886385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/886385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/886385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/886385/1"}}},"commit":{"parents":[{"commit":"df733398207cbc242328844d288dba1f02098c08","subject":"Merge \"fluentd: pin td-agent to 4.4.2\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df733398207cbc242328844d288dba1f02098c08"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-18 20:30:55.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 10:19:52.000000000","tz":180},"subject":"Fix cinder-volume: Add forgotten lsscsi rpm package","message":"Fix cinder-volume: Add forgotten lsscsi rpm package\n\nCloses-Bug: #2024018\nChange-Id: I7eb97c6bfb65f26b044f26fece09521836ac5a3b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 2e2900e00bb88a38ea51c925d2c6c0cfcd5841bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f35699e33c8a82855ce412f04b01db8b1a8c8a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f35699e33c8a82855ce412f04b01db8b1a8c8a7a"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
