)]}'
{"id":"openstack%2Fkolla~886387","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I7eb97c6bfb65f26b044f26fece09521836ac5a3b","project":"openstack/kolla","branch":"stable/wallaby","topic":"fix-cinder-volume-lsscsi","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-19 14:07:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7eb97c6bfb65f26b044f26fece09521836ac5a3b","subject":"Fix cinder-volume: Add forgotten lsscsi rpm package","status":"MERGED","created":"2023-06-19 10:21:21.000000000","updated":"2023-06-19 14:08:04.000000000","submitted":"2023-06-19 14:07:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886387-fix-cinder-volume-lsscsi","meta_rev_id":"da9be27c9cee49c5c9123551158cad3b2dc9fdbe","_number":886387,"virtual_id_number":886387,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-19 14:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":2,"date":"2023-06-19 12:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2023-06-19 12:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2023-06-19 12:31:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 11:02:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-19 11:44:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 12:20:52.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-06-19 12:31:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"7eca1f8224a7cceb32f65a9ec109a5fc9f5f2a86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 10:21:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d68b29a23dff22d7316efa042d0a84f22982ac9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 11:02:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8a4f5b013ce4445905991a595bb4240\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/002225113f334ba783c8ebf6641f9033 : FAILURE in 37m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9029cc5d52754dd3b0c9c5daa8bf2143 : NODE_FAILURE Node request 300-0021486004 failed in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a94081925a774f34814f018338ee7b45 : NODE_FAILURE Node request 300-0021486005 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c0e6e0c04b243b2bb95d2d4a241ab72 : FAILURE in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c4d0aeb373914ca69f52626760adc00ddd3f393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 11:44:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe7792a3b27e4445b2f3b778d3f9a08a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8b105f722889447c8fbae4e48c437fe4 : SUCCESS in 47m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0458fc0f27b14d30b3ce78f88727e82c : SUCCESS in 44m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8d43a4a00c104587acb2137986ee1a00 : SUCCESS in 1h 22m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3e5d463994924d63b6e17a5cf3e7a319 : SUCCESS in 1h 09m 12s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2e9b869e2044442fa4fdf0d85951eb59 : SUCCESS in 16m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5a920577aed045b2881b3850b068e122 : SUCCESS in 23m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/53797ea093074908a33ffe61821e84a6 : SUCCESS in 51m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9ca2c819fc943718d5130db987c5d25 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa041961822b4303bf1edab628b8a96d : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/958974a3364f4753bf5e81a64b8c867b : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96131e6a9aee4bf08914be41bf64ab98 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46285b2454aa49fe99bf322481d4f710 : SUCCESS in 6m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/57b07315641a4bc18e6c1ca91bbaf42b : SUCCESS in 42m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4112d39e50814221b07bd6adc2d390ed : SUCCESS in 18m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/26c9cb8e4fe442dbad8a45f50099af0c : SUCCESS in 46m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3382a848b5df49fd821717f33bfecfb6 : FAILURE in 49m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5693a993805010de0e0119c76aa8436f3fb48112","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-19 12:20:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a4db0fc3552e0e7a908d5edef287ffddc7060379","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-19 12:31:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c926ddfdd5720626f401aaf64a11568b430f24f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 12:31:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8be43fbf22439fe1909406d7e7c967c26e120a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 14:07:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/033565bf98644e4aa955a252f50892f4\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7369c4dbac504801b900edc87bec0917 : SUCCESS in 40m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/76620530f155482d9af71d5a56fabbab : SUCCESS in 1h 14m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e6ca85fed824db2ab6852f924cacffb : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04760381fe9647d296f4be29270b782b : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df45641140cc44fda711745d76c65bd0 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c97fdefd2de4acf8fae106b47946136 : SUCCESS in 4m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2143039d22548d5be570742cde14e28 : SUCCESS in 18m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b675f4cb99024ad3bbe0844ea95da971 : SUCCESS in 48m 53s","accounts_in_message":[],"_revision_number":1},{"id":"46b28920ef626543fe2a9bb2948a39a00505105c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 14:07:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da9be27c9cee49c5c9123551158cad3b2dc9fdbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 14:08:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d015c702e4344c0b846745a174ee8647\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/354a67f43ca94f12b00a7fde4579640c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a326f91b764bf3b72c14615936e6f433fcae63b","revisions":{"4a326f91b764bf3b72c14615936e6f433fcae63b":{"kind":"REWORK","_number":1,"created":"2023-06-19 10:21:21.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/87/886387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/886387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/886387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/886387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/886387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/886387/1"}}},"commit":{"parents":[{"commit":"4fccf664b4b64f2711c4eea616114ae60aa4f78c","subject":"fluentd: pin fluent-plugin-elasticsearch to \u003c5.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fccf664b4b64f2711c4eea616114ae60aa4f78c"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-18 20:30:55.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 10:21:05.000000000","tz":180},"subject":"Fix cinder-volume: Add forgotten lsscsi rpm package","message":"Fix cinder-volume: Add forgotten lsscsi rpm package\n\nCloses-Bug: #2024018\nChange-Id: I7eb97c6bfb65f26b044f26fece09521836ac5a3b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 2e2900e00bb88a38ea51c925d2c6c0cfcd5841bb)\n(cherry picked from commit f35699e33c8a82855ce412f04b01db8b1a8c8a7a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a326f91b764bf3b72c14615936e6f433fcae63b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a326f91b764bf3b72c14615936e6f433fcae63b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
