)]}'
{"id":"openstack%2Fkolla~886388","triplet_id":"openstack%2Fkolla~stable%2F2023.1~Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8","project":"openstack/kolla","branch":"stable/2023.1","topic":"rabbitmq-ppa-novemberrain","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-19 13:41:54.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-19 13:41:54.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-19 13:41:54.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-19 12:11:14.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8","subject":"Fix rabbitmq image build","status":"MERGED","created":"2023-06-19 10:36:17.000000000","updated":"2023-06-19 13:43:47.000000000","submitted":"2023-06-19 13:41:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886388-rabbitmq-ppa-novemberrain","meta_rev_id":"846a6e990e51c7711f12a67510d9a3e543bd59b7","_number":886388,"virtual_id_number":886388,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-19 13:41:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":2,"date":"2023-06-19 12:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2023-06-19 12:11:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2023-06-19 12:11:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 10:36:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2023-06-19 10:41:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-19 10:41:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-19 10:41:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-19 11:45:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 12:10:02.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"d558351e1a8eb0ed2adbdb11f6bd3c258c2b3db6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 10:36:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3fb02d52ad61f8330f9f86e4229ec462260208e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 11:45:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89e55e1e89b44cfcb87458685182eeb7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8b1fcffab1534747a572d9b3e5292c64 : SUCCESS in 37m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/24fe49e2d93049cc9f0188a421575dbc : SUCCESS in 1h 07m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a22c963d56254296a1930e3e7dcdb86f : SUCCESS in 21m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3976ca62d1b5413da6c1b332525146e3 : SUCCESS in 54m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/01e3f1d95ee745caa50a005bf8070f2a : SUCCESS in 1h 05m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d66b25e47894427baaea84c250dbc0d : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d669b2031c74bdd91bf344c7029b51b : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3247cd58200420693dfc4b4ce4ecafa : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba3c3601cb734c83a107c1af27713431 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbcea02f31f54de38574ce0fb6a2ffd6 : SUCCESS in 5m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e1269f9784764a9aa545b10ea1ddcfb9 : SUCCESS in 25m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/37910f6c659d4b6eb06756f4299d3bb6 : SUCCESS in 46m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ec543b363aee48258ac29220878aee76 : SUCCESS in 57m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/413a1c9edb94445ea457c150a098031b : SUCCESS in 37m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d2c702b329a74d818b24886f83b1b61c : SUCCESS in 46m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1ec3b8e6550b402aa92848a6802f28c1 : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1fd90a76774d8d53592ae0948d26a3330469e3cf","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-19 12:10:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9a942dd6a47d17d4b52ff98e49914b97e453d88a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-19 12:11:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a698c171a256d5b35be12c982a5b88ea4e1ce59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 12:11:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5517d8e6d438cf37e7bf993a06beded30aa922e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 12:16:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9125f5c0e50749ef8fe8d0bf6de1159c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/29df12711b58493da12479eb9455811b : FAILURE in 22m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a67384b3a51e4b318b14a341203d7cd5 : FAILURE in 22m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3cf54577d20047cfbbf91b7e894d053c : FAILURE in 19m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95eadc8268fb407184490d8f21658e15 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f32264c93b8248c0b4d89d5dc6309302 : SUCCESS in 3m 05s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f4e9f3da011c422cbdc75236eb0f8d1e : FAILURE in 1h 36m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ef9a8a9ce1e14d5888e43390991e3273 : FAILURE in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b591d4e69b1a5ebbeb51c4281345011f48fd4a2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 13:41:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d941a0fcb3cd40cf9e6f32e0aad63c06\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7a276702dc0147eea08143ae413fb8e4 : SUCCESS in 14m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c529857a235f4a5db6605372d9cbf952 : SUCCESS in 50m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a1229263b59b458395afb7bf9d1b4a5e : SUCCESS in 1h 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/119df37bb43c4b9e86ddac5bd468342d : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/becfebe113d54f52be9eb66abcd0f02d : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab42dddab5fc4b6781f87f36b7379407 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29be8524da0048e9a2908b72b0cd81cc : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/46d8a08295034aedb19fb478f50fbc7f : SUCCESS in 20m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f5c1aae01bc14003b90f3b0f00fd65ed : SUCCESS in 1h 07m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e44f825a12b340488fd04edbe52658eb : SUCCESS in 1h 08m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3d6a27ed0f574dd7bd55a75d4645fc4b : SUCCESS in 24m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ab5ddcbbb98b44f9a057558c69aed6d8 : SUCCESS in 52m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/954867a9edbf4e3b8a8767d22daf6207 : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":1},{"id":"160da7e73aedb52d6c12ff0cd0487c91bbf345c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 13:41:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"846a6e990e51c7711f12a67510d9a3e543bd59b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 13:43:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d10bf13202d944fdaa3277f4b437fd38\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ab6fd74543434bab5937ce51daad2e : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8dc32be6f1b99eccca5f4f6d15a4e654d4519950","revisions":{"8dc32be6f1b99eccca5f4f6d15a4e654d4519950":{"kind":"REWORK","_number":1,"created":"2023-06-19 10:36:17.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/88/886388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/886388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/886388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/886388/1"}}},"commit":{"parents":[{"commit":"2ca7ec16135bf62637ca1115e122f66cb1eefb27","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ca7ec16135bf62637ca1115e122f66cb1eefb27"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 10:14:02.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 10:35:52.000000000","tz":180},"subject":"Fix rabbitmq image build","message":"Fix rabbitmq image build\n\nThis patch is replacing ppa.launchpadcontent.net\nby ppa1.novemberain.com.\n\nReason for replacement is that only latest erlang version\nis available on the current ppa source.\n\nSetting Debian ARM jobs non voting for now, since the new ppa\ndoes not contain arm64 packages.\n\nThis repo is based on the official Erlang/OTP package\nfor Debian. The RabbitMQ team adapted the upstream package\nto produce packages for more/different combinations of\nErlang/OTP, Debian and Ubuntu releases.\n\nChange-Id: Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8\n(cherry picked from commit 4950c03e59b8d1af46018b81a49fa028e1b8296d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dc32be6f1b99eccca5f4f6d15a4e654d4519950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dc32be6f1b99eccca5f4f6d15a4e654d4519950"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
