)]}'
{"id":"openstack%2Fkolla~886389","triplet_id":"openstack%2Fkolla~stable%2Fzed~Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8","project":"openstack/kolla","branch":"stable/zed","topic":"rabbitmq-ppa-novemberrain","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-20 09:41:36.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-20 08:18:30.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-20 09:41:36.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-06-20 09:41:36.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-19 12:32:20.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8","subject":"Fix rabbitmq image build","status":"MERGED","created":"2023-06-19 10:39:15.000000000","updated":"2023-06-20 09:43:31.000000000","submitted":"2023-06-20 09:41:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886389-rabbitmq-ppa-novemberrain","meta_rev_id":"f7f8f537c015be0ff52d0b895e7e921d0cb299ab","_number":886389,"virtual_id_number":886389,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-20 09:41:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-20 08:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-19 12:32:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-06-20 08:18:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 10:39:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2023-06-19 10:39:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-19 10:39:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-19 10:42:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-19 10:42:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-19 10:42:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f8ebcef3d794346f43a8baeec80ede2b8101564e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 10:39:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e7e616aa4648d320642e23cec8f7d78cb6bcf13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:39:37.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/zed.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/kolla on branch stable/zed.  The error was:\n  \n    mapping values are not allowed in this context\n    in \"openstack/kolla/.zuul.d/debian.yaml@stable/zed\", line 12, column 19","accounts_in_message":[],"_revision_number":1},{"id":"6a1851c0f68ac9ccb45adb630c9ea2ab2669a925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:39:44.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/zed.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/kolla on branch stable/zed.  The error was:\n  \n    mapping values are not allowed in this context\n    in \"openstack/kolla/.zuul.d/debian.yaml@stable/zed\", line 12, column 19","accounts_in_message":[],"_revision_number":1},{"id":"3e624fc1ff7574a7e6b12405b512702eee63a095","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 10:58:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"61ba0b96656783673c3406597a9d6ce2ab450ccc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 11:39:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fedb7685647544e7be9dc83a8400007c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/055187161a6c43e093eeb59a2ff9187d : FAILURE in 38m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7288c262cd9d4d5f9b31cd8f41dacfe8 : NODE_FAILURE Node request 300-0021486156 failed in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4cab4cc1252b49b0af6e9c2bbcaec090 : FAILURE in 38m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cd47213fb8a43cb8a6b1b0809d820d3 : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d3c37866b3b4afe8b016d9ec151820c : SUCCESS in 3m 05s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/60bbceff1c064cd1a9259b63e49c1284 : NODE_FAILURE Node request 300-0021486160 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/981648a0c0534db5867457dca5d51b3b : FAILURE in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d688cbb89941123910a13dc5add91bab81531838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 12:24:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c8df5d16de94a2183442c0d484fed94\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ad84dacfe6bf499ab0ae2819be2788ef : SUCCESS in 19m 44s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a04e0a40190e474a8300351b7c3f923f : SUCCESS in 1h 02m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/859b60944e084d0b9336ee72a0c33ec3 : SUCCESS in 27m 03s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6b2ec8309dc548db81fab11c4fed3e03 : SUCCESS in 1h 05m 54s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bbd6f5f1c3bf498e86fcd712ee275684 : SUCCESS in 1h 21m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35b3d6a4a8c46a5902657eb54763b29 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0f577fbe7d84fa7bee716b80ce226a9 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86312963568b4f629b2ab20f05a85b94 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7d799789afb454ea6f4420cc0a884a3 : SUCCESS in 2m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8fe00e0d84d4d4eac1300249d121b96 : SUCCESS in 5m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1d7a5be3f9074950b292c45d29c79d79 : SUCCESS in 23m 41s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/e6c59064a2784c0c888a6adb976b9a37 : SUCCESS in 1h 06m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6dbb93bbd2d349b0936a44125f8d3bb3 : SUCCESS in 23m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a9a348c883b4ad09ff8d824805c3e68 : SUCCESS in 49m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/240e739af6504dbbab49cb36d793b93c : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":2},{"id":"227d8b6a58b18e17ecbdb524b6bad34dcfa699e4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-19 12:32:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2b814060b978df88e50b930debc5b64541d59a21","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-19 12:34:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"827f7e6755bcf7c048f066c930fbb88d12943daa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 12:43:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e19d56ce5a2c0a7d2f34983d2326c1568a44a9b3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 08:18:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e1377be725e6b5e34edeefe1aac43a6355965bad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 08:18:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a9ceb22863e1c2c5bcc03d8540eb242f379f81ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:41:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af6b2d61d9774dac92400f2a2dd5b032\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1fe1e89a270a4dbe9c0887cb78ab527f : SUCCESS in 22m 59s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b207f274cd004d71969520c432988653 : SUCCESS in 58m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a357392f8bac4af798566bd1fb6722c3 : SUCCESS in 1h 17m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e177738f694ba9829fbe21e67d8542 : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/095aaf0d30e947839684b1e9c357b2a2 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46eba013a1e042bf8d0de043c1961f9e : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8c7dff9270456fb1ce6fcde5095ec6 : SUCCESS in 6m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/72be28d6cade4b9bb1bc32ab4ba1bbe4 : SUCCESS in 20m 39s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/d53484ced47b49138f4ef6f98f7c586a : SUCCESS in 58m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fe6bd6dd4994498787ca1ca86fee1ba1 : SUCCESS in 16m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/28f3ed367a8a499387e7013381c654b3 : SUCCESS in 54m 11s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5b6f2227efe744cfb540c68d7f183206 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":2},{"id":"856e9a0ab41e2d88e85fdae2b02ca9f0caeba1bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:41:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f7f8f537c015be0ff52d0b895e7e921d0cb299ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:43:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e88fb9bf871475db98634e29d1e3b44\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e59c8982fae43349caf136f9167b02b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1314b297fd817c1ab68110e2c820bfa85afe4e5a","revisions":{"70ce6050e297888e94ec3c6c7c60e795772a6986":{"kind":"REWORK","_number":1,"created":"2023-06-19 10:39:15.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/89/886389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/886389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/886389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/886389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/886389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/886389/1"}}},"commit":{"parents":[{"commit":"fe9d33c64fff9b4d4607cabd090ead62b052a3ab","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe9d33c64fff9b4d4607cabd090ead62b052a3ab"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 10:14:02.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 10:38:59.000000000","tz":180},"subject":"Fix rabbitmq image build","message":"Fix rabbitmq image build\n\nThis patch is replacing ppa.launchpadcontent.net\nby ppa1.novemberain.com.\n\nReason for replacement is that only latest erlang version\nis available on the current ppa source.\n\nSetting Debian ARM jobs non voting for now, since the new ppa\ndoes not contain arm64 packages.\n\nThis repo is based on the official Erlang/OTP package\nfor Debian. The RabbitMQ team adapted the upstream package\nto produce packages for more/different combinations of\nErlang/OTP, Debian and Ubuntu releases.\n\nChange-Id: Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8\n(cherry picked from commit 4950c03e59b8d1af46018b81a49fa028e1b8296d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70ce6050e297888e94ec3c6c7c60e795772a6986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70ce6050e297888e94ec3c6c7c60e795772a6986"}]},"branch":"refs/heads/stable/zed"},"1314b297fd817c1ab68110e2c820bfa85afe4e5a":{"kind":"REWORK","_number":2,"created":"2023-06-19 10:58:03.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/89/886389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/886389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/886389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/886389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/886389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/886389/2"}}},"commit":{"parents":[{"commit":"fe9d33c64fff9b4d4607cabd090ead62b052a3ab","subject":"mariadb-clustercheck: ensure /var/log/kolla/mariadb exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe9d33c64fff9b4d4607cabd090ead62b052a3ab"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-14 10:14:02.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 10:57:46.000000000","tz":180},"subject":"Fix rabbitmq image build","message":"Fix rabbitmq image build\n\nThis patch is replacing ppa.launchpadcontent.net\nby ppa1.novemberain.com.\n\nReason for replacement is that only latest erlang version\nis available on the current ppa source.\n\nSetting Debian ARM jobs non voting for now, since the new ppa\ndoes not contain arm64 packages.\n\nThis repo is based on the official Erlang/OTP package\nfor Debian. The RabbitMQ team adapted the upstream package\nto produce packages for more/different combinations of\nErlang/OTP, Debian and Ubuntu releases.\n\nChange-Id: Iddec6ec555f98e13c2f199e5ca7dad7f4819b1a8\n(cherry picked from commit 4950c03e59b8d1af46018b81a49fa028e1b8296d)\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1314b297fd817c1ab68110e2c820bfa85afe4e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1314b297fd817c1ab68110e2c820bfa85afe4e5a"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
