)]}'
{"id":"openstack%2Fkolla~886431","triplet_id":"openstack%2Fkolla~master~I9b56e714d8dd7e2224fe19cd5349cbbb9dcf1a0a","project":"openstack/kolla","branch":"master","topic":"vtb/master-fix-gnocchi-setuptools-issue","hashtags":[],"change_id":"I9b56e714d8dd7e2224fe19cd5349cbbb9dcf1a0a","subject":"gnocchi-base: fix issue since setuptools 68.0.0 released","status":"ABANDONED","created":"2023-06-19 20:38:33.000000000","updated":"2023-06-20 08:17:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"f7de08dc76e82b6216d73d990c51f689745f7e1e","_number":886431,"virtual_id_number":886431,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-20 00:42:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-06-19 21:02:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-20 00:42:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ddb2af5d89a7c2bd5afcc15c4e1bdfc2149d98b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 20:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58a1c9fe1fedbc28bd0edc60ee7fd430d027656d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 21:02:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/add9160df5d54b21930ae17f049816c0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/614745a6093a4aaf8adba817b13fd9f3 : FAILURE in 18m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e6adeddb244241b49fed21c6516bb20d : FAILURE in 18m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/90c1f7457205401398b90fa1914b4176 : FAILURE in 22m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3060f605ad8849179d5ed62fd31dd1fb : SUCCESS in 2m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8008214b2b5340b6bd9af071184a1afe : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e1ccc2d826524d9a9a3a995009c2d002 : SUCCESS in 2m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c4aac0c3a8f941fda83604181c8f2402 : FAILURE in 18m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/be8a77fa9c4747fda0a6205b59989785 : FAILURE in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3099d97720dedfa7ad22363b2ddba0f2af8cc69b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 21:27:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b1323151cd619041a610c60bff7bd3a514a43263","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 21:54:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/756ab2f5faf84347b49303fcccc086f1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a37dbabea018420187b53f8aa1797769 : FAILURE in 19m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7cfbfa8d73774dbeb8597cb4cca825b9 : FAILURE in 20m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/02ae23b543d54deeaafd38380bcf4fdc : FAILURE in 21m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c75392d41afb4a57b8a974a1beec643d : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f51f7bdd94144bba3404fb99ef3973f : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/19fe78d47d7b4e138fa47940eedc501b : SUCCESS in 2m 04s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a4cfe78bb64644ffa9ea277f239860ac : FAILURE in 20m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dc14cd8820cc463a9be9c3522bd52456 : FAILURE in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"632dbe43017946be216b20474341df4e17093a7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 22:00:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73edcf6ea6cea6e3910c74610041c41663abc66a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 22:02:40.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"8c6c7396f798b4605d640de8a7971718efbfbb4a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 22:05:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3186c315e3393044779bfe77c927367bf45c9171","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 22:40:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"74b78c6c2b6fccc4843a9dc68d738d0a41c3a511","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 23:03:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a71b7f03ddf845bbb46c1c1f34474e69\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2f518b4e9be74c08ba9302fe328a62ed : FAILURE in 19m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/97fd86b88cb643c993d329e3c1fa70ff : FAILURE in 18m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b079e72bc3fd4d6da9a4bb57b9a85953 : FAILURE in 20m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58b1d24afa154e7bb887f677c7162e58 : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40076200669e445c9abad9544dd97dd1 : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1fbb7fc3ecf4883bf14a25701359221 : SUCCESS in 2m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1cd13b0798a84a6bad09e65f23e9acce : FAILURE in 21m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c36679ca695b491a851edd26ec5e237b : FAILURE in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dc3c3b176b7b22e3f9b78749bf0ece91c0bed3e9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 23:20:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c5278b915f229e56f99993e392b0c92de088f152","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 23:42:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e043d0f243e84e39938ee9b3cec7460a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/115f85525411443083bd56521f38532f : FAILURE in 19m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/26e4037f28834667a690a442a7a18012 : FAILURE in 20m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/61d35fc56f6c4a48a95d04e31acb8ab7 : FAILURE in 21m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/615a3ec14c34444596c1010bb8e29cb4 : SUCCESS in 2m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b69a1bcc796549adac4cfa39c3999b9e : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a0aedfcc0e6e4c8daacee2e13562158c : SUCCESS in 2m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7b1a1066e0354802a002c6791dc5ddc3 : FAILURE in 18m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/efade7c4dbb84a61959b79b264389ec9 : FAILURE in 7m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c3494638295223f6c9da099e9883a7d17c27594d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 00:42:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b975bb405a55435394c946976e2d65bd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/acfa86e4d3d74bd5a3b68a8372cf06a2 : FAILURE in 22m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5ff8537ec36c45e6a979460a0687d680 : SUCCESS in 1h 02m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/318a4f1476ea4a1eb1950e3d990c0081 : FAILURE in 23m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b4ba6216df51496ba875c6d5a06da66f : SUCCESS in 57m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9a027c00294c42939f3dc60c3077c5f7 : SUCCESS in 1h 04m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4ca00107624b949e92b14a4a46f341 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ded893e611414a66afee4c8cf005d6e7 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f0ee16ad6724b22a3d2e0e409b6205f : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88ab29823e704aecbcd6d5be297486fc : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c9297754f4641cc84e8bd7610251ba4 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/befb5eb498ee4f04b92bfea157a05f57 : SUCCESS in 4m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/97316067fc324bd18edcdafa375c4d3e : FAILURE in 17m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fb170deaf18844a9a505fe2154be9ad6 : SUCCESS in 1h 05m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e4499015b45345faa0a6dd92bf729f79 : SUCCESS in 1h 17m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/474259f17e394ae09c266a5f6819d363 : FAILURE in 24m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9daf618b3c484dc78c50f993966e9a6c : SUCCESS in 55m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/39abe910757543a1b26dd1d9aeea1def : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":6},{"id":"f7de08dc76e82b6216d73d990c51f689745f7e1e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 08:17:04.000000000","message":"Abandoned\n\nblacklist gnocchi build until issue fixed","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"13ff1f3914dbd91d24a0456f34d07cd472fb14fb","revisions":{"86c84847a2758599082b27debe5462314612566f":{"kind":"REWORK","_number":1,"created":"2023-06-19 20:38:33.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/31/886431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/886431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/886431/1"}}},"commit":{"parents":[{"commit":"232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba","subject":"Merge \"Fix cinder-volume: Add forgotten lsscsi rpm package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 20:23:06.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 20:35:27.000000000","tz":180},"subject":"gnocchi-base: fix issue since setuptools 68.0.0 released","message":"gnocchi-base: fix issue since setuptools 68.0.0 released\n\nSince setuptools released [1] new version 68.0.0 with breaking\nchanges the gnicchi-base container build failed with an error\nabout get_script_args described in the changelog.\n\n1. https://setuptools.pypa.io/en/stable/history.html#breaking-changes\n\nChange-Id: I9b56e714d8dd7e2224fe19cd5349cbbb9dcf1a0a\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86c84847a2758599082b27debe5462314612566f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86c84847a2758599082b27debe5462314612566f"}]},"branch":"refs/heads/master"},"18745a1e0b87c2640e4a67a5fd0e158b9668e9a6":{"kind":"REWORK","_number":2,"created":"2023-06-19 21:27:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/31/886431/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/886431/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/886431/2"}}},"commit":{"parents":[{"commit":"232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba","subject":"Merge \"Fix cinder-volume: Add forgotten lsscsi rpm package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 20:23:06.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 21:27:10.000000000","tz":180},"subject":"gnocchi-base: fix issue since setuptools 68.0.0 released","message":"gnocchi-base: fix issue since setuptools 68.0.0 released\n\nSince setuptools released [1] new version 68.0.0 with breaking\nchanges the gnicchi-base container build failed with an error\nabout get_script_args described in the changelog.\n\n1. https://setuptools.pypa.io/en/stable/history.html#breaking-changes\n\nChange-Id: I9b56e714d8dd7e2224fe19cd5349cbbb9dcf1a0a\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18745a1e0b87c2640e4a67a5fd0e158b9668e9a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18745a1e0b87c2640e4a67a5fd0e158b9668e9a6"}]},"branch":"refs/heads/master"},"422d253497857e9cecaef8171d5e04227204a2da":{"kind":"REWORK","_number":3,"created":"2023-06-19 22:00:58.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/31/886431/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/886431/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/886431/3"}}},"commit":{"parents":[{"commit":"232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba","subject":"Merge \"Fix cinder-volume: Add forgotten lsscsi rpm package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 20:23:06.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 22:00:49.000000000","tz":180},"subject":"gnocchi-base: fix issue since setuptools 68.0.0 released","message":"gnocchi-base: fix issue since setuptools 68.0.0 released\n\nSince setuptools released [1] new version 68.0.0 with breaking\nchanges the gnicchi-base container build failed with an error\nabout get_script_args described in the changelog.\n\n1. https://setuptools.pypa.io/en/stable/history.html#breaking-changes\n\nChange-Id: I9b56e714d8dd7e2224fe19cd5349cbbb9dcf1a0a\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/422d253497857e9cecaef8171d5e04227204a2da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/422d253497857e9cecaef8171d5e04227204a2da"}]},"branch":"refs/heads/master"},"a73b8dfaec424134c2857759173a0829bac2d4b3":{"kind":"REWORK","_number":4,"created":"2023-06-19 22:05:31.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/31/886431/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/886431/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/886431/4"}}},"commit":{"parents":[{"commit":"232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba","subject":"Merge \"Fix cinder-volume: Add forgotten lsscsi rpm package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 20:23:06.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 22:05:22.000000000","tz":180},"subject":"gnocchi-base: fix issue since setuptools 68.0.0 released","message":"gnocchi-base: fix issue since setuptools 68.0.0 released\n\nSince setuptools released [1] new version 68.0.0 with breaking\nchanges the gnicchi-base container build failed with an error\nabout get_script_args described in the changelog.\n\n1. https://setuptools.pypa.io/en/stable/history.html#breaking-changes\n\nChange-Id: I9b56e714d8dd7e2224fe19cd5349cbbb9dcf1a0a\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a73b8dfaec424134c2857759173a0829bac2d4b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a73b8dfaec424134c2857759173a0829bac2d4b3"}]},"branch":"refs/heads/master"},"58cd09604bfe01bc3b3e5052e487bd29b43c1da8":{"kind":"REWORK","_number":5,"created":"2023-06-19 22:40:10.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/31/886431/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/886431/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/886431/5"}}},"commit":{"parents":[{"commit":"232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba","subject":"Merge \"Fix cinder-volume: Add forgotten lsscsi rpm package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 20:23:06.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 22:39:59.000000000","tz":180},"subject":"gnocchi-base: fix issue since setuptools 68.0.0 released","message":"gnocchi-base: fix issue since setuptools 68.0.0 released\n\nSince setuptools released [1] new version 68.0.0 with breaking\nchanges the gnicchi-base container build failed with an error\nabout get_script_args described in the changelog.\n\n1. https://setuptools.pypa.io/en/stable/history.html#breaking-changes\n\nChange-Id: I9b56e714d8dd7e2224fe19cd5349cbbb9dcf1a0a\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58cd09604bfe01bc3b3e5052e487bd29b43c1da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58cd09604bfe01bc3b3e5052e487bd29b43c1da8"}]},"branch":"refs/heads/master"},"13ff1f3914dbd91d24a0456f34d07cd472fb14fb":{"kind":"REWORK","_number":6,"created":"2023-06-19 23:20:44.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/31/886431/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/886431/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/886431/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/886431/6"}}},"commit":{"parents":[{"commit":"232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba","subject":"Merge \"Fix cinder-volume: Add forgotten lsscsi rpm package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 20:23:06.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 23:20:34.000000000","tz":180},"subject":"gnocchi-base: fix issue since setuptools 68.0.0 released","message":"gnocchi-base: fix issue since setuptools 68.0.0 released\n\nSince setuptools released [1] new version 68.0.0 with breaking\nchanges the gnicchi-base container build failed with an error\nabout get_script_args described in the changelog.\n\n1. https://setuptools.pypa.io/en/stable/history.html#breaking-changes\n\nChange-Id: I9b56e714d8dd7e2224fe19cd5349cbbb9dcf1a0a\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13ff1f3914dbd91d24a0456f34d07cd472fb14fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13ff1f3914dbd91d24a0456f34d07cd472fb14fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
