)]}'
{"id":"openstack%2Fkolla~886467","triplet_id":"openstack%2Fkolla~master~Id6db7ffb22b3f63e9a406455d52078b76071672f","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-21 09:30:08.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-23 11:41:16.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-21 08:03:10.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-06-21 08:10:10.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-23 13:22:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id6db7ffb22b3f63e9a406455d52078b76071672f","subject":"opensearch: drop unnecessary workarounds","status":"MERGED","created":"2023-06-20 13:20:59.000000000","updated":"2023-06-23 13:23:08.000000000","submitted":"2023-06-23 13:22:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886467","meta_rev_id":"02905f1d3a53865e638a63c3cfab1fd5ea207576","_number":886467,"virtual_id_number":886467,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-23 13:22:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-23 10:08:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-23 10:08:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-23 10:08:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-23 10:08:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-06-23 11:41:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-06-23 10:08:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-20 13:25:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-20 13:45:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-20 14:01:36.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-06-20 16:14:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-21 07:15:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-06-21 09:30:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-21 09:33:10.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"b511504e1d992deb0bf8091b0583d07dc4a50c8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 13:20:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d804a5bff4e18d103d93e8e69dedad087b12489a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 13:25:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"05c017c8ae5314ddbbadf5b6dd506d4c95352496","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 13:45:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75f06d4e70b043d78eeaa35c6a5d65de\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/453f527906544806b0cde59c7cc31dcd : NODE_FAILURE Node request 300-0021495597 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/084639a4f8834b42aa3a36d806703d86 : NODE_FAILURE Node request 300-0021495598 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6008ad02121e4388bca9748874f532dc : NODE_FAILURE Node request 300-0021495599 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/40511cffd538485fa45a5d9aee099fe5 : NODE_FAILURE Node request 300-0021495600 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fad56ee4c12a4d16921d0cdd98aa4bd5 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1bf508f3d2be45c79ba6b0340c7ff7d1 : NODE_FAILURE Node request 300-0021495602 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/24c6535ccf954c66804769c07498f101 : FAILURE in 19m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/012eb9e2ff0e463faca9f837d7ee36bf : NODE_FAILURE Node request 300-0021495604 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79ef597e953a75f89762d7326ff47d3c24e6f9c9","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-20 14:01:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"52a1246546c6b6d8c9f5b8905b1bbff230bfa6b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 16:14:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0602378dad264c19a93176dba99b8f0d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/99016458b7964545b358ee7b46d8174e : FAILURE in 34m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/26f9517bb2e54779b2381e336ee14546 : SUCCESS in 1h 01m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0219a83ab759458f81bd4455915d2994 : FAILURE in 24m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9a349cf4e7ae4d1d8eb7d8df7ad7cace : FAILURE in 24m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5f6bc23f47914c06a93080ff967d385c : FAILURE in 28m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e216649f86fd443aa87248c999198e5a : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80eb85f1d05e43238f9a2d26aa9e26be : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f12dcab8668445858f2c4d8057c773c1 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/061765b40e334400be004c519724d5d6 : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c70e0d664ee14195976010f8ab944f97 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/452b66449b574f468b87c82a60e092bf : SUCCESS in 5m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2bae25626fb848e98e9531624a51be0f : FAILURE in 24m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9e0833f357c44291a4d3f42d7a50054e : FAILURE in 18m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/daeeda2e5b224c31811eaa7dca09e9ac : FAILURE in 20m 08s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3ef59b45aee544b28f7e5e005e815d28 : FAILURE in 18m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/35bd4989247a46d5ba3a676832132317 : FAILURE in 1h 05m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/89883d5013ce4443b5e58ddc581db125 : FAILURE in 43m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/65cf0bc5d1a249b099d8c6b978cd119b : SUCCESS in 1h 01m 57s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6aa46e8cd34f4f9da9bacd6ec8cc859d : FAILURE in 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad9242aecaeb8845cfa11bdecbfc840e6d6c3cf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 16:16:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1357c4f95b9ea1c6d6150418cc752eef759055f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 16:42:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d2454704c074c31a66321978bdb2db6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e931e39169a444a3aaaf405ad6e9f906 : FAILURE in 18m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/591d8118e5f44829bbc60841e133ce33 : NODE_FAILURE Node request 300-0021497590 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c381c0d3e8f5405bac81e89fdf9fbc36 : FAILURE in 21m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6918d1c6f5824277a44b2bd789e07e02 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1878bc67cba04d4ebf34087632f0db61 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35654f930c954a0aa970d43bc7efbded : SUCCESS in 3m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a3d3f20c538d451d9fb57fdf6606f933 : FAILURE in 18m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/006eaf6fc7a1482d9b295159c8cf2de5 : FAILURE in 7m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4eb732e9b07a1c7f0ab17685e94933680095cf0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 19:23:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/286bea7f8eba402c90682e74dd2e7110\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/116b10f4044f4cebbc1e99e81605969c : FAILURE in 23m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4587059d55a441e8af33f66f2a8901fe : SUCCESS in 1h 30m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f646d6f440694493b3566542686e2481 : FAILURE in 23m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7992f3f62a74440b89cb01ed84f0cddb : SUCCESS in 50m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6ae539024d794cbab3d89bd4517b7886 : SUCCESS in 1h 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/458a5768564848228d1a7a9a8480f05e : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eae098c0ea314f16ae829e04e27f2cc1 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c027945ebd745e791a89cb41092a23f : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2282de7a1b44d99b30b236f9f0b5a9b : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba50363edc1644dba745a17e8c099381 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/094f9a89c3f9468f9def6ce7659a29e4 : SUCCESS in 4m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/478a1bf8073149c1a15bbdd92558862b : FAILURE in 19m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8b24beb336ee4333a46d12b07deb06e1 : SUCCESS in 1h 01m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0a9c6a03cd21477ebbd3bc1e53ba11ad : SUCCESS in 1h 20m 58s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9263c70bbef94401a3654d7e2d335412 : FAILURE in 18m 42s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5e279e02b0504277a8203eef14fee630 : FAILURE in 23m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dd1ad56464c44904b5cd3e487b7fbaa0 : SUCCESS in 56m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ecfb500070ac4323b200add79b7eeb25 : SUCCESS in 1h 00m 01s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9025926d7f5e419393613fbb9394fa37 : FAILURE in 16m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6ca146f61755b696158518cab8c80dfc06b4691","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-21 07:15:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71314a2ea2fbd6fd3aadf5be8e151e1aa6a3dadd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-21 07:18:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"07f752333ab480ca40ef8c86929e360913122cda","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 08:03:10.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9097d4d81809be2d2c857a6ec5646f2246a3c213","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 08:03:21.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"15bb5ad7c83cdc31e85c6783c3e9c181c00da116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 08:04:11.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b6b309a3759c4af677094927711d544be5f309e2","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-21 08:10:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7071978df465c57989621c493a42f867171f5406","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 08:11:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f3ddf4fe2cd558721ad4537340bccab468eede2f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 08:20:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ebda67e2af3587fd4f53729a9e8eaccccb274ff5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 08:36:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/482dcd8beb274ec0be4b9aeafe5426a6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1ed3a474e5aa4e3e921b42f1cbb65ae8 : NODE_FAILURE Node request 300-0021503893 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d06b54ee6ad5467d95037d75c50606af : FAILURE in 29m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0f8a8d0e9a094434a35aa5fe28bcfc74 : NODE_FAILURE Node request 300-0021503895 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b27b06d118c0479e8b3050d60573bc1c : NODE_FAILURE Node request 300-0021503896 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2748dfc9dc5c4ec5b940c0851083bec2 : NODE_FAILURE Node request 300-0021503897 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/33ac8b9677e44db092bade6bf05b53b2 : SUCCESS in 3m 25s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2aa7ca31cac046c5bfe5c0ae39187fee : FAILURE in 28m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0fc7e7eee5c14b0497f0aebfcf25dc26 : FAILURE in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f71d02a5e33e9b918e0aa38c0fd2ae3d23b6a9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 09:28:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44de92d484bf493e8c5594c9454c37e3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7cf1057686fb41e9a1ea54d328adf11d : SUCCESS in 23m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/58d4b6c3e2b1493694acccd01acbe5eb : SUCCESS in 57m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a5c62801f28b4ef6b7eb546d56f05c4b : SUCCESS in 16m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/888bcd8c8a314d3b9948fa812577f30f : SUCCESS in 57m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/14aeec337e0048bab634ae057f3e4b46 : SUCCESS in 1h 02m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf70c42f42104f72aa5c45fac87a4a62 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/343319b82ecd437e95e1268f02fe9467 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7efaf0439d7e47c4a4aa4c4e642aff8d : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3ea100187e0439faf838d58a442bdfa : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6df453b6daef4adaa6f55c2117d17c91 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/230bc79de2564bc5b006c892a8e82aa6 : SUCCESS in 5m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/542472aadcd647a58945e4473083fbdb : SUCCESS in 23m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/51a5ad100b264d2fa4a0dcb8072c2f65 : SUCCESS in 1h 05m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cd306d295c8140fa8abe556e73a13a07 : SUCCESS in 1h 16m 26s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f0a23ce4ba2546128d726facefcf4def : SUCCESS in 32m 59s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/156d61d5d32f4359aad0d989d00100d1 : SUCCESS in 1h 05m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c22d963906b64dc18b8d2bf16718994d : SUCCESS in 56m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/39fcd6e27b4b48bfa431c173469c1710 : SUCCESS in 1h 04m 39s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/124ebe55216141f3ab4ddd362bb36cdc : SUCCESS in 50m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fb6681d9c69028ca155bbd058186edf59c241d45","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 09:29:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ca687e31be3ac5d36c55a63441cc8eb5a83a0c41","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-21 09:30:08.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"76d9584276133de6169a6a1591eb01a00989461d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-21 09:33:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f0f4ced90c5be052ceb455867664aa5d7c8716db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 09:33:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cdd5709a073ec6d16e3e7cdcb265be87236bb3cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 11:32:58.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55c787d5332c4f0b9cd717ad6d2dea03\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5bb3d774aa714bddae10fe71668d987a : SUCCESS in 16m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/347c97229bf14c61b8337effb9fca9fd : SUCCESS in 53m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2c2391625e1047f69f0709aa35213c11 : SUCCESS in 1h 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d0d5cf752b84148be520527981c6f3f : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80eee3cd26bb4c43b115d0d9aa20cd32 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afd910603feb468e82fa3396be4fd06c : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae51cd5cd7ca4503950d9f6f004b6b06 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0adbb8b2a7ff4476870626074232ff88 : SUCCESS in 5m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/13d559c1d5844d3ea295370fe6faf3df : SUCCESS in 33m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8a3f9531f21a492e90a496026252370d : SUCCESS in 1h 20m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/22df77a5624649c8b87ae261a2b58b0f : SUCCESS in 1h 04m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/57f5e6b5fd8f415e9b81927a10911fa6 : SUCCESS in 14m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5c1ac2bf45f04fefab7692a4f9259d35 : SUCCESS in 47m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/026f112b66224341879fc5c62fe0b75d : FAILURE in 53m 32s","accounts_in_message":[],"_revision_number":3},{"id":"af0d998e5ee3c53912942814b62c6f9837d3c0c9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-21 11:46:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7f8850de36dcf15f8b9a1ffdbf9226ab18d9580a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 12:14:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8d04673ca494a0dad033ea3a42c9e78\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/207bae2ab895414fa1a941b6d8bc3a7a : FAILURE in 23m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/17446f548e5949a39e7db95aac15fff4 : FAILURE in 22m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b37dafcfeb174bf293abd6a9e32ea790 : FAILURE in 20m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/657d5c6c77444f5b9a6b1e7c190a1306 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b65022059dc94207857cbaa2c11250de : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8bdce328b9ed454587de65b4b8d11aa9 : SUCCESS in 3m 25s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f6d7880279d649e18ffc4dfeaead58a7 : FAILURE in 22m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/211fe553cda24c5cb5c433addbe1f878 : FAILURE in 6m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"334bc79eb84a7340d9aac78265b196624771e79b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-21 12:20:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f09478a1ae0cbc7c2fd43a781b939d9c8dbc632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:42:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b98ead8a5b348be9e31493c29a91fdb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1abd643d81b549ab87b0a8404a388c74 : SUCCESS in 26m 10s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4904d4a357c5438896389245c1aa214d : SUCCESS in 1h 19m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d03073dcbaee4fc5951acecf0976e478 : FAILURE in 42m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/160c40d4d7024b3398787e66d3490087 : FAILURE in 39m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/181dea2c464a4f0bab24b149a8712979 : FAILURE in 34m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f2c2b8395744159b40726843f065d31 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15df49002de841c483084e37864cac4f : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c490f42bfbc4b96bc56f2f9102a107d : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a08f33f8e0c44efa1a346f997eef21e : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7b730ca74e24597a6a815d6252eee9b : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e18f92100e41c7b1eadc0e4f6199fa : SUCCESS in 4m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c03779f002594b88a1810bce694f56fa : SUCCESS in 24m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/97b4b221c4f24fd9b16a92b26675ba46 : SUCCESS in 1h 25m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/945dee49dd4e4c619e1420efa09994c4 : SUCCESS in 1h 41m 38s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1986dc8d327e4884ac57f717107a1096 : SUCCESS in 37m 22s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/99083551bbfb45278809aec77c4fa076 : FAILURE in 26m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/006e03cc413a4435bdf61df9145b4932 : FAILURE in 14m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b783e412eccc4ea3889024d79e324cbf : SUCCESS in 57m 30s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ccfaaddd4adb476cb6447d48f871d8ac : FAILURE in 15m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6be2fee66fbcd99f1eaca1b4d6e50f9dc1fadaff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-23 10:08:50.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"243838aaff1cf7d029294fc015d8ca789df25c52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 11:38:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73abedebb303462fbb9b1de14c303b78\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fe2ce3cad34641458edbb219c68656dd : SUCCESS in 23m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/62d5ec1656eb461ba86d9324590db5dd : SUCCESS in 1h 15m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c67a431bb0c3441f8844f69b283c178d : SUCCESS in 22m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/81230799c799496d8447a68ec8a38658 : SUCCESS in 1h 10m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e0ea34b494cc4d7cb5c4b80b7a922941 : SUCCESS in 1h 03m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/531f0f265ec046d6a7f02c329d4f20ce : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96fdfefea76c451db49b0b2105c9543d : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29cd4ec43c1b4357bf2930b50991f50a : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/322409c2dd0244f2bee6c86a5a94ac0c : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d907c2a563444d8680ef7e22b8b7edd4 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0353b7c35974268aeeda3e7863f6a1f : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ca47419db4b14ba9a4370897b1fbe3fb : SUCCESS in 16m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dc936321085f454e8cfd50d1fd22157a : SUCCESS in 1h 14m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dd685256d9bc46969d348bf4d07dfa31 : SUCCESS in 1h 13m 24s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/42bd2149a1a846fdab3411de45bf3165 : SUCCESS in 31m 47s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4ce62df40a2f4a44acf540db843803d1 : SUCCESS in 23m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/385b67ab0e4f4da1926898048e72f040 : SUCCESS in 1h 21m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d58a6ef2f3b94c6c81bff7e653b73396 : SUCCESS in 1h 16m 25s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d893dc43534e47668772b4e8e97ef50e : SUCCESS in 32m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d738a16d5f858bf0d2ed81896fa54a7ada88eb42","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-23 11:41:16.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"66ba41a465374eb1950a64f0a9f24e24dbc7750a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 11:41:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8649b18cff5d52eb45e91ea7f906edacb3d5a793","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 11:59:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7145b09010904683819977541fde5f04\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/974ba2b7d28e47b494b92c94d3dcb8ee : NODE_FAILURE Node request 300-0021526640 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b97c29ad86204bc69fa82fd6811a6932 : NODE_FAILURE Node request 300-0021526641 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d68f1aa4f1cd4f229f251259e8c5f1f1 : SUCCESS in 1h 47m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc2ccc67bf254463b6d8cfe9dde8258a : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2822ec40d7d4fb1a9aeaa41b926b805 : NODE_FAILURE Node request 300-0021526644 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f7d538fad5e446c3a957e5fff4a77c51 : NODE_FAILURE Node request 300-0021526645 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5c905d436d1849fa9e4d6c3e86401ce4 : NODE_FAILURE Node request 300-0021526646 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d65b3ebab81040fb941c80e62339740f : NODE_FAILURE Node request 300-0021526647 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4c9de53efedfd1c1c88b667ff1137136062f078d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 13:22:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf4ce54416144742b886ce5684640d08\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/90486bab7d5d46aa995875a367b61a0f : SUCCESS in 14m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f92fead63e4c454bbebfc3878b928a77 : SUCCESS in 1h 06m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c86b10b869134521b91ad34f06576fe8 : SUCCESS in 1h 14m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba763d2442534f6ea64657561d7773b4 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0455cada303743a3abae626ad98783d8 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9895227e369742168ed5a5a8b37b905d : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2c994fd25274381bab865824f3946c4 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f279cc99bc45ffb77a159e5e11da29 : SUCCESS in 5m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6a420e70d8b144c0b9af27e3dc55e4c9 : SUCCESS in 15m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f564522f83fd4e33b669f4c3434740e7 : SUCCESS in 59m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6624b23651ca4768bae5ccccbe8d214e : SUCCESS in 1h 17m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dd1a045040de4a84aa00a1ed12563bd6 : SUCCESS in 1h 36m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0f65170709594449a0d65f5c7fffc66e : SUCCESS in 51m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b725f475bc3e49628419a1075423dd58 : SUCCESS in 1h 01m 32s","accounts_in_message":[],"_revision_number":4},{"id":"9430b5b18834b83030ecff40e8488b999daf9384","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 13:22:11.000000000","message":"Change has been successfully merged\n\n3 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":4},{"id":"02905f1d3a53865e638a63c3cfab1fd5ea207576","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 13:23:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3104aa0cacee499ba202b39b4992285c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4017b1d8125b4e769de14c57d4973bcc : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b33167972d6e17b2eaab85e72043e3b4efa4b789","revisions":{"b3ce1fa1bd06e4a3a273a3e8d37c4d660a93de90":{"kind":"REWORK","_number":1,"created":"2023-06-20 13:20:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/67/886467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/886467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/886467/1"}}},"commit":{"parents":[{"commit":"232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba","subject":"Merge \"Fix cinder-volume: Add forgotten lsscsi rpm package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 13:20:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 13:20:57.000000000","tz":120},"subject":"opensearch: drop unnecessary workarounds","message":"opensearch: drop unnecessary workarounds\n\nChange-Id: Id6db7ffb22b3f63e9a406455d52078b76071672f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3ce1fa1bd06e4a3a273a3e8d37c4d660a93de90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3ce1fa1bd06e4a3a273a3e8d37c4d660a93de90"}]},"branch":"refs/heads/master"},"702a61789a2b873e691ac817b65c779e4705ef41":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-20 16:16:23.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/886467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/886467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/886467/2"}}},"commit":{"parents":[{"commit":"11f90f3746f6affd34885019a82eb94d3c098ee9","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11f90f3746f6affd34885019a82eb94d3c098ee9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 13:20:38.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 16:16:23.000000000","tz":0},"subject":"opensearch: drop unnecessary workarounds","message":"opensearch: drop unnecessary workarounds\n\nChange-Id: Id6db7ffb22b3f63e9a406455d52078b76071672f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/702a61789a2b873e691ac817b65c779e4705ef41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/702a61789a2b873e691ac817b65c779e4705ef41"}]},"branch":"refs/heads/master","description":"Rebase"},"d4eecaa3d53674a05bb8c0cfb916c341f98db942":{"kind":"REWORK","_number":3,"created":"2023-06-21 08:04:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/67/886467/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/886467/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/886467/3"}}},"commit":{"parents":[{"commit":"11f90f3746f6affd34885019a82eb94d3c098ee9","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11f90f3746f6affd34885019a82eb94d3c098ee9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 13:20:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 08:04:05.000000000","tz":0},"subject":"opensearch: drop unnecessary workarounds","message":"opensearch: drop unnecessary workarounds\n\nChange-Id: Id6db7ffb22b3f63e9a406455d52078b76071672f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4eecaa3d53674a05bb8c0cfb916c341f98db942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4eecaa3d53674a05bb8c0cfb916c341f98db942"}]},"branch":"refs/heads/master"},"b33167972d6e17b2eaab85e72043e3b4efa4b789":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-06-23 10:08:50.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/67/886467/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/886467/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/886467/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/886467/4"}}},"commit":{"parents":[{"commit":"f89cb612e089de71ac8283d45e41ab3d3ddc24c8","subject":"Merge \"Revert \"rabbitmq: pin to 3.11.16\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f89cb612e089de71ac8283d45e41ab3d3ddc24c8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 13:20:38.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-06-23 10:08:50.000000000","tz":0},"subject":"opensearch: drop unnecessary workarounds","message":"opensearch: drop unnecessary workarounds\n\nChange-Id: Id6db7ffb22b3f63e9a406455d52078b76071672f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b33167972d6e17b2eaab85e72043e3b4efa4b789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b33167972d6e17b2eaab85e72043e3b4efa4b789"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
