)]}'
{"id":"openstack%2Fkolla~886478","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I2a5a0d761223b4e99c8df59f80261108380d5038","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-21 01:43:32.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-20 21:58:11.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I2a5a0d761223b4e99c8df59f80261108380d5038","subject":"opensearch-dashboards: Fix permissions","status":"MERGED","created":"2023-06-20 13:09:26.000000000","updated":"2023-06-21 01:44:33.000000000","submitted":"2023-06-21 01:43:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886478","cherry_pick_of_change":883941,"cherry_pick_of_patch_set":3,"meta_rev_id":"37778cd9fc10a0d879c106c464cc3ac01a2d5fa6","_number":886478,"virtual_id_number":886478,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-21 01:43:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-20 18:38:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-20 18:47:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-20 18:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-20 18:38:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-20 13:11:31.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-20 13:41:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-20 14:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-20 18:47:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"e6a9e763536b4cc9f3aca611f9608edb71d745ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 13:09:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"041cd6a8049ac5f421a62f93cca12e23f8620c4f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 13:10:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"25fb046091f87ad0426d59992f7a1e30d79a476a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 13:41:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e989991edf3046e99b20b3bb90d0b0cc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/131fb07dcea1492f9bc3b2465f90d2ad : FAILURE in 23m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/62021c51cb5f473cb90ac66dba145fa2 : FAILURE in 28m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/97b06e964bed4d759078b2b4d47a9d73 : FAILURE in 25m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5b1058d8fef4cd39d022e511c01ca7e : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e514e6511d034fea88346eef39c49e20 : SUCCESS in 2m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/593a6c55d68d4e1e975090e0fd27f783 : FAILURE in 25m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/413c75e353404d3a8f13128869d122b3 : FAILURE in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf10ef89f35e0c8644142bbc9c4272e48cb43084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 14:50:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/555047ec7cab4822b105ff49d8b9bded\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6b9d0c4ab6c643adb80ce9a5d04637c0 : FAILURE in 30m 50s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2c55382dc0984033b3e03975b07ca495 : SUCCESS in 1h 23m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3f4744927b5842a3839c80c33d1f1240 : FAILURE in 21m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e10d156b655c48a488d8a906991fd58a : SUCCESS in 1h 03m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cbf2640458f04177840b7e163bb90d2d : SUCCESS in 1h 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892fc7b08c7440a1b6604b12bd7f46fe : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cebc1c4514648979e11d0a60e5ec6e5 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ae9b059d78e4155ab465ba9f0e322d9 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d54b2838d0b248a6a78e0a823dcc9668 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/684b6fe2da484ce1b7092e51eec2340e : SUCCESS in 6m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/80b08f146a2c4af994ca835a8e1dd038 : FAILURE in 29m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/28e4a52f886145ddbd25e02f40cc60f2 : SUCCESS in 1h 24m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bb2d1455d7ce4300b2e4af045ad7e4c2 : SUCCESS in 1h 32m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9eb23f8f727f40eca2a8e9f96ecc90f1 : FAILURE in 35m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9e98a643d9854c208738160dedc7a36a : SUCCESS in 1h 17m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1edff4e5ef504697ba72d19c0835e6bc : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3e9785aeb509397b8c57d12554144f7a59c8042d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-20 18:38:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"259f22e46ece2f7665eb61f07868e5e5dbccaf3f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-20 18:38:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7110789a83a47d03263d863cf18a1d264011d515","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 18:47:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c63cf974ea1ef54b069d6618ef9ec327ceabacbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 19:00:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4897ae7d8d9e4da0b6daa4e9b3b1c4dc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e033c93558e14cf38e8d7f05c0e10c87 : FAILURE in 20m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/519466beb0ff44f8a77b0c0f917e8258 : FAILURE in 16m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9d4a66044a2643e9a8a2f48384a5848f : FAILURE in 21m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6be2001ed754b47b966686e6330660c : SUCCESS in 2m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca9944b146fe43629831d671eba7c9c0 : SUCCESS in 4m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5dd81c900dcb45518af3765121d9d40b : FAILURE in 20m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6d1aaf46f695477187d916a20b5c679f : FAILURE in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a226d11e2aa17946a6a797d748571aea14b4d04c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 20:22:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19fe00ce59844de198fb3423e6bb8711\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/68cef22159c24bbba324c61b84da52d4 : SUCCESS in 21m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ef498ae0da224de1b8c8fa3946fa2090 : SUCCESS in 1h 13m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0a0c75b50f7944fd936daf4f8b53404e : SUCCESS in 14m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/54364a4c583f41498940d055de8e5430 : SUCCESS in 1h 09m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4b5ddb64b2a141e39ad2dd18da14e682 : SUCCESS in 1h 26m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eb7bd3ea5b44d04a40fd25209a33b86 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9033edd353543438f7bb50ab4cb35e6 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc6999d7f1814fbfa00e402246b3673c : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/674ffecd57e248d3acc8331a4340bd2b : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2c2216ef454908a00d6c228343e1ae : SUCCESS in 5m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bded294623f74eb9875ddd5e7b4c32e6 : SUCCESS in 25m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d30664a4dd6a4e99a938d053e4278701 : SUCCESS in 1h 16m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8c190390b58743248a75f8dcc7c7c952 : SUCCESS in 1h 25m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d924761f8efd4ce6a89273a3147d6d20 : SUCCESS in 15m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/61bd8acff3b4431c9fbeec43db970f92 : SUCCESS in 1h 01m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/44ee673088b8462bb235f87f6d07fb13 : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d878e76b2631664014dcc59addfb1d69feb0e658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 20:23:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1ce274c2750595ff73a10aa90c482273a9b4c8c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 21:42:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/beb03e0aae584cd0ac68be0251a39b40\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6463bb557e6d43fe8b88727a6b34a55a : SUCCESS in 17m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/25a3f73a55924a7d873df6115a06f0fe : SUCCESS in 1h 11m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bc2ce898532e442783a2c802a46dbde4 : SUCCESS in 1h 05m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/382e740f0ce84cc1978eca78f468c608 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e61164bfb28549ca9cf50946b4d7d464 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ae689c8cc824ece9fae394c77d5ef27 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba09e6b8a6bf4a8c87f62a91597e2866 : SUCCESS in 4m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6bdb8c1fd57a421f93b6cca0a8769cbe : SUCCESS in 21m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2a75eb0f6aac4ea3a23668f10fd69e44 : SUCCESS in 48m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cb30b4d809fe467abf8371457820588b : SUCCESS in 1h 12m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/acd1baabb55f4e60b0473df43390b974 : SUCCESS in 24m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0659202c05fb40519cd0d51ff2a367f8 : SUCCESS in 54m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e3b21589bb934c238c0e48852057054e : FAILURE in 45m 54s","accounts_in_message":[],"_revision_number":2},{"id":"fad51819835ebba70ffed24ba6f4267305e67c6d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-20 21:58:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3387b48810489d382b78a9b743d8becd831fa8ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 22:21:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39b5425fdb094ef9b48eb5c2f60de175\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d2f61413db9a41a7bbba6bb332c724a3 : FAILURE in 17m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9ccc5db0cda049ceb51b8bcc83bd243e : FAILURE in 16m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2e81b694486434eae21002223f74b90 : FAILURE in 20m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11a2fdaecfcb4a4fb4fb143158439be6 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2349617983d74c3b9df6212f2a126ad4 : SUCCESS in 2m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/12b625b7d8d5485d90757f313bff08bf : FAILURE in 19m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a6c59b29fea6472ca884bcdcc73e15bc : FAILURE in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad0a0899cdcc2b3dc1e4e2783d893dc19a606ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 23:45:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c943c4b6ee34af0b68d8f84fcdca572\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f345f2b7896948d99397191c87fbb868 : SUCCESS in 21m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cc44144fdc5a4808a5bee571fbcebc35 : SUCCESS in 1h 06m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5a251f2001c74cd0817c5f9da9879da0 : SUCCESS in 17m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/be2bf51f05934b949bc188c10c3b9f71 : SUCCESS in 1h 00m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3840576f62434359b30c0ec63ecb0e35 : SUCCESS in 53m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca4d847a7d84bdbb14d081896704efb : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47aec87cd25140a0a31381afe5e60c55 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c4467cbc25a488e9807bffa8cc03c1d : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3f9a1ce1cc246ca80cf8f02ed7d28f5 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caeebfe45b5f42f6853dd14dae712a06 : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/499c17d3a8d043979d89d3687c35df74 : SUCCESS in 23m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/12b739ed49204e038c0b4e89c8121fbe : SUCCESS in 1h 13m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/05888ccdd247430db1e5aa888a250b1b : SUCCESS in 1h 23m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b392f5f578194f74a6941de8edb0efd9 : SUCCESS in 21m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d71f89eacd82449287cdb8dba4ca4fa2 : SUCCESS in 1h 08m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/438f20c39a644e8e844f2f5981524dfb : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":2},{"id":"d5930437281a5167203a0a95e1f350c28300a04d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 23:45:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"329f0f2e5f3c0455ba1a3f1e9a0e39768a51ee3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 01:43:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4981fb11529d40449df27d7a99ffe33d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8319fab899b44a0488332209c089069c : SUCCESS in 16m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/931e54225eae43879491881225f57007 : SUCCESS in 1h 02m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/af663efa62bd44e997371a97f77fea17 : SUCCESS in 1h 20m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/973e59f5285d4fbf9d43ec7d0e3a08c5 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83fd2a9b35a347318d7906f56103fbe1 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b99a1a427d546c9825d66b7f9db4c64 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9198cd269d9842929fda67f77378b42c : SUCCESS in 4m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ce518b9f02964ac2b308e700f713d3a3 : SUCCESS in 22m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/23ead8acfd35430ea8f7b696a69c6922 : SUCCESS in 57m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/23bae146fe7f4ab489a4484a5f0f0426 : SUCCESS in 1h 04m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c3d493def3d247b086f18bdb23ddd8d2 : SUCCESS in 13m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c2017e7c8a3a49299b6d520e4b4aef25 : SUCCESS in 1h 05m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ad501ac28a2d4c06ba3ed83c11f97e4a : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5244849e19f5786ced822a1bef46e35969089e84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 01:43:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"37778cd9fc10a0d879c106c464cc3ac01a2d5fa6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 01:44:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/648289c41fef4ce88aeb50c27de89fd9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e1e9d3e5644da9bc366e8a05f0cc5e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f4600a14974163fb6e1a7b2de398185f01744e6f","revisions":{"bda65e83bef1a7bd9ea438772c498d94d06e0a05":{"kind":"REWORK","_number":1,"created":"2023-06-20 13:09:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/78/886478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/886478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/886478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/886478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/886478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/886478/1"}}},"commit":{"parents":[{"commit":"eefb79a2ed77f5f14b01a1be2d28a94990205dfe","subject":"Merge \"Fix Venus containers built from correct branch\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eefb79a2ed77f5f14b01a1be2d28a94990205dfe"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 13:09:26.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n(cherry picked from commit 6786e4384c1a19e4658310e109b8ab9704f5eff9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bda65e83bef1a7bd9ea438772c498d94d06e0a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bda65e83bef1a7bd9ea438772c498d94d06e0a05"}]},"branch":"refs/heads/stable/2023.1"},"f4600a14974163fb6e1a7b2de398185f01744e6f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-20 18:38:02.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/78/886478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/886478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/886478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/886478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/886478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/886478/2"}}},"commit":{"parents":[{"commit":"9efdc98d9fe8493735961696fbf95b87b499bd89","subject":"Disable build of Gnocchi containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdc98d9fe8493735961696fbf95b87b499bd89"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-23 14:14:21.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-06-20 18:38:02.000000000","tz":0},"subject":"opensearch-dashboards: Fix permissions","message":"opensearch-dashboards: Fix permissions\n\nSince we moved to using rpm/deb packages, the user has changed.\n\nRelated-Bug: #2020152\nChange-Id: I2a5a0d761223b4e99c8df59f80261108380d5038\n(cherry picked from commit 6786e4384c1a19e4658310e109b8ab9704f5eff9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4600a14974163fb6e1a7b2de398185f01744e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4600a14974163fb6e1a7b2de398185f01744e6f"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
