)]}'
{"id":"openstack%2Fkolla~886502","triplet_id":"openstack%2Fkolla~stable%2F2023.1~Iafeb9093340621d65a9c9cd126d65679d981dfb1","project":"openstack/kolla","branch":"stable/2023.1","topic":"disable-gnocchi-build","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-20 20:31:31.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-20 20:31:31.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-06-20 20:31:31.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-20 18:35:39.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Iafeb9093340621d65a9c9cd126d65679d981dfb1","subject":"Disable build of Gnocchi containers","status":"MERGED","created":"2023-06-20 14:58:35.000000000","updated":"2023-06-20 20:32:36.000000000","submitted":"2023-06-20 20:31:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886502-disable-gnocchi-build","meta_rev_id":"fb98ce4ec09bf041c65411b0acc52ec2fad7a50d","_number":886502,"virtual_id_number":886502,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-20 20:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-20 15:28:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-06-20 18:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-20 18:35:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-20 15:28:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-20 15:54:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-06-20 15:54:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-20 15:54:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-20 16:38:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-20 18:30:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22593cff04d2323e6aa184ff56627c0955354173","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 14:58:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e5eda07e526b3bc1ba994b680bbea18e636b2bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 15:23:08.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"88d8f937c853a52d9fbf8d8e743c9920af023c2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 15:26:36.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b1adbf3fa9ace943839525b6aae25a22a99e9af0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 15:28:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7ecf034e472c83830b3f85becae9cdf29389b796","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 15:28:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f62ab0161bb74b297c9934ffd1c917bf9c5d94ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 16:38:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/987d0d87a2084d9fb3893a566c79ba4a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ef792afb217345a5b068a8decb4eaa81 : FAILURE in 33m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/77cdca391e4349e89ee9321e74a21275 : FAILURE in 32m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5ef293a846844579b5a7fdb7283c43b9 : FAILURE in 1h 09m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/10777f05ad2f4009a6f85e757e06b839 : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b464a5765484136af1a8504761ba179 : NODE_FAILURE Node request 300-0021496897 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/52708a4e40344d2187dcb30018b1fc5b : FAILURE in 33m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7f4a61ddffbc43e7a738c5f557f66840 : FAILURE in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"52d5809927926185db21b70a5537a00d30dc3544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 18:30:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4becba8903964118af1cf90ed43d2c6e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2acf5d687d88453fa48e9c84348c980a : SUCCESS in 39m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/84eb89b5729f432faa26bd3553aaa6b0 : SUCCESS in 1h 06m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/da844443f4b84eba957d5d309ba87380 : SUCCESS in 21m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f5c2cfcb78ab47b1861be22cce3d9bf1 : SUCCESS in 1h 02m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a1a61d1765064f66b5963b4f2b1385e6 : SUCCESS in 1h 22m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2964d2cb410148c2b0facbf7c5567179 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01805fe22b754a18b74b840fe8a7387d : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dabf32c3209541df969a025e782bbcc6 : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d988f51d719b4fc0b58a3c99be4c0333 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca771a7bd13d40edaab5d978a7740791 : SUCCESS in 4m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7ee0519629144df2b514832c4f385fa5 : SUCCESS in 25m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1dc4099613df40ba8a59db7bd88549ae : SUCCESS in 56m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5846a9763ad64380903e87c62b40df11 : SUCCESS in 1h 03m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e47e9da1391d4c369c7408c66a550baa : SUCCESS in 18m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4669b05a833d409d869eb5f89fb10b8d : SUCCESS in 1h 04m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9a21f8bc580742d5b5252a77fb79d3e1 : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":3},{"id":"451528c3fd9690306841518b2956a90433746e17","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-20 18:35:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9860208a2faebe8aab1ab4e05cb6b880e6855182","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 18:36:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"363396e2db6124ae79ee0ece5bc82117f0b27199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 20:31:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b0f2d2c3ad46f5b90eca4970041bd8\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dbe4d188af7c4e6b92035591abffe4b7 : SUCCESS in 15m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4ec291dbad10441f87df9d6a01e6d1aa : SUCCESS in 58m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/08fc4623697340c1b1126d95629b8c31 : SUCCESS in 1h 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4d7044c02564c7d81a3a5fba2e2a70f : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0decbdbb38d94778954f76dad63d6049 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/138e11fe80024de4a1c93b819c667110 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1817a1317efc4c6f9c512ae27603163a : SUCCESS in 4m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5a7a26673ac3429982614224d92d9b7d : SUCCESS in 25m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ce3141e0ec3b4143b6c154567135a402 : SUCCESS in 56m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/516ec146dc004976ad52dffd96830274 : SUCCESS in 1h 10m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2cb95f99db9641e08e6ccb6861e6d31c : SUCCESS in 21m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8df9d630cd054a66b30fc5cf0e38f9c9 : SUCCESS in 1h 14m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e107dc4ccde94e269296b20e580fa259 : SUCCESS in 1h 16m 35s","accounts_in_message":[],"_revision_number":3},{"id":"720c4e592dc587e6e92a1c3d89c49e8ce4baf231","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 20:31:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fb98ce4ec09bf041c65411b0acc52ec2fad7a50d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 20:32:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/656bf9aaa5c74e108f0dfc878b9846e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4180e0426e924fe7bbe3c8561c00c719 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9efdc98d9fe8493735961696fbf95b87b499bd89","revisions":{"b38757d91c190d9b65c3336f7a3c662e87c2f69d":{"kind":"REWORK","_number":1,"created":"2023-06-20 14:58:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/02/886502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/886502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/886502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/886502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/886502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/886502/1"}}},"commit":{"parents":[{"commit":"eefb79a2ed77f5f14b01a1be2d28a94990205dfe","subject":"Merge \"Fix Venus containers built from correct branch\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eefb79a2ed77f5f14b01a1be2d28a94990205dfe"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 10:06:48.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 14:58:05.000000000","tz":180},"subject":"Disable build of Gnocchi containers","message":"Disable build of Gnocchi containers\n\nRevert this commit when [1] is solved.\n\n1. https://github.com/gnocchixyz/gnocchi/issues/1304\n\nChange-Id: Iafeb9093340621d65a9c9cd126d65679d981dfb1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 5505cd000ce7f12c272c5b62074c9cc6bf1bb7a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b38757d91c190d9b65c3336f7a3c662e87c2f69d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b38757d91c190d9b65c3336f7a3c662e87c2f69d"}]},"branch":"refs/heads/stable/2023.1"},"b2bb362e5011023170bb664e2045217e20f56884":{"kind":"REWORK","_number":2,"created":"2023-06-20 15:23:08.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/02/886502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/886502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/886502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/886502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/886502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/886502/2"}}},"commit":{"parents":[{"commit":"eefb79a2ed77f5f14b01a1be2d28a94990205dfe","subject":"Merge \"Fix Venus containers built from correct branch\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eefb79a2ed77f5f14b01a1be2d28a94990205dfe"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 10:06:48.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 15:23:05.000000000","tz":0},"subject":"Disable build of Gnocchi containers","message":"Disable build of Gnocchi containers\n\nRevert this commit when [1] is solved.\n\n1. https://github.com/gnocchixyz/gnocchi/issues/1304\n\nChange-Id: Iafeb9093340621d65a9c9cd126d65679d981dfb1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 5505cd000ce7f12c272c5b62074c9cc6bf1bb7a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2bb362e5011023170bb664e2045217e20f56884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2bb362e5011023170bb664e2045217e20f56884"}]},"branch":"refs/heads/stable/2023.1"},"9efdc98d9fe8493735961696fbf95b87b499bd89":{"kind":"REWORK","_number":3,"created":"2023-06-20 15:26:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/02/886502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/886502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/886502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/886502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/886502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/886502/3"}}},"commit":{"parents":[{"commit":"eefb79a2ed77f5f14b01a1be2d28a94990205dfe","subject":"Merge \"Fix Venus containers built from correct branch\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eefb79a2ed77f5f14b01a1be2d28a94990205dfe"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 10:06:48.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 15:26:21.000000000","tz":0},"subject":"Disable build of Gnocchi containers","message":"Disable build of Gnocchi containers\n\nRevert this commit when [1] is solved.\n\n1. https://github.com/gnocchixyz/gnocchi/issues/1304\n\nChange-Id: Iafeb9093340621d65a9c9cd126d65679d981dfb1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 5505cd000ce7f12c272c5b62074c9cc6bf1bb7a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdc98d9fe8493735961696fbf95b87b499bd89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdc98d9fe8493735961696fbf95b87b499bd89"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
