)]}'
{"id":"openstack%2Fkolla~886503","triplet_id":"openstack%2Fkolla~stable%2Fzed~Iafeb9093340621d65a9c9cd126d65679d981dfb1","project":"openstack/kolla","branch":"stable/zed","topic":"disable-gnocchi-build","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-21 07:07:18.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-21 11:12:04.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-21 11:20:50.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_14200\u003e by clicking the attention icon","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-06-21 13:00:14.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-21 13:00:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iafeb9093340621d65a9c9cd126d65679d981dfb1","subject":"Disable build of Gnocchi containers","status":"MERGED","created":"2023-06-20 15:01:54.000000000","updated":"2023-06-21 13:01:12.000000000","submitted":"2023-06-21 13:00:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886503-disable-gnocchi-build","meta_rev_id":"6d96e72138ceafa2e987d7f10a09e02a1d52d4f4","_number":886503,"virtual_id_number":886503,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-21 13:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-21 07:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-06-21 11:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-06-21 07:07:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-06-21 11:12:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-21 07:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-20 15:28:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-20 15:55:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-06-20 15:55:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-20 15:55:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-20 16:07:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-20 18:37:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1cbc94d0e0d183361d8aec77f99cb737716171f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 15:01:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0345952a96252eb3c37ecb7e38ef252d54a9e38f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 15:28:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a92ddf83ac17ccfb9adf1163387d947131ce5de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 15:29:57.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9512afadae42e96d4689da44aed87cf0498318ff","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 15:30:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aae1a251d4c07a5bb5885ff3901490f02140c2e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 16:07:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1f25df99585473da356de2db25de3a9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e25e05e2ca054120ab87246af30e65d1 : FAILURE in 35m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2b187159cd574ba881e68bb9571af859 : FAILURE in 36m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/86aca6ad321348129237de5409f54494 : FAILURE in 33m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b2fc8a4cac94d8b8b348e86fab912bd : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9207cf659d2488c9475b6f5923d1045 : NODE_FAILURE Node request 300-0021496923 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ba458977693040a8bcb2d847f02dc254 : NODE_FAILURE Node request 300-0021496924 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/16106fdbf50b4abe937fc5f36590370d : FAILURE in 6m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1f6d16bdf3c85f692a0ed662e66d9741bf669e3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 18:37:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2633475d7249708d96dd153f80e677\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/668c61fd47204b08ac667d991bba766d : SUCCESS in 23m 10s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/74a4e53f374e47eebe053015e43e4704 : SUCCESS in 1h 03m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4c29ec2e41844d1e9c06931a657a4870 : SUCCESS in 18m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ddd60a6118e04cfc998291946c68deab : SUCCESS in 56m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c7f694227c9a439585aa056d8ed73820 : SUCCESS in 1h 07m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddbdd34a24434ab49209c38487ca8be5 : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83ac4d848c3b4a35a36f259a74766773 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aed922ed7a47454cbc7dd60e96e9ac13 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61419de0d8154bcfaf30c1451ad2c187 : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e670c1f25a324b9db9d4e01931dd451a : SUCCESS in 5m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6441e5c9bee7449286270be10567c1a7 : SUCCESS in 30m 33s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/48efa7a917ab4ba08cea0a00d64380fc : SUCCESS in 1h 22m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ef21ae27b9e640fe859288c292f2b915 : SUCCESS in 22m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eccb346028044301a3f63b5c489ba7f4 : SUCCESS in 47m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3804bfc2575b45e7919619d76ca2b722 : SUCCESS in 1h 19m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f4a85d7075878f1715fbe579d865df8132eb6f01","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-21 07:07:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ebff7d2a61cde7d2c433627e1594eb5c9f0b16e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 07:07:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4f1a6527993cac4f4d6f86aa8319b3dd58bceedd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 09:10:54.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/60f09a29b48947c8b1d4414c8bef94d3\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3996f0c836f14d35a6d1ad579cea0c21 : SUCCESS in 24m 21s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5ad40c21de6540cba4ef698491a1cef3 : SUCCESS in 57m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1e8f67be666e4ba0a345876e7336699c : SUCCESS in 1h 20m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d6b24f159004461ad85f5fbc87d3eac : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6479d5167ce7475d96474b2033a407f1 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90dc8285453a44d8af7033e4216f66e5 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d2967bf0844283a8ff7353c006cb58 : SUCCESS in 5m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e6c4277b851f48f5b8d5563d25973e88 : SUCCESS in 1h 25m 26s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8546e06bf53b4b48aedf8b0106568a10 : TIMED_OUT in 2h 03m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ced1e900cff7427cb86bbdadb601cc73 : SUCCESS in 15m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b05bc9e39cf34242810c8b05e1575f17 : SUCCESS in 45m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5b193280517e48ab8875332f0b32c0b0 : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":2},{"id":"d7561eb45203a58088ef7cf9573cc1878e34f1e0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 09:25:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"afd347084b149dce887138df9b4b85594f1c93d6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-21 09:26:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5924db3ae86c120223eeaf5890610c91f7cb7c38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 09:30:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac251fb2170140e6b5c43abae9b8b6ad\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a89ed944d5724e8785523d16e751b7f4 : NODE_FAILURE Node request 300-0021504986 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bf08db2535694a35906384bfde371c45 : NODE_FAILURE Node request 300-0021504987 failed in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d3c3695cb8f4a81b847b4600a313da7 : NODE_FAILURE Node request 300-0021504988 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/112877db88fa430dac9631891b279a9f : NODE_FAILURE Node request 300-0021504989 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b995c00804b4efb9ab73122c7574e8b : NODE_FAILURE Node request 300-0021504990 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ce09c704253543fb886552af2d87ce39 : NODE_FAILURE Node request 300-0021504991 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/66e3888e09c442538fe8b31515df7f37 : NODE_FAILURE Node request 300-0021504992 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3221ae64864bd1ef26371ed3b9bebe82825508f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 11:11:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9f33a22e7b847b19bf76f51c60dd3b0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d1daadd9b3484c7da473c8d47fdd63d0 : SUCCESS in 35m 03s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b39af6c03dd24a2689397af74b9feac1 : SUCCESS in 1h 20m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d2482aaa595049558fcf47abf4301ee2 : SUCCESS in 36m 06s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3d55b538e17d41488e2ba1036cd1aeb9 : SUCCESS in 1h 24m 03s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3be5a2672649415cbbad00cb1b4a7ce1 : SUCCESS in 1h 07m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7092778a955242f7a4e7176852dc565a : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd5dcb19344340569207e668d1179017 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5df20965b5ed403d9df8aa85d9bc2f9f : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6b5c0537ef84f5db7b217ce67e2a853 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d85bf3a48f99443e8ec56f32ecf890e2 : SUCCESS in 6m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7a6e1c51af754cb4aabd1b1c3a88a64e : SUCCESS in 42m 30s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ac1e02c2b1bc44fbb44f6c0dc85efa1c : SUCCESS in 1h 05m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d02ac312e79c47bdae0310cc030b5cd4 : SUCCESS in 26m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d26b96e17ff4ad18a5007f3485a898d : SUCCESS in 45m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d0547ec4f6eb4f2d93953414d4b7d9d4 : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3a29cd9acdc077b2d0f6a28efec9ae42a376b010","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 11:11:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2b277beaa303c746289ddb92f33c9f73bf8bf9d2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 11:12:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6d80e8b643742145f7dae388d2cfa2996f9b3019","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 13:00:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feeb4c4aa4b54dfe89a34281416576f4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/51f1b9fc7c654750b6c115ea9125cddc : SUCCESS in 31m 39s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ba577f87c6e24e02bd6e85fc6ea81c78 : SUCCESS in 1h 09m 48s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c0ac027700ed49b28151a090f4e2a20b : SUCCESS in 1h 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e597085dd2d645f999211585a8b5dd85 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c11ff0c230e84e17bb36065db90433e3 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ddad53f5a984f5d98e8a25fa866e93c : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af235829305f4d7a9bc7e7f3caa2cbc1 : SUCCESS in 5m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d708dc34f46c43aca0e23d476b0fd6bd : SUCCESS in 34m 30s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/6d037a0fc64c43d791205182557fa6e1 : SUCCESS in 1h 18m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7cad774757d9458eaf458238f6f423a0 : SUCCESS in 32m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae51b43479f04ac8872392e62a76d512 : SUCCESS in 47m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cc8a182cd4884fb98871e77e21cc075e : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":2},{"id":"59ea9a41805cd275105ae63951974f2ab7666901","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 13:00:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6d96e72138ceafa2e987d7f10a09e02a1d52d4f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 13:01:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3031ce62849b47fd8f3c5d2b9e7b0cfe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ca0a33957c462dbcbb5faddfdf289e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"85c764a653b6cf8da92ce55ac9615068683f5212","revisions":{"4ba4ce2af2de3e08f5cbcd4db76982d44be23592":{"kind":"REWORK","_number":1,"created":"2023-06-20 15:01:54.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/03/886503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/886503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/886503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/886503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/886503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/886503/1"}}},"commit":{"parents":[{"commit":"647a5aad7fdc1b89615d1d91c42dba7b93dbac8e","subject":"Merge \"Fix Venus containers built from correct branch\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/647a5aad7fdc1b89615d1d91c42dba7b93dbac8e"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 10:06:48.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 15:01:29.000000000","tz":180},"subject":"Disable build of Gnocchi containers","message":"Disable build of Gnocchi containers\n\nRevert this commit when [1] is solved.\n\n1. https://github.com/gnocchixyz/gnocchi/issues/1304\n\nChange-Id: Iafeb9093340621d65a9c9cd126d65679d981dfb1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 5505cd000ce7f12c272c5b62074c9cc6bf1bb7a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ba4ce2af2de3e08f5cbcd4db76982d44be23592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ba4ce2af2de3e08f5cbcd4db76982d44be23592"}]},"branch":"refs/heads/stable/zed"},"85c764a653b6cf8da92ce55ac9615068683f5212":{"kind":"REWORK","_number":2,"created":"2023-06-20 15:29:57.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/03/886503/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/886503/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/886503/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/886503/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/886503/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/886503/2"}}},"commit":{"parents":[{"commit":"647a5aad7fdc1b89615d1d91c42dba7b93dbac8e","subject":"Merge \"Fix Venus containers built from correct branch\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/647a5aad7fdc1b89615d1d91c42dba7b93dbac8e"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 10:06:48.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 15:29:54.000000000","tz":0},"subject":"Disable build of Gnocchi containers","message":"Disable build of Gnocchi containers\n\nRevert this commit when [1] is solved.\n\n1. https://github.com/gnocchixyz/gnocchi/issues/1304\n\nChange-Id: Iafeb9093340621d65a9c9cd126d65679d981dfb1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 5505cd000ce7f12c272c5b62074c9cc6bf1bb7a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85c764a653b6cf8da92ce55ac9615068683f5212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85c764a653b6cf8da92ce55ac9615068683f5212"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
