)]}'
{"id":"openstack%2Fkolla~886511","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I84c7633765749a5e9ced983505ef290c32154ccc","project":"openstack/kolla","branch":"stable/2023.1","topic":"cinder-volume-nvme","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-21 22:05:15.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-21 20:23:52.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-21 22:05:15.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-06-21 11:30:00.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-21 22:05:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I84c7633765749a5e9ced983505ef290c32154ccc","subject":"Fix cinder-volume: add missing nvme-cli rpm package","status":"MERGED","created":"2023-06-20 15:42:10.000000000","updated":"2023-06-21 22:06:13.000000000","submitted":"2023-06-21 22:05:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886511-cinder-volume-nvme","meta_rev_id":"ac0df1f007d81d06b72130dbde81c33259150c5a","_number":886511,"virtual_id_number":886511,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-21 22:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-21 20:23:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-21 18:14:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-06-21 18:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-06-21 20:23:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-20 16:02:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-06-20 16:02:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-20 16:02:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-20 16:02:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-20 16:11:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-20 18:50:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-21 13:36:05.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"b547dc2137a9f6a7d5a89b5be876a0e44e999a15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 15:42:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f9a64bb771df574e31db175f8a0d2dd04f10449","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 16:11:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fc480b2688d47f5bba0c3b8f0c9478f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9c9e26a4a327492482c5785293bdcf2b : FAILURE in 28m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a75983e2cf5641e999958f65f46b0c7a : FAILURE in 26m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9a624fa7e782474eb9a8ce478bd67bfd : NODE_FAILURE Node request 300-0021497059 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a5c9773ce674fc9a2984be120199ce8 : NODE_FAILURE Node request 300-0021497060 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/587ef1c037484a139550a08660335a94 : NODE_FAILURE Node request 300-0021497061 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a534a7964b03494aa328ebde5682b7ca : NODE_FAILURE Node request 300-0021497062 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6bf2f4e000814b4190f487a95a31ed1b : NODE_FAILURE Node request 300-0021497063 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0caec1559a91848d733a391c10da86f60048880c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 18:50:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7891ac101f44b4e8b60ae03804ebcc2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d72afabdbce74b5e8dac7030391f31f8 : SUCCESS in 21m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/04efb5cdd2414ef08b9ae6b2c06d7983 : SUCCESS in 1h 08m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ff3473fc42264c4dabc3cb8d802926f7 : SUCCESS in 22m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f6398c4c0e1144c6ae306da80d5df9fe : SUCCESS in 53m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f3010b7adac84a0ca61c5458da651a5e : SUCCESS in 1h 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bdf1fcbb62348278df260cc1380abfc : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc182ad2dcc04c6882c3651979e1d136 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16d5947c4d5b4e078670134affac329c : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3192563f789c43e78f13fbd26f2f3533 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6cca0c59b6d4280a60f6108a5933b5d : SUCCESS in 4m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cbcad826933b4c749af86c8cf9a4b446 : SUCCESS in 31m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2424b30ea60249e99fea28b439142b3b : SUCCESS in 1h 07m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d9ce8464833647fa8eff3d78de7318ce : SUCCESS in 1h 29m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c279eab1642849bf8afbe9908ffaee5e : SUCCESS in 16m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8886394886344b05bc49217fc8dd3655 : SUCCESS in 59m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6e362bd4bb48431f947006df92bd7dde : SUCCESS in 1h 09m 16s","accounts_in_message":[],"_revision_number":1},{"id":"938eda88508f06d9f3162b1ec7ab749b2922b972","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-21 11:30:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d254d9bceedb281c4a3299d37ff0ed849eb95f7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-21 11:30:10.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"11f57876990a955116c244a45677295affb77090","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-21 11:30:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37c4fef42dca9c4791a3d68d11e722678edf5f39","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 12:31:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0fa2678e1e60585046236aa324150433f88a687","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 12:32:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"822afc0a346609903822b759a1c772680e86e0f4","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-21 13:36:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0000292938f797d86a55ffbc2e5e82a04c6caf75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:25:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2a8aa701ea940d382322af2a8ce797f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/afcd9c620ad0481a937c98bc2f47dffc : FAILURE in 16m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d4ce19a6045e455b8bf422e138e29e76 : FAILURE in 18m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/245bb9ffa961408eb437bc32acc0993a : FAILURE in 36m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/729e724ebf3045edafbf9debe05dc891 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3b377914c504ce08b1240a1b5cd7dd3 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb77c69cbcd24e99be7e91fd1e114b9c : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ad52119c6640e1a5906e37a36238ce : SUCCESS in 4m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6c0fe65fffb348b0943c0bed8d54c8dd : SUCCESS in 27m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0629f97aac904162933e520aed733460 : SUCCESS in 1h 19m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/eb7a40d1eacd4ec6a42e900e11de0102 : SUCCESS in 1h 11m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/42d755653307425cbb035fc044dd3441 : FAILURE in 37m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/37119468d1f74a059ad36ba3c2afe05d : FAILURE in 30m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7185799c145d4a9987835b5b2b6a54b6 : FAILURE in 32m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6c919f8400da705775345a5388c5b2cbb655063d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 14:29:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"848ba1f9e908da4146c0f1f16a23557c50ca8439","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:58:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbf8e5a4a7dc449dbb0a18eaef645c1e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4f767f3bca1d408d8f8d1475c96fc9e5 : FAILURE in 25m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fd8b10ea11f04014820aa6fe994a4f5e : NODE_FAILURE Node request 300-0021508350 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/32b5876e72fd48cf8b731434de083df6 : NODE_FAILURE Node request 300-0021508351 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff955db4eba2472ca0866b82f5c533eb : SUCCESS in 2m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d71d3e1bd5b94244b905525f2dcd876c : SUCCESS in 3m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/de6d66deec9a4789bf2f705965ae9771 : FAILURE in 24m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5b5a8db5afcb4ad6bf7e79a708522a0c : NODE_FAILURE Node request 300-0021508355 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24c23961dad0081fc24a0974dc673bd3b399111d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 15:58:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4d2023080cbd554572db589987f0730da2079ad1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:27:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4aa3208f6b94d9097dd5598428c8b58\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/da3b07fe0d9c426daa7f5362c7950bf6 : FAILURE in 18m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/81cf35ff91a3446c9abe5287f8d1a91b : FAILURE in 18m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b6c00acd951c468089f8d66639870a37 : FAILURE in 28m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/761e28d0000d4332b773cbfad76ea2b5 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b440ad50a9ec4f518c7a6001b4eb9014 : SUCCESS in 3m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c2aae67d17504fb495cb20322e3bee94 : FAILURE in 21m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a721be2fde5044d1a0975475edd4dd7e : FAILURE in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"14449d5861a35be74c4e5cee92da73162ae22df1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:36:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b631bab558e49098e199af4bd8640db\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7e4f3f62c889494bb1505a1a81744df0 : SUCCESS in 30m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cded6642f3204e96a1286857ca21e927 : SUCCESS in 59m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/af5cef1812584d00b531ccb520f066c7 : SUCCESS in 17m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a4a15cb46c024015932292f21037b8a3 : SUCCESS in 1h 15m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6298665bf14e4a81b3f829ece53d489a : SUCCESS in 1h 05m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83946135a23944e9a595af25449bfabb : SUCCESS in 11m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c9ceb344c064b58a22e378a7a4c3fac : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af3d67c3be904a348c30596c11a6837d : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67995ee26f6b4ed5bfe4148157b57d03 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31ee7411cfc4e7b9231d8502c02eec0 : SUCCESS in 6m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e0c61704186f4be8b505a244ce6cefcc : SUCCESS in 32m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2b55767dcd724e4ba82e9bc04dd433ae : SUCCESS in 1h 10m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c18b408323b14d05b6b065776feaec58 : FAILURE in 1h 06m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8df3ca322da241ea83c70f42eac3ea6b : SUCCESS in 17m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6e7cd647ef43408f8bfa2222e0509ec4 : SUCCESS in 58m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/54ebeae8de7d4b07b83d6c5d0eabaa59 : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":2},{"id":"2219897e24534e43f1f44c9f28a5b96face17099","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 17:39:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cfd17d094801f03710b15d4f81d2cb7995e74945","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 18:07:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9158b8f4d1174826ba029800c313fd06\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ee42ce3a57fd409798bebc787869d1c0 : FAILURE in 19m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/02e2241749e14e19892851a1ba311380 : FAILURE in 20m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8d5cf4366161497ea36e44637ddc4d75 : FAILURE in 21m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9d15b2e65c9458e81bef80d49b0a211 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1ff3bb8823fc4f69a7a82aaa1fa6f557 : SUCCESS in 4m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c43d937d33724ec0b425ae09d11f5e4c : FAILURE in 24m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9487e388b269460e8782a4e158a6566f : FAILURE in 6m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"891f1b897837260a60b18b6ce86b2fdca1e9eecd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 18:14:49.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"23bcf2590bf6365bd7983efc7bc569abfa54d4ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 19:02:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5506d0031344da28ed140d90242fb7c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d067be2739074612a54fe1e2db0d61cc : FAILURE in 18m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/997e94fb06ed413b8c7f9666a40e5fdf : FAILURE in 18m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bd65c3242fe344d5950320f88cef0326 : FAILURE in 21m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83ed21d8f6f64b908a878a387eee1afd : SUCCESS in 2m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2247e0c0d68b4ce6a680a0bc046a1101 : SUCCESS in 2m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/093276e3f6c34cd384c3fe7fdbf1e98e : FAILURE in 44m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/084f6d7e5ace42c7b8cbc083ce5a0e1b : FAILURE in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"06fe0ed97c468df9f4503bb173f6304d92021ff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 19:49:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0907195738ec4366944b41dac92b7f2f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4bd69dde8228490fb5f1f4fb1fcd89eb : SUCCESS in 18m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d94cb142934b44c186144ad40990aed6 : SUCCESS in 58m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/57ec5393395b4a52bd4dd49e4f32fe74 : SUCCESS in 16m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/131bbb2e7ce14fe38f360b93086c4cee : SUCCESS in 56m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b2afe4a1d7924238b6d1930ed28a3e9b : SUCCESS in 1h 00m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0530cd93911e4e799f1efba686815e69 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80e92037bb8c40a4b847b67cfd5b187f : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03e35b9dcf084e568c80b9cd41ddf367 : SUCCESS in 4m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9cee43e13fc4abf9cb7bbfe4ea85499 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c603701c45406abe9df53a06f48f57 : SUCCESS in 4m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f97ed7ff5ebd416faed6c3081e619374 : SUCCESS in 16m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8ac4573003e540fc9e231c56c296aa6d : SUCCESS in 1h 00m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/841729e7eaf74c1894a722ddcc0397aa : SUCCESS in 1h 09m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c0d8931e206c4a008f1aa1c0641dbd4f : SUCCESS in 29m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4c5a7c38229c4387bde47e2c3384c3aa : SUCCESS in 59m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/19e84d5d48124f41a557f1cfa74b5cd2 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":3},{"id":"39aca4578d340116840e6b3259f1ca46e16c08c4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 20:23:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2598cd76d4936189425bbc998970f3b9992e9f68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 20:24:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8ac680dc6fe443bae4a4919b2086f6f44a2d545b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 22:05:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fee863a24d444a239ef5a2d1bdf37c56\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3b32cd376c314b95a0edf0e7d1b68e29 : SUCCESS in 33m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a5697d98947d40218306aa72f1d44572 : SUCCESS in 1h 06m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/25ec6424b1854161ba6366725635964e : SUCCESS in 56m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d069cad256f74a91b82bb73384a46003 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b632bfa5504642a8a7ead81276f5495e : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de9c512fe33a494a804bc60e9862d1f5 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71fe201fef684c7786896fb113129602 : SUCCESS in 8m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c5e861f7f2f544708b9b35e462718f49 : SUCCESS in 20m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dd58b35d9eca47fdadd54817a827eec3 : SUCCESS in 1h 18m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b708949203a84db8a0b1e9494e59c18c : SUCCESS in 1h 05m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6737d3c6c9dd4436b3341619498c62e9 : SUCCESS in 23m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e9e56768e0cb457f87bcf1982dd5801c : SUCCESS in 53m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2f5f5f44c7b142be9daff64d08d080be : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":3},{"id":"4e9d7d2269f514be11f687dfde6a8b790fc06ac4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 22:05:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ac0df1f007d81d06b72130dbde81c33259150c5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 22:06:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac4bb26fdcd24fcfbc372421ad681d3c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c3d3df47b294b64a340a342aa9e176c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5a67a0162d13959eba795c9464a106fdb55287cc","revisions":{"8edffc0a56cc3563d361f925e378337a0fc3f3e6":{"kind":"REWORK","_number":1,"created":"2023-06-20 15:42:10.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/11/886511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/886511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/886511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/886511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/886511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/886511/1"}}},"commit":{"parents":[{"commit":"9efdc98d9fe8493735961696fbf95b87b499bd89","subject":"Disable build of Gnocchi containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdc98d9fe8493735961696fbf95b87b499bd89"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 18:14:11.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 15:41:37.000000000","tz":180},"subject":"Fix cinder-volume: add missing nvme-cli rpm package","message":"Fix cinder-volume: add missing nvme-cli rpm package\n\nCloses-Bug: #2024390\nChange-Id: I84c7633765749a5e9ced983505ef290c32154ccc\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 11f90f3746f6affd34885019a82eb94d3c098ee9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8edffc0a56cc3563d361f925e378337a0fc3f3e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8edffc0a56cc3563d361f925e378337a0fc3f3e6"}]},"branch":"refs/heads/stable/2023.1"},"d80c1f25b77b4715c4eb66b4923829099020ec54":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-21 15:58:01.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/11/886511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/886511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/886511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/886511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/886511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/886511/2"}}},"commit":{"parents":[{"commit":"f4600a14974163fb6e1a7b2de398185f01744e6f","subject":"opensearch-dashboards: Fix permissions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4600a14974163fb6e1a7b2de398185f01744e6f"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 18:14:11.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-21 15:58:01.000000000","tz":0},"subject":"Fix cinder-volume: add missing nvme-cli rpm package","message":"Fix cinder-volume: add missing nvme-cli rpm package\n\nCloses-Bug: #2024390\nChange-Id: I84c7633765749a5e9ced983505ef290c32154ccc\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 11f90f3746f6affd34885019a82eb94d3c098ee9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d80c1f25b77b4715c4eb66b4923829099020ec54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d80c1f25b77b4715c4eb66b4923829099020ec54"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"},"5a67a0162d13959eba795c9464a106fdb55287cc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-06-21 18:14:49.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/11/886511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/886511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/886511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/886511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/886511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/886511/3"}}},"commit":{"parents":[{"commit":"9efdc98d9fe8493735961696fbf95b87b499bd89","subject":"Disable build of Gnocchi containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdc98d9fe8493735961696fbf95b87b499bd89"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 18:14:11.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-21 18:14:49.000000000","tz":0},"subject":"Fix cinder-volume: add missing nvme-cli rpm package","message":"Fix cinder-volume: add missing nvme-cli rpm package\n\nCloses-Bug: #2024390\nChange-Id: I84c7633765749a5e9ced983505ef290c32154ccc\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 11f90f3746f6affd34885019a82eb94d3c098ee9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a67a0162d13959eba795c9464a106fdb55287cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a67a0162d13959eba795c9464a106fdb55287cc"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
